)]}'
{"id":"openstack%2Fkolla~958020","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I400e9a9a3bf03053589835f933023eedee504f08","project":"openstack/kolla","branch":"stable/2025.1","hashtags":[],"change_id":"I400e9a9a3bf03053589835f933023eedee504f08","subject":"Enable Barbican build on Ubuntu ARM64","status":"MERGED","created":"2025-08-20 07:22:36.000000000","updated":"2025-08-20 13:03:23.000000000","submitted":"2025-08-20 13:02:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958020","cherry_pick_of_change":957677,"cherry_pick_of_patch_set":3,"meta_rev_id":"81387147c7db932a8bed43b2ad7992b2b760a6c3","_number":958020,"virtual_id_number":958020,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 13:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2025-08-20 11:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2025-08-20 11:17:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-20 09:04:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9d1aff1f29d983f434c1091289f1c10e23a4e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 07:22:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5432bcd0798517cbe196ce15a4710e3650a4b2aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 09:04:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb37062e75d4e62a28b76f71b38083d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/38e65536c0b740ab888e58d4fe306ea3 : SUCCESS in 14m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/282ef69ff2234831a67a20235fc77fb0 : SUCCESS in 51m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6afa13987fb148669711b3c745c7dc8f : SUCCESS in 14m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/795deb67e04e4a2caa6804e3d755f5db : SUCCESS in 34m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e5628b405d6048c18ed05b6a63e530c6 : SUCCESS in 1h 09m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e32b074ddb5c4f108f8b781666c1f683 : SUCCESS in 1h 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4aebfe22974d95a95327849070a775 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a35a9da0d3fe4a7c8a4cf6504831d41f : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2384a72750e54a72aa1c3ab163fc8e30 : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c741d6738f934f0f83879880a188491e : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38e1f07a1f534107a427da7017e6f044 : SUCCESS in 9m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8a59f4a1481c4441aadef42d4c828cda : SUCCESS in 3m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0c9125387fef444d9c6dbb5703731fe4 : SUCCESS in 15m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4cbda06d803049bcad3cb265f0e7f0ba : SUCCESS in 1h 19m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/268bb38e5a5d44b6ad3a57d445bc5ddf : SUCCESS in 1h 04m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f2f72d86b9f748a7a4b825b209e2d9db : SUCCESS in 1h 36m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/667539be04f44b6eb8d7cc6098e90a9f : SUCCESS in 19m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7c0c6783e7a442d189f0811bd9e8c526 : SUCCESS in 27m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e719f5513e04a0cb01d1d95ab6b502d : SUCCESS in 1h 01m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8e4e105bf138413e954e8bc976a8f87a : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":1},{"id":"cbbf3ed248ee8ae657259657d3689f01fedb1a11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 10:48:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f752d504c2714a5d9d42b09b677bdb55\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/24773ee517b946a0922d4406aa4e73ac : SUCCESS in 1h 29m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4882a71c1dbb4c2290d9c659031fc3a5 : SUCCESS in 1h 24m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8360db7b92c04ee799c631dfa58e6320 : SUCCESS in 2h 27m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/962223141bd14448b6da56d9793c1d4d : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/933e0c49826e40799127a8f81325e453 : SUCCESS in 6m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/12e52cf6420a48639cd9f3c37c469d57 : SUCCESS in 1h 16m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0212933359e949b2ae1d691ee370336e : SUCCESS in 1h 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb647d586fafb7279df248e84fe4e36c7cc495c9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 11:17:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb2446e382a7d5c466d03682a375fd55cfd5cffd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 11:17:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73a842fa26cbe5bdb56197638bc96220bbce65f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:02:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b27fdf4d5cbd497ea1f44e9cbcd8cc81\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/65cee9b7944a44cfafc414e0a2649e9a : SUCCESS in 19m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c0306fe236dd4ccab935d684e9f4a515 : SUCCESS in 30m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d8de59f4d712425f9cad05399ce22447 : SUCCESS in 1h 07m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/53733b2a9b4742199e3d840418bc3c21 : SUCCESS in 1h 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381e17e442ad4be3943ead66adc0bad4 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/740a85039c12403a85f5983c04bf7d67 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fc9c5530a1146109a57fd1782ef492a : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c26500ed49f46b3bb4428c5fb6ca29e : SUCCESS in 8m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8cdf8785fb3b445a9d6bc15640dc906e : SUCCESS in 5m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7a9e74c2742c4ee09f6cbfd840f5decf : SUCCESS in 16m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/df6c7255096a463d88f0fc791b8bf3b3 : SUCCESS in 1h 30m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cde0d45df48b4dcca7145dfedff99f70 : SUCCESS in 1h 01m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/72f28a21df5a4ba9b279525710f4472f : SUCCESS in 1h 40m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9845a37094d6498c9ee4939edfb5ccb0 : SUCCESS in 20m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e2a96d39459046f48371e6c6b938ab3a : SUCCESS in 30m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e6bf86fb9f554a6bb0ee445b2864084c : SUCCESS in 1h 05m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/07b2fad2e0384f5cbc072c0b6d8e22f7 : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2c24aa3272a21b1b5dc56c703921def00252bc31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:02:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"81387147c7db932a8bed43b2ad7992b2b760a6c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:03:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0511abd59c6b42d285421093343a7bac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec1628189bb44e784e8697b886be4de : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"171f725a9faa0965d79bba6a4e0bb31e0d3a8912","revisions":{"171f725a9faa0965d79bba6a4e0bb31e0d3a8912":{"kind":"REWORK","_number":1,"created":"2025-08-20 07:22:36.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/20/958020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/958020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/958020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/958020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/958020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/958020/1"}}},"commit":{"parents":[{"commit":"293ef73bdf02a8f5b1c6f1ac45a8eef5c90d0735","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/293ef73bdf02a8f5b1c6f1ac45a8eef5c90d0735"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:26:48.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-20 07:22:36.000000000","tz":0},"subject":"Enable Barbican build on Ubuntu ARM64","message":"Enable Barbican build on Ubuntu ARM64\n\nRe-enable the Ubuntu arm64 build. The previous exclusion was due to\nuWSGI issue [1], which is no longer reproducible on Ubuntu Noble.\n\nChange-Id: I400e9a9a3bf03053589835f933023eedee504f08\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 85878fe6a89f1a881ac276ec10dbfa0c093fadb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/171f725a9faa0965d79bba6a4e0bb31e0d3a8912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/171f725a9faa0965d79bba6a4e0bb31e0d3a8912"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
