)]}'
{"id":"openstack%2Fkolla~958081","triplet_id":"openstack%2Fkolla~master~I92f3095b2d51bd6ac7ff48b0c7d79852eccc9803","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-25 10:38:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-25 14:13:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92f3095b2d51bd6ac7ff48b0c7d79852eccc9803","subject":"[release] Use Flamingo Ubuntu Cloud Archive","status":"MERGED","created":"2025-08-20 13:19:26.000000000","updated":"2025-08-25 14:15:11.000000000","submitted":"2025-08-25 14:13:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958081","meta_rev_id":"4c5b6ae83656b5c41263eeb9ccca64bfc36618e0","_number":958081,"virtual_id_number":958081,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-25 14:13:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-25 12:59:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-25 10:38:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-25 12:59:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-08-25 12:59:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-20 13:22:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-20 15:17:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-25 12:59:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"169cd18f9066b477b609d3c5f94517b3f439725e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 13:19:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e535dd0f84a385bacdeca6e99255a1e83a5353ae","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-20 13:22:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"68b4417eb6024ef31aa9b69e080c8bef4f8893bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 15:17:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8eadd285a93f48fba21c3bdc98afe9cb\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5facdec7be7d45d0a026a45380cc06b2 : SUCCESS in 11m 25s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/44658c5bb956489a8549cfd663327957 : SUCCESS in 55m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/66d4c4c3cc514b15af0043e779d4631c : SUCCESS in 22m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0590dc74f3294f8e9530d1235d727252 : SUCCESS in 16m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6994f90878ac4e25911adc24ea97ba68 : SUCCESS in 35m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7c20a02358d14682916e415b6f7faeff : FAILURE in 1h 44m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d5e2f3d5b464acbb3f7c3c93356dd6d : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3af697b26d50483fbb1cfaa7f6d07bc4 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bb4519e6dc340e89039e74e2c36b641 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6264ee0798c4ec382135e111f1f8df4 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81b216759294439681be41397866f67e : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037310eb6ac94aa2addfd62691c8b926 : SUCCESS in 4m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5f37b1bf96da4a028e06af9b9b3f79cb : SUCCESS in 3m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e9ea7b2221cc45fe9ad0b078b6bc2b3b : SUCCESS in 16m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d869b6c712b2463eb03f2f6b8755ef48 : SUCCESS in 18m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8fd963352b9240a582215e0719c592da : SUCCESS in 1h 01m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/93b29f5ad01341a7af526060da5e012c : FAILURE in 1h 44m 48s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/bc97503b04184d2fa608f8b9deaca499 : FAILURE in 1h 39m 18s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/22dc7c87d067420c9fb10fa565a1fe9a : FAILURE in 1h 40m 13s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d1cc9dc93035485c83f67211717f946f : SUCCESS in 1h 18m 29s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/83d5cf402df94e6fa6f8e0d70cc98052 : SUCCESS in 17m 57s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/78b906ba8d7b4eb991f5c3971a01d003 : FAILURE in 40m 30s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/bc357d052f9d45a6a9c8efbc0f7e02b2 : FAILURE in 26m 21s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1a9f9dd8e25f406399c06365728010fe : FAILURE in 1h 12m 38s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8de01cbf53ee4254a021f90e80dca6ed : SUCCESS in 33m 01s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/022b816e6e5c426a8e40f72e11fa90d6 : SUCCESS in 1h 55m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab0394fabd5232043c8bf0bc4badd14c043469a0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 18:14:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4b0b0d1b5a5a521da5a368b8bbadbd3acece14f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 20:55:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa816102f30464a954ee629bb59e69a\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/fa8aac4eca594781bd3cfbd0b1607aed : SUCCESS in 1h 26m 46s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/2e76625ca55a4b4cbd23c124103ba35b : FAILURE in 1h 51m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2c446b9f36214a909d01c04f0e522455 : SUCCESS in 1h 01m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/291638ede992405aa5512c38d12d730c : SUCCESS in 2h 24m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52f8907fa36e4c7a996663789555a20e : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/870d2d90b4c14aec90fa577715e01d04 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b046f0c68364fdba5928524966b3b71 : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bfa908f515b94cd8b9605d1c20b30778 : SUCCESS in 1h 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"715b1753a8d0910c70d8c712d1f59f36c2e25cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 21:01:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/724f520224154504ab17ea9d97507d58\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/af1d88076327448f8e552a12752eb36c : SUCCESS in 16m 17s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/570c3f55ba7a48249ef21096431a89d3 : SUCCESS in 1h 14m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d00da23e9d96412394b18a2032f674ba : SUCCESS in 20m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c1294e4e0d8a4e49b33ae1b15ae5a82b : SUCCESS in 32m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/70b0c59bf004425597d5bad8444081d4 : SUCCESS in 47m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d577ef25e8c64b2f9eefb659d5719855 : SUCCESS in 1h 07m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42fd256a960146f8934fb5f400d2fb34 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0abe840e0a8d482884eb766db410dcd4 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d93e13fe443d411c81ae347677c58d53 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6f8af7bb02b418d8bc6b079d278b6f2 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/470aae353fda421582a5ffb5727ef7bb : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7793f6c7c4c41c697e2bc3486bd66bc : SUCCESS in 6m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/770674b9a9a04675a8f1d54bc8a8f161 : SUCCESS in 1m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3fb9472d46694598b02f6fa2d07806d9 : SUCCESS in 17m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fdde11dd1fd344d38cb3a68098d7f7c1 : SUCCESS in 28m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9f4b980c94b24db9844fc2b2cd67c8e8 : SUCCESS in 38m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/31320c8cdb5f4c048713b8dcf3ad992d : SUCCESS in 1h 04m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0c45f731feb54eec8d08516fdc8fabd3 : FAILURE in 1h 16m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4476a171583942ff845385bedec19669 : FAILURE in 2h 43m 37s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/34c2bbe09ace48d989a558065721c3e4 : SUCCESS in 44m 58s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/85d4d29727414f11b0e73143af0eff81 : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4c31103159b548f9bb3cb09ca4ff06f4 : FAILURE in 35m 27s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ace2a83846e143e2bc26d1aed9867395 : FAILURE in 30m 04s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/81e9d3d1eb454107bfceb6e287b30e2f : FAILURE in 1h 12m 11s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/86f15bdb95ad49d5af0c34502fdf6244 : FAILURE in 28m 43s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/f6776e7cc4f8470f92100eeb32eb4e8e : SUCCESS in 2h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a3f0c0661372752a2cd8855fec14c63bacaa4d1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-21 07:45:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcf466be97eb116efcc547b6b619e2c9d4d8c723","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 10:28:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bdba6132be0444c9f8f565c1ebfa229\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/8bcda6101b404d96905eb27e114d6deb : SUCCESS in 1h 46m 48s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/386e467c3d134e8f90041e21f86f23c3 : SUCCESS in 1h 41m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/601aa0bdf46344d48f6e7c7e043f0f6f : SUCCESS in 1h 40m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d10892e9aca046f6a0471f8d21fb9a9d : SUCCESS in 2h 11m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f73297fdede04cb082fb294f09b1f782 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6165ab1752e44db89b655c725a8785a : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9beb87a00bd64fa4a2ca47009d94efc0 : SUCCESS in 8m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8b9ca8e127474a43a33872eca411a440 : SUCCESS in 1h 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"981647a4b680f297c1f187b8d791eec13c78b2fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 12:49:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32adfb2c189e458d8dace79b53dcf2f5\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/29f3cdfe87364806b7a3cbb3e1d62826 : SUCCESS in 15m 38s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/900c1c9123b1404084642180db5224c1 : SUCCESS in 33m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4531aff8c6d84a7ea407b93987797eaf : SUCCESS in 20m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e4ea2e032ef143ac9d8e7864e8e2703f : SUCCESS in 29m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/301d4ae67253479d84df496ad063aa08 : SUCCESS in 47m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d6419867348744de98df2d7c79d69b79 : SUCCESS in 1h 09m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc3443c93e44b51acf7f8b98997a794 : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57c5834b4d2644c3982cf528f2c25607 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7539071e9cb4be58e1f6b98eef12520 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/897575208b804db689880cab7c89bd6e : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ed19fcbcd954a4484bb798f0bc41890 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6631b3794c43d3affb67c62f04d3d6 : SUCCESS in 8m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/164ad4157cac4cab92c8ce6d927737cb : SUCCESS in 4m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6c3332f621154803bfb367d18b0bbca4 : SUCCESS in 17m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/18c5ebf2757749c4b6e5b5c9c3b0f6fa : SUCCESS in 12m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8ff6db84b48f4c50812d82cf4d9501b4 : SUCCESS in 31m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d06ed49adab64d138d60b59093ea38ed : SUCCESS in 1h 30m 52s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/43c4d2810b49491480b9a7c482076797 : FAILURE in 2h 08m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bc6268d0a1a64529a442f6879c57f3ec : FAILURE in 2h 12m 18s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/2e8ba60782464338bc2ecb76a15efb98 : SUCCESS in 58m 17s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/956f4c7386ca444abc9446c6d5dbbc9d : SUCCESS in 28m 40s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b015bdb943b34865ab61bd33b47dddca : FAILURE in 11m 11s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2ab510c4f41244b5a66cc13fc9afc3ab : FAILURE in 26m 30s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/734a21b30a2f48019b12de746ec3b53c : FAILURE in 1h 12m 00s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea26dfa34c54469684fa8af9a80d1a48 : SUCCESS in 32m 25s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/6a84ccad625f49aa8fc5bfb2744b0562 : SUCCESS in 1h 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aab0d880a0aa1977a3104f4914bc9c56f4a337bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-25 08:15:23.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d41725194b51287c3b7863e7f7047660eb47bdeb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-25 08:35:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b4a60e7e2c65396d0a808905a42230ebe0a4cefc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 10:29:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb2cb23d624e4a86aa5850f6ccb29727\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e9b3020bcc594c8e8d5c49e7ce86dae9 : SUCCESS in 49m 49s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f961e2fe06bc41f8b6418275650ebfb1 : SUCCESS in 2h 11m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f47687daa8464846bfaf8f881cf254b3 : SUCCESS in 48m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c7f036efe55241fea59903bba4b0bdd0 : SUCCESS in 1h 32m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fe29324755124b4fbae3fa41980efbcb : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/960b5ca66b7d4e988f5a1943c8970b0c : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f378d3c5dc9640c9997a3dae9b48b215 : SUCCESS in 9m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7e2bf040a17d4580bfe7e35ddedb4571 : SUCCESS in 51m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6e5a4f9c3d3ea164a38f3e417c3a06de8328856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 10:31:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6788ba5f8caa4fbc94d73fe76f256f3a\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/6521f46ced3b4aac879da997a10b45d1 : SUCCESS in 11m 50s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/65e8fb3bf13c4e76b30ed8d96788be7c : SUCCESS in 1h 17m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/419969602e3b41699841f67e94420db9 : SUCCESS in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d3e4a1874340421e9b3a88992224be37 : SUCCESS in 31m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/58cf1af0012746a2b554a2a6fb517819 : SUCCESS in 44m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8e07f4c257614f76b931e522ed711101 : SUCCESS in 1h 04m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449393ad3c974710930b8b5e054169e5 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f94089c50e4d467c89de5f63296ce94f : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/733f8e5a9bfa4d278ad81d10366c4557 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3d7a8e362d74ddfa2e87bee9dc5b6f4 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/039ca43b49dc418da18ced2a74994240 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/540608c6a6c04333b030fe54da397a74 : SUCCESS in 6m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dc61d029ba254f568a4fe413d635125a : SUCCESS in 2m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aea70fd22e8949b3bf485a3000569cba : SUCCESS in 20m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5e68cbf8c5a34fb18a99836915d9add1 : SUCCESS in 25m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0ba35a2a5fe54abe8ff9863e46802944 : SUCCESS in 48m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/55e19fc6924a4d8984ed2b478583c3e5 : SUCCESS in 1h 09m 45s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fb59bb09926c4c35a2b37511f5ca23a3 : SUCCESS in 1h 34m 52s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b5798101d6a04f11a3c4c5416003be48 : SUCCESS in 2h 14m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1835af3740cc48d08c68875667282720 : SUCCESS in 43m 24s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6234134dc9ed4bda81bdee2f95109730 : FAILURE in 24m 11s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/308d68ffc7e24adbbf84ed120415c5e6 : FAILURE in 45m 02s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/149d0044f033464a9e2e019155306ac9 : SUCCESS in 50m 16s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2ffca6393bfb46c39a438e42f268c6ac : FAILURE in 1h 00m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9fceefcacb7d448780d082fa0804bcf9 : FAILURE in 27m 25s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/408d07f2dbfc4e51b2893202ec7540ad : SUCCESS in 1h 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5636e968e72e3f156b2643c34ca2ed588bd4b24c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-25 10:34:08.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"55f7ec2951d8c3ecca10032de7d43e505915fd8c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-25 10:38:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fe874b0f0701a331d04120c5099d413fb39839f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 12:56:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4de2bec95341e6b9b0e5424394e0f7\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/826dc77f033640e288a666e340696ed8 : SUCCESS in 25m 58s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/eb9539cc085d4cec9bb3800cfc2fd239 : SUCCESS in 39m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8c269f7efb9449ffa4571378aa435bbb : SUCCESS in 18m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b131543f20724dcd91e6776b12875756 : SUCCESS in 31m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f50776c58c1648c8a6b79a2e17450abc : SUCCESS in 49m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f69976d6b57a4e70949e0b966bb31730 : SUCCESS in 1h 11m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b612e837a78413f99d9e41fe121eed6 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59f3ef9024e34f66a60b40ee764f7aea : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6709546985d544f38f0c244ae0a45bda : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2e2b61bbdca42aabb8ec77a20f0c25d : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ede159dc5e464e934e788b4b1a9afa : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cab0e65d7b3460f80602bb86b4752ef : SUCCESS in 7m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8799a404900040ba9160380fd6544739 : SUCCESS in 1m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0464eb10186b43a79b92f71d0e173006 : SUCCESS in 15m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fee79a35394b43979948f7e20f411566 : SUCCESS in 26m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2507bca21e714bf180cf83ee63a1815d : SUCCESS in 54m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2663556f62024b77bc03c049df02805a : SUCCESS in 1h 16m 47s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/62cf7ec907834a98bc7ea83c660a9a4f : SUCCESS in 1h 33m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6865713f381d43da8a3fc2a5a93a97d0 : SUCCESS in 2h 20m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/b908ad2191c74f6e8c289e6610ed232e : SUCCESS in 1h 17m 44s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a546542170434dcdbd483a4374d05dc1 : SUCCESS in 22m 51s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e64ad06e333e46809e1efa4ed6b8e68c : SUCCESS in 47m 57s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ed63236982b9447db2b18ac06679743f : SUCCESS in 49m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/52434ebf115a41a8af3b3aa6ecb01a34 : SUCCESS in 1h 15m 38s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/37fd6b0442f9419b95f4f152bbeb98eb : FAILURE in 26m 10s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/8791259b23854466acc5c2ea13fc0f07 : SUCCESS in 1h 58m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4743c7c86bc817712065232131636d3bf28d014f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-25 12:59:11.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cef281ae412c7034b903aa3335b8698fc5607493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 12:59:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e18f1893e0fc9b4bd7367f5f8581f3a335f9accb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 13:25:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2282416e784e499a88af3a2f7c83bdf6\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/9a28ca4ddc4342b885044441bc2551d3 : SUCCESS in 39m 28s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/19556b0f16934b11b14e9a2f6a2d8d38 : SUCCESS in 1h 23m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4442e0b2735e41ecbd3b0a58a79bb439 : SUCCESS in 34m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/86b46f98a325450bbce9a05b34e2982d : SUCCESS in 2h 13m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b872868c9ff5407eab45f006986c77ee : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9275fc7272b14a11af850a114d18cc74 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b78f7587dc814beea2dc01c707cdc380 : SUCCESS in 12m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/60e019dd2a434a26a0d46b3894e43ab0 : SUCCESS in 40m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a231eacaad86a8a482f92a4f112f000349e990ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 14:13:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a22a94e5e18942189a4b594a0454a269\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9171c5cb0d9f460d8bb892c2c359a279 : SUCCESS in 24m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8347d4675eb248f697858e8a6d06b050 : SUCCESS in 15m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a7582bedf3ef433dbf463cca16271754 : SUCCESS in 42m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32749247d2d948b492650d3ff452cc11 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba7a0b6608664f77bf880d6436d8a730 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8244cdc0ec9448eb8264c33a5c4e81f8 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb20080bcb504eebbb3c0abe7eaa65a7 : SUCCESS in 7m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8530e04014a94cd0b5e6734cd27424c3 : SUCCESS in 4m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e677a4a0ebbc4bb39dec4522a07f641c : SUCCESS in 22m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cfdf36014d99405fa49647bfab2dfcd6 : SUCCESS in 20m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d846eac55a6046f2aa031777af61d3d7 : SUCCESS in 51m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/37851f320f4045bebe7e549a158a72aa : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ceb171ff5a6af0debb8af08357fc0a760cf2dac5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 14:13:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4c5b6ae83656b5c41263eeb9ccca64bfc36618e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 14:15:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e00bf3b9025473399df4c79c60ee761\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4103f1629ee84fe8a539d3d9a097eb28 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cabc3627e84e93453858eac1a310e89f918d6c83","revisions":{"9d67565689d78375dd18c6bbc667db1fac6cd128":{"kind":"REWORK","_number":1,"created":"2025-08-20 13:19:26.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/81/958081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/958081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/958081/1"}}},"commit":{"parents":[{"commit":"a40ad7ea0a6418cc0a37acb196dbfc52bbcdf32b","subject":"Merge \"Enable Barbican build on Ubuntu ARM64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a40ad7ea0a6418cc0a37acb196dbfc52bbcdf32b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-20 13:18:50.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-20 13:19:24.000000000","tz":120},"subject":"[release] Use Flamingo Ubuntu Cloud Archive","message":"[release] Use Flamingo Ubuntu Cloud Archive\n\nChange-Id: I92f3095b2d51bd6ac7ff48b0c7d79852eccc9803\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d67565689d78375dd18c6bbc667db1fac6cd128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d67565689d78375dd18c6bbc667db1fac6cd128"}]},"branch":"refs/heads/master"},"3b38b7b66bafdc40d3933801f12328777e3c0b45":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-25 08:15:23.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/81/958081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/958081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/958081/2"}}},"commit":{"parents":[{"commit":"a40ad7ea0a6418cc0a37acb196dbfc52bbcdf32b","subject":"Merge \"Enable Barbican build on Ubuntu ARM64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a40ad7ea0a6418cc0a37acb196dbfc52bbcdf32b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-20 13:18:50.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-25 08:15:23.000000000","tz":0},"subject":"[release] Use Flamingo Ubuntu Cloud Archive","message":"[release] Use Flamingo Ubuntu Cloud Archive\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958070\n\nChange-Id: I92f3095b2d51bd6ac7ff48b0c7d79852eccc9803\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b38b7b66bafdc40d3933801f12328777e3c0b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b38b7b66bafdc40d3933801f12328777e3c0b45"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cabc3627e84e93453858eac1a310e89f918d6c83":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-08-25 10:34:08.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/81/958081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/958081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/958081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/958081/3"}}},"commit":{"parents":[{"commit":"a40ad7ea0a6418cc0a37acb196dbfc52bbcdf32b","subject":"Merge \"Enable Barbican build on Ubuntu ARM64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a40ad7ea0a6418cc0a37acb196dbfc52bbcdf32b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-20 13:18:50.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-25 10:34:08.000000000","tz":0},"subject":"[release] Use Flamingo Ubuntu Cloud Archive","message":"[release] Use Flamingo Ubuntu Cloud Archive\n\nChange-Id: I92f3095b2d51bd6ac7ff48b0c7d79852eccc9803\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cabc3627e84e93453858eac1a310e89f918d6c83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cabc3627e84e93453858eac1a310e89f918d6c83"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
