)]}'
{"id":"openstack%2Fkolla~958088","triplet_id":"openstack%2Fkolla~stable%2F2025.1~Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-21 07:03:20.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7","subject":"Influxdb: use stable RHEL repo","status":"MERGED","created":"2025-08-20 14:03:38.000000000","updated":"2025-08-21 11:28:32.000000000","submitted":"2025-08-21 11:28:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958088","cherry_pick_of_change":957680,"cherry_pick_of_patch_set":3,"meta_rev_id":"b6a7e0aa0efd21ca72d424e97386d8e17f607c28","_number":958088,"virtual_id_number":958088,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-21 11:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-08-20 14:16:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-21 10:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-08-21 10:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-20 14:16:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-20 16:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b227df346e8e0cf51c629940a80c253b7524253e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 14:03:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/base/influxdb.repo","accounts_in_message":[],"_revision_number":1},{"id":"1bc3ba95de823313211ca90cb1741822e78ad7b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 14:04:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"43f59f5e47b383012f078b1ef4dcaf4d89eb81d0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 14:05:21.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"d87fd0e6240edfd3553d6a59cbe9ea369c861bd8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-20 14:16:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fcc354acd2db514121c735be75e29d78750a4bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 16:14:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd7a91eef3eb4c6ab6bb5cbd2fd278d0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/01a855d249364bd08aa9226c66453780 : SUCCESS in 15m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f983f065ba1249b1941022ab458b74d8 : SUCCESS in 33m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/eb4180ebe3aa4424ad51607a511a15eb : SUCCESS in 16m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d2a8e564924c4d4286ccb842c6e0686b : SUCCESS in 25m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/40c1edc7459142809a9268738abbb5c1 : SUCCESS in 1h 11m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/44e9397b88944329bea9c4b399eda803 : SUCCESS in 52m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49bc66315eed48d19dcf4baa638795f1 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/369e15b1c3a647ee91958215dfe4c3cf : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82272a22246a4584a0128ba6ad8ee9ab : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5954bd841bea48dd90b924bd163f5005 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2373e9953f874ba0ae166c932220d4a2 : SUCCESS in 3m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/60d3e34ee6be4839943ce5d5ed531229 : SUCCESS in 2m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bfcf101987d54e688f3857142e45afa6 : SUCCESS in 27m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dbbacc7c4eec47cd8cfec769ff8f0770 : SUCCESS in 47m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d2132b2a2cea428a98fe3037714d3f7c : SUCCESS in 37m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/26b99d1ba6ee4a59ba6974b32bc918b0 : SUCCESS in 1h 23m 33s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/825c854a62774df8a16175ae5fc5cfd2 : SUCCESS in 1h 52m 48s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4d459d36648c4438a81644cab079ed35 : FAILURE in 58m 25s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/92bf199b825b4015acbae3149173dc51 : FAILURE in 37m 57s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/1542189a42ff4d8c806c2b2500bf6c0f : SUCCESS in 18m 34s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/cc5072aae4894ab9a8625b91fc6316ad : SUCCESS in 51m 25s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1f86f3b4c5d14932a242660c40b5e579 : SUCCESS in 1h 15m 15s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/481c48648ed142aa8f1b18bc9460d613 : SUCCESS in 1h 37m 49s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5f8e16f727464a998f67d9c46f1f8c7b : FAILURE in 14m 58s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/265334873c754182824d10f5cd42b244 : SUCCESS in 24m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d5ccc2d1460c48fda902eb9e027e6f1b : SUCCESS in 21m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/694d6802fd934b499ca463ff08204a47 : SUCCESS in 53m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ea24928fbd3849e4b5ee40cae4d96c0a : SUCCESS in 52m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/aeeccc73188e4ff79a1295e1d5ff49cc : SUCCESS in 1h 10m 30s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a9c37fd7a64f467fb9fc19506e5cddcc : SUCCESS in 1h 34m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/cfcdb30d83e848e9a5c4b32ed4261b57 : FAILURE in 37m 50s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/02799e8b547a4e28a55b9ad54264cb1e : FAILURE in 26m 42s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/38efb958081d456e9028ecb07de60453 : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8ec2a71153ed4dcaa49b184fc0bb674c : SUCCESS in 33m 03s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/65e0c66d5bfd4a94950602bd175e29f6 : SUCCESS in 1h 44m 45s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ec2b8c0844fa489b886b7b15a77b12cb : SUCCESS in 30m 21s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/35071f97e95b4c369b45a505ec5a29b8 : SUCCESS in 2h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"284cbf54e51e3343c6b998ad2df0539e93e369ef","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 18:15:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3405ce308c2a60aaf317bd214ac1edf9a6f3303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 20:53:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74c8d13c721a4ebbaa9c27263837a907\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e4eb3601be294c87b670ff0c5f45766b : SUCCESS in 21m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0ad427b22a0b46909e7561c6e8235219 : SUCCESS in 58m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8fa80c390ef84717bf9f1a3ff4a468d1 : SUCCESS in 19m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/33aa661806164b5e80850a883443763b : SUCCESS in 21m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/64d13d45c5d6471c9dc23868a5f4b68f : SUCCESS in 1h 05m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4fe46acfa1ae495caacd8994ce18d0bc : SUCCESS in 1h 37m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43018c74da9444a0aa7a1f1cc30de95b : FAILURE in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65332d169ab148a8a156f343074b8695 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/547c26f75ee541268c4e90cb8f4c75b8 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f59c5ece8a94bafa8e431431a601144 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533a4bb137e145728f664e23378cb76c : SUCCESS in 6m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/39777deef84d4fdea4a06c095b0b4b5e : SUCCESS in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d30c011f9f684da09ee2e62badf1f24a : SUCCESS in 19m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1d573864e7de43448642d3d36da1ecc0 : SUCCESS in 1h 14m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4d80eeedbb67477ca8710c9c28e57437 : SUCCESS in 1h 10m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2c4f3a405d2a4c4089127dedb644fd28 : SUCCESS in 57m 53s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0452c829753144cd8862550cbfa80074 : SUCCESS in 2h 15m 44s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0128f7742b4148488a5c47f75a1ce4c3 : FAILURE in 1h 08m 46s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/fbff3c8bc6c04b079d889a92a460e8b0 : FAILURE in 40m 19s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/16399696022d4b1fa8bb33c874ed7bcf : SUCCESS in 28m 32s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/570a6a71c4c64bdd92bc072798d28d76 : SUCCESS in 1h 00m 58s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a2275e5c39f84b50a311b88985356c5e : FAILURE in 57m 36s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5a81434f182e44aaad41bdb204c7658d : SUCCESS in 1h 38m 43s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/19e674ab433d4422b4e95932230616f5 : FAILURE in 27m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8c00963d29964acdb4c1a975b1e5438a : SUCCESS in 15m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/55da622a9dcf43bdaa35e71a5a6ba61f : SUCCESS in 24m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8b2b2513cac848419c242e75b52acc24 : SUCCESS in 1h 03m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6a4060000e60429eaa31a2f26dff63c1 : SUCCESS in 1h 24m 20s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/cd9a591d12534931b525e03654e2be31 : SUCCESS in 2h 00m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/89aa1c3fda8d432393a6713df00475e5 : SUCCESS in 2h 31m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/72166689778b465bb11e31afa41b62e9 : FAILURE in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5946a5522a224e5fbd6123e7845b1450 : SUCCESS in 28m 07s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/07daf082ffa14ee5b672264ab9ce40f1 : SUCCESS in 34m 57s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7174f10dcbda40e99597dbbac38deccb : FAILURE in 58m 47s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9009f1ddae2843da97d59f825e248bc7 : SUCCESS in 1h 38m 29s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e1b85b36899f4b4698f86469aed26f5a : FAILURE in 18m 11s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2e2390e5a94a49ec95730e9faa8a9e7a : SUCCESS in 1h 52m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef15168ce60bc1e4e612de240419750a99752cec","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 21:07:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03499387b406602f186d966cff51256eb63d7d0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 21:19:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55086e427a3140eca43a3c8d19710378\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f650c7f3acac44a2940e6c9a0014fe7d : SUCCESS in 55m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/873ff74923fb495999e2490c564a4376 : SUCCESS in 1h 28m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7071fc6d31464b2aaac22af39ec8fcc2 : SUCCESS in 1h 26m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d2dc1fbbda04dc2b167bba9b4739c2e : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53bacd9850034824ae38c5fdc2de29de : SUCCESS in 4m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3d23dd61da0a41b8af599b3969859dd0 : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8dea7c5ffaa9467ba3cdd2ab3fc82b5d : SUCCESS in 1h 46m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"593ae3361b659c4e776c8bd960dd4c9661a18ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 23:04:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b183eadce85d46328d31ff9e304755b4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f59426b2648547a58a5466f4c55d261e : SUCCESS in 22m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4ebeb13ca7ef4b3daa167a13548d8209 : SUCCESS in 49m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2a8d764ab86e468a9b225646ce7aa08c : SUCCESS in 14m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5518f472a9d046c8810df09b03365059 : SUCCESS in 31m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9649a8d82dba4546ae54509519f32592 : SUCCESS in 1h 09m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/751c0955b44d4b6bb2030470d1221c87 : SUCCESS in 52m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41ba0e9061847019d93550da606cff3 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/153efcc5f6504e079e4b67338bfd617b : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64aad98a572d4938910030738db49838 : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41a85682637b48d289f9770ab88fe090 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c25d0223dba48f0abb947ed9c5f9e1c : SUCCESS in 7m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f10aaa0531964b31831907351713af39 : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a5b85d82f7ce49689ef3a34b6de7a868 : SUCCESS in 21m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5e9973f979ba44a6913c830f59bb0274 : SUCCESS in 1h 10m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/82e6bf6cd1834c8fb66c1b66cd5c400d : SUCCESS in 52m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5f4faa80c23b472fb093961330e2fae3 : SUCCESS in 1h 36m 00s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8719ed1cbfde4a8a9d7e2ad75b13a9b0 : FAILURE in 1h 48m 28s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0b0e743151094dd2a2b36617f35169ed : FAILURE in 1h 01m 46s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c3c3d127e8de4a959093d492d032c432 : FAILURE in 1h 06m 01s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/540226124f0140749dc42c9e5f380bca : SUCCESS in 26m 26s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/cbf5ee60d646471caf3a0f3afe1e4c24 : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a65029370d644e37b15cfcf986daf314 : SUCCESS in 38m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/cfe660e501da4f5286e74e642d32887a : SUCCESS in 1h 39m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea2942c580124c0fb56f06802dea438e : FAILURE in 19m 11s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7807029d6a974ac1b991334207c4b76b : SUCCESS in 21m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/18d07c53e8c04fe8a3222aa037b0a151 : SUCCESS in 17m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7b86cbc4fcca4ecc91cf6892ca3ee8f7 : SUCCESS in 1h 07m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7b50e60abfd74208b7cd3442d06bb502 : SUCCESS in 53m 09s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/18ac8065a91647acb18285bc7041bcee : SUCCESS in 1h 52m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d9fd4306f7e4485ab89e84c47e5cfffe : SUCCESS in 1h 28m 42s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/785723ada1f041ce81e1492b198a5da3 : FAILURE in 1h 09m 52s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/20b9f56bb5f540e0bad75c4a283caf6c : FAILURE in 24m 16s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/81d1aaa3045745dfac226a01719d0833 : SUCCESS in 1h 01m 11s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/648836f7bdc042d3a665031267c98077 : SUCCESS in 31m 38s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/587e98a00df7493c87a13ddbb4978c13 : SUCCESS in 1h 32m 24s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c821d4fc8040415cbcbb6508e65d6bc4 : SUCCESS in 32m 53s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/d1d9f367a5704bce9bcd2c5fe5c2912d : SUCCESS in 1h 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c33013dde731799c384e4eb06d5d0809f9456a3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-21 07:03:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a795549d81a5594e7566cf9b40b7be5c831b75f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 09:33:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ceb94698584c25b6e2c2467702d0df\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0f656b47c1034c84924390943543cb62 : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9efd8c2a66a64bad8611c898e73705af : SUCCESS in 1h 34m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/30d40d0cdd8a43d2b8c33f1aeb990449 : SUCCESS in 2h 27m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c9cf798a76e494fa2ad4c84a2690aa8 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c9911ce5e424e91825614e5d6b4c85b : SUCCESS in 6m 03s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/aa207fbabb6b4c4aaa9f097cfbf56358 : SUCCESS in 1h 07m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c4c3afbd92964b48857b7c69d91fbc3f : SUCCESS in 1h 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db4032d0e104cce600f46b9680f1d67e8d6cb9eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 09:59:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6267b3a859594e04874d577c23da8915\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9f0047a2fd9e4eb89f098f687bd5cc03 : SUCCESS in 23m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e8ea1482e22c4081a9288409c3a30fc8 : SUCCESS in 1h 03m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4441cb4b48254f3386ccf2a87a6ee7b5 : SUCCESS in 10m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/84eeb941ee2a4978b1df75bb71e7c3ee : SUCCESS in 14m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2bb41f25465a44679d9762e1da3385ed : SUCCESS in 30m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/75aa81d29c084f118eb0579a87946349 : SUCCESS in 1h 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a97e4da421ae43218c42814d5312050e : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e28a8ef73e5432186a51348dd6db8df : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f092fdd3ad64f3ab31b6b9fe2d7a1df : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48496c47bc3d49759933d5a0e37642ba : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1904d20d03d40d7832d7c5afc96a314 : SUCCESS in 3m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/331c77a481d642ce8d6dcdce9f9393df : SUCCESS in 1m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e9f0e8ef9f6b4b27b549e59baf0aa80c : SUCCESS in 21m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fefc10e405ca44d782314ca19dbc3700 : SUCCESS in 51m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/82a8571335d048549b7c207f64ddf595 : SUCCESS in 1h 07m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3cb950551b974940baebca756a93920a : SUCCESS in 1h 34m 58s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1fca21e936514f549473fc11f8fa80d9 : FAILURE in 2h 07m 49s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eea66ed46bd54ca6a94c8148b7e6f5e0 : FAILURE in 1h 20m 20s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/cd94d7f166174205a89c197588aabad1 : FAILURE in 48m 03s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/a284da9e90594382829131c444c04680 : SUCCESS in 22m 49s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8882e39a97eb4fd4866271f4db4adedc : SUCCESS in 1h 11m 54s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/44c921cd170d403d88544abb29c5e479 : SUCCESS in 1h 13m 01s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c9ad344d98f046a0a339780a0d35456e : SUCCESS in 1h 19m 47s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6a8044ba23154a958ca1a65dd864c5df : SUCCESS in 24m 47s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6f70c3f2c44548389e31a18c464793e5 : SUCCESS in 10m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a48e0c08cde44fdd94c97d5955f8b840 : SUCCESS in 26m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/98c1ab4eafb04767bd2bbee47a0e4d2d : SUCCESS in 39m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bb1b5d895fb84d30a4c98489bda5e27c : SUCCESS in 1h 50m 54s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/026af2fb95fb4b9c96007fd9e1b586a1 : SUCCESS in 2h 04m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1e9eb7ee3f9c4b1d9801163ffa61f806 : FAILURE in 2h 18m 36s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/34da46d5149a4dcead5a03f7f44f2abb : FAILURE in 30m 14s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5b03e984f2b04c0aaca1d6ba1d2bbc6a : SUCCESS in 26m 12s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c3f299156d9845eaac1991820b0e3e1c : SUCCESS in 33m 52s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/371a1cce7fae4a58a322221f9d35850c : FAILURE in 1h 30m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2790bb5df6d34ad9a1658a989e083d2c : FAILURE in 1h 00m 16s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/51e3775943e345578b18bf1e5ec79b65 : SUCCESS in 13m 52s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/86591404aa2843e3a55aa0115b6f5580 : SUCCESS in 2h 08m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eb9e1e0cd6d356c880dd0269eb600b1fc3b43147","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-21 10:12:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d3b2958d4612852344cbfc2423415ac66d51b7b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 10:12:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfed7c18bdd4678ca37331ae7786cef70f46cbbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 11:28:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67c23809d973428eb213813a3045e867\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/95d4ce332d384c85a39524abcfb1c647 : SUCCESS in 9m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d2f45394d1944f76a893a1b8034107fb : SUCCESS in 32m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bdba814336d542ff9871ba4ec9d04e1f : SUCCESS in 1h 06m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4775648210784159820eaadc126a3204 : SUCCESS in 47m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a269ed85104508819ddbadd705336d : SUCCESS in 2m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71be8b55f4aa44cc831858ef038cb606 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08278d6383db400188bd2a89d866d87c : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c506022c322548cb9ff809a19b61047f : SUCCESS in 6m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/106689f2bff9440d98d8321385c3b368 : SUCCESS in 2m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8853b6be212c448b91cdbf5e70613385 : SUCCESS in 20m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3dea9dc88e1d45bbb90840972e53cde1 : SUCCESS in 52m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/13e5a541ef744a9a8d5b92ade6adebef : SUCCESS in 1h 09m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/70c4f88e80c1417691ecde0a748e2b85 : SUCCESS in 48m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c68709a937ad420fbdc530141d797f24 : SUCCESS in 21m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4eddcb1b22ce466f8df3c6cfec4defdf : SUCCESS in 16m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/87fdd6fa45e64b2ca73127ca73802154 : SUCCESS in 1h 10m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/36165876d8374b10b45376f665342b6e : SUCCESS in 49m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b6a7e0aa0efd21ca72d424e97386d8e17f607c28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 11:28:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb723cd26cc9251781c4581967ecdb5975a5ad6e","revisions":{"1767128396d66e0452f251a3ad5a93d5db5b7bfa":{"kind":"REWORK","_number":1,"created":"2025-08-20 14:03:38.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/958088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/958088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/958088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/958088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/958088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/958088/1"}}},"commit":{"parents":[{"commit":"171f725a9faa0965d79bba6a4e0bb31e0d3a8912","subject":"Enable Barbican build on Ubuntu ARM64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/171f725a9faa0965d79bba6a4e0bb31e0d3a8912"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:56:54.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-20 14:03:38.000000000","tz":0},"subject":"Influxdb: use stable RHEL repo","message":"Influxdb: use stable RHEL repo\n\nSwitch the RHEL repo to the stable channel as per [1]. Use new GPG.\nTelegraf arm64 builds are enabled again, as binaries exists.\n\n[1] https://repos.influxdata.com/stable/x86_64/\n\nChange-Id: Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 444a81b62bb5ccb571f82e34f4ce0c48d8f639bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1767128396d66e0452f251a3ad5a93d5db5b7bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1767128396d66e0452f251a3ad5a93d5db5b7bfa"}]},"branch":"refs/heads/stable/2025.1"},"eb723cd26cc9251781c4581967ecdb5975a5ad6e":{"kind":"REWORK","_number":2,"created":"2025-08-20 14:04:57.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/958088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/958088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/958088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/958088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/958088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/958088/2"}}},"commit":{"parents":[{"commit":"171f725a9faa0965d79bba6a4e0bb31e0d3a8912","subject":"Enable Barbican build on Ubuntu ARM64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/171f725a9faa0965d79bba6a4e0bb31e0d3a8912"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:56:54.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-20 14:04:57.000000000","tz":0},"subject":"Influxdb: use stable RHEL repo","message":"Influxdb: use stable RHEL repo\n\nSwitch the RHEL repo to the stable channel as per [1]. Use new GPG.\nTelegraf arm64 builds are enabled again, as binaries exists.\n\n[1] https://repos.influxdata.com/stable/x86_64/\n\nChange-Id: Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 444a81b62bb5ccb571f82e34f4ce0c48d8f639bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb723cd26cc9251781c4581967ecdb5975a5ad6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb723cd26cc9251781c4581967ecdb5975a5ad6e"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
