)]}'
{"id":"openstack%2Fkolla~958284","triplet_id":"openstack%2Fkolla~master~Ia71614e6f838ecc121d303a49ffd8e2d5a15c1aa","project":"openstack/kolla","branch":"master","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-22 18:56:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-09-01 10:29:41.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}}},"hashtags":[],"change_id":"Ia71614e6f838ecc121d303a49ffd8e2d5a15c1aa","subject":"Revert \"Added systemd-standalone-sysusers package for Debuntu\"","status":"NEW","created":"2025-08-22 15:15:13.000000000","updated":"2025-09-01 10:29:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1f9bc811b389dc32e83db99b58e889b67b8d5245","_number":958284,"virtual_id_number":958284,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-22 17:48:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-22 17:48:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-22 18:56:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"07eb5bc5d912d2b90ec28d2bc5fb2b5d18921a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-08-22 15:15:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec87c1d99a87a6658a9fa9f0f90a9af0bf21d347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 17:48:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76d14809d1ce4a74983501c545bdc997\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/1adf44a3e6a84571a9e18dc51b6f6166 : FAILURE in 1h 00m 19s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/ac515c2586a6427a8141ac74da7c8fd7 : FAILURE in 1h 03m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/809248550e8a4d1281a5f54cbcdf0ef8 : SUCCESS in 21m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b1df5451da7a44ee96e2f6042f89169d : SUCCESS in 31m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/53c04b3b291e43d0a5d13c0bf6dda424 : SUCCESS in 44m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8d9e088feeca4cddab1e4ca341296c22 : SUCCESS in 1h 04m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e58c0ee846d04db59eaac0208f925817 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccc5d1b8ceef409a8e12c60207c824f4 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/023e61ff832d4bf3b8b1ad5e0514a485 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7472d1822800448084aae15485986d8c : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d92172e68b941d5b4b19b0cfd613fde : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70852919abd24c04a85b2f8a908262ce : SUCCESS in 6m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b02ad6f9fdbc48eeb76168acca691a63 : SUCCESS in 3m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ebeab5001fb74d5fb0e47ac1c102d696 : FAILURE in 22m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e08bd4aee8de4f03b1d480485c8fef35 : FAILURE in 30m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/012776d816f74af9a804f7f35cb60ace : SUCCESS in 1h 13m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0c2c38cb3ec84905b436d0bcafba0721 : SUCCESS in 1h 30m 51s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2f1227f253314d5288ed8d6762df0cca : FAILURE in 1h 38m 10s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/64682a0cd2104d4d9a9b71ac73a06f81 : FAILURE in 2h 27m 02s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/792cd89f04bb4396b1e4c19c6f2a4b28 : SUCCESS in 1h 16m 38s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6db36929fb54447b942c365c5af76711 : SUCCESS in 15m 46s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/bfa51520483449e09b5691e5d255bf20 : FAILURE in 51m 52s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4faa9fca85e144e996ec8ec7c3a8ac41 : FAILURE in 39m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fb4f810063db4b0bb466d802c09a2875 : FAILURE in 1h 15m 33s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2c53b7f84b9b487991ca40c3731ff108 : FAILURE in 16m 22s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/a5a17706a5bd42b4be2ee2df5dddf53f : SUCCESS in 2h 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf4d23a5b38ce8d2afb914d9c5f59530d52be79c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 21:56:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9eb2ab638c04149960c147e59a6b746\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/9ddda8540df344d58e82aaa07a0cf255 : SUCCESS in 58m 44s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c3af9a6df6124311b6b7260e07ad1884 : SUCCESS in 1h 50m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/837887bdc1ee4c0f882393e0011e8842 : SUCCESS in 45m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25ebe2b4c64d4db78b78bc27e1abcce1 : SUCCESS in 1h 53m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/51a7bf5c19b04f7fb8670bd512284eab : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2a7ce1666184bd9bc030393fca22229 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/529dbccfc07c4a48a8f3601a2153ead5 : SUCCESS in 7m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/20a485f2f7d84e1db6eb7ea8e0063618 : FAILURE in 45m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f9bc811b389dc32e83db99b58e889b67b8d5245","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-01 10:29:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74b6827add5eba2464e7d9b0445754003644ba8d","revisions":{"74b6827add5eba2464e7d9b0445754003644ba8d":{"kind":"REWORK","_number":1,"created":"2025-08-22 15:15:13.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/84/958284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/958284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/958284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/958284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/958284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/958284/1"}}},"commit":{"parents":[{"commit":"c93cf1be3e54accfbbfdb0600a02fb64637ec98f","subject":"Merge \"Update information on RabbitMQ and Erlang version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c93cf1be3e54accfbbfdb0600a02fb64637ec98f"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-08-22 14:57:37.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-08-22 15:14:45.000000000","tz":60},"subject":"Revert \"Added systemd-standalone-sysusers package for Debuntu\"","message":"Revert \"Added systemd-standalone-sysusers package for Debuntu\"\n\nUnfortunately this causes the removal of files in /usr/lib/sysusers.d/\nby removing systemd. This causes various groups below 1000 to\ngo missing in the container, and other groups to be mapped in their\nplace.\n\nFor example, systemd-journal is mapped to GID 999 on Ubuntu Noble. Using\nsystemd-standalone-sysusers results in group \u0027input\u0027 taking GID 999. This\nmakes it difficult to grant the Fluentd user access to the systemd journal\nfrom within the container.\n\nThere are other ways of fixing this, I\u0027m open to ideas.\n\nThis reverts commit 55168f909d3c95b6dfc386d3fbee25fd20a7cae8.\n\nChange-Id: Ia71614e6f838ecc121d303a49ffd8e2d5a15c1aa\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74b6827add5eba2464e7d9b0445754003644ba8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74b6827add5eba2464e7d9b0445754003644ba8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
