)]}'
{"id":"openstack%2Fkolla~958638","triplet_id":"openstack%2Fkolla~master~Ib9805fb6a93a0da887918174d9994caa5f1306d3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2025-09-10 08:22:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9805fb6a93a0da887918174d9994caa5f1306d3","subject":"Prometheus and core components update","status":"MERGED","created":"2025-08-27 11:30:35.000000000","updated":"2025-09-10 08:24:39.000000000","submitted":"2025-09-10 08:22:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958638","meta_rev_id":"14e6514f4ec867391b73b84d7b54e01deb0af7b0","_number":958638,"virtual_id_number":958638,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-10 08:22:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-09-10 07:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-09-09 16:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-09-09 19:08:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-09-10 07:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-09-09 16:55:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-09-10 07:00:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-27 12:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-09 16:55:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-09-09 19:08:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-10 07:00:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"15d4666aa207e0cf7e1caa505c4fdcd4ea071bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-08-27 11:30:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1649aab2fa3b869ba380750d551df0398acd967c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 12:44:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c654a2ee3ba74d139e6bed18b7cebc34\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5ebfeabe4e6d4ac1987bb86eca985ec6 : SUCCESS in 16m 27s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/3ea28bfae3e64aff970c8fe6739d984e : SUCCESS in 1h 07m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1f15a1030bae48a58d96ad2a5510f99e : SUCCESS in 17m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/815bef2160794524a5e1b0892b977305 : SUCCESS in 24m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/afb29b1a0eeb4ddb82e8006830773843 : SUCCESS in 51m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/40213e4d4df94466b87590f92d544037 : SUCCESS in 1h 12m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f38cf8164e2e49d39dafbd4921c22ca9 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37ae4d46c6224b78a6e7e3479ddac665 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5fb41b13713d48e8a7173b2e7e5427a8 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63b85e1657fe4707b56f289b8e25547a : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c54a2d6550ed4295a4ec76d14edbe220 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4909ff1e31704bd085cd1b7a2d614fa4 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76e9c62a637943d39933bf957cd23554 : SUCCESS in 2m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3d4751b4afbb47f9b119fd29bf29dbb3 : SUCCESS in 2m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d2029c66bc6842de86ac4bf84b56671a : SUCCESS in 11m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3aed864a049a425194a33df759c77866 : SUCCESS in 26m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/261f5b7c55734e81b16a0a410b09ccef : SUCCESS in 48m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9b134d61292748eda47d95f8b5efe521 : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f18276a457dd640ca2b77a470d55116003d90954","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 16:02:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb2e51c9e4014bca9e35c2dd63f001d5\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/25b5e974495c404880bf0a92744014ba : SUCCESS in 53m 23s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/df015cee147c482a864b9b24e522102a : TIMED_OUT in 2h 33m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/73266e74c3ab423784f6b060bfbbb18c : SUCCESS in 40m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/32df1ee0644b41e0bd51c7744e487e2c : SUCCESS in 1h 33m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b6243a35c2264418aecdfb6b08486c7e : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/26e341985c1b4909b3f07e65683b7052 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/57f473c7896147bb993d21041eea7b4a : SUCCESS in 10m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cdee9c8506cb4052833fc12154606349 : SUCCESS in 56m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e9844551b83d08ab8612453d8605b3bc30e6ed6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 16:55:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a41c363f45de8b60fb419e20a4ec87f18fb19a5b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-09 19:08:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc67a04a6e8419383bca78b001ea27791d505502","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-09-10 07:00:14.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6019ab11347af6257ec92117ad3f69becfed9576","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 07:00:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4da46b38866e8c8f667feaf44ad7c922b9d4380f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-10 07:13:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ad8acdfe1f0e07d27ecab9ec776397426dcc044","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:22:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f2f45800e5647549f6033ece6e92439\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/db40a736765c4951ac39beb4c169a004 : SUCCESS in 11m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0c5245465e0c4a1491764b7607fc9d4d : SUCCESS in 26m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/92c8c92e6ff04b50aa749ba4f348f9e9 : SUCCESS in 46m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5f9c7d5c164844e28c2108958acc7688 : SUCCESS in 1h 06m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db562828493479392eb98a858e9f2c7 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a78ff3eedd6408faae8ae7d9b79478b : SUCCESS in 1m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51027ce0c6364007a7cfe33032e911f0 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a7f34059cb40f3ae20364b19e34edd : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69719ccbc22544cd9bdc629deecf9413 : SUCCESS in 1m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/50cabcfebf5548a09736c0bc7c1269fa : SUCCESS in 2m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f779b78f801e4d2ba54c3949ecd2dc48 : SUCCESS in 21m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c50204501b424bdfbb29a89ab997799b : SUCCESS in 27m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b1495d2bc2af4ac5b34e953db8856df1 : SUCCESS in 49m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73b88093d51f4c13b20db56a5393f4dd : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":1},{"id":"da6128e3bd0f29909b194bfbf26e37403dfbb5b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:22:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"14e6514f4ec867391b73b84d7b54e01deb0af7b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:24:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442d3fba8776436fbdd3c74da068b131\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d124af97ee64633b8c332bafcfc039a : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df703516517b49c4971aea2b602a4171 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67e37b703ad6745aee4feec1080ee44e8b01ce74","revisions":{"67e37b703ad6745aee4feec1080ee44e8b01ce74":{"kind":"REWORK","_number":1,"created":"2025-08-27 11:30:35.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/38/958638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/958638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/958638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/958638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/958638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/958638/1"}}},"commit":{"parents":[{"commit":"d3e3f6b4145bc887afcd550cda67388c0f82771f","subject":"Merge \"[release] Use Flamingo Ubuntu Cloud Archive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3e3f6b4145bc887afcd550cda67388c0f82771f"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-08-27 11:20:45.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-08-27 11:30:16.000000000","tz":120},"subject":"Prometheus and core components update","message":"Prometheus and core components update\n\nUpdates a range of monitoring and core infrastructure components\nto their latest stable upstream versions to leverage new\nfeatures and bug fixes.\n\netcd: 3.5.16 -\u003e 3.6.4\nletsencrypt-lego: 4.20.4 -\u003e 4.25.2\nprometheus-blackbox-exporter: 0.25.0 -\u003e 0.27.0\nprometheus-cadvisor: 0.49.2 -\u003e 0.53.0\nprometheus-elasticsearch-exporter: 1.8.0 -\u003e 1.9.0\nprometheus-memcached-exporter: 0.15.0 -\u003e 0.15.3\nprometheus-mysqld-exporter: 0.16.0 -\u003e 0.17.2\nprometheus-node-exporter: 1.8.2 -\u003e 1.9.1\nprometheus-server: 3.2.1 -\u003e 3.5.0\n\nThe Prometheus-server package has also been switched to\na Long-Term Support (LTS) version, which ensures extended\nstability and security support.\n\nChange-Id: Ib9805fb6a93a0da887918174d9994caa5f1306d3\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67e37b703ad6745aee4feec1080ee44e8b01ce74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67e37b703ad6745aee4feec1080ee44e8b01ce74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
