)]}'
{"id":"openstack%2Fkolla~958826","triplet_id":"openstack%2Fkolla~master~Ie7e607c5a9c6f8c2e36aea5d52dc2ffbe97dddb3","project":"openstack/kolla","branch":"master","topic":"kolla_drop_inspector","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-19 11:58:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7e607c5a9c6f8c2e36aea5d52dc2ffbe97dddb3","subject":"Drop ironic-inspector","status":"MERGED","created":"2025-08-29 05:57:11.000000000","updated":"2025-09-19 11:59:06.000000000","submitted":"2025-09-19 11:58:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958826-kolla_drop_inspector","meta_rev_id":"5488291041acb4652d4e5435036d14a71e1d696a","_number":958826,"virtual_id_number":958826,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-19 11:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-09-05 07:23:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-09-05 07:23:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-19 10:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2025-09-05 07:23:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-19 10:45:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-09-05 07:23:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-09-05 07:23:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-29 06:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-29 06:35:26.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2025-08-29 07:39:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-29 12:05:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-09-19 10:45:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"13f4212ccf59b1d6c35ad0fb9ed7c348e7c83feb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-29 05:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f094d82e60577c44e7ab5b3eaf6f7c006c0a0c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 06:03:40.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951864 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"dbdef2801af59af0b134f4947a4a30352c285be0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 06:03:43.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951864 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"00e3a7db14dd62e468f16e4a87e2101dc6ec5a6e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-29 06:03:46.000000000","message":"Topic set to kolla_drop_inspector","accounts_in_message":[],"_revision_number":1},{"id":"2ae3cb8cb4d5cc83637a50786205229a1dff478b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-29 06:04:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ada473138617034e055375a4178bd6ea702955ab","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2025-08-29 06:35:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf788eec600041f178c7f7203e78263e682ac680","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-29 07:39:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"79c048c9b8ab9c70ccab6faa76d4a7ff8ed69283","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 07:40:22.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951864 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"c0bc8f8bd7188836efbbb9dad8a6e7cfc2305600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 07:40:29.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951864 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"d661cf0081a65c1a0a2f2b680b7e2c5195cc5535","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-29 07:53:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1463a4120133f2284130431e8cb2ba59b36d17c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:14:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae66d6b6e7043e9af8e4b7e398c3a60\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/8d9e97bc2f534162b4984faa82ffd50f : SUCCESS in 23m 10s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/f2b0aff99e9f4927921185c68f99fb9e : SUCCESS in 33m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d1771eb808ee4b7aa6b224b39d39347c : SUCCESS in 20m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d65b0511c2644fe0a80ce1e0face0449 : SUCCESS in 23m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/70418b621c97463486b90431eff2f6cf : SUCCESS in 48m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1f28d3e00de04d49a8d0c83cc49b076f : SUCCESS in 1h 08m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb37180e5e1d431ab810a66029496ec4 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01498c11f92a49de94b6b97ec7ec1cc2 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe4a818d55254d2eb3db96a03a0ae164 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07f18e5a2dee4d72a532d590602c70c1 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18ceaebfd88b418c9b042db434fa4d05 : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90050d263cce414d8abff71db6536547 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71aa3f25bc98474a84ac9636d5fbfd6e : SUCCESS in 2m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c70c4394e80340d1a93275380689d257 : SUCCESS in 2m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a5698df4669145bc9c2ec95a11f580b9 : SUCCESS in 25m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a86d1e1aad624a66aef6d79caf2577b1 : SUCCESS in 15m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/82bc77b7452f484abe9b86f9af0041ce : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/53fdff2b847e4d69a935f79608c495ae : SUCCESS in 1h 16m 36s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/363e4251737c434aaa5ccf62912fa3d2 : SUCCESS in 1h 03m 38s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/db00d871383847fe9b2cfba4129d82ab : SUCCESS in 12m 43s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/9381a43a224e497db5d0a71a6b5e618a : FAILURE in 1h 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18049eb1995c06bd5c9b9f7af0aed7a1332d13c0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-29 12:05:20.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"a3480b9cdbfa8176cabf3afbc865b33609d6cd4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:45:44.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951864 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"daa066d92b029096ec5043111f95dd945195afbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-05 07:23:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ded2eaa4a5269cd8eefe577842efe47dc60d4075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 08:44:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe90585045ed41fdb3c27e8af12bf708\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/97663e7bd72244279b7a733aefc76bb3 : SUCCESS in 12m 41s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/dada4c6deb374c938a95382bd95295a6 : SUCCESS in 1h 15m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7fdca74523db496b80188f4b4dbc0f72 : SUCCESS in 19m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/41ee3004d2424f9e9d56d3ca26c5724d : SUCCESS in 30m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f951ba51f475495a81ae2ce1df2cd181 : SUCCESS in 46m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3c208781c4b94b21b479d3bbe66c930d : SUCCESS in 1h 05m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c4c6e38fd3f4d9c8c4adac21644a88e : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45a4b744d0814c619dfecf5505e57d24 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38d1d6caf2ed4606b96b889474ecc863 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2fd7cb6081e4ff0b388ac6aa8871c7e : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b998b2dd2006401192eaa65de9859d4e : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d403619caa1f47f2bb533e7ab19f3a31 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef9f93bfcaea4790ace26fd919bfa665 : SUCCESS in 3m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3adef04597ca40648a76a429e706f2da : SUCCESS in 2m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e4142e9f708642ce9ef03abfe6b13473 : SUCCESS in 22m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b60da79e00b434bb1bfc9e86243815b : SUCCESS in 31m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ec2e7067a1c44a588c8553896c1c7ad1 : SUCCESS in 54m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/77173344cfca496bb51abc35eceb0cf6 : SUCCESS in 1h 12m 29s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/eb8691e5f7384ca18b830383f99570e2 : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/f559c8b855d640979f0bec53348fc3b0 : SUCCESS in 21m 54s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/524bb3c318454090ac551a6ec743d739 : FAILURE in 55m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e4a9a41a582c8de9d925d6134927c7d13b352910","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 09:27:02.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951864 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"9b12f737620182ae009236e68b70ba17ac86e226","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-05 09:27:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"acb21efaed8aace5048cad2fabc1bb53cdc33cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 10:52:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f092cc6f3443b78bdb12f58c4e277b\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/8fa008ca30dc4a7ea99e2755d87da588 : SUCCESS in 20m 12s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/6dfe86c7ec4348a79e619971c34b29bc : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/adc5af07994b489d854920021bb60374 : SUCCESS in 20m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ca16b2d08bc44c1ba541ceb58cfc55b4 : SUCCESS in 25m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d6e9b5eff81b404fb93aacf6ac4df38b : SUCCESS in 43m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7ff4e86461fa44b0a6a2fba8dd7a112e : SUCCESS in 1h 05m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd556737b93472eb9c4d50d83a8ebee : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8b924bdbaa84a8eb9187cbe0635ccb0 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/574ec5d57da94833803ccd91ad810fd1 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/025adcf30d004b338b393d9d71949473 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cce0cf4ffe034d4da6ca1fa44cee1aa3 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229f1729eb9e4274bb0435ba0b369c19 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1174b4170524fb2b00973625f4e3c38 : SUCCESS in 2m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/243dc605d76442d48958c388082c896e : SUCCESS in 3m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4cbd0d8a6dd64b609f713f3ddae065ae : SUCCESS in 13m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f5da1f4665cf4215b57599ed608a5d06 : SUCCESS in 30m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/63edf78b987547109bc77bc05605ad86 : SUCCESS in 49m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/290ce26c7e3547538794e8b8460df5b6 : SUCCESS in 1h 13m 23s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8f05e4c6586f4a2abb76e8bb3c82b8a8 : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/6d506d77c66348bab32e9b1e372f6608 : SUCCESS in 24m 01s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7edcdc2708ba4ef58b487a6ebb3d6479 : FAILURE in 1h 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"499d3b96dda90f6ca615f0156ac692444d1d4f2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 12:28:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebd97e46a14b4b588fb1521d45602e63\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/0d26ea6f6d554358ae9f39a049b22e89 : SUCCESS in 48m 49s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/4f242d571b25468887b936a7e78cd379 : TIMED_OUT in 2h 32m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d005b85731674a78bdd655ff44fb7892 : SUCCESS in 1h 08m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/636d88f38e5d4da5b23f197e0a2262e1 : FAILURE in 1h 56m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb2dec885da548db907889f05660e42d : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e31d948eb5b46ba8d51634c43db67fe : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2ae4aef7ac7a477a8853117997679c3b : SUCCESS in 10m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e54075db9f9b4813bf0f77ff214df57f : SUCCESS in 1h 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b62da220322c85b3e607fa5ddfdf45736fc42698","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-19 10:45:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aed51621e08ec1295a49549568b87b78846ec493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 10:45:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"717d35dbe0f436aae8567a02d985a4033e87fc30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 11:58:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d0f2d81def64e0ea8a911b89407f5a2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/10350a2a991648699a1798cdec3b82bf : SUCCESS in 14m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/66ee17cbeb1945c4a4df2fee250ff34d : SUCCESS in 23m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b334307fca1341a3adf4810b799207b1 : SUCCESS in 42m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/29adb7761e3848afa157d20d1e07a6e8 : SUCCESS in 1h 05m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fa5a0d18064412b8622fe956a871d20 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/810167a708ca48d2af01d9e920647af9 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aef91bafa4f44a658cc1fa2fbe2a8360 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4507be797e8448ef98d76a3b891aa188 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4717ddcbb34244999f81c709c93dabc0 : SUCCESS in 3m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b60add88024048f5baaee6fffa0dcb16 : SUCCESS in 2m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5f6afbad80e5486a83e796a0af4951b7 : SUCCESS in 24m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/832504c0f3fd4d42a10a9342721d1a0e : SUCCESS in 32m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/35681373972e48b6a59aba390009b581 : SUCCESS in 48m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7c32030fa7a54bd5b82f591eafbcdab8 : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":2},{"id":"df49819dbd47c8ce9762a56da658150c3b5f3372","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 11:58:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5488291041acb4652d4e5435036d14a71e1d696a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 11:59:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/944b718b37ee4add870aabaf36ad3643\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a27d2c48c94983994a593bfb5cfced : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fead358827c4731912d9d33be6b5ecd : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97899bce586d3a1713a9db20cffc2345625f064b","revisions":{"dceef910dc8c236a6b18ee46534f7968fdace076":{"kind":"REWORK","_number":1,"created":"2025-08-29 05:57:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/26/958826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/958826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/958826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/958826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/958826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/958826/1"}}},"commit":{"parents":[{"commit":"d3e3f6b4145bc887afcd550cda67388c0f82771f","subject":"Merge \"[release] Use Flamingo Ubuntu Cloud Archive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3e3f6b4145bc887afcd550cda67388c0f82771f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-29 05:56:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-29 05:57:08.000000000","tz":120},"subject":"Drop ironic-inspector","message":"Drop ironic-inspector\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/951864\n\nChange-Id: Ie7e607c5a9c6f8c2e36aea5d52dc2ffbe97dddb3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dceef910dc8c236a6b18ee46534f7968fdace076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dceef910dc8c236a6b18ee46534f7968fdace076"}]},"branch":"refs/heads/master"},"97899bce586d3a1713a9db20cffc2345625f064b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-05 07:23:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/26/958826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/958826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/958826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/958826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/958826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/958826/2"}}},"commit":{"parents":[{"commit":"20f371e01dc0c3629006e2deebdee3d1b86cc01d","subject":"Merge \"CI: add weekly COPR RPM build Erlang job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20f371e01dc0c3629006e2deebdee3d1b86cc01d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-29 05:56:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-05 07:23:23.000000000","tz":0},"subject":"Drop ironic-inspector","message":"Drop ironic-inspector\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/951864\n\nChange-Id: Ie7e607c5a9c6f8c2e36aea5d52dc2ffbe97dddb3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97899bce586d3a1713a9db20cffc2345625f064b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97899bce586d3a1713a9db20cffc2345625f064b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
