)]}'
{"id":"openstack%2Fkolla~959135","triplet_id":"openstack%2Fkolla~master~I3ff95aa75f1f1cccfa40f3ab7f216cb221a6af29","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-03 12:43:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-09-03 13:48:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ff95aa75f1f1cccfa40f3ab7f216cb221a6af29","subject":"Use versioned Erlang 27 repo on aarch64","status":"MERGED","created":"2025-09-02 12:28:19.000000000","updated":"2025-09-03 13:50:00.000000000","submitted":"2025-09-03 13:48:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959135","meta_rev_id":"f46ce9d896d5951e1f6bd8950ffaa0549828c9cd","_number":959135,"virtual_id_number":959135,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-03 13:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-09-03 12:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-09-03 12:43:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-03 12:43:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-09-03 12:33:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-09-03 12:33:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-02 14:42:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-02 14:44:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-02 17:52:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-03 12:33:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-09-03 12:43:16.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"}],"messages":[{"id":"faee354b7f5524a5d33bfcf260c8d53b6310beda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-02 12:28:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1a66d5c67fb4a9ffab9bd88b6d60a9eaf87f802","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 14:42:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aee8c26c21d45a2aa0f57a02b555c50\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f318dd2c98c64de585c41c5182009a4f : SUCCESS in 45m 54s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/d6ffa8b6f1cc43e3b65ec447f7e9348f : FAILURE in 1h 55m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/842ccb00f4ce45e0a570c37d1f6f6d7a : SUCCESS in 44m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/58427548ddf743f39e01e6d9b47ea22d : SUCCESS in 2h 11m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/92551c09aef447e096972ad74cfaaae5 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bea03d7531424abaa073ffa2a9304c1d : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4be58b48669b4b8eb27d6f73e091d008 : SUCCESS in 9m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/051af30e15a341538aa16cf8d854344b : SUCCESS in 50m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8babf6389a6fc88cfab8851a0179b06b09c10b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 14:44:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45a59b6c5cde4628a2e7f7ac2c20ad7b\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/2f17f9260b484c4d87df32df0db4bb6c : SUCCESS in 10m 53s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/7756664e2d3b49feb91f7973038effc3 : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/638c6d07b69542648872084ff1736c1f : SUCCESS in 20m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8aba97686a9b4ec5bbd16c5a0129e450 : SUCCESS in 22m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ece8f0d704714085b390f23cde9f21d3 : SUCCESS in 44m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6bb7381527a3412f8d5d795e2feb4fc0 : SUCCESS in 1h 06m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d955617b9b59445b94afd17f163a7b7f : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a87c86fe70104411aeea6f77d7b84002 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26dfed4a32504ca4b7b3696521ee562f : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0cfa33ad0734ec5bec96b196202b1b0 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2239e288951440109226c90c719f88f1 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6eb4a9d6f264a97bf296987980b528d : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a86a3915e1fc40c3be72538b82c440da : SUCCESS in 1m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bc71e7464dff4af2984347c96148384a : SUCCESS in 2m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0aa67ed72a7a4578a26cd116e1957935 : SUCCESS in 20m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5b4c2457c23640908f487823ab0f4910 : SUCCESS in 31m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/57976a741d694288b0049b2a138f9dfe : SUCCESS in 48m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d061e45b70ad49399347856d07fde6aa : SUCCESS in 1h 13m 28s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/3ab9dfd1a77a4bc68d3def34aded7fcb : FAILURE in 1h 24m 21s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/969b5ee44b2846908c48d74231551a23 : SUCCESS in 2h 12m 51s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/53c73f536ea14eb8b33d27dee0578df3 : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/bac6ce0911bb4bc3ae2cc55da6dab026 : FAILURE in 22m 40s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/376ebf2d76b643e388f4bc76a1fad661 : SUCCESS in 27m 39s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c3f22c93df2a4bc892521e4ca6b38cc8 : SUCCESS in 51m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a5a0f3b0c4d14199bed47388287d118d : SUCCESS in 1h 15m 54s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d9ef085e9e694c1aaeda99824f700ed7 : SUCCESS in 32m 09s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/0edd0a37b8364f9fa8bb75f593544531 : FAILURE in 1h 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d03551c9eaf8f087e6060c33ba26306b302bbcd7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-03 09:57:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be5e1230b5f9d139e0f44ab1301d788021ddca7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:17:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdadb8cad09b46ba82d51abfe2e8eb53\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/ca0a345c632847cdb10e73215e523650 : SUCCESS in 15m 57s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/7ffc1c10ac0048cd8fac6bbc7707494c : SUCCESS in 59m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/87984a627ffb404392618cda78d17b77 : SUCCESS in 20m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f758b1fc51ff424eb9af638a43462b4a : SUCCESS in 21m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0b8b62b805b04947ae94ce5048f1ca78 : SUCCESS in 46m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d58b86552e9e4741ae3b44f42df371ea : SUCCESS in 1h 06m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b894b07665794af3a72b3c5879417e7b : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/658c44eb3c4e435b91ac913365e2ac08 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53bb078dce19442cb026c377455bce3c : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd4f309ea9574c488e2d71363d35187e : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0c7f3cdf49248b8b40b34f3da2fe861 : SUCCESS in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7078e155274411a7db9dff7a2fbd04 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5452e423a45e46b485278239612d5767 : SUCCESS in 3m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bc46c16b90594d19a8568129826059bb : SUCCESS in 2m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/161363a2e85a4d73bb9edb261999baeb : SUCCESS in 25m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f4516490d2294c95bb90c1d8c66f3222 : SUCCESS in 27m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ba47742b336c4af987fe77fb92d21ead : SUCCESS in 50m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/20a2b368fb2542f9a32dcbf98f390b49 : SUCCESS in 1h 25m 04s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/88f7d438cf434206806cad34d6840ac9 : SUCCESS in 1h 36m 17s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fcc67a6a312c44fda3390441f218347e : SUCCESS in 2h 18m 56s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d3b4a5fcaa3441fda6a7d7f17936beb7 : SUCCESS in 1h 28m 45s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/09504b09f167440c911f7a201cbe10ec : SUCCESS in 29m 58s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7d6f15f20f57407fa01ed8f9098b4047 : SUCCESS in 39m 02s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3f4a227eafa5407ba368d2272082e6f2 : SUCCESS in 54m 25s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9616c5c103b1429e9a87706a956bfa55 : SUCCESS in 1h 20m 26s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/945aff8870af4177bdb8e74788f40b94 : SUCCESS in 32m 34s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/8f038d52a8354e6680ccde68125ad1cd : FAILURE in 1h 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee7e92ca0a045779438be004c7b6dd7b4394b232","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:29:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b742743b8859497b9a8bb7e28103078f\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5cc0fe48a198497ea57bbe4b51e2d6f6 : SUCCESS in 51m 13s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/91482b12f96444e985e440157f38c428 : SUCCESS in 1h 56m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/870fd8747c144ba8bdecf0e6b9816f63 : SUCCESS in 51m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/08559a9e94b04d1e91f2a195f8482eb8 : SUCCESS in 1h 25m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0dcf8864182d43c59389f178c74dd56d : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/26cb50dcb4de440cbeab168ce7b49327 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6a84c99ded544984b335bcd575461403 : SUCCESS in 10m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b563ef63d05c4832952097023f91e3c4 : SUCCESS in 58m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ec255fbb4528aa9d96078d179629789ec1de617","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-03 12:33:08.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"287659f7d8f05920d1eba807b804a4be7119c225","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:33:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89584927e7113cf5e357d0405807f3d83c482e45","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-03 12:43:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"726ef6cbba75f919251edf612af3f51c8f48722b","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2025-09-03 12:43:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a8754d6063044785b8b46708b65302d955fac1b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 13:48:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76e4c2b741d04f8b8dafb230c8bb34f8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a96415c4c75c4f57a08169f0844b6cdb : SUCCESS in 16m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b1bcc4d4d976442c87fcbf8178738544 : SUCCESS in 30m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/204707e8e12c405c89838bbb3fe8b139 : SUCCESS in 43m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/669f172bee3848428a32258a2098414f : SUCCESS in 1h 04m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c6584ca8bb4433bc871e10d2b23c57 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33c81723770d4a29810c449954ccaa40 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/022993dcf58a46bbaf7b70aecc60e76d : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869d9722c9e24ea18eef4058490f3497 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/466b750754ce4d658c3cc68cd09b09da : SUCCESS in 4m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e45d025063bc41c5bb82d3affd60aff6 : SUCCESS in 4m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6842a3342608444a81fdc055d1dfc20d : SUCCESS in 32m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d8a16ef417334c639accb042ca0e653e : SUCCESS in 17m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/671f69d0620842168d67ddfacce5d5ae : SUCCESS in 49m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e375a49392a34ee48368d78f2d87fe01 : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"125095f9ab185fd71e258acbf7032106726012d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 13:48:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f46ce9d896d5951e1f6bd8950ffaa0549828c9cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 13:50:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67a5d8a1b56e4333ba0c9c3f820b04e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3752261b20284f2987d15f60f288fcbf : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1a7fcd98da249febdcf637afdd7ec94 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2ca2d38fe522b6b38133e56e6898172b8a5607c","revisions":{"d2ca2d38fe522b6b38133e56e6898172b8a5607c":{"kind":"REWORK","_number":1,"created":"2025-09-02 12:28:19.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/35/959135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/959135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/959135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/959135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/959135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/959135/1"}}},"commit":{"parents":[{"commit":"cc328dc8656399ec4a68156573e91e6930fa1948","subject":"Merge \"Revert \"zuul: make failing job temporarily non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc328dc8656399ec4a68156573e91e6930fa1948"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-02 12:27:31.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-02 12:27:31.000000000","tz":120},"subject":"Use versioned Erlang 27 repo on aarch64","message":"Use versioned Erlang 27 repo on aarch64\n\nFedora COPR build system cannot handle storing multiple package\nversions in one repo.\n\nChange-Id: I3ff95aa75f1f1cccfa40f3ab7f216cb221a6af29\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2ca2d38fe522b6b38133e56e6898172b8a5607c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2ca2d38fe522b6b38133e56e6898172b8a5607c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
