)]}'
{"id":"openstack%2Fkolla~959300","triplet_id":"openstack%2Fkolla~master~I6df65b911a7b3f0027da9eeda98e35a29c8418da","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I6df65b911a7b3f0027da9eeda98e35a29c8418da","subject":"skyline: Add mysqldb","status":"ABANDONED","created":"2025-09-03 11:02:34.000000000","updated":"2025-09-09 16:52:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"423208d6d13546d5df9cb2215aab5821c062e1be","_number":959300,"virtual_id_number":959300,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-03 13:01:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"reviewer_updates":[{"updated":"2025-09-03 12:36:44.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"CC"},{"updated":"2025-09-03 13:01:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-04 06:58:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"c3299026ef3e6956ea8e70c764fdbd0711b30c91","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-03 11:02:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"547fe604634128758a67268bd7f2cc1bfb5f4e57","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-03 11:44:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"48ffed58e29131d1e34985b75272964045606d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 13:01:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/add2876331d44b1cb5800b9aebc48d3a\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/1275a0ebd323434dbab3a6eeb92a0b73 : FAILURE in 16m 54s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/306e8aa227f44d4b9a1e495b05d55ea3 : SUCCESS in 59m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1d0ded90261945f38e16f53c7f03cc2e : FAILURE in 21m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/13fb7c900fc043a3b8b2b62c862ab4ac : FAILURE in 17m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e15d056e7c6245108872458e2b4a5798 : SUCCESS in 48m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dcb7d28c6da74bb1ac5e4f51924f2ab5 : SUCCESS in 1h 09m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9efe2e68844c4ccaa8a277e001434ffb : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0b3a29fd9654a9f84177519deeb2f42 : SUCCESS in 1m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa7b0179af7944738cbdabfad0bfe473 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e50058c478e94ea3916ca02a86cab391 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e743d6793ca4686a27de89c2bec7955 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c155f1dccb654985b2f645b27737efb8 : SUCCESS in 3m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/baa83b8db4c741afa206d79f38cbe4c4 : SUCCESS in 2m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d0144b7ec54245b9a1a28141a459d69d : FAILURE in 16m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c3863a379ea346ca9a68af840ab91c1a : FAILURE in 34m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b8f56a0c28fa408d8be82604fe6a8af4 : SUCCESS in 53m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bae1b9778c664896bb4c0827fa9378db : SUCCESS in 1h 15m 03s\n- kolla-ansible-ubuntu-skyline https://zuul.opendev.org/t/openstack/build/c5df71d240f04a14b39f73ceab2411ce : FAILURE in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fee7dd6876a45bee75454104a6467584128c65ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 15:21:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22389af0a670442d8d43bef6c365473a\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/b54c06be00b347c69b0f2c7cd859bc29 : FAILURE in 56m 06s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e665d07f1a4f43cba0ea79c247fe85af : SUCCESS in 1h 51m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a6c0c703a2e74fdf9e81cc2de7052536 : FAILURE in 38m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/660445f422f04134b10edc3384eee382 : SUCCESS in 1h 58m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/872457ce73f24368acd065e9221367bb : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40c52d14721c4087a7d391d4100d89de : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/100352ef960e497a97e168689a988b2f : SUCCESS in 9m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6db751309ec84df29c9f93885639fa07 : FAILURE in 47m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"423208d6d13546d5df9cb2215aab5821c062e1be","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 16:52:17.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af290ddcce7bccd74f1542d5ae339b45ab01c34a","revisions":{"31ba6c5e1419c8aa05ca3c7b3da087d6f0816ddc":{"kind":"REWORK","_number":1,"created":"2025-09-03 11:02:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/959300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/959300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/959300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/959300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/959300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/959300/1"}}},"commit":{"parents":[{"commit":"cc328dc8656399ec4a68156573e91e6930fa1948","subject":"Merge \"Revert \"zuul: make failing job temporarily non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc328dc8656399ec4a68156573e91e6930fa1948"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-03 11:02:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-03 11:02:31.000000000","tz":120},"subject":"skyline: Add mysqldb","message":"skyline: Add mysqldb\n\nChange-Id: I6df65b911a7b3f0027da9eeda98e35a29c8418da\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31ba6c5e1419c8aa05ca3c7b3da087d6f0816ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31ba6c5e1419c8aa05ca3c7b3da087d6f0816ddc"}]},"branch":"refs/heads/master"},"af290ddcce7bccd74f1542d5ae339b45ab01c34a":{"kind":"REWORK","_number":2,"created":"2025-09-03 11:44:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/959300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/959300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/959300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/959300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/959300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/959300/2"}}},"commit":{"parents":[{"commit":"cc328dc8656399ec4a68156573e91e6930fa1948","subject":"Merge \"Revert \"zuul: make failing job temporarily non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc328dc8656399ec4a68156573e91e6930fa1948"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-03 11:02:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-03 11:44:57.000000000","tz":0},"subject":"skyline: Add mysqldb","message":"skyline: Add mysqldb\n\nChange-Id: I6df65b911a7b3f0027da9eeda98e35a29c8418da\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af290ddcce7bccd74f1542d5ae339b45ab01c34a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af290ddcce7bccd74f1542d5ae339b45ab01c34a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
