)]}'
{"id":"openstack%2Fkolla~959323","triplet_id":"openstack%2Fkolla~stable%2F2025.1~Ied94e0fac36e60cbf227eb9105f659adc1500545","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-04 09:02:21.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-09-04 11:46:58.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ied94e0fac36e60cbf227eb9105f659adc1500545","subject":"[stable] Use versioned Erlang 26 and 27 repo on aarch64","status":"MERGED","created":"2025-09-03 14:08:03.000000000","updated":"2025-09-04 16:08:06.000000000","submitted":"2025-09-04 16:06:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959323","meta_rev_id":"36fa84561b6cd70546b96b708d8e96e8f3dde1f2","_number":959323,"virtual_id_number":959323,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 16:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-09-04 08:25:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-09-04 09:02:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-09-04 08:25:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 15:55:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-03 17:00:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-04 08:25:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ca78307806eacd3007db00f7d603393d3c206cd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-03 14:08:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db00c207660107469115b508f7e90a210be770ef","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-03 14:08:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"82eb81f94ab982fe597f5ad24e92525ac7dc8906","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-03 14:09:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d102e509e56e58cd1c89d88676698a0e6f974eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 15:55:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48e5c68f8a8645d789c64f09d3ed8589\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9f3116f1bd6e4d2bb320fe1b92ccac67 : SUCCESS in 23m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9330578fbc8945158e57ea2b9c47458c : SUCCESS in 1h 07m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6bf81c3aabe84117a2be576c827fc070 : SUCCESS in 15m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d6473f19d6584b9b87a6c1793df31fab : SUCCESS in 22m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0fb5cc00cdc3470ba9681e05cb9d2ce2 : SUCCESS in 44m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/89270e77501b4f6d85d21b10ca758f48 : SUCCESS in 1h 07m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623a49e98bf0406f890202fac62cf16e : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d5eed4964744531ba83d579c4d5167f : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67caac7985af4a049f3f9f80afcc2b60 : SUCCESS in 4m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df5a7085cded49f8bbd4168b82ffc80e : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794182fd88084ef28e9074241bbf2424 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b95e611f8364025b1d9f654a33317bd : SUCCESS in 4m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/905b015a00db45f2a3c2517618931bb3 : SUCCESS in 4m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2037a6001bb14301b18fc06bd904c2f8 : SUCCESS in 14m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e4c9cf0674094c32a2381a0b02f7192d : SUCCESS in 44m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2d9531fd76534d5cbe5f0a34e3a2f635 : SUCCESS in 35m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2d466cebb0f74bb2a86d19aca8432f6f : SUCCESS in 1h 43m 10s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/6fea103fdd694c0c807bd9aeae755558 : SUCCESS in 1h 13m 46s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/84b92fff913a4d44bfc65313f1fb70f3 : FAILURE in 1h 34m 31s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9edb2824daf54346bf64fb4083ac72f9 : FAILURE in 1h 03m 35s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/ae2ec79b203a4f468538b6578ccd24d0 : SUCCESS in 25m 28s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ab1337c2f4a341709ebaf25fd0258adf : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/eee1335c22ea444d8c5b4f7d45b634c0 : SUCCESS in 55m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0782618154a541efadd8e6ca2babc282 : SUCCESS in 1h 34m 12s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5d965d4b84774113b3b82f6e3bdebb6d : FAILURE in 23m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f2e952b912d14d0f9805d4cf7542ee3b : SUCCESS in 18m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/37700c6734284499aa4623611d6c4821 : SUCCESS in 31m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6ffa6c7fca054a14be1e0625a920b2d8 : SUCCESS in 46m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/216da52907da4d049567178776908b3e : SUCCESS in 1h 13m 54s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b680f137ffda4d1da0c51acca6dbb3da : SUCCESS in 1h 35m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fa286ee0821c4cbd9522f2a2c3f339d5 : FAILURE in 1h 19m 51s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9765a2b4c3c644f2b6b54bc668760e3f : FAILURE in 1h 00m 33s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e661987ad1474a41bacea6e3216d7c46 : SUCCESS in 27m 02s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9424460a7cd9428e96f0ad2c169cb64d : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c9f6a363c9ad4c0fbdd50ce401bb1909 : SUCCESS in 52m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/62cc331009084757a71f11ade78526e6 : SUCCESS in 1h 19m 30s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fe1c6f8e1a0a401ea9f2fc85b6b6607c : FAILURE in 26m 40s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2c45c6e5e2fb43458eb5629518866cb8 : SUCCESS in 1h 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df78f753daa67fa2e5bb4fbb1af5471378107334","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:41:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9290970230c4372bc4dc7532a25cae4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/22280e5d58e346b8a31e2cb46952804c : SUCCESS in 34m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a390810d6aa44d119b05fc4c6eefa358 : SUCCESS in 42m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a66f723475d64e8da35c32a11f4e22ca : SUCCESS in 2h 35m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa2525ae43154eca9e4698e879f6d2f8 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b2c0fcfc54a42f5b4fc5f0f144b0035 : SUCCESS in 4m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9862a4efdc6c4874a82191e2d76bc1e2 : FAILURE in 56m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/df9491a15d7c443e817b06a1054fba61 : SUCCESS in 1h 07m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"295416c9c4683a92776b06a6a47c52759fe604d3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-04 06:52:05.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"f34e4076238197d00bc10bf4d5796fa327d7f13b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-04 08:25:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"575a1fcdb11699abdc7880ee8694ea7f5d91cb0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:25:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2060e1f5e1891815c4ae3df3b3e18baaa36089b2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-04 09:02:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ae1a190ce7fa0afdefd4c3a3206c1f388f3266d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 11:27:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f48349c471c47a9a426803add20ddfa\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5fc1e0a62c924d7b9f1f359525f86543 : SUCCESS in 9m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/06e653b78bcb4a3daaa257e07ad034c6 : SUCCESS in 14m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/af63693208bc442c80acb5144790087c : SUCCESS in 55m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4e5441b3cdb84c539dd470453cf39f6d : TIMED_OUT in 3h 01m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907d999add7f42f2ab64ead606d534af : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/622cd46158234778bb24032feec8d657 : SUCCESS in 1m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b5e2e7cb1ed4b96b03d76b9fdfddb70 : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9b829d2a164d0b84bc706b848409b3 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a44e55e4f69d4c669433305f04890857 : SUCCESS in 1m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/88741e9b4e554974b44a29dfa3eedabf : SUCCESS in 2m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b693530f5fd54905b14175e91c6b1a5c : SUCCESS in 17m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9b5a1c197203400ab4aeb66891c2fb6d : SUCCESS in 46m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/242ad1c2ecc94bd7a77ca771c3b75fc4 : SUCCESS in 1h 06m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4e1b0316aa4f481c9e5193e0fb7005b5 : SUCCESS in 1h 24m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/685ad98be7cc4833b3ca61772b0f2ba4 : SUCCESS in 10m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7aa94ab9dbe048b78f26d6a603bd172b : SUCCESS in 31m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b032159063a2433e8c5ccfebe8c18f1e : SUCCESS in 56m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/329f06dff894499987f891592712b7e1 : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":2},{"id":"42047312af5c08e1cdb59e3ce658ab4f1e3b1a90","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-04 11:46:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7edab6547ee0ca19c02699669bc2d113cbda852b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 14:23:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22bfa171244341dcb1d470c27f73225e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/14953ff3e23f4b72a016bc2fd5e599dc : SUCCESS in 12m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b3a8ada92f6343b4b01a8df9e460b05c : SUCCESS in 1h 05m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0e3073b474274ae6bccba30f9a791892 : SUCCESS in 20m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/74876ecd0950482e8fc86158b42d9364 : SUCCESS in 27m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/56eb5810112044e89a26c1bf775e04a9 : SUCCESS in 46m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ec00c59e2d71456a91d89d394545c2b5 : SUCCESS in 1h 07m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006aab84f47f4e9692006267722a2ec6 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f0a14e5e9b4437d824230af48ee91c3 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29fa65b2101b457ca4c125b1e980d8c8 : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02af006eeab24c639317c3a382ae201e : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ca5e3e1a344195b790f665911115b9 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a573c612ca6b4ce1b2dfbf9dc9ee2ab9 : SUCCESS in 1m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/211b780305ee48edb617b453522517c1 : SUCCESS in 1m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/77312dac5c8c4f399763397954de11c2 : SUCCESS in 23m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/85dd034b6c2642f586fd6cc7669e4aa0 : SUCCESS in 1h 21m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5363709c1f0945338d1e1a1c68c67371 : SUCCESS in 1h 06m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2be181d492e3413086f65d56f2052fc8 : SUCCESS in 51m 42s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0c274746e5ec406f866404463013de66 : SUCCESS in 2h 03m 59s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3188c58004f7438f9160afdbdb732afb : FAILURE in 1h 35m 15s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/54fa2f7819b241d79bd26ad544191cdb : FAILURE in 33m 48s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/15d20f76d06d46a0988fcc27f58c6649 : SUCCESS in 23m 29s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/009f5a05279b4b6f8632c4c19b7587c5 : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2e729a1ad5bc402ab4665da9e4233b84 : SUCCESS in 1h 00m 14s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ce33a1cce9c644beac9528798e7eb8da : SUCCESS in 55m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4df4acb823354d72bd4914e92ed2f5d5 : FAILURE in 27m 22s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d186b2f5bc06402eb8000a1791d47a41 : SUCCESS in 24m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dc97df9cc75c499c9bab90a26f8911c4 : SUCCESS in 29m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d34e26fb034b4d52a20251e8e8913fff : SUCCESS in 1h 09m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a819baf78b334ec29874e1a6be374fda : SUCCESS in 1h 33m 28s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4db8b6ac5a334eefb76112d1375994f4 : SUCCESS in 2h 02m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/69548d31381c4833a469dafd46afc769 : SUCCESS in 2h 28m 58s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d4eeaf1ffe714e21bc155ba550d991de : FAILURE in 1h 03m 17s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/12c9dc3c2271410ba72bf466b5beafec : SUCCESS in 24m 37s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/496998ca1b694f7199378c91e92e1d7c : SUCCESS in 49m 48s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/5764510f7b73477785899792a9af2b31 : SUCCESS in 1h 12m 43s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7783f2038d824f0e88ba6f1e7ce34a4e : FAILURE in 1h 19m 01s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6808f572123c4002931be4af0c836a1f : SUCCESS in 26m 31s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/d570864a9e094b18842d5be607ac6103 : SUCCESS in 2h 05m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a4be6fb6f7fd2978352ba1a82e96e7ca13dad5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 14:24:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aed4a3498aba14c8a168ffb8e5633712b05c5e6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 15:40:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a43c1625f447eca383a65950ee1c12\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d10ef42673f0481485d8b7fde463006f : SUCCESS in 1h 05m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b50542b8a3a24cfe9bf33c8f3ce383e9 : SUCCESS in 1h 05m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a4fcab69f2b6467ba4513c1b97197499 : SUCCESS in 1h 10m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9cbe25fb77564835835103db3165b98c : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6b28fab9b8849a986becc4cb65fa915 : SUCCESS in 5m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4d6123be120d4db5ad59bd3f45dc9199 : SUCCESS in 31m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/27905be930244d6d85c9e1f895918bec : SUCCESS in 1h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"949a5b7727df2aa4e0ba1c8c6ab852ca657f6934","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 16:06:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c97a2b3540084f3ba207f406071c2d85\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7f7bcf997ca641ff8e05c66a9db79090 : SUCCESS in 20m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f182f3d252fa47209165cb186c9dc7dd : SUCCESS in 32m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aa0360bc82e942a286385c11b80f3625 : SUCCESS in 48m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/53bf1015b05f4fd99342108df94139ce : SUCCESS in 1h 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b045d52f9a947829ae4d267b3fd8c08 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db19c17234734fb98d6ba304595652fd : SUCCESS in 1m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f866cae80904b2dac4903b8177e68bf : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308d187bfa1d4cf995c1b736b326cab1 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/010935c2aa3349928337714f16ed59d0 : SUCCESS in 3m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1a460229f8374a5e92d22f38dba19782 : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/15acd6a4c4a34f59a963a7b7fb334550 : SUCCESS in 21m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/657b29cba9444cb88f170a58b0d2e8da : SUCCESS in 1h 36m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/94b3e955b72b48ee857056fb9dbbeb6c : SUCCESS in 1h 07m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0f1b5c8e4b0247bc9ce1d33d67d1130f : SUCCESS in 1h 37m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/793c7f3d577d42619d9aa259c2f654e3 : SUCCESS in 12m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/06928179def842fca186a24bacf023cc : SUCCESS in 31m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5a1ff8d04b0d41079ea1cd8bf801af7a : SUCCESS in 49m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/af5c17820de8463ebba7f41ac9cfe6b4 : SUCCESS in 1h 13m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5706a2b17d32ba5f49be3a0c346e52e32d17265f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 16:06:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"36fa84561b6cd70546b96b708d8e96e8f3dde1f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 16:08:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8f21f53ebe94df8a926631d193c58ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b731cfa9b6714cf5aaccc1c773ebbbb2 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef10d505199e4fce9aeb14f15e579095 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"688c3c7ed898684b2ce9f729198642fb38ab2f13","revisions":{"aea1c38ce13c8297ad47bf9fe5b92d3e77dc2806":{"kind":"REWORK","_number":1,"created":"2025-09-03 14:08:03.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/23/959323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/959323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/959323/1"}}},"commit":{"parents":[{"commit":"6c08b17423de465e0a6daf578576053c9eb9f384","subject":"Merge \"docs: Update contributor guide\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c08b17423de465e0a6daf578576053c9eb9f384"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-03 14:07:01.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-03 14:08:02.000000000","tz":120},"subject":"[stable] Use versioned Erlang 26 and 27 repo on aarch64","message":"[stable] Use versioned Erlang 26 and 27 repo on aarch64\n\nIn stable branches we need both Erlang 26.2 (RabbitMQ 4.0) and\nErlang 27 (RabbitMQ 4.1).\n\nFollows [1], adapted for stable branches where both versions are\nrequired.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/959135\n\nChange-Id: Ied94e0fac36e60cbf227eb9105f659adc1500545\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aea1c38ce13c8297ad47bf9fe5b92d3e77dc2806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aea1c38ce13c8297ad47bf9fe5b92d3e77dc2806"}]},"branch":"refs/heads/stable/2025.1"},"688c3c7ed898684b2ce9f729198642fb38ab2f13":{"kind":"REWORK","_number":2,"created":"2025-09-03 14:09:47.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/23/959323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/959323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/959323/2"}}},"commit":{"parents":[{"commit":"6c08b17423de465e0a6daf578576053c9eb9f384","subject":"Merge \"docs: Update contributor guide\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c08b17423de465e0a6daf578576053c9eb9f384"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-03 14:07:01.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-03 14:09:42.000000000","tz":120},"subject":"[stable] Use versioned Erlang 26 and 27 repo on aarch64","message":"[stable] Use versioned Erlang 26 and 27 repo on aarch64\n\nIn stable branches we need both Erlang 26.2 (RabbitMQ 4.0) and\nErlang 27 (RabbitMQ 4.1).\n\nFollows [1], adapted for stable branches where both versions are\nrequired.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/959135\n\nChange-Id: Ied94e0fac36e60cbf227eb9105f659adc1500545\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/688c3c7ed898684b2ce9f729198642fb38ab2f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/688c3c7ed898684b2ce9f729198642fb38ab2f13"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
