)]}'
{"id":"openstack%2Fkolla~959523","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I5cc6172156c908fbb968a3e3020f24ad8d9e58bf","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-04 16:42:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-09-05 09:02:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5cc6172156c908fbb968a3e3020f24ad8d9e58bf","subject":"erlang: use proper distro for noble","status":"MERGED","created":"2025-09-04 14:01:04.000000000","updated":"2025-09-05 09:04:06.000000000","submitted":"2025-09-05 09:02:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959523","cherry_pick_of_change":959426,"cherry_pick_of_patch_set":1,"meta_rev_id":"a918686735680823e4c4cef2d05b979e180e00df","_number":959523,"virtual_id_number":959523,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-05 09:02:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-09-05 07:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-09-05 06:55:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-09-05 07:15:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 15:29:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-04 16:41:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-05 07:15:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a2849ec581e96a0b0146182dc8e88fe83fd049d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-04 14:01:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"379e003ca862c924d68ac07cab411b77d6e3014e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 15:29:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ebaf28650c54ed389fcf35167e64457\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/69ae6f58ae72420fa6ce6d84796eee5c : SUCCESS in 27m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5df81259f16d4de5b8a1384c69f4e2c1 : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/23ee93cf27d24ae5984ec1832e254ed9 : SUCCESS in 9m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/017c140d54e6407c9959d667648a57ba : SUCCESS in 32m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6bf203332ae04b5eb29641a79d4c6e5e : SUCCESS in 31m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5083c08261734af29498ea89b4232e3f : SUCCESS in 1h 26m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d26e44bb9d464a2693d7a4e826d35bae : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7f39ff5e52149368fc4b66bc26b4f22 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c9a538b04a64da1a9ca9f9c91b1dff0 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5a07b6845544a51ab5496fc54c26522 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01f04abab6a249c28729b5f7d976f5bf : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32dca036b5db4f149eb03aee9b9f6b4a : SUCCESS in 2m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4c8bfb3769c64f2085d3804d28ad6d4d : SUCCESS in 2m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/716b7fa4f4e646d0a340bf0a24a917c3 : SUCCESS in 11m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ecc54fc0029a4c03aa559fa53464aebd : SUCCESS in 39m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5e8f064c0b724aa494869799126ce644 : SUCCESS in 39m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f47edfa08afa4d6db27cebc12891044e : FAILURE in 52m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f8f67d6795894ec086ee1c2da07f886f : SUCCESS in 21m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ab64acf4a7c844dc9d2d170f98ecf713 : SUCCESS in 14m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/db00b527be174ad9b8526ee456ed3103 : SUCCESS in 1h 01m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c8ac80b80ea949659ace2de17008a2b8 : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0d0b0455c17a9ce36c070dcb91195705752a136a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-04 16:42:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2242dd4b4f77e2b4810cd1438896e6615df4c48","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-04 16:53:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35d555cd7c421aebf6ec7862d0bd4a48d6f7915c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:20:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e65db45183e4f23879fc1627c15be47\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cc8fa03963c94daebacfa9e835f3ccf7 : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/05555e50f0b1420fa5a729310ac032a5 : SUCCESS in 1h 07m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1ec5bca9d99d487787dd04d9dfa4d6c0 : SUCCESS in 11m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/88b5e843b33e407688a3b651b7edfc54 : SUCCESS in 17m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/adc169dc91c04fe3a7fc59cfb3b783ef : SUCCESS in 32m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0aee4c6d89334cacabadad10c998033a : SUCCESS in 52m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439a572b6f8f40fb95b6554d8beb2eae : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54d0a8b660b24c7d83d86f5712b7385f : SUCCESS in 2m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a835f3992cd243f982efdfd506d2629c : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6a02770d2e84b4890e4dd2caf9a09c7 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61abfaeed35a4779a76f0fb2ab3bdd5a : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65a01cea86749e7ae6a35a19c6e9af9 : SUCCESS in 3m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/77369712259f4376a053d3963d906551 : SUCCESS in 2m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/31e1aba089f348d59cbd4f33599ea59d : SUCCESS in 21m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b266f7ba9cc14249866a27bb722d4189 : SUCCESS in 22m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3e00acbbce8c463ab9718cab0bc80ca7 : SUCCESS in 1h 02m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2833461bd20b4d5e8d59ad2f897c0764 : SUCCESS in 1h 31m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/73b2c2f63b844ecf9b8c53d04a529e41 : SUCCESS in 21m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/be227ddff3ac4b298cf6643fbffada12 : SUCCESS in 25m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f889198af8154f2c9b96f105170272ad : SUCCESS in 1h 06m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/833096ef49fa405485066b59a6980ce2 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"04384a27538fa7af0204a57d375763c0207850c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:10:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/786c78422f7a437ea35dd934373673ba\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9ff6bd32c230433099bad0fae2ccd35a : SUCCESS in 53m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/df54ce5908a04e5cb0d0b70feea16433 : SUCCESS in 58m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/88ae9856365e41bcb438376f5da76906 : FAILURE in 2h 20m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e2e04a592bc4eb89c6f151a92b9a634 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/29877d117fa742638ecd5c0b424ab12f : SUCCESS in 5m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8c2965ad91fd4a418b8f066cd1e813cd : SUCCESS in 39m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/566c5231bce944a3b933c2990209adfc : SUCCESS in 52m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c56d6bad219f212b875132436cb46369f8b3e1bf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-05 06:55:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2e30b433bcdd7e40c5f73f235055ee1778fd5408","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-05 07:15:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7881551e0a264f6f55d9b186c08ec70e05283581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 07:16:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfcb6589f02a35d37a6a61f1bf58cac77151d43d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 09:02:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de799c24e8c44d02b3f4f920a4ab6d21\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9201e4a65aab474ea3880b2378732c31 : SUCCESS in 16m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9cf4607ea6344cbf8d8b13f0725a19b4 : SUCCESS in 29m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1149ce83241442228dddf1a3d706588c : SUCCESS in 1h 10m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ae005bcdaae748b28d6f06f5c9c5d43c : SUCCESS in 1h 42m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd30f538cdfe48c5b34418ac77fe8241 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/959c461ba16e428996c349eedb515015 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5639d92df23848beaf3a8002a67d812d : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63545878480c4be39f6a39feea4f03f0 : SUCCESS in 4m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4c599994c39d440db43454f035ac81cd : SUCCESS in 2m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d09272bdcb9441cc8bde435e5fadeff6 : SUCCESS in 19m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5add8463973d42f6b7fcd94966f2bb7f : SUCCESS in 30m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c5d105ceff1041f989df097927ca2c91 : SUCCESS in 31m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/31693ad9adef402aacb5a2efabc91b5b : SUCCESS in 1h 18m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/80eb482f25d643c0aa46b14aa9a9c9ae : SUCCESS in 32m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c972e1842a474f47a52dd8cdcd427f78 : SUCCESS in 30m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/da2a317e9e8344b6b71b1a8c0d68cb11 : SUCCESS in 1h 07m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7ec659807bb64c74b487170d2adb0e86 : SUCCESS in 1h 21m 31s","accounts_in_message":[],"_revision_number":1},{"id":"83da13ba9aabcfeebd3572d743328403684c9b53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 09:02:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a918686735680823e4c4cef2d05b979e180e00df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 09:04:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca8f5bbd3df4d9ba00262addbaf7729\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21a3b7e1b364eff8475031c6879c32f : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26ca750344a1c2a20e9cd1b35f8e227074ffc9c2","revisions":{"26ca750344a1c2a20e9cd1b35f8e227074ffc9c2":{"kind":"REWORK","_number":1,"created":"2025-09-04 14:01:04.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/23/959523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/959523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/959523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/959523/1"}}},"commit":{"parents":[{"commit":"dbd6614b822ec2766a58e56991cb8056f087db65","subject":"Merge \"docs: Update contributor guide\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbd6614b822ec2766a58e56991cb8056f087db65"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-04 10:26:36.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-04 14:01:04.000000000","tz":0},"subject":"erlang: use proper distro for noble","message":"erlang: use proper distro for noble\n\nIt was altered by mistake with [1][2].\n\n[1] https://review.opendev.org/c/openstack/kolla/+/951938\n[2] https://review.opendev.org/c/openstack/kolla/+/953858\n\nChange-Id: I5cc6172156c908fbb968a3e3020f24ad8d9e58bf\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit a637db7b9345094858d4497b7d49faa802853447)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ca750344a1c2a20e9cd1b35f8e227074ffc9c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ca750344a1c2a20e9cd1b35f8e227074ffc9c2"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
