)]}'
{"id":"openstack%2Fkolla~959692","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ied94e0fac36e60cbf227eb9105f659adc1500545","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-09-05 15:17:23.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ied94e0fac36e60cbf227eb9105f659adc1500545","subject":"[stable] Use versioned Erlang 26 and 27 repo on aarch64","status":"MERGED","created":"2025-09-05 10:33:52.000000000","updated":"2025-09-05 16:23:43.000000000","submitted":"2025-09-05 16:22:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959692","cherry_pick_of_change":959323,"cherry_pick_of_patch_set":2,"meta_rev_id":"68f64998df56040ecf3a30ce01b2a03bcd78babb","_number":959692,"virtual_id_number":959692,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-05 16:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-09-05 11:06:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-09-05 15:17:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-09-05 15:17:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 11:06:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-05 13:44:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a13476aef3e62d256a40fec1b6105b5f6ad40f62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-05 10:33:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"233a3df1423eecb20be62889ba92bae13c336e1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-05 10:36:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a1e19f4b8d4207112339a77f71ced1f9261915e5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-05 11:06:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"96ad370283d8959f12d658736aa14a3627a2ce03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 13:44:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/570a7ddfdd204c0a9a27d69701632640\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6be581f2f97f49dea1aeb0a83c833913 : SUCCESS in 17m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5552069f592e450dab94ada885f259c6 : SUCCESS in 1h 07m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5bd196750ada4ef1898af948c33a7119 : SUCCESS in 17m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b3be825006fc4359823d8f47818b971f : SUCCESS in 29m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ae66c7679998471f8d5c91ac64000365 : SUCCESS in 33m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb2d75a108f4f9d99bd7957ed7a7313 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf7a61e6dfd14561a20ff51512ff3917 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b853f48966e48c1b418862b5b6693a2 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9439c4c4ea2246a49dc3cf9277e300f8 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1450244e5096496abc32a354859df1e1 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11df1ebf1e024064b3373b38e8db42cd : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20230d8b2b224f68bd4b5ec034920cbb : SUCCESS in 4m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/99a9e4c3503c45069c7284de5c179427 : SUCCESS in 2m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f954551ee45546bc86628fda03f33cb5 : SUCCESS in 16m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1f643df78e364d9ebcc3d1dbf3c55814 : SUCCESS in 28m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4dc6b4dcf8204b0a8133218aed708df3 : SUCCESS in 1h 05m 43s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/27aa9f6f54c14423864ff36ebb04d4eb : TIMED_OUT in 3h 02m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3ac33e02b97146139cd45d29dc14c0fa : SUCCESS in 1h 18m 29s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/b8269b99c6fe499e92b273636f754da6 : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/6f709ebc36284a408acf133a99f29be8 : SUCCESS in 26m 37s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d03ec01351b54037ae3b7977a33e6908 : SUCCESS in 31m 34s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5a865364f546436aae531eff5d5d3e1b : SUCCESS in 1h 11m 06s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a28924255b634b0ab206625157991539 : FAILURE in 44m 40s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/93208614dfbc4cbc97a9eb35e037c226 : SUCCESS in 24m 50s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5db457523a244cf180cedd23552ea4e8 : SUCCESS in 9m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0916283291e2435497201084ad47e21d : SUCCESS in 21m 12s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/c4cf517b0a9b4807ab833cd59364b30d : SUCCESS in 1h 00m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/11ef1c00b78c4ee8b8f590d6d51a1c1a : SUCCESS in 28m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d2c09dc1648c22d5ae8dceb9b2180228a6ceab35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:06:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/137246c01ad84abeb7fd4013ce5dba9a\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7566113eb0b54cfe89216686f7d1f28e : SUCCESS in 47m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3a0117bb96d14acf97be19274d89f3e9 : SUCCESS in 1h 48m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/199cb07bb5404ab3b206a062ce39e2ac : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4776e18260ee4463aa823127e055d1b5 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ceaade69eb2b4e7689c8111bcabc5e7d : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"39e4eb2eaa515b22cb5a4027e4d718fa935482d8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-05 15:17:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"95e5b9a6ce007ebd65f30308fdedfa4ef5f14dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:17:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4876a4e5705060cd6d59367d380607cded1581a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:22:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b920985bcf4a4ce68051bb10dee2c7f2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d832c9d55adf4d1b9c2f24514d3d1b48 : SUCCESS in 17m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1b9513c519804f97a1fd482e4824d717 : SUCCESS in 15m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ad46889b71094127915ab9e08c4c3df4 : SUCCESS in 30m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bd3619a49ba4c7386c709cd2b748c5d : SUCCESS in 1m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/603301b634b2466399a1879585ed6817 : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19a4d4b97ed54b468d45bd557dde9d43 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb694795b81f4266b752985bc48982be : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27f91fd8f6694299b4d786b0b0d3acf2 : SUCCESS in 3m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a19f7732ce4544a196013c22a4331b5d : SUCCESS in 2m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/99d984445938426c96cfef29274d4a00 : SUCCESS in 19m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1dc810b82e8e4943b15a862431261ce7 : SUCCESS in 32m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c5503e5e024c404a946119598a53d11e : SUCCESS in 30m 06s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93624b5648174f50b17b9fecb67c6456 : SUCCESS in 16m 57s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb0b88bacf9245a49f1c81bc3584768c : SUCCESS in 18m 15s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/7261402f18894d11a1e7614dbbb16d13 : SUCCESS in 58m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f711da21283444ad8c2c385a848a9dec : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b05d37b33e4514185bf45154d41cf0365d9e6dd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:22:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"68f64998df56040ecf3a30ce01b2a03bcd78babb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:23:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2d4b739c54943fe8f0ba45325b1394f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906977f09acc4214b49d68d04bbf7182 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a4a476d0e594c1094aff8659e364ee7 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37d68ab91b3184919511fb8941373c14f0fe17cc","revisions":{"8d55626520b21a33352ac04f2ad61705bea9c7d4":{"kind":"REWORK","_number":1,"created":"2025-09-05 10:33:52.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/92/959692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/959692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/959692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/959692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/959692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/959692/1"}}},"commit":{"parents":[{"commit":"d0870105a2749a9e03e55709611c68f05d99d209","subject":"docs: Update contributor guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0870105a2749a9e03e55709611c68f05d99d209"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-03 14:07:01.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-05 10:33:52.000000000","tz":0},"subject":"[stable] Use versioned Erlang 26 and 27 repo on aarch64","message":"[stable] Use versioned Erlang 26 and 27 repo on aarch64\n\nIn stable branches we need both Erlang 26.2 (RabbitMQ 4.0) and\nErlang 27 (RabbitMQ 4.1).\n\nFollows [1], adapted for stable branches where both versions are\nrequired.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/959135\n\nChange-Id: Ied94e0fac36e60cbf227eb9105f659adc1500545\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 688c3c7ed898684b2ce9f729198642fb38ab2f13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d55626520b21a33352ac04f2ad61705bea9c7d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d55626520b21a33352ac04f2ad61705bea9c7d4"}]},"branch":"refs/heads/stable/2024.1"},"37d68ab91b3184919511fb8941373c14f0fe17cc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-05 10:36:14.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/92/959692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/959692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/959692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/959692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/959692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/959692/2"}}},"commit":{"parents":[{"commit":"d0870105a2749a9e03e55709611c68f05d99d209","subject":"docs: Update contributor guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0870105a2749a9e03e55709611c68f05d99d209"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-03 14:07:01.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-09-05 10:36:14.000000000","tz":0},"subject":"[stable] Use versioned Erlang 26 and 27 repo on aarch64","message":"[stable] Use versioned Erlang 26 and 27 repo on aarch64\n\nIn stable branches we need both Erlang 26.2 (RabbitMQ 3.13) and\nErlang 27 (RabbitMQ 4.1).\n\nFollows [1], adapted for stable branches where both versions are\nrequired.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/959135\n\nChange-Id: Ied94e0fac36e60cbf227eb9105f659adc1500545\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 688c3c7ed898684b2ce9f729198642fb38ab2f13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37d68ab91b3184919511fb8941373c14f0fe17cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37d68ab91b3184919511fb8941373c14f0fe17cc"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
