)]}'
{"id":"openstack%2Fkolla~960437","triplet_id":"openstack%2Fkolla~master~If1a078d2704fbd0b119d6c1ffc5a10aeff0a150a","project":"openstack/kolla","branch":"master","topic":"kolla_uwsgi","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-17 13:50:13.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"If1a078d2704fbd0b119d6c1ffc5a10aeff0a150a","subject":"gnocchi: Bump to 4.7.0","status":"MERGED","created":"2025-09-11 07:52:26.000000000","updated":"2025-09-17 20:29:42.000000000","submitted":"2025-09-17 20:28:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960437-kolla_uwsgi","meta_rev_id":"eef27e90593c150adfb5a76b2c13f3bb2cee8f33","_number":960437,"virtual_id_number":960437,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-17 20:28:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-09-16 12:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-09-11 10:26:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-09-11 08:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-09-16 12:28:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-09-16 12:28:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-09-11 09:24:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 08:14:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-09-11 09:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 09:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-11 10:26:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-16 12:28:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c012763ce2460e23842c536169d47bdf59aee593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-11 07:52:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"305b42c2cda889a02f4ce43fcf34ff6668438291","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-11 07:52:41.000000000","message":"Topic set to kolla_uwsgi","accounts_in_message":[],"_revision_number":1},{"id":"17f68352bb467ff74093035bb44b5f201dfe9fbe","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-11 08:14:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85d2a732c2d200fea983cd7a9b9b71dc13b3c190","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 09:01:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3673e8cb417242b59299b54f6f763ed4\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/46e9a40d7ce64b3d96fbc17229829568 : SUCCESS in 40m 02s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/356df04fcd45466f85305eda1e523754 : SUCCESS in 1h 07m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7c828d9e05ef458bbf1b0f221421a813 : SUCCESS in 40m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e315843895c49ff8404afedcbd2db66 : SUCCESS in 1h 00m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a1f3788cb774506ab65f7e7e419d373 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c1a91692ee3474093e8e37a68d2555e : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b0e42ce710a40ddb422964233bcdced : SUCCESS in 8m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/074bd8af848f49639e7fffcac306b3a5 : SUCCESS in 38m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c6cffef143b4d340579e63608bc94458d13617c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 09:11:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ed5aee7a53419b97b1282e36e3eea1\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/44c18820937a491c92bd5453124b0de6 : SUCCESS in 11m 19s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/cb7ba576c61345e9ba6708dd3e5e649c : SUCCESS in 52m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e63ea073167d43eaa2db2a37a180477a : SUCCESS in 16m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8108607e31fe4b888acec71c7d99e22c : SUCCESS in 14m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2c165d83c4334f33a9f658305ceed37d : SUCCESS in 49m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1d73b86a47c845609a57b96ca2f178a9 : SUCCESS in 1h 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3234c0e7ac124602b248cb651af20d3b : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cbde05e2e8448ffb50daf1fd8027660 : SUCCESS in 1m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c1f9cc997df4714b15afc76c9de5859 : SUCCESS in 1m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9c708335a19489da31e33455d7b2cd1 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88532b4c6e95459f97b5472b1d881679 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e346e9c48d46088ab10bd9a344c486 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ccd56f91f16480c8b67d2aacea9a1cb : SUCCESS in 3m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4b690d1c813f433e83005b29a6e3e564 : SUCCESS in 4m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/63bdf4b7113b4c14957dc580fc8c2836 : SUCCESS in 18m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b6c9ef4ad6ae45b2bcbb09355b858f28 : SUCCESS in 14m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a370ad6aef1c4556bf74f6ba258b8254 : SUCCESS in 53m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/048236e5cfef47b694faee74e1e10fc9 : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7a41477ebd47720aa78ef8a3cac1e9e7d15251b6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-11 09:24:10.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"2a6b28603228fb22e706f2b6ebf580f6da6a4c52","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-11 10:26:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fcee778e8b2a05854498bf6bc31654a9ecfeb7d3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-09-16 12:28:44.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2edb6cc833f524a628a848edc47369cf31cfd0e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 12:28:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89937a6679832ec9935a4754dfffb4c158df2dc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 13:42:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca2db7d9194646a7b43bfb16d76379bd\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/04360a8a3e1041c3ab8ba2f262c89442 : SUCCESS in 18m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a9debfc067ad48a3b7012b65aa1149d2 : SUCCESS in 28m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/423bda92bbd2420dbae057c8591a4d31 : SUCCESS in 46m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/20ec6a3ae84947c6a8113fc9133e5203 : SUCCESS in 1h 06m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f13108e3c2148f5a9e8645a4d2bec9f : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3085f56e4304ccbb45cc20066867b70 : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0ce9612e5b54c7ea6e639954ed89f6e : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4ec29d7b3c4cf1846494073a0d15dc : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5166961a6944be2b307699bafcb739e : SUCCESS in 3m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a460a0d7bb0c43cfa0a3eb777bb3cec8 : SUCCESS in 2m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b32a797ab5624344b5db8286f36a453d : FAILURE in 21m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/05b1f9913b884ae8b84c577bceb0d081 : POST_FAILURE in 23m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/52d737bee3094999b9c2898a6202c259 : SUCCESS in 50m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6ed1a6ed04cf4f45bea1578d264b3f5e : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7ce421c9f5222b3633b58bc954d968e7d353d288","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-16 13:44:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5aa7bc25a1bb1f2c76e0170b7da943d1bd7956a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:00:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54c450711cea4752a6b47dbbc45d8b4b\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/00e480cf7b2546b2bcc27b6d8a42f2eb : FAILURE in 13m 57s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/661af7cf47e545c8a596079f9efffa39 : FAILURE in 14m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8b05dfb662ba439691eedbe69c217d88 : SUCCESS in 15m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5e88cce8a7074165a11538c6d0d321cf : SUCCESS in 28m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1759db7fe56546feb56849f8acd6b96a : SUCCESS in 48m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a14e174dfd6d461ea26502f08363649a : SUCCESS in 1h 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f063577c1f465c800fa31b9d6d8d88 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/823c9a71038740fba9523eaff7344b36 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e88250f2b1d14c21b1d03d62f87ec0bc : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b97c6f4abca74f9a81124f76e9d437dc : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/329a8355586e4c8ab0bee0eba45fe675 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48735cb8d6114222b1813cdaeecdd94e : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1d0e619e1c04780a6c65613ea2dcd3b : SUCCESS in 3m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/69d756d4a6a349b49b642582943acd2d : SUCCESS in 2m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/30691c49d8cb485093a2958a59ea2b25 : FAILURE in 22m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/416c053039444dae9582cd7c043cc2d8 : FAILURE in 27m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5ee9d4628ea74d07a92941d1fd839558 : SUCCESS in 47m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4e7864f519e04e8ba12b879a849870a7 : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a1270ea9c207f895ef27292c2ab8db1e9d0055e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 16:35:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db0d1d93bff945b08128c25699f71347\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a97956b16edf4e8bb8f0b9e0615e7412 : FAILURE in 20m 46s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/20b0efc0187b4f27804cd50bacf0786c : FAILURE in 17m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/68751363209f40709abf33adb9ff45fd : SUCCESS in 35m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/555bf4a0aee7448783d0ecb4652119b6 : SUCCESS in 1h 58m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cfac8faeaf92428eb8eb889bf01ff604 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5889ee2a25544bc791fe398368244f53 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4fade54322348638c6d4e0d3b2f5022 : SUCCESS in 9m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b90e339525964ce58c821cacf84c028c : SUCCESS in 38m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c242814e5e44b1953e3cc962bbc2632ee57b2cc3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-16 16:51:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fea4877064f3d32b42b1e6552448b3d3ccb794e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 18:12:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5eac9f83fa74950bd38ebaf6b0b88d3\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7b6b12d8fde74c5a8c106a2e0a169695 : FAILURE in 15m 23s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/d88273e7c1d84424839d12490922b96b : FAILURE in 10m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8a566ead397348ba81d39028532bd458 : SUCCESS in 21m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/13ee2998c23045a2aa438ec7883b494f : SUCCESS in 24m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6372d3aba4194dfd9cc74df4f664ad9e : SUCCESS in 46m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/584e1fd9e1404b85b5384d703589beb0 : SUCCESS in 1h 06m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02dc1f2f5045438caf594b8ba6e1ed6b : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f15ddb757da4770984df721015b518c : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8405aa5f2ce34c0da0d68932e88992f5 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8d6e4090f264cfd9f6886b6de660ca2 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be65365286354b9db6522e04fe49a55d : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc8538b4ff541dbbe76026c55f4781f : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1107ec4ec272481a82b04a924640b5b4 : SUCCESS in 2m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8fa36a5eb3764163b5ef345d12849605 : SUCCESS in 4m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/97f7317c9d4947df9e4f9f4eff0b5008 : FAILURE in 20m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8f2e94b5d06847a29a6a28c12d8a73c1 : FAILURE in 30m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c64fd31d83c748e7a18b8e34cca721e8 : SUCCESS in 49m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c5e243df3b3441c086f9fe34c2c0503a : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":1},{"id":"99e28aa6421b480320f167e106c5b06bed86ee0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 19:06:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/973b729bfcdf4c5a88fe69fadbd35841\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/1a8527a2407d49e6b5331a02448d59cc : FAILURE in 43m 09s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/530cb5797ddf458690d9ed293d880c1a : FAILURE in 35m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1fe3828813d14c37aa9a4eff45b6e430 : SUCCESS in 41m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/64f3c5375581444892c43da0bfccfa52 : SUCCESS in 1h 40m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e64446f1198b412a8e193e66ddbab747 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/299654097b0b4da3bec7035a70b994bc : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f74223c06cf14111bc09f1315b9508e6 : SUCCESS in 9m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2909c290fc344a12b01f35249952d462 : SUCCESS in 1h 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79b55c3d83baeb3656e492164169296546ce143c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-17 05:17:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71d8a0970d073aadd23b50cbd4c3e40798b5d895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 06:30:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3d5dc4529214bdfbdfbd9befc6497e2\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7786b1e7d22049dbbe66ecff5a357170 : FAILURE in 14m 34s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/bd1d32027dbb4182addd650397de9bb8 : FAILURE in 13m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c8c4edf11e1647b581746bac0a144128 : SUCCESS in 18m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6e55400a149e43f589bce4a5d62e4bb7 : SUCCESS in 20m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/83b5d37313994406ab5be06ae93dd88c : SUCCESS in 44m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d6459dc9e38b42fba428ecd570fa2c44 : SUCCESS in 1h 07m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2060eb8f68594f20a024f97fe129c781 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1af4ebe2c5b144a8aadac742f382a41e : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78efd128795b49f9a9dba901b1941696 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cb5ab05ff2b4aaab44663965920b132 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/107795d88ba147df9a3fb3b6f96827ad : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b9dcc3b888480bafebc5c71719144c : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23e7309854b948fc83dee043dd8c7fe8 : SUCCESS in 4m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5ea4e5bc115f41999c02b7d9225fb97f : SUCCESS in 4m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/75c6089a40d342d5955908850efbada5 : FAILURE in 19m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b860a47c86b44c26a4da90ac6def482f : FAILURE in 24m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d2f3ba0ea47f43e8b98526f0fb82d957 : SUCCESS in 50m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/72f9591e970446e68358d15bb8925086 : SUCCESS in 1h 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f7e77f8b8c16293409eb0412d8a726f9200b962b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 07:05:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7900363e965042f4bd616c84fe07e018\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/ef9fe5a735e04ea7a1a8e1f352b6ae8a : FAILURE in 25m 10s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/b1d7ad774f704d88bbc8d591070fda63 : FAILURE in 26m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ae53580d83e4e2aaf5f4f6f7b9b2905 : SUCCESS in 45m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/df09c5ff467c441aab974bdb09c3e7e2 : SUCCESS in 1h 45m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d0b5556edb54d70921c1e8f57448743 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7a62d38b30847d59c95b35e3f7287ce : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4094ee8cb8274740ad691d3115dbede9 : SUCCESS in 8m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/180f2f808d604b249545e6263ca14bfb : SUCCESS in 53m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6783fc81ebd3d1b5447500a189faac9fb23a71d0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-17 13:50:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95606b2e928ff31203289a463658f2d00042ae95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 15:15:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2461489aea1b43a68cb401983cff0dc2\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/eaae99aebe6d4503b1a5753a5d1ae0ed : FAILURE in 12m 57s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/42323b05c2b040788a8858310e53ea6e : FAILURE in 18m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/be4390f7009b4087a352a9b776b5f658 : SUCCESS in 15m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e65145e20caf45a19f3339fa10007221 : SUCCESS in 21m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bc987b1ce1974c2d888b5516755dce01 : SUCCESS in 44m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e84d8be891dc40a18c9eb406771d9db2 : SUCCESS in 1h 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c295b9fd14f4ef7bf98fd67e66be2ac : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6eb6da3f461944a5904ea2901ea24ace : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e89a5cef569840a08e2db4be1dcc502c : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d4a99a3543b47729402ba9d25f05cb3 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a8cd92ae9784f419830a9b7933ecefb : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8980acdd76c447a8b130d792e554bcd : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb83f4991fc940659c7ba249c025f035 : SUCCESS in 4m 27s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4e0dd3f6ba6841078fcbd830e588ce31 : SUCCESS in 2m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/28618979c86149bd97f7dd71085a0634 : SUCCESS in 24m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cc18b7fe7b434368a9038eb732a397a0 : SUCCESS in 26m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/839bc4a8c0c8464e8a26ec063f039c02 : SUCCESS in 55m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e7fd8815d1494ffc9e5cf0ea974d4d8c : SUCCESS in 1h 24m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0e776a3d7c4018c420295ae77b24d3900961c5f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 15:16:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54b865cbc12aa797438dc218b3d38143d405ce18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 15:57:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f859fd253da451a90da6de10e27bbf1\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/4f63798f189e43d98bc2f97b86778cf5 : FAILURE in 26m 05s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e07d034f0fdb4c3e9291c5b5c2ed7f2d : FAILURE in 29m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d9c6e78e0454aa9bb2098cb4633c2f9 : SUCCESS in 44m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e31308f5b7254d76973a4907ea9074b1 : FAILURE in 2h 00m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/46826f86b8494399bb25fc4c900dd9d3 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e3476705e0b44239005001d5c752e72 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f43260543d3044e29a69d5207220a620 : SUCCESS in 8m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4f4cb3a57cab4dc2832a5ec64c3a45f9 : SUCCESS in 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc3710e6a32b0f768c5d14e46e271c8b5e57cf04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 20:28:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe86f01d80d4d6fa694fab718a42098\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dcccfc406f8c47b5be81672feede491b : SUCCESS in 18m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8f7d0e3cfd314358a2ae2dc1fe48613f : SUCCESS in 28m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7dc668749db443649425a2fe93ce1b84 : SUCCESS in 44m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a8a5b03cec59466ca1ec2a5775baa524 : SUCCESS in 1h 06m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c575acc1b34ea49601bf47d3af6067 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cba6d34351945fba1e35c2ea7a850fc : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae0eb3dc302744b7ad437f1a3c690cc6 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc800775b344d5e9089521e31edb93b : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa5060fba9924f9b8798e858eff03022 : SUCCESS in 3m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/aa4327562fcd47868eff4adb63eeaf21 : SUCCESS in 2m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/364ee596269e4af1a5f697d58f8e5e2b : SUCCESS in 16m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/03802fa5526a47ea977baabb03455d75 : SUCCESS in 29m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7007d59609e74f2181d7f2c077a92551 : SUCCESS in 1h 11m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1222ce32e3394632bef4c8eed9501cc5 : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a47232a0f3fbfdcde532c15bcdfda8f373a22181","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 20:28:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eef27e90593c150adfb5a76b2c13f3bb2cee8f33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 20:29:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ec27b086d9742e28de0407cf11f67bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abed5782746647cf9ce9a12d6b1cd793 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f60edba736f4457c9322c93a3dee34c1 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f9d8c0e1ced1e558f30b56b31f4fbeab0e78d09","revisions":{"4f9d8c0e1ced1e558f30b56b31f4fbeab0e78d09":{"kind":"REWORK","_number":1,"created":"2025-09-11 07:52:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/37/960437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/960437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/960437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/960437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/960437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/960437/1"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-11 07:50:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-11 07:52:24.000000000","tz":120},"subject":"gnocchi: Bump to 4.7.0","message":"gnocchi: Bump to 4.7.0\n\nThis is required for uwsgi module usage\n\nChange-Id: If1a078d2704fbd0b119d6c1ffc5a10aeff0a150a\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f9d8c0e1ced1e558f30b56b31f4fbeab0e78d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f9d8c0e1ced1e558f30b56b31f4fbeab0e78d09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
