)]}'
{"id":"openstack%2Fkolla~961438","triplet_id":"openstack%2Fkolla~master~I80070b3c3658948c7157564fb8a79e6dd427f1d1","project":"openstack/kolla","branch":"master","topic":"kolla_flamingo_release","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-18 09:22:17.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-17 15:58:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I80070b3c3658948c7157564fb8a79e6dd427f1d1","subject":"CI: Use UCA mirror for Ubuntu","status":"MERGED","created":"2025-09-17 07:22:14.000000000","updated":"2025-09-18 09:23:05.000000000","submitted":"2025-09-18 09:22:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961438-kolla_flamingo_release","meta_rev_id":"8340524e942e1350c618eea84eea8d5d7417e3b9","_number":961438,"virtual_id_number":961438,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-18 09:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-09-18 09:10:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-09-18 07:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-09-18 07:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-09-18 07:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-09-18 07:53:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-17 08:20:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-09-17 09:55:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-17 15:56:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-18 07:53:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"e97a43f6f0bd222a0457db87348af9dc434643cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-17 07:22:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f37f01ebdb8a2be35420a3bf2cad289aa3f5760","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-17 07:46:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"afc0df684173e5387edf9c21a8a3940529fe7c0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-17 08:15:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e46fdbdfbae2049925935a366606a90f72f908df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-17 08:20:22.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"39f61436ca37cf11f67634ca4577e17d6db0f49e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 09:55:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8d5e4c84274408bbaa6f2020b00b78e\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/d170ceeb203b466d968ea1a5db2f138f : FAILURE in 9m 04s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/3772025044434aaca8ed11848110583d : FAILURE in 11m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4fb07dfb1a0f409ca42e2ed49c651931 : SUCCESS in 16m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9bfd8f42deae419288cf1c010ef26f31 : SUCCESS in 30m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fcd2d0ab6db1450fbb348eff977c1eb1 : SUCCESS in 43m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f6a9128b5dab434da94fb96232f4b042 : SUCCESS in 1h 05m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb023c04beff4885ac6bf7ff4dd9305f : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13b5bea78e3a4fa497ad3a010ad51b7c : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc0fcfb4bb1349d08b8d8785922071ac : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1752dc0b52ea4e51a904445cd5847765 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bac6d2a18bd4c73a2eda5a48b30d972 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d3ea4b044d40aeb769c9317b176b9d : SUCCESS in 6m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fcd354f4b48a4235a0e61b251d4b3301 : SUCCESS in 4m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/edf89942533b470a84f2f8fcd1695a62 : FAILURE in 5m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/05e110a7aaf243138801602e7be37b11 : FAILURE in 5m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/228a6b2bb97641dc8510fd612bc2a064 : FAILURE in 6m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b6aa36d869f3408db0e76a9585c06b71 : FAILURE in 6m 26s","accounts_in_message":[],"_revision_number":4},{"id":"cdfe1caf19786913f1a3f96fe69373f26ba285ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 10:07:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9844073658ef4ed1b1c42b9b45fe1f0b\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3dbe9f61d0004d59b126a05c11f9f400 : FAILURE in 19m 56s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3a8fc7f897104e2aaffd958bb8633e43 : FAILURE in 24m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/08a2f206b7c841d0a0830b6953c1e3ff : SUCCESS in 37m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fa443782db824e67b81e06262f4726eb : SUCCESS in 1h 44m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/635ff337549e48c99fe37afe43e90428 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0444cfa2ddc49ef9b7237a4725c4178 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/31b77df5ede34bf9a6f36e2512db6156 : SUCCESS in 9m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/04a06124f7f8435a92d1c378d70d83ec : FAILURE in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4fbaeee0c0c8d111ce018b934d1f04838421c240","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-17 15:58:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"10dbeb5ec988f7b2a8a1915778a95813db69cf6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 17:35:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dba7f39e13c14ba5bda8d618e795c9d4\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/a57da04b86dc4ca9b4e60eea9cbb3845 : FAILURE in 11m 16s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/b42ea2b850184dec89fbd1f6c6a0868e : FAILURE in 13m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/89055aaa89114eceaf79429dcec6c404 : SUCCESS in 18m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eb57937ec3864e249b853ddac50262dc : SUCCESS in 29m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a1e2b9ef4f7b4daeb9558154c2bd8539 : SUCCESS in 43m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f48682e34c8749548c3aa4ccbb70eaf3 : SUCCESS in 1h 04m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43eda9e03a6d4f168705307de989007d : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fe01dc9bee94bd797d1a30c198fcc10 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7998e5e773d4422dbe8fa67b5b7c02e6 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51d82eabd7f541edb70a64fcfceecf59 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ed8f99bacd84b7a8cf87d1db8971d27 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b05413b1ee4f56883ca80078508403 : SUCCESS in 6m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/797e4e58e3244b05b781caaec79ec2fe : SUCCESS in 3m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3f28dd7b9ad54a23b9c556fcacd857b5 : FAILURE in 6m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/339f1e898afc4287baea5a19140ed778 : FAILURE in 7m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dc01fb480fbc4594b1b7b5f28b5926ea : FAILURE in 6m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3ecb6fce3bb144f5a5e82a879d06df25 : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":4},{"id":"70a1e05bece19e90b6a8305a93adba895efc7b6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 17:59:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bbc32afb9574c12b8ee1957170291d3\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/4049100c0ca34b65887570303e96dea5 : FAILURE in 32m 10s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/995a6ab57b4d4d13804c9ced8136007e : FAILURE in 25m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f1d145180c41463f86e79af5653b254c : SUCCESS in 50m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5572ee572c064ca19b8fa6c8c7d1097d : FAILURE in 1h 46m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e910b5f57ee54cf5aaa8b96e21c0dbea : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf2630c85eb244e992a2690d4605b8c3 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a91df8c1f7f84f38b974461808a8f72d : SUCCESS in 11m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7cefb1dbb8b94b6ea8517f650fa45806 : FAILURE in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ad1a1b7d57f955a6b2420460d668e9420d965b8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-18 04:46:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"933ca9480d469566ca230ecf5ab875c689c544c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 06:08:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/344530017d7443b2972c36e2fc7560f1\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/115711237bd2400e9b1e219ea91ca7d9 : FAILURE in 14m 15s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/5a6d7103c152464e9f09651909477b93 : FAILURE in 16m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4d8830d6d2414a0fb997e33fd6fc396c : FAILURE in 21m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/60e0a5d4c66a43bc81267a7072647a96 : FAILURE in 30m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/990b3cc97ccf4fe6b4f4379a7b2d262c : SUCCESS in 45m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b4ee73c12d66451ca417691d37ff2505 : SUCCESS in 1h 08m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c867313873431d995f085bd6a35970 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06f975be618f4a71a1eb536935912074 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0b50e99f24648dbaed4ee515b2b00b5 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ac6a260f7e54783abaff874c9d3a3f7 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3480fd8f28042e4a282f15a979995b4 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f9fc1abeeff4485b7646f40bc582d06 : SUCCESS in 4m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/38a62d99c49b454fa123af519c74f5a9 : SUCCESS in 1m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e88799c247584b18b45c0dfbad52e124 : FAILURE in 21m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/84981f24a3ee4f5182dabd1c0344f004 : FAILURE in 27m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5d32897acbb74a9dabbb63f9ba8586d7 : SUCCESS in 52m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a4483383fb844dfe9c188e87f53da112 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":4},{"id":"b76d6850031f59c2e457c5f2c46fd8795b0e85be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-18 06:19:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f4a39c2502cd8b5afa2157c33892af353158cc04","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-18 06:19:47.000000000","message":"Topic set to kolla_flamingo_release","accounts_in_message":[],"_revision_number":5},{"id":"b48af69f940d18a97864e8d6e656081d9e40a164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 07:44:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/917ae610ead549bb84a3ac70a5743cff\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7d9faf115c1c401a97543714fbd59b73 : FAILURE in 12m 19s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/c18f970892e84f5abf490809e261455b : FAILURE in 10m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bd1c6deb1093411db7b4a6a0acadbfdc : SUCCESS in 20m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/476b97a33aa44b08b439140f9f1f8064 : SUCCESS in 22m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5dc600250cd3463bae677c3d5268d940 : SUCCESS in 55m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e99632bb9bde47c6a93ce7dada62ae54 : SUCCESS in 1h 06m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fc69bcd045442fc8e9e5e10a7587914 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d838944e05c34df2a140782794190fc4 : SUCCESS in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/151dfb6c1cdc4390a14e8627b36b2455 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a96159b5d2d1430c9fd6734e1811ad3a : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c502390bfdd43b1bf10a4fdea10193b : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/571efc980cef40df8dcf8f4a50fc2741 : SUCCESS in 7m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2edbd4cea1014607acd8e2e14347aa1f : SUCCESS in 2m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/280ae9ed91034ba9942527738b99b0ec : SUCCESS in 20m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8f127429f8974ce5ba7a381976c49fde : SUCCESS in 15m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/83547722e2314408a0b48653c7c23de5 : SUCCESS in 56m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a5d5ddd26f0c48fcaf0e926f4fe62927 : SUCCESS in 1h 23m 56s","accounts_in_message":[],"_revision_number":5},{"id":"016ed0fb0d47e31dc10c5ba184f2653652239d1c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-18 07:53:21.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"46fadab63e902955da70f44780d9200605001ff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 08:01:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f2b9af0addc836b48e962b3e2f2d8c2cda061f71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 08:28:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b27cf204fb4dd7a74b80986100d67e\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a5467eef0da44ce1a83d42a1656c2d7c : FAILURE in 24m 36s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f862e39dcedc4a7bab45f953b170b355 : FAILURE in 26m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aaeb7ec877fa40ac981ad66455c9d9d1 : SUCCESS in 39m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f76b903b23e8452c865aec24f7b96254 : SUCCESS in 2h 00m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e1aaf0c32204c3581fa5684f64249b6 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/334c0de3b1d34c55b680a39cebec9ee0 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/86821689002e402c96e82397a265c6c7 : SUCCESS in 10m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5716a76376204cbfba2ff031eee133fd : SUCCESS in 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"add08c0773920eeddce6fa68f6cee66e11cb38ad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-18 09:10:32.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"823315c3c2886693621599c8a6cce574021a59ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 09:22:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d7c22b68af14bedb4f4032dfefa1f1e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f3c4268205814c21a96706198c81259c : SUCCESS in 9m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0ad9d39181134918be7ec35b4f5d33bf : SUCCESS in 28m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7502f546d5b04abc85a12fc5b20620c1 : SUCCESS in 43m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6b3ca039a94c47ba81cccc4bef81dd45 : SUCCESS in 1h 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1a11609c584d6c8a89b789b8833dc8 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4b7ad7955b946bb814c8cd0b9fadd53 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/474be776e1b34e13b5cc9bab7f7aa636 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47417cd7f8f44f1980d21ff275221cce : SUCCESS in 6m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/82e534d95f8142b8bc4fbe64d42b35e9 : SUCCESS in 4m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a90d38e5040643c5b1fd256a18ccd103 : SUCCESS in 19m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e91506d0ad784d4f9f729a8df09d3f80 : SUCCESS in 24m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e669fe95d511451a927bb670eca99f56 : SUCCESS in 59m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/353c5141e8ae4a6db0bf55f9196f7d79 : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":5},{"id":"33fac63d3b3469397327debe084cc7d7ed25e223","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 09:22:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"8340524e942e1350c618eea84eea8d5d7417e3b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 09:23:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f435731b8c7247538436eb4dbbfd00e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe117e3248541fe9d73c00cc460ba1b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"46e89ee3b4d43e5723978b48330433226d5dace1","revisions":{"b80ede6b340ac42d657d9a5ee9efb54b2f25acb1":{"kind":"REWORK","_number":1,"created":"2025-09-17 07:22:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961438/1"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 07:19:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 07:22:12.000000000","tz":120},"subject":"CI: Use UCA mirror for Ubuntu","message":"CI: Use UCA mirror for Ubuntu\n\nChange-Id: I80070b3c3658948c7157564fb8a79e6dd427f1d1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b80ede6b340ac42d657d9a5ee9efb54b2f25acb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b80ede6b340ac42d657d9a5ee9efb54b2f25acb1"}]},"branch":"refs/heads/master"},"2a22f783491356f6028b245111326469392a0ad2":{"kind":"REWORK","_number":2,"created":"2025-09-17 07:46:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961438/2"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 07:19:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 07:46:01.000000000","tz":0},"subject":"CI: Use UCA mirror for Ubuntu","message":"CI: Use UCA mirror for Ubuntu\n\nChange-Id: I80070b3c3658948c7157564fb8a79e6dd427f1d1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a22f783491356f6028b245111326469392a0ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a22f783491356f6028b245111326469392a0ad2"}]},"branch":"refs/heads/master"},"cd814f53a5dcabc32bb40e1a5770aac6d5b78e8c":{"kind":"REWORK","_number":3,"created":"2025-09-17 08:15:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961438/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961438/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961438/3"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 07:19:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 08:15:51.000000000","tz":0},"subject":"CI: Use UCA mirror for Ubuntu","message":"CI: Use UCA mirror for Ubuntu\n\nChange-Id: I80070b3c3658948c7157564fb8a79e6dd427f1d1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd814f53a5dcabc32bb40e1a5770aac6d5b78e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd814f53a5dcabc32bb40e1a5770aac6d5b78e8c"}]},"branch":"refs/heads/master"},"44cc6a34d8eb8788aab2a2c7198c75c41b7edd4c":{"kind":"REWORK","_number":4,"created":"2025-09-17 08:20:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961438/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961438/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961438/4"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 07:19:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 08:20:22.000000000","tz":0},"subject":"CI: Use UCA mirror for Ubuntu","message":"CI: Use UCA mirror for Ubuntu\n\nChange-Id: I80070b3c3658948c7157564fb8a79e6dd427f1d1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44cc6a34d8eb8788aab2a2c7198c75c41b7edd4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44cc6a34d8eb8788aab2a2c7198c75c41b7edd4c"}]},"branch":"refs/heads/master"},"46e89ee3b4d43e5723978b48330433226d5dace1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-09-18 06:19:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961438/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961438/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961438/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961438/5"}}},"commit":{"parents":[{"commit":"baa9d4ed9ccf0dc09bbe03f52aaf3ae7d4017611","subject":"cinder: Remove wsgi script handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa9d4ed9ccf0dc09bbe03f52aaf3ae7d4017611"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-17 07:19:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-18 06:19:34.000000000","tz":0},"subject":"CI: Use UCA mirror for Ubuntu","message":"CI: Use UCA mirror for Ubuntu\n\nChange-Id: I80070b3c3658948c7157564fb8a79e6dd427f1d1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
