)]}'
{"id":"openstack%2Fkolla~961638","triplet_id":"openstack%2Fkolla~master~Iaf3c4de859986629f9768563a319fe0385acff6a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-10 07:56:51.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-09 19:03:12.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2025-10-10 07:56:51.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-10-09 11:36:51.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Iaf3c4de859986629f9768563a319fe0385acff6a","subject":"Update Fluentd from v5 LTS to v6 LTS","status":"MERGED","created":"2025-09-18 12:12:16.000000000","updated":"2025-10-10 07:57:55.000000000","submitted":"2025-10-10 07:56:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961638","meta_rev_id":"78fb4ee2e9569abf274c4df80f9c86460120c78c","_number":961638,"virtual_id_number":961638,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-10 07:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-09 14:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-09 19:03:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-10 07:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-10 07:06:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-10-09 14:32:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-10-09 14:32:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 12:56:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-09-18 13:14:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-18 17:11:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 18:02:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-09-29 07:23:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-10-09 11:36:38.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-10 07:06:19.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"80c160882dbe4c27f5cf4054da43bbee57142a7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-18 12:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53d88e6c76ffc64baabc626e6ba48b5e5da0f163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-18 12:55:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d0a5fceffc2501838301ae6e4dc6c19520e3d6f1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-18 12:56:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74b57458790b98b041dbd212763ac4f6b4eb6e17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-18 13:32:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2bdbab0ef62d1544d8588becfe7a34e02ffb35a7","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-18 13:34:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"373421344899c5bd5de9090b9e2b25aba61ed016","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-18 14:24:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c930dab92f37822555e0b0de594f9d7053ec7d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 17:11:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb41c86b22724828a43d12389eacda16\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/afa51638569f4125ba99f11625beca02 : FAILURE in 4m 41s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/29db46b1eedf4112b5c969338b72ee28 : FAILURE in 8m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b1dc5ee9b9374e9e868a4b79ba40fbb6 : SUCCESS in 19m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ca4b195e9076425d93bb7ca5ae2633b0 : SUCCESS in 30m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b02f28a1fc334822a4bd58f6507e6dcc : SUCCESS in 44m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c3e854f3f47443c49491c14e4a8b28cf : SUCCESS in 1h 06m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/173a1a4481cd42ea832c1e41c4fa8935 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d0ee9851b214f83a2136f1869bbd9f9 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ab40cbd70f64500bddbdbf3220302b7 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1533c5540524b7ca33f431c5641ddf7 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85af4406f6d341589d39ae510cc6e24c : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af77ba5b56554b67a19833a05f94a071 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1e8b65461b04cc09e48fda0398aeffe : SUCCESS in 3m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c85242c14e2741d78503f75cc8ca6d0a : SUCCESS in 2m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0ca71e291e634d8f851a7402d0628045 : SUCCESS in 18m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/032c63d4e053490e9a5bca37c50e463d : SUCCESS in 28m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b893a2b8ef08400899457698f3afc097 : SUCCESS in 1h 05m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/59a576a32cb847aaaa761537ffe2a10e : SUCCESS in 1h 28m 34s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0a96e09bbfbc4ca4bcf81a23df7dc31d : SUCCESS in 1h 31m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/195ec071323f480f8862bace57a5cf65 : SUCCESS in 2h 42m 39s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/045ad4395a4c4ee885bbfe07f7a292c0 : SUCCESS in 1h 05m 20s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/979e91f4a73545229cdf787a574967fa : SUCCESS in 26m 07s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4467f13e182046d699359275dd7bbcb9 : SUCCESS in 1h 04m 32s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9aa71c34cbf64eeeab6ae7d5b353765f : SUCCESS in 49m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4cc97a31d61a45e58d329abf057f5dc9 : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/95a522002ea947bf86d7202b14a645d2 : SUCCESS in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c3d0c637e831e08052220130c132e3342803fc8a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-18 18:02:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"925422bdba8683f80232b0049c74c010aed63f51","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-18 18:38:08.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4a25f495e5006acd4d9a5f3ffa2b1e266524adfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 19:29:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a3e751454a46a59838ca74960d6736\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/50671f6a7f3e462cad5eb5cd9dccdbe2 : FAILURE in 9m 36s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/69109c4d8b484707a044fd3b5f4a97f3 : FAILURE in 15m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ef49e24622d478c961d51b351fb6eff : SUCCESS in 28m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/76d48e8cc0ab4527b39f2fa4fec9990b : FAILURE in 1h 42m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b2f1c4320b284dabb33d7e12883a2126 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aae9d98775a64b5baa925c2db08f973e : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d16043ea00549b999d3baa29708249d : SUCCESS in 12m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9b60be3d884a4e72b3293852cf04cfd9 : SUCCESS in 1h 08m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0e861efda86da6cac594c9cd0eebb02cbd7aa63","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-29 07:23:16.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"80e1744a44301c976c77b52eea27f8258f8b78da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-29 10:02:46.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"53bdf8329503e2affe739eed9cbc56890a522d6f","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-29 10:03:37.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d7d7a9d6b7bd22a688c793fd9b491f81d0bac771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:18:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59cecae250094aa88c6991adef179889\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/25aebd8b032b49d5a261027a81042eff : FAILURE in 2m 18s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/7afd47425f1c499990177f46c2e400ba : FAILURE in 5m 11s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/0bfe1999e10b4cfaa0e9c4efed2dcb29 : FAILURE in 3m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc650704484344a9b1b3608294b5501c : FAILURE in 3m 28s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d97e45b166f945c6b217138c67815059 : FAILURE in 3m 15s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e6e248f072e046fe93f91711bd2f60cf : FAILURE in 3m 56s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/865eb002cd8747b5b030f04c021b5098 : FAILURE in 5m 39s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/065a2fd4bc0e493e9c966c34a56b5852 : FAILURE in 5m 35s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd30ee5c1a1f4913a055cfd4e1364b4f : FAILURE in 6m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b943eb373e4449a2bb34f0fbf9fba9a3 : FAILURE in 5m 54s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/57db120dc2e740a996490aff54def0d4 : FAILURE in 6m 46s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0ac0e7a985004b08ad69905419634c8b : FAILURE in 6m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d05e0b2b7d874d0a899917fd8fe65709 : FAILURE in 7m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/aa8744fb12564c8c81a7cf37f7a0466f : FAILURE in 6m 47s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/b258df278fb94b539fd8b7b509660e60 : FAILURE in 5m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/09c5369a0cc44a88a18e0d427c686ecb : FAILURE in 7m 02s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/9a93503eeaee4761bf20866e6aee5b9b : FAILURE in 6m 16s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/69d821ca64cd40459833232b0de76c8f : FAILURE in 8m 56s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/3e62f834b131476da2f77467c87ca6a5 : FAILURE in 7m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9063d38308bc42deb0b408a5aa95e079 : FAILURE in 7m 54s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/55f421f2b11c4ea2ab24f57afd26fa1a : FAILURE in 6m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/edc74f2c0095452599bd3983e73f1fb7 : FAILURE in 6m 27s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/0be124e645ef4e2097e09a6b9397cbdd : FAILURE in 6m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/45cdf98613a440078c0ce8e19d1891a7 : FAILURE in 6m 43s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/69f85ad7936a42ee929f2ab7d960c63b : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e7c92bf37045b49d8ce632c40d95f5 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a40706661312499e82d611b3dc3c985b : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7e73a30e3bf446f8e7da9e3370b497f : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/532ee613f93247c7a2db767b26ab5981 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa24414fe6d4481f8cd15eb20a4b2b33 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70adb7da2514d798d91d515b975d101 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c14df01c7374962afe5d73e7788a891 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":5},{"id":"af79a3d31fcdc333a37c1c63b2c0b822b35067d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:23:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f203589e2cf4d3f9bff16fc200371b3\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b9b1cea49b12401f813f0a1a675bf01d : FAILURE in 6m 05s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/b35c65afe9d647dd96f3608718c52d8f : FAILURE in 5m 28s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/da79ecfdc36d46228215516ea008df19 : FAILURE in 5m 39s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/efbd92797a6c4cda9e2eae3413a3ead8 : FAILURE in 11m 34s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/df953a13fe104c3c885edd8a8e4a0eef : FAILURE in 11m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5052729242f8436eb82223f0b4ae9a06 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc2f02abda7b43c28f62b034c73cb8fc : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84494e8f5eff4774997d7b0458439815 : SUCCESS in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3f629705cf7ef574714a2587f3848a5c1d842586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-29 10:41:26.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6da32d0f63db258793e68ad86cdc5c0c8d7eb40d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 12:09:01.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1ee713b92180197307f55b60e878f69bbc179ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-29 12:13:08.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"fce4bd81c11e635927d199d2cd64a73db2d6fc2c","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-09-29 12:13:28.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3b3546c36ae4c5128e5c75274022256789b33ad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 13:32:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d885dadc9a1a4af381f680bf7920fb79\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/c47de851ceee4eea85522c37dde4cd0c : SUCCESS in 40m 22s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/fc364bd623974f05a992e471866c7864 : SUCCESS in 28m 54s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/aef75b5c1b22412496e0a5d207f01fff : SUCCESS in 34m 03s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a05d603ee3724bc7abb2a1aeeab5a03b : SUCCESS in 1h 05m 37s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/8ddaf6dff6554a6889e5f57e82b1a175 : SUCCESS in 1h 16m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e44639c9b794b7fa3967fc43af12aca : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8b78cded8ea42fb9828c214207d54e6 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a26ccf789dd94bd292d2359ee7f015a9 : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"69a05bd9d4882ae440b58575c9c741ec077afa57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 14:55:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76107a3c37d746109eeab0d992b0f2f5\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/18ce0ed4a35642098d5357b4aac81509 : SUCCESS in 10m 13s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/83957de006c648c0b8eb8c5ba71555c5 : SUCCESS in 10m 43s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/fb64815418864c3db2d5fc583c892258 : SUCCESS in 29m 02s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd456e28f7764670854a0a13563ee9d4 : SUCCESS in 10m 16s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5dcac645a2bc4c6691f595afeb0bf034 : SUCCESS in 22m 31s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/cafbc327ca654e8588b279ed0ddfd111 : SUCCESS in 33m 21s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/f6f3ba63da244be4a42dd69737674a54 : SUCCESS in 43m 26s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/fa7d3c7fe8774983b24a8419f4960eba : SUCCESS in 1h 04m 09s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3068b05791d4c81bf7c7844e780fec5 : SUCCESS in 48m 36s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/975e94d9d02443c7b849ba1d72626230 : SUCCESS in 1h 07m 48s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/039a0c228df4494794c691d88684f067 : SUCCESS in 1h 31m 59s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/912cb1ffd9fc4bf696fea5126c1d6e30 : SUCCESS in 2h 25m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/25fdbe1b55e842bab2709987380fe403 : SUCCESS in 1h 53m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/bf892e1ac1be457486c36f1484165e06 : SUCCESS in 2h 41m 23s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/aad39ff7b24e4838b09edbe1ab812dc7 : SUCCESS in 1h 05m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/46031f0468684ad8b8c66d44ac82147b : SUCCESS in 1h 17m 12s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/9b70848d03654759bd0b7def174b81f2 : SUCCESS in 21m 17s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a1180dbdb7b848d7a0fd4dcc34f2edc1 : SUCCESS in 15m 16s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/775414ce219e41098549b0459f4b9eee : SUCCESS in 55m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/b489c6518a7747c2aae3a989290297da : SUCCESS in 1h 00m 50s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/2bdc408625e7440ba519aced0ecadc03 : SUCCESS in 49m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/5304332eab7c4fb69f4ea1579e80913c : SUCCESS in 53m 17s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/4e6a2a372a014824b237b74530047843 : SUCCESS in 1h 16m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/777a7c69b9db48e1bd7262728759374d : SUCCESS in 1h 22m 07s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7756d73cd518485f84fcf8d3429cbe6f : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e0fda7bf80f48a1bf59000ca0aa9c20 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/137adf68a0a04608b7ccd73492455554 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e72615fdc520491b947954701651e51b : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/337dfc501da944248329a019aee11177 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de914870d33b4e3286dccc783461165c : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bdfc8a9b98e4e74bc8ced2f1a036121 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ece51c5586e45259005df5f60c6cff4 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":7},{"id":"60f25985a5c6ba838a8df9dec724291d54a8abd9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:36:51.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n* kolla/template/repos.yaml\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"30db4fc256d2f67958a03fd1d66d469f66c77af8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:37:58.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"7846ee24d6dbaeb0f0bfbe121e560f5b22304faf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:38:37.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"ff146c6c023f339b7f7b2e2dbafea94e08b6ca29","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:38:48.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"e80256cb2b4fcda876931154c52133467b3265d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:39:10.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"d766678419f6bd3d71bf87725ae03d62d7858027","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:39:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":12},{"id":"0e7c96a82227eabd06e2a6cf3ffe0ce1dbb1a1ee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:40:07.000000000","message":"Patch Set 12: Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"f3a889c57a03542f46a1199914506b937ba048b4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:40:11.000000000","message":"Patch Set 12: Backport-Candidate-2","accounts_in_message":[],"_revision_number":12},{"id":"42131de6b2ab03341e60159de3824e2b1f800e80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:43:35.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":13},{"id":"521c5f05198461e0aa7443e5e538b39f0d66bb54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:45:48.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":14},{"id":"e505e19412f6200395a60b88b4a2fcb3642bd1cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 14:27:03.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40678169da194ce9a185dd766596beaa\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8cd566af143e4dfc882e4b5dbced80a3 : SUCCESS in 19m 40s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/869d684828e3433b98a705d4644cfb8e : SUCCESS in 21m 48s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/378b12601ba04c85bf5c1386c30a4f34 : SUCCESS in 29m 32s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a072e494c4144924adb30d9fd476a170 : SUCCESS in 18m 44s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c34d2c6c1c5a4a02bf5ce76f94eade0b : SUCCESS in 48m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad0f42780cf5497982df7ca1e4833bb3 : SUCCESS in 18m 41s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5a0b2ef9cf6d4448b98ae84a08cb6fa1 : SUCCESS in 20m 38s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/bf0b6aa1d46b4d65b4cbdd3e71b971f1 : SUCCESS in 2h 07m 06s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/122ef75d75944722b63185fa3f14abff : SUCCESS in 2h 40m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/8fd4ae174c5543b09dba7f7bf9562325 : SUCCESS in 1h 50m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d46da7e532f5471c981a201e4964ec83 : SUCCESS in 2h 32m 19s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/f7242ed64ea14a889a9ca6a34305c3aa : SUCCESS in 1h 29m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f0e90495170949fd932150cf523da6c1 : SUCCESS in 1h 22m 47s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/706f4516ea9b411682751d895c234f8f : SUCCESS in 22m 22s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/4c47e0ed44354d6aadde716b27747391 : SUCCESS in 15m 40s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/424c8395a7dc46ac8531dee5fa6055aa : SUCCESS in 56m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/4bb2734b235849f886948ce45e1572ff : SUCCESS in 55m 12s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/b9985a6711e247e5a12b6692c4feb255 : SUCCESS in 47m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/862e4160a7fd4b7baabc483db5a489b7 : SUCCESS in 49m 46s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/a8d6458dcbec4888a523577ab75bed47 : FAILURE in 1h 14m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/709b06fc95314dc5bf2fcb1d38422adf : SUCCESS in 1h 31m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f2c9c2e98e6b4b7592d612e543644091 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a23253e965954ed39af5f70d689ce6e3 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa082fe70f2c45b89f1b94012877ec90 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c7bd888bbee448f8c6f86c0715b64ba : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fab8b0d36e348ea89fa5b3b0e61fd4b : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5eadc82878848a383f135b5d6b5ac6c : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/799141328e124b2487851c44d6f7a548 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61a2f94b271744fe8efb197365ce680b : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":14},{"id":"5728dccd0ced35d3854d0775e2462a2d04ffd5b9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 14:32:07.000000000","message":"Patch Set 14: Backport-Candidate-2 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"7b7f1cdd6319096a83ea8b328b21d2eb33b83da1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 14:45:28.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6645e3edea654aa8ba5b25662a962d62\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f08c9790c1ed483a9d9e2b27e8eacee9 : SUCCESS in 52m 20s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/183e888ccbdc4634a0d036ea7d200869 : SUCCESS in 53m 04s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f4d8936d001148dcb3a545a6f45dff72 : SUCCESS in 37m 34s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f46f8e6c0d904ccdbef180f45ba6d224 : FAILURE in 36m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0fb280949cb940e8a9dca8a602806bb6 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1cab24d7d1784e47a2de9ff81cd77953 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62ed2be223b14aa3ac149106ff3d15f2 : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"37b1833d3c296977921dc859d4a674842c98ccc8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-09 19:03:12.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"37ebcf3ae437b4fc6ca1b26a56ea7ff8d56cf62c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-10 07:06:19.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"45c966a06ae4cc1b4344f45a82017c29c1d82b89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 07:06:29.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"bc3cb0c8228ec524fcb36ae7921fd6e1b1a6c5bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 07:56:51.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7fc1f59c4744c92a5dbb43f76d2794e\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/5c1e4fde40894f86839f2f862e2d1d54 : SUCCESS in 11m 58s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/a82f0eaa6ad3489eaeee0ce2ed80bf25 : SUCCESS in 17m 03s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/21469eb428e64ae49968258865ede335 : SUCCESS in 20m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c33d3de31a4d4167b8e6e4b6e463a44e : SUCCESS in 49m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5caa13028a064eadb0c1370a97f437ef : SUCCESS in 22m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/095649a0fc7843fe9b8f0f007e171d9b : SUCCESS in 26m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bfd6a5257fbb4102952b450ccee7489e : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdbc5778e9d44d9bb42ee00be65173df : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d8599d15bbb4d0aa2ca3a2ed61177f6 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eae9a2b87da640bca029c4caf7b9f5d7 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e39c99cb27d402d90629192c609ac44 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a66e45bc9e64c81accae65d75265d9f : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":14},{"id":"e03f28aa1292d66323457a84c225dcb5477329e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 07:56:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"78fb4ee2e9569abf274c4df80f9c86460120c78c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 07:57:55.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e9d5535898848b9bd48dbf83ea385e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af6d5e01939142cfba2e3f85d7765cf2 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42accee918f2476282959ac91077351f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"79726bd9d5e3d23558b4495a84be9e12b0714b3f","revisions":{"4987d085454c67b444cc74eb1094f4505f91befd":{"kind":"REWORK","_number":1,"created":"2025-09-18 12:12:16.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/38/961638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/1"}}},"commit":{"parents":[{"commit":"46e89ee3b4d43e5723978b48330433226d5dace1","subject":"CI: Use UCA mirror for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:12:13.000000000","tz":120},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu Noble.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4987d085454c67b444cc74eb1094f4505f91befd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4987d085454c67b444cc74eb1094f4505f91befd"}]},"branch":"refs/heads/master"},"a9dd37ea57a4823a0c355c7be850785ace818272":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-18 12:55:52.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/38/961638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/2"}}},"commit":{"parents":[{"commit":"46e89ee3b4d43e5723978b48330433226d5dace1","subject":"CI: Use UCA mirror for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:55:52.000000000","tz":0},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu and Debian.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9dd37ea57a4823a0c355c7be850785ace818272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9dd37ea57a4823a0c355c7be850785ace818272"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e06049e4cf32db50def28c55646875a2f4d4a07f":{"kind":"REWORK","_number":3,"created":"2025-09-18 13:32:53.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/38/961638/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/3"}}},"commit":{"parents":[{"commit":"46e89ee3b4d43e5723978b48330433226d5dace1","subject":"CI: Use UCA mirror for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 13:32:32.000000000","tz":120},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e06049e4cf32db50def28c55646875a2f4d4a07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e06049e4cf32db50def28c55646875a2f4d4a07f"}]},"branch":"refs/heads/master"},"f67264c73e4b8f8b7779c210bc1aa67714ee1c04":{"kind":"REWORK","_number":4,"created":"2025-09-18 14:24:25.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/38/961638/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/4"}}},"commit":{"parents":[{"commit":"46e89ee3b4d43e5723978b48330433226d5dace1","subject":"CI: Use UCA mirror for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 14:24:12.000000000","tz":120},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f67264c73e4b8f8b7779c210bc1aa67714ee1c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f67264c73e4b8f8b7779c210bc1aa67714ee1c04"}]},"branch":"refs/heads/master"},"b8e8063fa63bc169f3c77d7b0cc46fa1ce4ba740":{"kind":"REWORK","_number":5,"created":"2025-09-29 10:02:46.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/38/961638/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/5"}}},"commit":{"parents":[{"commit":"46e89ee3b4d43e5723978b48330433226d5dace1","subject":"CI: Use UCA mirror for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-29 10:02:31.000000000","tz":120},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8e8063fa63bc169f3c77d7b0cc46fa1ce4ba740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8e8063fa63bc169f3c77d7b0cc46fa1ce4ba740"}]},"branch":"refs/heads/master"},"3641312ee23cf78a75cfe235367f00b7de7a6275":{"kind":"REWORK","_number":6,"created":"2025-09-29 10:41:26.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/38/961638/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/6"}}},"commit":{"parents":[{"commit":"46e89ee3b4d43e5723978b48330433226d5dace1","subject":"CI: Use UCA mirror for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-29 10:41:15.000000000","tz":120},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3641312ee23cf78a75cfe235367f00b7de7a6275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3641312ee23cf78a75cfe235367f00b7de7a6275"}]},"branch":"refs/heads/master"},"c676598cabfd98baca3493ef7ff6cb9290aac568":{"kind":"REWORK","_number":7,"created":"2025-09-29 12:13:08.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/38/961638/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/7"}}},"commit":{"parents":[{"commit":"46e89ee3b4d43e5723978b48330433226d5dace1","subject":"CI: Use UCA mirror for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e89ee3b4d43e5723978b48330433226d5dace1"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-29 12:12:55.000000000","tz":120},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c676598cabfd98baca3493ef7ff6cb9290aac568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c676598cabfd98baca3493ef7ff6cb9290aac568"}]},"branch":"refs/heads/master"},"b72cac6789bfd9ec71317c3d46827d8cf56801b7":{"kind":"REWORK","_number":8,"created":"2025-10-09 11:36:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961638/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/8"}}},"commit":{"parents":[{"commit":"90d255daa32be1c80621eb94cba87a2bded570b2","subject":"neutron: keepalived wrapper script does not rm containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d255daa32be1c80621eb94cba87a2bded570b2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 11:36:51.000000000","tz":0},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b72cac6789bfd9ec71317c3d46827d8cf56801b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b72cac6789bfd9ec71317c3d46827d8cf56801b7"}]},"branch":"refs/heads/master","description":"Rebase"},"e1572173a00b1ed01090eca671451cc3944eebad":{"kind":"REWORK","_number":9,"created":"2025-10-09 11:37:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961638/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/9"}}},"commit":{"parents":[{"commit":"90d255daa32be1c80621eb94cba87a2bded570b2","subject":"neutron: keepalived wrapper script does not rm containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d255daa32be1c80621eb94cba87a2bded570b2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 11:37:56.000000000","tz":0},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1572173a00b1ed01090eca671451cc3944eebad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1572173a00b1ed01090eca671451cc3944eebad"}]},"branch":"refs/heads/master"},"4e2fcfca5cd5710fcddeca3495ed5acdb5a09fc9":{"kind":"REWORK","_number":10,"created":"2025-10-09 11:38:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961638/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/10"}}},"commit":{"parents":[{"commit":"90d255daa32be1c80621eb94cba87a2bded570b2","subject":"neutron: keepalived wrapper script does not rm containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d255daa32be1c80621eb94cba87a2bded570b2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 11:38:37.000000000","tz":0},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e2fcfca5cd5710fcddeca3495ed5acdb5a09fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e2fcfca5cd5710fcddeca3495ed5acdb5a09fc9"}]},"branch":"refs/heads/master"},"8035bc9e8d1d18a3d4ba6b0a630537fc837b648f":{"kind":"REWORK","_number":11,"created":"2025-10-09 11:38:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961638/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/11"}}},"commit":{"parents":[{"commit":"90d255daa32be1c80621eb94cba87a2bded570b2","subject":"neutron: keepalived wrapper script does not rm containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d255daa32be1c80621eb94cba87a2bded570b2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 11:38:44.000000000","tz":0},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8035bc9e8d1d18a3d4ba6b0a630537fc837b648f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8035bc9e8d1d18a3d4ba6b0a630537fc837b648f"}]},"branch":"refs/heads/master"},"5bace178cfc3df647628ca1e21641e284d13970d":{"kind":"REWORK","_number":12,"created":"2025-10-09 11:39:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961638/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/12"}}},"commit":{"parents":[{"commit":"90d255daa32be1c80621eb94cba87a2bded570b2","subject":"neutron: keepalived wrapper script does not rm containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d255daa32be1c80621eb94cba87a2bded570b2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 11:39:10.000000000","tz":0},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bace178cfc3df647628ca1e21641e284d13970d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bace178cfc3df647628ca1e21641e284d13970d"}]},"branch":"refs/heads/master"},"cd192da437d2fd122875f1abcf4fcb53306401ac":{"kind":"REWORK","_number":13,"created":"2025-10-09 11:43:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961638/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/13"}}},"commit":{"parents":[{"commit":"90d255daa32be1c80621eb94cba87a2bded570b2","subject":"neutron: keepalived wrapper script does not rm containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d255daa32be1c80621eb94cba87a2bded570b2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 11:43:35.000000000","tz":0},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd192da437d2fd122875f1abcf4fcb53306401ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd192da437d2fd122875f1abcf4fcb53306401ac"}]},"branch":"refs/heads/master"},"79726bd9d5e3d23558b4495a84be9e12b0714b3f":{"kind":"REWORK","_number":14,"created":"2025-10-09 11:45:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/961638/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/961638/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/961638/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/961638/14"}}},"commit":{"parents":[{"commit":"90d255daa32be1c80621eb94cba87a2bded570b2","subject":"neutron: keepalived wrapper script does not rm containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90d255daa32be1c80621eb94cba87a2bded570b2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-09-18 12:10:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 11:45:48.000000000","tz":0},"subject":"Update Fluentd from v5 LTS to v6 LTS","message":"Update Fluentd from v5 LTS to v6 LTS\n\nFluent Package v5 LTS will reach end of support at\nthe end of 2025.\nFluent Package v6 LTS will be supported until at least\nthe end of 2027.\n\nThis change updates the apt repository configuration and package version\nto use Fluent Package v6 LTS for Ubuntu, Debian and RHEL.\n\nNo changes to existing Fluentd configuration files are required.\n\nChange-Id: Iaf3c4de859986629f9768563a319fe0385acff6a\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79726bd9d5e3d23558b4495a84be9e12b0714b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79726bd9d5e3d23558b4495a84be9e12b0714b3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
