)]}'
{"id":"openstack%2Fkolla~961926","triplet_id":"openstack%2Fkolla~master~I3193a1da91d89599d9287c12c332c20b56a89e43","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I3193a1da91d89599d9287c12c332c20b56a89e43","subject":"DNM: Test firewall settings","status":"ABANDONED","created":"2025-09-22 11:34:19.000000000","updated":"2025-10-14 11:28:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"ae510c4b88a3482e42fe46881411a0e468f8c7ee","_number":961926,"virtual_id_number":961926,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-22 15:27:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-09-22 12:59:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4264c3c7a4faf798c0d537bda6512e7512c805e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 11:34:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19f1ab9b1a3812c1f0e61167269866300e538828","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 11:47:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"158237c2f55e2f2b1f6b9d3781e518a19026164b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 12:59:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/069f927afbfe4547b8e64a1d82ae2845\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/fadc967f3ce64683928522c273991b23 : RETRY_LIMIT in 1m 33s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/1600bcb536a34df48ce272eb63b130f0 : FAILURE in 18m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e73c723de1a443cbab3d5a4b14c48848 : RETRY_LIMIT in 1m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/abbb02f8f8f34ffe997881f2e54bd276 : RETRY_LIMIT in 1m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/304218de0c4a49c2b7af4b038c89979f : SUCCESS in 47m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d3822c979aea46f7bca4089f05b2420d : SUCCESS in 1h 09m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5152f10b24488cbb8776e6b827b486 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a41e1ad37575453a83d35cd271fbd07e : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3de48127d0af4d12807e54e4865070b5 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f7ccc7351c14e2d9d31beceb8872f44 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96c437700e124d04a84af56443b0924b : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d9527c387e4beeb0d727edd6331c7a : SUCCESS in 6m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ed19c4c6f8cb44f9bfd8ce8c139391e0 : SUCCESS in 3m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f54bb28d08894a868bb02b5d52199e96 : RETRY_LIMIT in 1m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/97c826fce71b4397b8e0742baba55d42 : RETRY_LIMIT in 1m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e220e201f9ef408c9cb06affadcfc8a4 : SUCCESS in 54m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/89bf7724c38840bcadd756521ebff6df : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":2},{"id":"76656d40b49144738dd55ded22b167bca043886f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 14:10:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ddc971f10f74aba86484bbd7caab8f8\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/688ed966a32e483881365be313c50124 : RETRY_LIMIT in 2m 51s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/ed11e05c13794cafa16d3e3ed9fd34c7 : FAILURE in 13m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0a61e6ce199f48b8bc303d5d611d653e : RETRY_LIMIT in 2m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/11527ccf884040efb9fc7a3d080383a3 : SUCCESS in 2h 03m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0572b36b79c47959ab59a0ba1877fde : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c6c1479168c4e5d994994ba37dd9e5c : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30b89ada841f40d18e76f45fe963c279 : SUCCESS in 5m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e3413d884e194b2c98f0e11d6140c44f : RETRY_LIMIT in 2m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a72b04a695a1a8e89132fa4280b029b8360af40","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 14:16:11.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b8a3c3ce9ef771ba77482269fb1252ba8bafa4f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 15:27:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9b87ff9811c435f867900db46c491dc\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/79833c55f9f14865bf9eb42fc35fb520 : RETRY_LIMIT in 1m 34s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/46b397f7fee94ec2977494c670e10302 : FAILURE in 10m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/abd587b8da214520815c61d6465cab24 : RETRY_LIMIT in 2m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/62fd28b69ea1430098d3db628d31229f : RETRY_LIMIT in 1m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c8fa5bf791e14d26997084e6b7803e04 : SUCCESS in 46m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/400ac652fe234e72a8b403138c45794f : SUCCESS in 1h 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1224de1c6b004bd79676a6c874ae0fc4 : FAILURE in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13cd56ec82c2441bac5087a00269e10a : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7d79423a9884c9dbc1d0c6691aa2e1b : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38b54ca1fbd94be88b45c9b91a6ed7af : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/522e858bee1146859df250f2f0e27a15 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53388a3f8310464c8a7996a5fdf44878 : SUCCESS in 6m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3012376a8b224295a577f6a181a7a9ad : SUCCESS in 3m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/09a53f165b1746e88ed7b7a5fba427c6 : RETRY_LIMIT in 1m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/beef0812907740eb89d54e14c0566142 : RETRY_LIMIT in 2m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/287d40203a7a45b585bbb2e446619bd9 : SUCCESS in 46m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1f711eed3c514e8ea749768192ce1bee : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":3},{"id":"62ec45b5ff9d014a6508eee235600dfe43ab652a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 16:03:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84df7d608d0d4e539f051260f78849ca\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/d5d078176f314e0e8f9920dbd3182637 : RETRY_LIMIT in 2m 00s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6a0232ca03114474a955d2e89667a3ba : FAILURE in 21m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/15b9db23798641c78b0d2983d283c5f6 : RETRY_LIMIT in 2m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ab982a2f847445a5a5062529c0c4ef6a : SUCCESS in 1h 43m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0664599eb8934d1191430cab09e8fa9e : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/48b7b16e2d3343c2b258947f010fa796 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/838d6a24d34a46cd8e4a516375f73755 : SUCCESS in 6m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9104626432ad41b2aca33cb2d34d8b32 : RETRY_LIMIT in 2m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ae510c4b88a3482e42fe46881411a0e468f8c7ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-14 11:28:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cdbcbcb81957b2b8bb4df84cf67faf12f5dee8a9","revisions":{"23cec28e40dd04d8c4c4c038a986037e303c0820":{"kind":"REWORK","_number":1,"created":"2025-09-22 11:34:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/26/961926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/961926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/961926/1"}}},"commit":{"parents":[{"commit":"8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e","subject":"Merge \"Drop ironic-inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-22 11:34:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-22 11:34:17.000000000","tz":120},"subject":"DNM: Test firewall settings","message":"DNM: Test firewall settings\n\nChange-Id: I3193a1da91d89599d9287c12c332c20b56a89e43\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23cec28e40dd04d8c4c4c038a986037e303c0820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23cec28e40dd04d8c4c4c038a986037e303c0820"}]},"branch":"refs/heads/master"},"645eae7dd9399eb117e742662e7773797824b33a":{"kind":"REWORK","_number":2,"created":"2025-09-22 11:47:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/26/961926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/961926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/961926/2"}}},"commit":{"parents":[{"commit":"8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e","subject":"Merge \"Drop ironic-inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-22 11:34:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-22 11:47:59.000000000","tz":0},"subject":"DNM: Test firewall settings","message":"DNM: Test firewall settings\n\nChange-Id: I3193a1da91d89599d9287c12c332c20b56a89e43\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/645eae7dd9399eb117e742662e7773797824b33a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/645eae7dd9399eb117e742662e7773797824b33a"}]},"branch":"refs/heads/master"},"cdbcbcb81957b2b8bb4df84cf67faf12f5dee8a9":{"kind":"REWORK","_number":3,"created":"2025-09-22 14:16:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/26/961926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/961926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/961926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/961926/3"}}},"commit":{"parents":[{"commit":"8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e","subject":"Merge \"Drop ironic-inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-22 11:34:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-22 14:16:11.000000000","tz":0},"subject":"DNM: Test firewall settings","message":"DNM: Test firewall settings\n\nChange-Id: I3193a1da91d89599d9287c12c332c20b56a89e43\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdbcbcb81957b2b8bb4df84cf67faf12f5dee8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdbcbcb81957b2b8bb4df84cf67faf12f5dee8a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
