)]}'
{"id":"openstack%2Fkolla~961930","triplet_id":"openstack%2Fkolla~master~I35548a02f652c035c17aabe149752322ffc869c4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-23 10:24:48.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-22 14:45:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I35548a02f652c035c17aabe149752322ffc869c4","subject":"openstack-base: Fix psycopg2 install on EL","status":"MERGED","created":"2025-09-22 11:39:39.000000000","updated":"2025-09-23 10:25:36.000000000","submitted":"2025-09-23 10:24:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961930","meta_rev_id":"7e8ea8e7268af2e5d98e8e269d750dc8efa7d983","_number":961930,"virtual_id_number":961930,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-23 10:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-09-22 14:45:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-09-23 09:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-09-23 09:52:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2025-09-23 09:52:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-09-22 14:16:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-22 11:56:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-22 14:01:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-22 14:07:44.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-09-23 09:52:38.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"3cece0f06d38c74cc27f47b95f69d60e72d0b7bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 11:39:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8957c9e21ea615e37714b84c119bb52fc9f2199e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 14:01:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0846e89f41d4dfc91b79def0d71a2fe\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/1320ed1f9c6e49a98b5084f0926e9059 : POST_FAILURE in 16m 36s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/7337f053b11f40358c98cd7e248c3196 : SUCCESS in 40m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/42bf5f8763cc49b9ab6cb9dbef3ad227 : SUCCESS in 18m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/251f361b2e3b4f1fb929c9750686a54d : SUCCESS in 28m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bea9eec89b524695abc677f5d7eefb57 : SUCCESS in 44m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7beb65bdc8f640d49b7f560f3d43a235 : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8af58bc08fb4d4aaf109451963dc3ad : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05a06056dccb4a108a708e959e4e954c : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/093d4df462d540e08eb1e1828470d7ab : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fff6cf43e32422a96126d08778bbf72 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0464db375cd34886b2fc1eb719fe78e4 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe5708af925444d9b60ba5ffb873e628 : SUCCESS in 6m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b1955e6d17c94b56a7a12b026f68b559 : SUCCESS in 3m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cda53076ec8f49db89ec8b2b0ab12eb2 : SUCCESS in 22m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3894e15cfc7c4576af511556737cc4b3 : SUCCESS in 22m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ee193dfa4615489da3fe6086b75d8d8d : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d38d87965d6843d2aebc00c676fda4b3 : SUCCESS in 1h 24m 56s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/98e59638d76c4876b1738f29d2f3aed1 : SUCCESS in 1h 47m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a41691faa8d046879f97c2757256a1cd : SUCCESS in 2h 21m 14s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/430fde25c02c48caacca31d98bbe8cad : SUCCESS in 1h 12m 01s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4ff42854f54a4534897303db07efad26 : SUCCESS in 59m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d1c33b60af2841bfbfba46f0443d33f6 : SUCCESS in 50m 53s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0ff84475cbeb44f1a389570288199d67 : SUCCESS in 1h 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21e6e934133d097f2dbf55e70c2eafeb4ca517ad","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 14:07:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"955e0a459696b858d0d06e0a3f5ef05301cd7832","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 14:07:10.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"81ea0c8df088f08a3c12404f2694447851e130c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 14:15:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abb3728de9784b7897269bc6e906bed1\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/632d06445b654c88914ffc8d836f4000 : SUCCESS in 37m 13s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/1ad82b32c97c419ba248fae65f3d45a5 : SUCCESS in 2h 24m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2dafce959caf4c4d8398aaeff310dff5 : SUCCESS in 38m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/39b83234cf9e47a3a98796afa44cca0a : SUCCESS in 1h 58m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c8cb33bd826948b28ff6f741d7a886b9 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a45d99d8834b4f09b5f831e0ac7d11a5 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10790bc39f0742979e7b4b55740a0bd4 : SUCCESS in 10m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/17d8aa695b8040398032abc04a71ba55 : SUCCESS in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44cbbda01c792cbdd0978a93defd81e09d7a190e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 14:16:42.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"b1c795aa98ea979ab10b81a6ca244e9783f5acb3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-22 14:45:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"57681700683ccea76a96c5bba62ebab70eb0a355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 16:21:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d6965915e73421a97da6d90a269499d\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/1f4e287c2fd44fcaaca6a5486a3beb99 : SUCCESS in 21m 51s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/6601cb05e269423282b09359ed1dbc7a : SUCCESS in 1h 07m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0a8884235cbd42a89a15b70419723a6b : SUCCESS in 14m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/073f857023b54fdd95e2158f476c284f : SUCCESS in 32m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b4116f12a73a4c2ba07958f3e9f7326d : SUCCESS in 48m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/285c27a9bfc44ca992a88f65969b854b : FAILURE in 1h 02m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad1ae72e27144ed9d2c126a78734135 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5640f4a4343240f4b3bc7dd5308c3873 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a587a9a77bb40bb97c2efe85cdb2dfd : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb56bc0a41cf45b887cf1840985f4c78 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5a43f01ee234af2bfcdfecc5691d7e6 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab8192ad04b4e7fae5f77237e942960 : SUCCESS in 4m 19s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5b80dc1844d7442f905fafab2c78d65a : SUCCESS in 2m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5706ba271e4640399adde3826af41d1c : SUCCESS in 22m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/084218f2b5574333b6b9ce236eed39d5 : SUCCESS in 28m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9a3d2f605e6d4ca2a29df994f97e27bf : SUCCESS in 53m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d70638a801184085acd255ec192ae194 : SUCCESS in 1h 13m 29s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b748316168fb4de5a8891cf3e744953f : SUCCESS in 1h 34m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a211cd55a6704c92b14ae82efb587203 : SUCCESS in 2h 13m 11s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/b041bdb84d23461a8750c1be2a4a5d97 : SUCCESS in 1h 19m 41s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/70e3edba37894c039b0db784f13fe79f : SUCCESS in 1h 10m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3d311e7f1c9742c0998ffbb8ebec49ac : SUCCESS in 59m 39s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8460eacb827c4641a03ec3503468b716 : SUCCESS in 1h 23m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eccbe166d3e73a98f405a70140a835b22c0da5b9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 18:31:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aea1998028cdaa1d757c30c87c860ad109546627","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 20:27:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ae39d7f5aa469e878130e2c748b87e\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a47c348a7eca43fca4545baf0afff279 : SUCCESS in 41m 41s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5600386bdc534a369dce33cc977b07a9 : SUCCESS in 1h 53m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/929856f4caa645b681d7bbe934989021 : SUCCESS in 30m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b618d9a476ae4882b48e71362d88d0e1 : SUCCESS in 1h 45m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36b6432552e840058118e76ac383a7ce : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9026ef7f8794190b4a04d0c43308eae : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1e8a00f8c7e149eba3f046b72abaa8a3 : SUCCESS in 9m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/01ccb576206247e8b236b25e814fb09d : SUCCESS in 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0623707d0498c1036d9fa8b2913b0e6ebf4500aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 21:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6717346e97e644868f087808443df818\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/c2c7582b74ad4135b1ae11fef1efaba1 : SUCCESS in 18m 05s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/a01d2891239c4a8785743296a6cc1d8b : SUCCESS in 50m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/83039c03ba8b4056a8b9d2cbf2ae85ee : SUCCESS in 20m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/58b3a5882a3442e99c68b0d632f89957 : SUCCESS in 16m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4d17de2546e943918c19f8e3577148f1 : SUCCESS in 44m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a5eb0522466b4bac8e6598112aa6142f : SUCCESS in 1h 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c7c5e908236411da675fa27f737f614 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/573d6d0f756644f89c28258875397771 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de5ab20c4fdc4fb98ace0da80142556d : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7be7cf40de0468b9b13d18011eef386 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d036b55400b2479794b2e9bb969e1f9b : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2e374a78a243f2a545c5511e22904c : SUCCESS in 7m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6def80fd2fc0433aaa16e09010949c89 : SUCCESS in 2m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f25ab47753f445b2a4cb826aa5731f6b : SUCCESS in 16m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/299f9afc1b3646d2907aa948c4351d33 : SUCCESS in 21m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b159db37abcd48eea6f51dbaeccab868 : SUCCESS in 1h 06m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3e880019320a41ac9ee505c97c22415a : SUCCESS in 1h 31m 53s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/3d73a7d8657042d0bae2c8feeef0df80 : SUCCESS in 1h 54m 58s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d47e6d344d264a34b2869555225787f2 : SUCCESS in 2h 37m 45s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fc13ca5b80c24fe696cd7e8da4a06c6c : SUCCESS in 1h 21m 52s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/15f86997c14c446e9fc29b99c359a7aa : SUCCESS in 1h 01m 15s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c2652640b7af4aa588ad3168b534f9dd : SUCCESS in 56m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4d1d3566ece74bcd967237f9cb208e3a : SUCCESS in 1h 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a67f39a55d1c1a7d4bc0f3c143e6baeaf639f77c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-23 09:52:38.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"247ecd7ccfab12df5f2a7272896007932fe614d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 09:53:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"944cac6feff6ac9aa117a2e1db5cb7075c17fcb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 10:24:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e014e9b2949f49f68167b6107ad2f5bd\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cfa55279310c4f22a9f132ab429a0f31 : SUCCESS in 19m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/86438adfa42541fc95ad423ce9939625 : SUCCESS in 26m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b4f0eb659940b48333d43b59c5caa5 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23f679eb06084e70a60fcf384f9155c8 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8a4ab10741548c79c19319c9a86b1d3 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc5a011641d4ac18263af099dec6689 : SUCCESS in 5m 49s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/86ba2d6716c24df587947df01e279c7e : SUCCESS in 1m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d9617166240c421db686d22465edee55 : SUCCESS in 21m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4d2677c19dc54041b02a248306ed0a70 : SUCCESS in 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e70a047067be926ba360de29fd46f55523e0f6de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 10:24:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e8ea8e7268af2e5d98e8e269d750dc8efa7d983","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 10:25:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b5f45e95f2d492995dde4e0eb3e47e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3085413c43db493e8f9df126f0955ffd : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c35fb9ce5ca12890e3c6fc4b501ffb2510789211","revisions":{"c35fb9ce5ca12890e3c6fc4b501ffb2510789211":{"kind":"REWORK","_number":1,"created":"2025-09-22 11:39:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/30/961930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/961930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/961930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/961930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/961930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/961930/1"}}},"commit":{"parents":[{"commit":"8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e","subject":"Merge \"Drop ironic-inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-22 11:38:29.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-22 11:39:37.000000000","tz":120},"subject":"openstack-base: Fix psycopg2 install on EL","message":"openstack-base: Fix psycopg2 install on EL\n\npg_config has been moved to postgresql-server, postgresql-devel\nonly installs a link to pg_config which is unresolvable without\n-server package.\n\nInstalling postgresql-server-devel to include headers which are\nalso required for building psycopg2.\n\nChange-Id: I35548a02f652c035c17aabe149752322ffc869c4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c35fb9ce5ca12890e3c6fc4b501ffb2510789211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c35fb9ce5ca12890e3c6fc4b501ffb2510789211"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
