)]}'
{"id":"openstack%2Fkolla~962657","triplet_id":"openstack%2Fkolla~master~I7816edb6ebd8c95412a6833ea20b32193c2d1415","project":"openstack/kolla","branch":"master","topic":"drop-mariadb-clustercheck-support","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-06 08:55:26.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-10-03 13:48:50.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-10-06 09:46:45.000000000","reason":"Change was submitted"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-10-06 09:46:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7816edb6ebd8c95412a6833ea20b32193c2d1415","subject":"Add MariaDB healthcheck.sh","status":"MERGED","created":"2025-10-01 10:46:38.000000000","updated":"2025-10-06 09:47:30.000000000","submitted":"2025-10-06 09:46:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962657-drop-mariadb-clustercheck-support","meta_rev_id":"c9b9c02f5c42d1c0ac3c010a95644c9752a21e11","_number":962657,"virtual_id_number":962657,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-06 09:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-06 08:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-10-03 13:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-10-04 06:02:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-10-06 08:55:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-10-06 08:55:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-01 13:51:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 11:10:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-10-02 11:56:11.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-02 14:46:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-10-02 15:09:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-10-02 15:14:28.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-10-04 06:02:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"ab97631af033d7db9816fc0c7e4a08548be3ddc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 10:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4af224c06c53cb92531a5a2fea3e004b9978e872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 10:55:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"36b7e948b8207b8a7d7851c59a15d75dbf67941b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 11:11:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f87e2b5fdec10c17c1815170a88bac351562a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 13:51:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cb3f917c5b544bc86d5b015daea15cb\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b21959ba8a1f4af8a1a3a8558960d6ab : SUCCESS in 20m 41s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/bae83b6798b746e5adf9b326d662b27c : SUCCESS in 20m 27s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/8928353f143c402da327c6ba38b29597 : SUCCESS in 30m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a52968e59f14d418594279a85c31e39 : SUCCESS in 18m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2567c55224a542d2b9f00487bff60967 : SUCCESS in 21m 45s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/8688665daf8541c8b9725dd86219c9f2 : SUCCESS in 1h 03m 36s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/6736651b58544d4bbcb26cfd8979880f : SUCCESS in 47m 23s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/4e8f29c9b2614e7989543dd7ded0049a : SUCCESS in 1h 05m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0d357f8811c4a58a2b2ad292af105ec : SUCCESS in 50m 39s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ba8f0928ace745699beaf02a0a75acd0 : SUCCESS in 1h 12m 30s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/d0a9d2fbd2ef428c91581c537fa60540 : SUCCESS in 1h 38m 17s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/385f4e77a8d7424aba1f4d9ca684004b : SUCCESS in 2h 38m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f67217db4b6c4830a35c80cda2008a06 : SUCCESS in 1h 44m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5698d7c3ca0c4807a64ff825ed12a8aa : FAILURE in 2h 04m 39s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/f017d425dfb54348befba9b962d63b3f : SUCCESS in 23m 21s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/0199954337fa40368962419e9efe4e14 : FAILURE in 28m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/05fafd0455d8461f9c7fc9f46b0ff90f : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/724e96a933f14866b2e9e4efbefea6e5 : FAILURE in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/732d7a82db6a405abcc253b27854bd0b : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8feb9dacfd864456a7275139285bd30c : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ec455fb9b764c888152123676e8a638 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9adfb01a1e6248238bd35c83c87ca717 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61bb7311c2024667a2e63a94ac1f6760 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6933a359a922438b8600cc40370d6799 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":3},{"id":"3cd44df63a9493b6667eb12a7f79dd1df4dd1f7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 14:21:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8424798e54549d7aa070456c7d80c73\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/2ac601d918a44993bec5bde00c81f4e8 : SUCCESS in 45m 14s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/6e2d2d9661644592a26c9648bf964816 : SUCCESS in 23m 32s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/155781d200554f38a8e36afab24db6c4 : SUCCESS in 36m 02s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7ffa5aa2b85744d5afdaab5b1607a645 : SUCCESS in 1h 04m 59s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/ff52ee91298f432eb8b4dc2f07ce87d7 : FAILURE in 17m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8141e1fd36b4281b9b048d2852d338e : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/464c1304025f4f75a421021a0e70e27e : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9df235c7151c4e9c8db09b7548a1a1f3 : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b3c339e92e060d455ee833a2bb4e9cb44aad5b09","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 14:40:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"717ddb82e0343ccdce24ba949384671d49f9e375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 14:40:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fd358a5828e63e847784124297fa359e7c6e9d15","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 16:38:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fabd5cca8bda3d77ad0a0828cd543c18cde5a8e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 17:21:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0d45b04d1f14f0bae82342a2d3dda5e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f53239e5bb7b46a6bf5210653bbad81b : SUCCESS in 17m 45s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/14dc204dffaa40a6ba0f5a3798d95b0a : SUCCESS in 21m 43s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/85d5b97d3eaf490496187312439563ab : SUCCESS in 23m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcc87a8bcfb8468391cd00af7e433a2f : SUCCESS in 18m 12s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e3b3c15089284f25852413801e6f9321 : SUCCESS in 21m 08s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/20c343e0a9074a6db718a0b0e92d3e37 : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/f86e13f2c44a46578e241d04e8f7ab24 : SUCCESS in 44m 34s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/c934070a2310403aad41ceb70be94861 : SUCCESS in 1h 09m 28s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cda4466b9ac14ea096eb0c182839611e : SUCCESS in 49m 26s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/aca17808b45f449fb804635f4129ec1f : SUCCESS in 1h 10m 15s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/2deec8b598cf4297b206d818e1dec220 : SUCCESS in 1h 35m 28s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1a26c7c27be545b688d6608069238cb3 : FAILURE in 58m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/c262d48bc765446e9352da63fbedb347 : SUCCESS in 1h 54m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/01a6ae48245146758b4390bf274ac8db : FAILURE in 2h 39m 13s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/8e18094ed5294b4299af38c32cbff1b1 : FAILURE in 21m 30s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/aff49fad8f164afb86b3db24e7320b82 : SUCCESS in 29m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe4779079fc84bfaa73a393935805a95 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec735faab768426cb181dd5f5f4f1d82 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/283d3ec0891d4de3bdc564b1f499b0e4 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f29370089a64d3db8553dac5246108a : SUCCESS in 1m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90994d6df8ca47b6bb27c24eab6905e2 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d06a42324084134928d3761e979d785 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f90c3d53e4c4f2d93e5371aaa471ba3 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bfa4f19a1d542eeaf742897e9c10973 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":4},{"id":"e95330f1abd37f76fca46d34f6251caa8ea8d3b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 23:37:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3c3add544b24c7ba4c23f0dbbed6552\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/44b61478ed33439aa76a320a54a0ce38 : SUCCESS in 30m 56s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/2c710ea5e9f54082b321dffee9a470d1 : SUCCESS in 42m 02s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/a6885c5f57784344a89fc5c6ce95851c : SUCCESS in 47m 24s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/500649dc99cc4b78b4588bcc810a251f : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/283c7ffdc7144eba9dccaa216fcbb667 : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99733d98637b4fc497c9a42377fd44ae : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/010554eb53e5407588236f7a68d458a2 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/434baab00b28476a8cfc87e6ceb2813a : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2a06a65e44e49e9dd4b39a1e692dde6f3c4470dd","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 08:09:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cfa824549d0f9cb0460ce01f76a594be6cee5d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 10:58:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501cab9e9831442a8c025cd4803b1c16\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/aca02a4b1f8f473a912fee896ac926b4 : SUCCESS in 9m 54s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/6715db240c9f4c7691c0e303a436a735 : SUCCESS in 17m 48s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/61e87e9d6ffd480db8cc45dc5b122ae1 : SUCCESS in 26m 22s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f924ac72a94c48e1897e7be19505dbfb : SUCCESS in 26m 08s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a39a1e6e5dac4cf8b159525b9f73f039 : SUCCESS in 20m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a11df26424544bcc80e911c79932406f : SUCCESS in 25m 40s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/07f6622e0c064a59b9399cd2459d2c2e : SUCCESS in 1h 48m 39s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/66489c7f413b4c7fafe2d6b7b4b45c33 : SUCCESS in 2h 41m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/3e1a5948833e45afb31c91a3c1071d19 : SUCCESS in 2h 01m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/42bf52d2d30e4ba9914ee18e12478541 : SUCCESS in 2h 47m 27s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/1e5217c123264ff59d6ddb73d76532f2 : SUCCESS in 27m 17s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/02fabb9430be4134889915abdba1aefb : SUCCESS in 22m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7879b46eeff141458e8018931c5f9821 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eec7f8ea2894c5f93065dddd85a8525 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/571a9b40a01f47c4929b2281380bd4a4 : SUCCESS in 1m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52d02b0e1a5f487f82e031d76415eaa5 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee72d9b33fad4de6b582d7b0802c45b1 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0be06f7a20ed46a3b84c4caa23317190 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc86c330763452497f433c94391678d : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9d30b7f044c413da4ea781a984bd2be : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":4},{"id":"709a649b3f04e4388a418e66995c2169402dafdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 11:03:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3384921e8d9a44c89929e3361d47bf59\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/4ce215b83a52439ab48eea59f6a038c8 : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/03760e823a4a4fb18816caba536a1582 : SUCCESS in 31m 07s (non-voting)\n- kolla-build-rocky-10-aarch64 https://zuul.opendev.org/t/openstack/build/09e84a27b48645c29f044e9c100d90aa : SUCCESS in 1h 22m 40s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/b36a3ddbe0f54dcaa789c42e2be5e466 : SUCCESS in 46m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0038f4ba42394e90810c6d0a2b82b9fb : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbf8086ff8ba4ed5b3e46c632b53cb09 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af07249f68d84e18b3f59bcd834abbfc : SUCCESS in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fdb897836f61161094e714d04b2475d3606f0f03","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 11:10:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c785623c3dc85b57adfe676a12db4dfa07888b1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 14:44:55.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"45198aa87c63633be3d349689b69f7d15404be54","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 14:45:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9a46c9c3130dd22a9d19e4dc6da3865d50269995","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 14:46:55.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e26ad6249ca47acb10de5a29d80958820ce49064","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-02 15:09:04.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bacbaba8f503b2a490263e936c55239436f605be","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-02 15:14:28.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3eee81f94a16e4bd07fbf3223ee78973be4e80fa","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-02 15:14:59.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"880de87ac04c76709e67e1ac9de844588172c081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 15:17:43.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"4880b5c7490dfe09c337924101c84e39e2bd3334","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 15:19:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c7b4fe5761448b219885fe94e0d570c09c6f4b5d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-02 15:24:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ca4cb853bf1f0d8cdb486b5b535c0d461b4b4f70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 15:58:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"def8d543f310ea8ff02a403e3ddeebf81d3a6e97","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 15:58:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"44eeb3061f4ea1e61286ce8c3abf404dca9d94c8","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 15:59:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7aab0b4f92c8f52331b98f19122234c1e50e3b3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 16:17:36.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7ae2688fea4b6e4ce502764ed433eb0ad8100832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-02 16:39:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"22dd4a4692b726f2f50699fd5f4694127f1d037d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 17:28:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33896f3dee024580b0913966248fe56c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4fef74aee26b438299f736e7a43af6a2 : FAILURE in 20m 21s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/e8db0f6b4cb9455eb49a2001a3468a43 : FAILURE in 18m 50s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c38bd8b591744de095ffcb394d3d4725 : FAILURE in 27m 37s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/14b5e9f2c08f473cba25b79e4c769883 : FAILURE in 13m 51s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5bf93f66a05c4e0192ad88e89baf739f : FAILURE in 48m 14s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/484458b17e014d509e54b78def55edcf : FAILURE in 23m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94a533daf3af44e3b8d10e3f2ed572f3 : FAILURE in 28m 05s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/8c9162f04eed43308a771b02d5037284 : FAILURE in 20m 04s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/104ac722ccc943d783b9280fdf70406e : FAILURE in 19m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/683d407551c74ce293472823917fe2e1 : FAILURE in 20m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/70fabee51dae4d848dd36437ca725dc1 : FAILURE in 20m 07s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/02986398f6b544baaeddbd0f16bb0294 : FAILURE in 11m 02s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/977ee5cf8de64cc4bb79f8542c88f7b9 : FAILURE in 10m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/135c500d196449de96ca77139a1fe049 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0b84ffcc2ca488c8621725a72766e70 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4defbb9580a4263a4573db35cbd785f : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed93432f33c34de38ee8e9e1634da401 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00cca808bc7a48e1aa7f93f9effba843 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3849eb2c2bb44714b3ceddfca241ad03 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea70180d5f5f401ea208972cdbd265f3 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7bf47fcf0654d9fbc235622c0b4d379 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":9},{"id":"e7786aadcd098351075a78a3a5f4434356b78654","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 22:51:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53bc94de8994179bc451810f8ef07d6\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2f8dcecfd42548a58c3a5eff83114f6e : FAILURE in 57m 52s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/b8fc9c66fb9d4daf89129713a73a80af : FAILURE in 42m 58s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/41e6bbb6f1f64a38bc0f52174a3fa9cc : FAILURE in 29m 07s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f2d4c5b936fc422f8c6e1d25698ee9dc : FAILURE in 49m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/88394e12c50341599b456ab8a37fd046 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65c6a2fcd2c6426b8277a18152cddc7d : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/46155496e46f47d7b3ba65ab54e0d6c7 : SUCCESS in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"66b8e26f3462f8e539c9f4c043f309ce0374ca68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-03 10:08:46.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"ded3b758bb3c01cff4659cff9921666cdf8cd745","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 11:08:13.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dea38f90a844bb9822925d9366021a1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a1f1ee7761794784b683cbb589d15f6f : SUCCESS in 55m 29s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/0292fa7ca4084b21acdc78d99ee0ff1e : SUCCESS in 55m 41s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/95cc17efb17e4f67b9930683cf2d515d : SUCCESS in 56m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e031fcb204564c5aa41e2f38432f7ea9 : FAILURE in 26m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e0a41a7cf15440cc89a1943b1b6e70a1 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1d7f6d5dd6d8474388ba8b0d6a4d3d17 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01f4affc87be474399f5fc5a9f37fd21 : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c6013b599aafd02823f580a9fedb19d4b430dbf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 12:35:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d510e9445804959b67376b7649afb0b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b8378d09de2d4263bb519adb8314bc2d : SUCCESS in 11m 12s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c186615008474fddb4fbe8197b28b75b : SUCCESS in 12m 32s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/bc4ba31c57c048d18f37e66a95c6c6a7 : SUCCESS in 33m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/87720b1a980f443891ed81e4ec1de019 : SUCCESS in 13m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/105931b0db1f42e3a22a3403281ada1c : SUCCESS in 35m 27s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f3d2868c3f3463781735402a0f732ed : SUCCESS in 19m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d7225391a33144de938a439d06082803 : SUCCESS in 25m 46s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/87d63e4765d64ceb9ec94ea368839aa4 : SUCCESS in 1h 45m 20s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d470d8ce86dc4d84a85122e1dc42a34a : SUCCESS in 2h 20m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/bbaa6c5290ec4cca9fc4c012233070b3 : SUCCESS in 1h 44m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/537bfb0ef8a4404fb2347ade2c82cd2c : SUCCESS in 2h 25m 35s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/8a82b52b155a408e81db5de52b364220 : SUCCESS in 21m 40s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5a51443132d6474a871c3fb1b341b938 : SUCCESS in 25m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7c22b7394bb64202bc2f7453a9f44b0e : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53d892ad9a8447c9878e38020303b27 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17d9d5198c2f40d9906b0b18d65b2c61 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2399ad2574f14188b54bb9edb0fccfa0 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06b98a6f99e74840b017433f7c2081c9 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99d26361b9de4845808e48046b8b7a4e : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f087d287089541baa60200e33ce0388f : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9091a1027624e9ca14f231cdced6586 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":10},{"id":"837bab0f9e5a89d4735f01c1f17a861d21e80286","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-03 13:40:02.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"01731435085f53c3fcb1741e5d95e85bcbdd4fcf","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-03 13:44:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9c902787a208ea0fe89f01c71450437dcd8d77cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-03 13:47:26.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"149703b9563ce212b49d4d6c1a35305c24db1609","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-10-03 13:48:50.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e6155f56e55cac6688736485d8555ea2e19e7974","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 15:02:54.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e433764e9a7f404f8882b23bb2175b2a\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/666b2b91197b4d8bb49e8ef22c6fe3b1 : SUCCESS in 46m 17s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/12c29ad4ce904aee8cafbf0d5f88ab7a : SUCCESS in 42m 39s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/3e16c7315acb41a49b12347783d6ade9 : SUCCESS in 27m 28s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/72daf508535d458e96d0400e80f5f230 : FAILURE in 46m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9c7d02d3e0c41afb1bf775a48d12d62 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9ae1f134d59448f8e7e5a6f25f712f2 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0237f58aa6a4f52a6cda06d30415db5 : SUCCESS in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a3b1b4b54d8d5c6ba3cf894929b632a2e8df7ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 16:20:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4acc4c0790d411eab0b6afd4a064468\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/298b8fec173d497e8b3cc2eee19b4b6d : SUCCESS in 14m 01s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/463a57cad27e459fb93fd0d2423a6eb0 : SUCCESS in 20m 46s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/a9024b9882554dc993c7d347ae8c61ff : SUCCESS in 37m 40s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8963d6051801452082ccc3c68737fa5c : SUCCESS in 16m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ec2bce6f0ebb4d1f8fb6a225f8da9a91 : SUCCESS in 35m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b37a69ee7027432198bd8c3397361343 : SUCCESS in 24m 03s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a3b002e56faf467f99c62c3800a6b2db : SUCCESS in 27m 40s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/10a75ab09b0b401799c5d1fea7212068 : SUCCESS in 1h 51m 11s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e3e2de79585541cf9cb8dfef0bf5fbef : FAILURE in 29m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/c9329fce507a49209f44814a9c6c78f8 : SUCCESS in 1h 51m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0acdb223a9f5483e8ee51408de4ebef8 : SUCCESS in 2h 29m 56s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/4627e3dcc2234999bf628c3565e70a98 : SUCCESS in 21m 48s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/73d40c298b114ccea0be3b1d48101586 : SUCCESS in 25m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8d5c7a2be72a46ecb6404adf6253b78c : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d54bbbd5ef24959b44a8f7665dffc9d : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62acd8217fbf4de7b758881976a4ce1c : SUCCESS in 6m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79814357f87c4dbe934372525d21b39c : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5586d45a05e242299873e6c2e61c0485 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85115e1511d643c487eb3b92a4aef242 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f58e66689814eafb76d1e9cb1781d45 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/978ed00fa0154e97854f6b5dcaafd101 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":11},{"id":"470c98be8efa7250a412e3226c69c002055a2225","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-04 06:02:23.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"abaf2437d0abf04035edaddce9e12d71dc225be9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-06 08:55:26.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7f0f9bd1b99ef2b74fd9f85b92dc6c78dc57bcef","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-06 08:55:55.000000000","message":"Patch Set 11: -Workflow Backport-Candidate-1","accounts_in_message":[],"_revision_number":11},{"id":"12bd8ef98eb7096dc7591b66a29f33f383dd21d8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-06 08:55:59.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"560e2557ce9211a7e78ac9bd1fe04cbddb2fc22b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 08:56:09.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"be95a5b3788a6c619179de05f2008efe06185272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 09:46:45.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f445316910f54b168d169eb835770bcd\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/2aecd33b930743d69a7e2d3085f5e1cd : SUCCESS in 24m 42s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/4a8a8b059fe0460dabc0c26fbce0a26e : SUCCESS in 17m 28s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d50e5e690f114d9788db6a714793a6cc : SUCCESS in 15m 21s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/08229dd9e5654456a8a5150612078109 : SUCCESS in 48m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a6bf1fe3460492d8f26c1c6dbeaf368 : SUCCESS in 20m 20s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/568a491bea594b54af629c96b95e0f21 : SUCCESS in 27m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cd27870a1f524ca482fe23aca5397ce4 : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/764dbd54a9ea461eb159a79c5cabb5e6 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba10b7cb9770446aa0c67ca637abdab7 : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b76806d4356546d08c1435e188b308f4 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b65fc52bb04f4cac95bdaa2e526dd4 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f38dba754d4145e9afddb11e74b6bf23 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":11},{"id":"2a0e4a62eaf0d5106c3ae0621acf141789fc16fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 09:46:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"c9b9c02f5c42d1c0ac3c010a95644c9752a21e11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 09:47:30.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581bcadf45c843e892756d915a1d1f38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a512222500ff474b93b8908e060aa640 : SUCCESS in 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1a47455363444f2bf677076649c6e70 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"683ae79657532243a31ead1b35e82877f1778991","revisions":{"dfb0da0cb93b6e265f7af7bf61ab2eb3146ae45b":{"kind":"REWORK","_number":1,"created":"2025-10-01 10:46:38.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/1"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images.\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfb0da0cb93b6e265f7af7bf61ab2eb3146ae45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfb0da0cb93b6e265f7af7bf61ab2eb3146ae45b"}]},"branch":"refs/heads/master"},"6710732473e20dda7db454fc2ee6fe9fe533d335":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-10-01 10:55:12.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/2"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:54:31.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6710732473e20dda7db454fc2ee6fe9fe533d335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6710732473e20dda7db454fc2ee6fe9fe533d335"}]},"branch":"refs/heads/master"},"fc7867d5eb7c334c8140d46d51586bbb4da6720a":{"kind":"REWORK","_number":3,"created":"2025-10-01 11:11:29.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/3"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 11:11:17.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc7867d5eb7c334c8140d46d51586bbb4da6720a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc7867d5eb7c334c8140d46d51586bbb4da6720a"}]},"branch":"refs/heads/master"},"1f2e4e2173f8f2ae7633ac4ce580bf2bbdb6df5f":{"kind":"REWORK","_number":4,"created":"2025-10-01 14:40:28.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/4"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 14:40:22.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f2e4e2173f8f2ae7633ac4ce580bf2bbdb6df5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f2e4e2173f8f2ae7633ac4ce580bf2bbdb6df5f"}]},"branch":"refs/heads/master"},"af5a7216e1d348b4d7fcec792c2142e50af3effd":{"kind":"REWORK","_number":5,"created":"2025-10-02 14:44:55.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/5"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-02 14:44:38.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af5a7216e1d348b4d7fcec792c2142e50af3effd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af5a7216e1d348b4d7fcec792c2142e50af3effd"}]},"branch":"refs/heads/master"},"d4f366759764f31d7090a8b7d676f1c1bfa7e2c6":{"kind":"REWORK","_number":6,"created":"2025-10-02 15:17:43.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/6"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-02 15:17:36.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4f366759764f31d7090a8b7d676f1c1bfa7e2c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4f366759764f31d7090a8b7d676f1c1bfa7e2c6"}]},"branch":"refs/heads/master"},"5f0bd3aab816725adea289b44dc0028ab5f5173f":{"kind":"REWORK","_number":7,"created":"2025-10-02 15:58:26.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/7"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-02 15:58:19.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f0bd3aab816725adea289b44dc0028ab5f5173f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f0bd3aab816725adea289b44dc0028ab5f5173f"}]},"branch":"refs/heads/master"},"a4fa8b8eca530fdc15bed7db6e9dcd15cdd76de3":{"kind":"REWORK","_number":8,"created":"2025-10-02 16:17:36.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/8"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-02 16:17:30.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4fa8b8eca530fdc15bed7db6e9dcd15cdd76de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4fa8b8eca530fdc15bed7db6e9dcd15cdd76de3"}]},"branch":"refs/heads/master"},"fca2f819b8422dda3976c071262e485c429ca56f":{"kind":"REWORK","_number":9,"created":"2025-10-02 16:39:31.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/9"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-02 16:39:23.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fca2f819b8422dda3976c071262e485c429ca56f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fca2f819b8422dda3976c071262e485c429ca56f"}]},"branch":"refs/heads/master"},"dbad9866c1fece11cb232f5a9db354a6b79dd8de":{"kind":"REWORK","_number":10,"created":"2025-10-03 10:08:46.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/10"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-03 10:08:25.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbad9866c1fece11cb232f5a9db354a6b79dd8de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbad9866c1fece11cb232f5a9db354a6b79dd8de"}]},"branch":"refs/heads/master"},"683ae79657532243a31ead1b35e82877f1778991":{"kind":"REWORK","_number":11,"created":"2025-10-03 13:47:26.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/57/962657/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/962657/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/962657/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/962657/11"}}},"commit":{"parents":[{"commit":"8b121529cbcfeb20f75114873b0d2b46ffde1e68","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b121529cbcfeb20f75114873b0d2b46ffde1e68"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:38:42.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-03 13:47:05.000000000","tz":60},"subject":"Add MariaDB healthcheck.sh","message":"Add MariaDB healthcheck.sh\n\nMariaDB clustercheck has not been updated for a long time.\nSince the script uses ``mysql*`` commands but MariaDB 11.4 (next LTS version)\ndoes not support them, it will not work once MariaDB gets upgraded.\n\nAdded alternative MariaDB container health check script ``healthcheck.sh``\nfrom official MariaDB docker images. [1]\nThis script uses ``mariadb*`` commands, so it\u0027s prone to MariaDB\nupgrade.\n\n[1] https://mariadb.com/docs/server/server-management/install-and-upgrade-mariadb/installing-mariadb/binary-packages/automated-mariadb-deployment-and-administration/docker-and-mariadb/using-healthcheck-sh\n\nChange-Id: I7816edb6ebd8c95412a6833ea20b32193c2d1415\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/683ae79657532243a31ead1b35e82877f1778991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/683ae79657532243a31ead1b35e82877f1778991"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
