)]}'
{"id":"openstack%2Fkolla~962788","triplet_id":"openstack%2Fkolla~master~I261a2fb64b873063914542760fa33cad5c2dd26a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-03 14:54:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I261a2fb64b873063914542760fa33cad5c2dd26a","subject":"base: dnf.conf - add repo_gpgcheck\u003d1","status":"ABANDONED","created":"2025-10-02 12:40:36.000000000","updated":"2025-10-03 14:54:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d223904b21a1910675ce4cb35eae3d17a434871","_number":962788,"virtual_id_number":962788,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-02 15:17:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 15:17:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc917030766e57ee84a124592c9050c9e948634e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 12:40:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31472490b540c7ef7be4fbaffb76570a9f52db1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 12:40:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cbb30c8e141c8812777078c3e5325a4502178bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 15:17:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47880a70a37e4bcb92b9cda60f57ba62\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/aa2a6fdea86d4c149b9828336364e1ca : FAILURE in 5m 20s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/770720083cd64a8ebb48143b42dbbf9b : SUCCESS in 19m 28s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/549785d124234b5cbcd13fb936dc73ec : SUCCESS in 28m 01s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a8e8aa0644114181bd89248d2c598887 : FAILURE in 6m 04s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/30238d49d89b46eab0f40e5eef6f4c03 : FAILURE in 11m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aa8e8e2b70f4eaa82f75770d3b4a16d : SUCCESS in 16m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e8468b4b49c6437e9c8a9d0bec46b355 : SUCCESS in 25m 12s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/f826699fa1924d7697545f100b444423 : SUCCESS in 1h 37m 54s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/fa8d45c0c7ff4ca2a7959e2a22ec5a30 : SUCCESS in 2h 34m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d180a4e285ec48bc91f44bc64ac74c97 : FAILURE in 1h 11m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2d834a88e30c461a82903a791ca63338 : SUCCESS in 2h 32m 10s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/50e9773ab91f4017bd9b138ec0a20717 : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/a62e9c1af46048a18fa8bfe4e3dc431a : SUCCESS in 1h 25m 19s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/41c26ca3cbb74b75be9f09178d1c0d56 : SUCCESS in 24m 30s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/4f792fa095994778be6a8a55187be59e : SUCCESS in 25m 41s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/98456e8ace9740c28abb9f7438b0fe30 : SUCCESS in 25m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/75e8c044a102485c91aa541b409c1669 : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/59086a67111142738d8e8b546464454f : SUCCESS in 52m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e32ce97da2ab42bab230b8424e119a55 : SUCCESS in 55m 42s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/7cb3292e3dbc4aa0ba15a0d2891f8348 : FAILURE in 1h 22m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/9733a76bf01446e6979b9c2c5ef76ad1 : SUCCESS in 1h 22m 57s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ad5753d103ec4cac863bf242d62c85e4 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0207ade9f03545159c5a54e2e8580c1c : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b84b70409a54df888e5536d4a42dd9b : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25f2f938dd8c42549d78d9fc354a5bc5 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4a38147631c43a7886cbaa4e69452a4 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0acf1f6397414d519f5c0091c6cd832a : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b3dd80005a42889ec63246ec6c4b1d : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8e0a5f0213ffd702de94600fe275cc1cdbd45d9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:15:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec748cf7bc674a7baee13221b9a00754\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/706735346bfe416ca084dcffd0839403 : FAILURE in 9m 41s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/00726e66756b4950a4daadcbd1c070eb : SUCCESS in 56m 17s (non-voting)\n- kolla-build-rocky-10-aarch64 https://zuul.opendev.org/t/openstack/build/175c35fdf0a64cc4a8316b5ce93b0778 : FAILURE in 7m 06s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/e215d58d860e475ea2072911e1076697 : FAILURE in 41m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f75be220ea8646e6a450bfa2cfe8572f : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80e1eba7dd344e269ad2c7d4e4504dab : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0effddee6323483fbaf425623bb3a902 : SUCCESS in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d223904b21a1910675ce4cb35eae3d17a434871","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-03 14:54:17.000000000","message":"Abandoned\n\nepel doesn\u0027t support repo_gpgcheck, I don\u0027t want to wander deeper","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22d83e1e29f349f5599e3fb58315f92ef359797c","revisions":{"fc7e4ad17f7dae96bbcc121cf80e8cdc0e572f73":{"kind":"REWORK","_number":1,"created":"2025-10-02 12:40:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/962788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/962788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/962788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/962788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/962788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/962788/1"}}},"commit":{"parents":[{"commit":"f17a23747f31305eed1be84569aac907fae15c19","subject":"CI: Fix rocky publish jobs - version missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f17a23747f31305eed1be84569aac907fae15c19"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 12:40:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 12:40:33.000000000","tz":120},"subject":"base: dnf.conf - add repo_gpgcheck\u003d1","message":"base: dnf.conf - add repo_gpgcheck\u003d1\n\nSort alphabetically as well\n\nChange-Id: I261a2fb64b873063914542760fa33cad5c2dd26a\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc7e4ad17f7dae96bbcc121cf80e8cdc0e572f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc7e4ad17f7dae96bbcc121cf80e8cdc0e572f73"}]},"branch":"refs/heads/master"},"22d83e1e29f349f5599e3fb58315f92ef359797c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-02 12:40:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/962788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/962788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/962788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/962788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/962788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/962788/2"}}},"commit":{"parents":[{"commit":"5c6ffd745794e1c8e22f1d861fc0450ff2491cd4","subject":"Stop duplicating repos in repos.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c6ffd745794e1c8e22f1d861fc0450ff2491cd4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 12:40:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 12:40:47.000000000","tz":0},"subject":"base: dnf.conf - add repo_gpgcheck\u003d1","message":"base: dnf.conf - add repo_gpgcheck\u003d1\n\nSort alphabetically as well\n\nChange-Id: I261a2fb64b873063914542760fa33cad5c2dd26a\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22d83e1e29f349f5599e3fb58315f92ef359797c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22d83e1e29f349f5599e3fb58315f92ef359797c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
