)]}'
{"id":"openstack%2Fkolla~962789","triplet_id":"openstack%2Fkolla~master~Id1019fff2eca58f76f690d2c823a2e4fbc27e350","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"34034":{"account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"last_update":"2026-02-16 14:13:19.000000000","reason":"\u003cGERRIT_ACCOUNT_34034\u003e replied on the change","reason_account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"}},"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-06-07 19:53:17.000000000","reason":"Change was abandoned"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-07 19:53:17.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-02-16 12:16:39.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37820":{"account":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"last_update":"2026-06-07 19:53:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id1019fff2eca58f76f690d2c823a2e4fbc27e350","subject":"openvswitch: Fix log file permissions for fluentd access","status":"ABANDONED","created":"2025-10-02 12:45:22.000000000","updated":"2026-06-07 19:53:17.000000000","total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f7d9d310e7914548c36171f5e74de2be825a2cf6","_number":962789,"virtual_id_number":962789,"owner":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-16 13:26:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"all":[{"value":1,"date":"2026-02-16 14:13:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-02-16 12:16:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2026-02-19 10:26:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":-1,"date":"2026-02-16 15:58:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-16 11:54:13.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 13:04:20.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-02 14:15:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-04 05:55:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-09 08:08:21.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"state":"CC"},{"updated":"2026-01-09 12:51:00.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-02-16 10:37:23.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-02-16 10:41:29.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2026-02-16 14:13:19.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"state":"REVIEWER"},{"updated":"2026-02-16 15:58:06.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-02-19 10:26:01.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"}],"messages":[{"id":"89d8230f82ffeb90305e4490452c28770bacdc78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"date":"2025-10-02 12:45:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad0103086f3a76af9e38906783400987268d9591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 14:15:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0dcaa2a1bb2438d881aeab5c4766ac6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4f41248b089b4109bc6e13615f8f1b6b : SUCCESS in 19m 36s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/5d7ad18bcb804901b7b0e8652a8164f3 : SUCCESS in 19m 47s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c4d9288b09974d8eae5a53f237d7deb4 : SUCCESS in 32m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/cfbc2ec9e6a544fabf74a9d9ee8af135 : SUCCESS in 14m 52s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/12c1b755c4ae443191ba17f00d54ab25 : SUCCESS in 53m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37b6092e7a094a8f8bd9ccd495c22762 : SUCCESS in 19m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3878e200e7646d5b864aab81af5bced : SUCCESS in 19m 09s\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/f7737bb4cf3d400ab532e771599d0d5b : SUCCESS in 48m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/ac687b700b0047eb830d1a12ab6e52fa : SUCCESS in 51m 55s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/73be1e67470946509c6e40d80fc941fe : SUCCESS in 1h 21m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/5d15a04c55a64952a5fa5be2cfd0642a : SUCCESS in 1h 26m 52s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cec71c56979f4eaf917e82a544fbf08a : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272afb039b10455bb20822a5852a213b : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fd95a94ced0432d854f2cf5fe960526 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8657b6862bd4219b849a167d6fa8377 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ba3550763734ce999cc20993cc9b03f : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/554b0b78c293445c8adfc33568ccf8cf : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936b0c618497479eaa1c562a50295deb : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"81d398ca0fb558eb89d29ac3c4ceafac6f1d20ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 19:19:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f64f4c506ecb462e8ca9b8138d7bd3d3\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/97e3e9d6cd3d470da80765fec731fbc5 : SUCCESS in 30m 08s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/134c98a888af477b9d3bd9ed8d2827d3 : SUCCESS in 54m 18s (non-voting)\n- kolla-build-rocky-10-aarch64 https://zuul.opendev.org/t/openstack/build/93f0e7629a9742238d444227cc5a9000 : SUCCESS in 59m 32s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/0a10ce2d15b446859de6ad4de4def65d : FAILURE in 1h 12m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9871874f020a4d46a8d40e73f5509d7c : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c5e85497418430ea95f03c62d52d689 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21984bb7b84444febddf96cf77c06d97 : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"007d1ed780917d0c8d87129ad3dbdd0543d28a80","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-04 05:55:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c006ec28a1c0643c873e718b7e7725f1b66f67ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"date":"2025-10-20 08:48:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e2ed2169167589c02345519b6107755862f9d3ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 09:30:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/079305d62cb742388fc243cb5962867d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/04a7b6c9d94e400e8682a7e1095b37fe : SUCCESS in 34m 27s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/61649e2c6d2942749d74bfdf8d85049c : SUCCESS in 24m 52s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/23378dfea9604761b102ad47fe5cd0e2 : SUCCESS in 34m 05s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/680bf832030d49f48fe7b1e561bc3469 : SUCCESS in 39m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/385b541ce206486996d691ab55877392 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9e366ddcddb4706a1f588f80b3af822 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c31bc54de7bd4ad2a28e1ee2a8992006 : SUCCESS in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d882a22bff7bf1368b7430ae5087ba4d11f1e2e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 10:10:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/071b417f088444a9a601af01e6f205c5\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4e5ad27372ec4330a31062bd1e8267c1 : FAILURE in 3m 45s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/d1df0d0e19b745eeaa8d68d2b69e768d : SUCCESS in 19m 22s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/69d549ed3bcd4e12bd11b8b5a9eff80a : SUCCESS in 28m 42s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c0c869d587764b8394480343cdabd036 : FAILURE in 2m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e200eb3a968b41cc938a2fe66898a058 : SUCCESS in 1h 04m 17s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bfd4223785249bdae581c1a3d29467b : SUCCESS in 20m 54s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e8223dd2df98407ab66eaf5ba5747a29 : FAILURE in 2m 11s\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/7b14866594e445978d20cbda9af6e0d8 : SUCCESS in 51m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/3e46258f025849769e713bf0bb7db5c2 : SUCCESS in 54m 43s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/4e85cf20a92342159c0f923c98af8018 : SUCCESS in 1h 16m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/309dee0df7fe40f5a61b44304611c769 : SUCCESS in 1h 20m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/084346d3e988423391644ad4427f4684 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcdcceb605ed4e6fb248daf7479aa728 : SUCCESS in 1m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8730d5460b849e594a2f10bd8b56a1b : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe9c08f3dfee41fea2ab0c55ebd29d28 : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85601a5a5a2f48af842b2fab5801e110 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a15f5401497d4f2b97b9c0f1202713e7 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a5610732e314ccba4aa5387f66e80adae8c04fde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"date":"2025-10-24 11:17:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d308d8fb4050c736ff8f32f5118905c563446c8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:00:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca240c5428474be9ad8650b118f6a994\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/415e0db66e464bd3847ec76e09bc193b : SUCCESS in 33m 30s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/570ccef22a264b8a95649432feffd7c0 : SUCCESS in 35m 32s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/de0a12c7586947d78a8c830b372960e1 : SUCCESS in 38m 03s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/a6994cbd335b4280bf2f927409b7d5df : SUCCESS in 27m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1348c670cf048fb830b45e393fd2a58 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4ce2b490350c45f59ae7facef5d2539c : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b3f933687a1d4a519ea96c42d9193af6 : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2ad64e9c88a54d13d85fd6b61efb09ec2a2ac15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:03:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89e5e4755ffd415387ebaf3f4d0ed3fb\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a3cca6a57680456293a467855abc5cff : SUCCESS in 11m 54s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/a0539ebef4c14453b0b0f4a619f1399f : SUCCESS in 10m 25s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/901df46f5c04402ba5448a289d5ac442 : SUCCESS in 15m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8eb5da73364f4565847c829e74c1b276 : SUCCESS in 12m 17s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/bc410629dafe4d60b1ac6a3562261c8d : SUCCESS in 51m 52s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a054d154f7124687aa053b18296dcb48 : SUCCESS in 19m 25s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d652691c59f742b3b678d226b93ca20e : SUCCESS in 27m 26s\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/00780129ccea46e58e1066f85980d7f1 : SUCCESS in 48m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/6bd640e7769349cd8e3d4f08d1da3aaf : SUCCESS in 51m 40s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/f815c801ab8547789c5492a220db9e78 : SUCCESS in 1h 44m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/11092ab9930a409e9c66d65a24e6c112 : FAILURE in 1h 30m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/605a6192660c4e909441524992ca71ac : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c985f054c6794a2e9be9251a39c71959 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2b292f09ac44a7f843095bf1ba251c8 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a789e01294604e049e76330f77283010 : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9abf3bdab04249f28d3db503c74490f8 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc52590c93f4fef8aff1a647f4795f0 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"5c995e7e08acbc588e05fa66bbf37370e7610e90","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-01-09 08:08:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65f4655db1544bbfb42a83f59c993286ff668b0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 09:36:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee42e3fe8f0b40be950afc8a9a34d801\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7b316e07126d4e4b911dad39f4343823 : SUCCESS in 41m 33s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8b8e5d971c334a6a9f28089ed40d5fbe : FAILURE in 38m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3a6186d8154243d28657a78461606981 : FAILURE in 12m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/90feb27f15ed4b72ace95876bc0d004c : SUCCESS in 52m 36s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6d6bd68938a34213a57a7be6376fe1c7 : SUCCESS in 50m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d24a337f3136401892dcef6a8203e2e7 : FAILURE in 51m 20s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33032b22a67b4120b6d568cc375c87c1 : FAILURE in 26m 15s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/030483494c3d45818756fa053cf84a98 : SUCCESS in 1h 10m 22s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/eccf24e99ab1403d8b0b104929a55d9e : SUCCESS in 51m 39s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/08107bb3a1a345009478b352e64e1150 : SUCCESS in 1h 07m 33s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e82d87bdcff740329ac6cbe2711bb67c : SUCCESS in 1h 01m 22s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/6ac0708bda5b469fb4e576259510fda9 : SUCCESS in 1h 19m 24s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36f18dca054541c1894024c0f6835d0d : SUCCESS in 54m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3441d343e4f246389ce4905ff77a04ee : SUCCESS in 1h 14m 33s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/de62a10bc7bd422799dfbf788aa046c0 : SUCCESS in 54m 07s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/289a060dbf654dd9a611f94b75dd9c7a : FAILURE in 1h 21m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/223648dc4c2e4f829e110d8891f52dee : FAILURE in 1h 20m 19s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/809f2be972904fffac7112088959bb46 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/823fe2d3b5434c92a68d2def4ddabb97 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7e3674f9bcf4a619323ddeba7670779 : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ac1cf327c6442ed82540abac333b152 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41310f95446f44e09ec096d273939891 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46da00641c8742ad8875a84b3bfe05b7 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":3},{"id":"e6e346a1a0c1c8f1ac6facb933de33fbf5186eeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 10:24:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ae4a7582da444eabe7229c19dace695\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/be6ccea7f06e4ac4b95ab1bc3326241d : SUCCESS in 1h 34m 09s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3b33c336eb6e480ea11acd91a9bbc09f : FAILURE in 1h 23m 28s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/dd8f66e5913947efb501d4ba3a128b51 : SUCCESS in 58m 14s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4871e8ef182f4d6488387cceaf11a9f7 : FAILURE in 1h 03m 29s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/884f1f3c31ff425484b852bb3066eaa4 : FAILURE in 2h 03m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5129e77d605e42039f0ed348ae267b58 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3a69c1d93024908839fd252db53cf23 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1cc1b7682bf648849e3440c4208d9b68 : SUCCESS in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"06035fd7a81a2720483b3ffaf3a76e8548bbda94","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-01-09 10:34:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e62358f2f81d4bc9d762b873777ffad0f5a781bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"date":"2026-01-09 11:03:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2ae84239b26d877759618c1402e54d9dd5ce2406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 12:47:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4deb1decaad7402a9f34d08361333055\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c4b2ec6d8f2a4f3ba5185a9aabe43293 : SUCCESS in 57m 51s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ecc87a40217b47f7ac46a66218052ea3 : FAILURE in 51m 32s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/91225e3935dd4ea4965fa4477944e134 : FAILURE in 22m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/648c99082feb4876b5f13162ee157117 : SUCCESS in 54m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/173a76e615674cb48e1633d7ea498cda : SUCCESS in 1h 06m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/599dfec0925646d1afa106e19929d8d6 : FAILURE in 46m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c7ae1dd5f5e54b868b3437e802f93597 : FAILURE in 30m 41s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e56f2197ce304effbc0caed720c28d89 : SUCCESS in 1h 27m 54s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/21974857e90d4e91a91d925c067d74f7 : SUCCESS in 55m 51s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/e9208734c0ec40418595c2e51370bbd1 : SUCCESS in 1h 29m 37s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/711ff4ded9d047a1a9fdc9473e1b73d1 : SUCCESS in 56m 53s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e909e4a67b2241ec855c2d29422400b9 : SUCCESS in 1h 37m 44s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c31eed1de4e4451197fb36aff42629eb : SUCCESS in 1h 15m 01s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b2faae3350e444f3a351854c72edcd2c : SUCCESS in 1h 16m 45s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e823d53a75eb4d7fae980adaf95bd66c : SUCCESS in 54m 49s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/15cd850b48b8444f84bf4a2e739d7c72 : FAILURE in 1h 38m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/1b61d17e5dcc4a48964a2982411e8169 : FAILURE in 1h 34m 17s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/461127f76f7a47ce91e99b5612dffeea : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3522f86869ec44eabb33b379944b159c : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8895c2cf3955443fbf8edbf89a6fdfd2 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34c6089515f74f0295dd3304c47054db : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b738635cad384d148d73ea0d46f386dd : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4435cb39594fc7826de763225bf899 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":4},{"id":"4bb0875d057326ee651b3f2bdc238cbb31687ae2","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-01-09 12:51:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a0638ca7c4c1a6b3308665a28147c54ac6cef4e","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-01-09 13:33:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d20122979edf0bdc06de314d67d64d0906cb043c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"date":"2026-01-09 15:31:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d07f381c07027b61b0ba07e825f7166465a4ed19","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"date":"2026-01-09 15:33:27.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1f03e92ad111aaf32756ffc2cb0147782bfe5dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 17:08:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69e34fb25f7549489bcf4bff61b7c5b8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0dd85d2bad374e20bb6a7fc6892f17ec : SUCCESS in 38m 54s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5306addf214b4fe395058a195a8d14ac : FAILURE in 48m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f4bcf9ca7cda44a5951dc9c4311ce9bd : FAILURE in 26m 30s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/afb1014c06914aa3b10f2ff9b7624154 : SUCCESS in 54m 21s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b1c10f5c35e84348a31f9c6f21fb1cec : SUCCESS in 37m 21s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/769ebda2476b4a5f8c8883ab76019e06 : FAILURE in 39m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6ed1eb16d7f48c9914f7fc57a4cdc09 : FAILURE in 27m 51s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/6b224eaeb228442e9e167e450644dc9a : SUCCESS in 40m 49s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/80dfae91877c43aa80ead2ec619a2ef1 : SUCCESS in 1h 00m 38s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/2461287a03c04ee18de72b41c633c6a4 : SUCCESS in 1h 19m 00s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/34d3f62e842e4a20a6186f9b31cbacb3 : SUCCESS in 58m 56s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9e1085bc20c245c3a95936d8d06b50e0 : SUCCESS in 1h 16m 39s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ca428f3236449a7b64c45653a72953b : SUCCESS in 1h 02m 37s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/36eaeeede0eb4b7c92d90abe150b6f84 : FAILURE in 1h 17m 32s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/68169e7c92b846b2abbdcb7a21ab1196 : SUCCESS in 54m 22s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/3fb83340ee2345b68194219261a3840a : FAILURE in 1h 27m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/8e299b919f11421796fb57b6744153c5 : FAILURE in 1h 27m 20s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3d28be717df140278fe512aedcb93bf9 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ede1bd6f29a49a99cb5e572db7ca13f : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f2e9403fd9943ca971f234005fd3c4a : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bfda5cb56eb47e696987cba28eb7a23 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f419b2399cd14b6f9d175c3fd2a2c793 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7353df2052cf48d09433e5ba02aeb1a0 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb61e5460be54b5ca956e2e50e9416a4 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":5},{"id":"3808cc5471dc22033e49f71d9973ac2046fabfd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 20:30:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92012c3028954b5bb04f3aa768527a41\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/6cfca107124748e496017e834b86088c : FAILURE in 1h 02m 25s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a0cba7e0d89d419181002ecb23020c8b : FAILURE in 57m 03s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0b55b11faffa491b9ef5a1723df8df0a : FAILURE in 54m 16s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/48a0bf1ed6204edca9bc1b43756b33e7 : FAILURE in 1h 06m 06s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/2981218594584549ad810fc8c5f6c182 : FAILURE in 1h 11m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b66330e44a549fa9aea52c5d2753e9a : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37f51e6d2dde4da8a0592c3eb4699a71 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/afb4d6b8a9d348b3b48e479730db24ab : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"da27ca915d1d75ae3f0e203688df3f9caa9f2a18","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-01-10 19:47:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"636b8036bb2219e18fb5f1c22c5ca75f0d23c884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 21:23:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23566d3c73c748a1956123544c534e49\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/34cfd5038bf14a628c36b7d1d81149a4 : SUCCESS in 40m 14s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/edbfbc3c18e34ab996e0ead331b073bb : FAILURE in 41m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/0d46b826d39246749326a7a50e3b4653 : FAILURE in 21m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c38a25c7a2714dd3b53ab9c306af8427 : SUCCESS in 40m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0600cc2e74734a4faf991fe77f9fc687 : SUCCESS in 1h 13m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/293a2b5d2ebf424e88d177bfb2cfaf97 : SUCCESS in 48m 08s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb22c866aaa548729a8bd7e4ba113a29 : SUCCESS in 29m 46s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/8a07264970bc485fa3ede148a8a8b5bd : SUCCESS in 1h 26m 38s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/d07b60e3b3f043069cfbaa6555238f54 : FAILURE in 16m 58s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d6a1960206814b05b8ab95c1d57febef : FAILURE in 16m 03s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6b4c033fc37a43bfb33e1890c3cf91dc : SUCCESS in 1h 00m 45s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2183851a331247959ddcb98c46b72a6a : SUCCESS in 1h 25m 54s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a871e9460d1a4128aca07ae55676e03e : SUCCESS in 1h 02m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/80790a4900b1493fa3f3c688158ee03d : SUCCESS in 1h 20m 11s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/7b54b44acc134830bc2657d1cf4dad85 : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/37d1c83bb608483d819b3190a8b189b4 : FAILURE in 1h 27m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/e2a6d5b3548c4893a7ff61be549ff7a1 : FAILURE in 1h 18m 23s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fdf749067ec44975bd7eeb4fc32e5101 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e888d47872c46abb51a84625fda751b : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a43aed6fe154fbf9de65ab303e12a0c : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f0e12d1b6e44de5aeaa826206487c61 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec352a63580647e5b290c589d590af55 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd700f177c74a72b25097459bb495a4 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ed2c3e1ef9b4de5b802039f37f0daa5 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":5},{"id":"be8ecda0fc6c358d5ee03406851f3e2f16b22440","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 21:28:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00b77f4382ae4cae8147d648c37c5a33\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a539cbe626c94470aa2bc3920b0a926d : SUCCESS in 1h 37m 05s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8ab2ffb757a94d8fb4e5f6b20ba3da99 : FAILURE in 1h 10m 54s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d9bc299ba1a347a896a14f171efc2db9 : SUCCESS in 1h 11m 15s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9f00809cd05a43bd91666f558162f9a0 : SUCCESS in 1h 11m 00s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ef0d549c58e04dbea8e0bbc08439b82e : FAILURE in 1h 17m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5381e4481f0b4ebfa79a0334286963b9 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ad6626c838e4ff289a4d53d07f349db : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/34097d8937d64531a9d72293494aa346 : SUCCESS in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bf54dc680c6d11325c1ce55dc52351a19673711e","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-01-28 14:03:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"af1b852d1b49c20d0853c47a9c20ef0a2d10ce48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"date":"2026-01-28 14:54:49.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1a2650b47eb0be8bca9a2d4c49763d36776b2de4","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"date":"2026-01-28 14:58:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4ba66493460238ef72da982534debd0bf4ddfa96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 17:04:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aac49ac4cba46aabdf4faf2d864cfe1\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7b1c9d8232a54156a3f5dd370238691a : SUCCESS in 39m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d4d8acda725e40fca6be7e86ef7b7a84 : SUCCESS in 44m 54s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9b86af9b20344340b7a56bf27738fe8e : SUCCESS in 38m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/eb42d6c1efa448239a0109fff182f0cb : SUCCESS in 56m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9735773e65f14be895f3b959ae19d331 : SUCCESS in 41m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f0a3118d4ea43c7bcf2c0fca4bb5ff5 : SUCCESS in 57m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cbfda3627c534dbda7bdc2bf6f18c5c9 : SUCCESS in 29m 32s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c2a9e87b3b7343db9fbeb63670bfe2c4 : FAILURE in 39m 22s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/448c07bca297411190b604d03f7a4b14 : SUCCESS in 58m 21s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b4cebe58673e48a3b832a1932338620e : SUCCESS in 1h 39m 55s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/9226523b629b468bb7ea70d21c86c11c : SUCCESS in 1h 02m 45s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fa1e8e88d33b4e6a80b9c21a60905814 : SUCCESS in 1h 07m 26s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67826cc150274fe1a74e3c92e09ed84a : SUCCESS in 1h 19m 32s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6419de5153ab4ed1a7d648834354f31b : SUCCESS in 1h 31m 03s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/4b7c3e96840a426f9a4ce9fc3844d253 : SUCCESS in 59m 31s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/936e768a59f54ae3baf5065e91d51d5b : FAILURE in 42m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/4b160954ce0040caa80029e6796891bb : FAILURE in 44m 32s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1379063bd06d491aad001ea263f9365d : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4be32d17faf4b0dacdf87dd47c57506 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ebcc2d5f9ca4cadb82723ad66e784ac : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39044b63f49e4c73891f10af0cd93ad6 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b390b3efd00447ce9035b5bb15a72ca4 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c99fb0f64ce4130a019c9a360af218e : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5317e3c0ff8c4a73903f3db166584361 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":6},{"id":"63cb83439632f1b1ea839d4c4699693256ac281e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 03:15:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04b8d70fab11478fac1611ef15e1b7d8\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/abe8c2662aaa4844adde9c3920e80415 : SUCCESS in 1h 45m 18s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/45c86ebfb85b4114a1f5e4844ba1ea8b : SUCCESS in 1h 43m 04s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/03474653c42847a9836c60f301cc4c43 : SUCCESS in 1h 42m 36s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/134f1a0d26ff45f99c56b203180e0205 : FAILURE in 23m 07s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/8f75eadb0af749219ede63cda0e6c51c : FAILURE in 1h 42m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/196e361f4baf463bb0fadff7930d7e32 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e828d5ae485c4eda8632f1ed3b9f31fb : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e17a531882ac45c09f5b68a33b1d2084 : RETRY_LIMIT in 2m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c0e70d52b1dacb668f6762f810fbc14539bfbb8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-16 10:37:17.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_37820\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"}],"_revision_number":7},{"id":"9de66596b2164d09bfbc01683f663a6ded68763a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-16 10:37:23.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"369702ac0803e2e0df3c42b7553b22ef82e37805","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-16 10:54:27.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"e0aafeb1c344c7053e5d2e9160b96a7941eb381c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-16 11:54:13.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"2583a55218f3e4f4e6a026ab5d0726a684b99641","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-16 12:16:39.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"8bcd52389aa1ad91c3dc60546148015d7b3bddd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:26:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db962d5f74704f10a8be507e34cafb97\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c14222c111c44763b6b7a6b58110f96a : SUCCESS in 46m 31s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e475a15734634033ae29c4e61e3dcb5c : SUCCESS in 47m 09s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/42f4ab4bc70848189b60f72ac6c679d9 : SUCCESS in 14m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/18b1c521c1534291b18a6f563be6c34f : SUCCESS in 48m 38s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9d95c60ee9fb46eaa74c4ab13e856995 : SUCCESS in 37m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd43b37707e64206b4dc3f7d10135cd7 : SUCCESS in 43m 33s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/86b3e193cd67446b897bb288d26b9006 : SUCCESS in 15m 40s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/eef76ed688854f248b971ff062b60bf8 : FAILURE in 43m 00s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b65028d37acb4fe78e14949fd2356200 : SUCCESS in 52m 34s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/2127a0822e8f4b5b9424eef6b35d68c1 : SUCCESS in 1h 14m 50s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a478ff60b44d4aa9ba8acded5b4e6410 : SUCCESS in 1h 00m 54s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d58016f8eb4d40ee907b0e085457c3da : SUCCESS in 1h 20m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72cbe02cc2444fc38a6e174961befed3 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/039771017a1d4cf1b03e1bc3cb3131fe : SUCCESS in 1h 21m 23s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/75913dabe1934f91ba2f15ac5d01760f : SUCCESS in 1h 01m 40s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/921f8a9fae1c410fa851387a78554729 : SUCCESS in 1h 30m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d80a6c3e7e774f368e6022659d034a7c : SUCCESS in 1h 24m 52s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/577d998262f0465595950422fcbecbc8 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623a2148da3c417ea783e09c79f8ddea : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ee77e51899b458c9ae590b712a1a301 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea379ac816cd4fc9a0a16c60a1e190f9 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e10b34d53e24ea384a87ca1c9b8908f : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9da752e8a84395b8b86b7144446e3c : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b3526ac9adf4cd29c36e88e9fbaf4e3 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":8},{"id":"85d19619747581501851bad311d7f2c3c02efc01","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2026-02-16 14:13:19.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"eaa7c86d885116051b95ca09e2667844d4a8d466","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-16 15:58:06.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"24dab7d64c5b45eb53a677ab16eb9768ece905bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:58:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e2027aaa406479ebc66d86acc71ad5f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e65f2d3286ae4fc09a934d6dd6bd104e : SUCCESS in 51m 24s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/77a54c0c51134296b26f1be6288f0d7f : SUCCESS in 54m 05s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/998e2dc2ec5f4ab2bdecc600f2160850 : SUCCESS in 49m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b61cf417e75f474b9184362dea0845cb : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/41ebb0a98e6c4d7aacc858b2369f0e46 : FAILURE in 1h 19m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d20a0fc22e6a4718802d238abcf69583 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/815d90eabfa64573b4a4dae6fc601c1c : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c3d91acdf7fc44399b4b6ca87222a07d : RETRY_LIMIT in 2m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3253f368d1b8c9f1b672feb4e2bfeda586d62307","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-19 10:26:01.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"f7d9d310e7914548c36171f5e74de2be825a2cf6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-07 19:53:17.000000000","message":"Abandoned\n\nNo updates since February, please reopen if there\u0027s still a problem in supported stable branches.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6d393bca2ed349b7ff33977e6ebd774db893dacb","revisions":{"2bb505ba1006e6fba34c1d8eca6ed0d5d5386841":{"kind":"REWORK","_number":1,"created":"2025-10-02 12:45:22.000000000","uploader":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"ref":"refs/changes/89/962789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/962789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/962789/1"}}},"commit":{"parents":[{"commit":"f17a23747f31305eed1be84569aac907fae15c19","subject":"CI: Fix rocky publish jobs - version missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f17a23747f31305eed1be84569aac907fae15c19"}]}],"author":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:31:24.000000000","tz":120},"committer":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:36:40.000000000","tz":120},"subject":"openvswitch: Fix log file permissions for fluentd access","message":"openvswitch: Fix log file permissions for fluentd access\n\nThe openvswitch log files are currently created with permissions\n0640 and ownership openvswitch:openvswitch. This prevents fluentd\nfrom reading these log files, as the fluentd user runs with the\nkolla group but is not a member of the openvswitch group.\n\nCloses-Bug: #1795609\n\nChange-Id: Id1019fff2eca58f76f690d2c823a2e4fbc27e350\nSigned-off-by: Norman Ziegner \u003cn.ziegner@hzdr.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bb505ba1006e6fba34c1d8eca6ed0d5d5386841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bb505ba1006e6fba34c1d8eca6ed0d5d5386841"}]},"branch":"refs/heads/master"},"4f4041bc46b67d3cc54c014d2f1749931d976a99":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-20 08:48:34.000000000","uploader":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"ref":"refs/changes/89/962789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/962789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/962789/2"}}},"commit":{"parents":[{"commit":"f71ab2be350a3366a5113ac04248bc8cb368dc52","subject":"Merge \"Drop support for building MariaDB clustercheck\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f71ab2be350a3366a5113ac04248bc8cb368dc52"}]}],"author":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:31:24.000000000","tz":120},"committer":{"name":"Normo","email":"n.ziegner@hzdr.de","date":"2025-10-20 08:48:34.000000000","tz":0},"subject":"openvswitch: Fix log file permissions for fluentd access","message":"openvswitch: Fix log file permissions for fluentd access\n\nThe openvswitch log files are currently created with permissions\n0640 and ownership openvswitch:openvswitch. This prevents fluentd\nfrom reading these log files, as the fluentd user runs with the\nkolla group but is not a member of the openvswitch group.\n\nCloses-Bug: #1795609\n\nChange-Id: Id1019fff2eca58f76f690d2c823a2e4fbc27e350\nSigned-off-by: Norman Ziegner \u003cn.ziegner@hzdr.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f4041bc46b67d3cc54c014d2f1749931d976a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f4041bc46b67d3cc54c014d2f1749931d976a99"}]},"branch":"refs/heads/master","description":"Rebase"},"ae685f6ed4b82e4d82cfe2a21af241e82fe63050":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-24 11:17:52.000000000","uploader":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"ref":"refs/changes/89/962789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/962789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/962789/3"}}},"commit":{"parents":[{"commit":"2fdd26aa3823c12198f32405fbc3992d6be1333d","subject":"Merge \"neutron-metadata-agent: copy wrapper script for haproxy processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fdd26aa3823c12198f32405fbc3992d6be1333d"}]}],"author":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:31:24.000000000","tz":120},"committer":{"name":"Normo","email":"n.ziegner@hzdr.de","date":"2025-10-24 11:17:52.000000000","tz":0},"subject":"openvswitch: Fix log file permissions for fluentd access","message":"openvswitch: Fix log file permissions for fluentd access\n\nThe openvswitch log files are currently created with permissions\n0640 and ownership openvswitch:openvswitch. This prevents fluentd\nfrom reading these log files, as the fluentd user runs with the\nkolla group but is not a member of the openvswitch group.\n\nCloses-Bug: #1795609\n\nChange-Id: Id1019fff2eca58f76f690d2c823a2e4fbc27e350\nSigned-off-by: Norman Ziegner \u003cn.ziegner@hzdr.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae685f6ed4b82e4d82cfe2a21af241e82fe63050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae685f6ed4b82e4d82cfe2a21af241e82fe63050"}]},"branch":"refs/heads/master","description":"Rebase"},"6d8d6bd83aa5495e0c98c0190b4348de004b2ac2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-09 11:03:32.000000000","uploader":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"ref":"refs/changes/89/962789/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/962789/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/962789/4"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:31:24.000000000","tz":120},"committer":{"name":"Normo","email":"n.ziegner@hzdr.de","date":"2026-01-09 11:03:32.000000000","tz":0},"subject":"openvswitch: Fix log file permissions for fluentd access","message":"openvswitch: Fix log file permissions for fluentd access\n\nThe openvswitch log files are currently created with permissions\n0640 and ownership openvswitch:openvswitch. This prevents fluentd\nfrom reading these log files, as the fluentd user runs with the\nkolla group but is not a member of the openvswitch group.\n\nCloses-Bug: #1795609\n\nChange-Id: Id1019fff2eca58f76f690d2c823a2e4fbc27e350\nSigned-off-by: Norman Ziegner \u003cn.ziegner@hzdr.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d8d6bd83aa5495e0c98c0190b4348de004b2ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d8d6bd83aa5495e0c98c0190b4348de004b2ac2"}]},"branch":"refs/heads/master","description":"Rebase"},"d0e5c42f0d254510ae846540be77e78a2e0c912f":{"kind":"REWORK","_number":5,"created":"2026-01-09 15:31:13.000000000","uploader":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"ref":"refs/changes/89/962789/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/962789/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/962789/5"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:31:24.000000000","tz":120},"committer":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2026-01-09 15:27:37.000000000","tz":60},"subject":"openvswitch: Fix log file permissions for fluentd access","message":"openvswitch: Fix log file permissions for fluentd access\n\nThe openvswitch log files are currently created with permissions\n0640 and ownership openvswitch:openvswitch. This prevents fluentd\nfrom reading these log files, as the fluentd user runs with the\nkolla group but is not a member of the openvswitch group.\n\nCloses-Bug: #1795609\n\nChange-Id: Id1019fff2eca58f76f690d2c823a2e4fbc27e350\nSigned-off-by: Norman Ziegner \u003cn.ziegner@hzdr.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0e5c42f0d254510ae846540be77e78a2e0c912f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0e5c42f0d254510ae846540be77e78a2e0c912f"}]},"branch":"refs/heads/master"},"65bf17802af60ab6f3febd519d13f42a922af030":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-28 14:54:49.000000000","uploader":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"ref":"refs/changes/89/962789/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/962789/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/962789/6"}}},"commit":{"parents":[{"commit":"f6f52cdcc2857bde404a4abe7d772f96cd0d61a8","subject":"Fix Misinformation on ProxySQL version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6f52cdcc2857bde404a4abe7d772f96cd0d61a8"}]}],"author":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:31:24.000000000","tz":120},"committer":{"name":"Normo","email":"n.ziegner@hzdr.de","date":"2026-01-28 14:54:49.000000000","tz":0},"subject":"openvswitch: Fix log file permissions for fluentd access","message":"openvswitch: Fix log file permissions for fluentd access\n\nThe openvswitch log files are currently created with permissions\n0640 and ownership openvswitch:openvswitch. This prevents fluentd\nfrom reading these log files, as the fluentd user runs with the\nkolla group but is not a member of the openvswitch group.\n\nCloses-Bug: #1795609\n\nChange-Id: Id1019fff2eca58f76f690d2c823a2e4fbc27e350\nSigned-off-by: Norman Ziegner \u003cn.ziegner@hzdr.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65bf17802af60ab6f3febd519d13f42a922af030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65bf17802af60ab6f3febd519d13f42a922af030"}]},"branch":"refs/heads/master","description":"Rebase"},"db13b4d39bfc9cfa8a98417c05b2733a797bab64":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-02-16 10:37:17.000000000","uploader":{"_account_id":37820,"name":"Normo","display_name":"Norman Ziegner","email":"n.ziegner@hzdr.de","username":"Normo"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/89/962789/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/962789/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/962789/7"}}},"commit":{"parents":[{"commit":"28501022bc17ca40fb6b8dba879d5df45bcab24e","subject":"Merge \"Stop installation of zake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28501022bc17ca40fb6b8dba879d5df45bcab24e"}]}],"author":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:31:24.000000000","tz":120},"committer":{"name":"Normo","email":"n.ziegner@hzdr.de","date":"2026-02-16 10:37:17.000000000","tz":0},"subject":"openvswitch: Fix log file permissions for fluentd access","message":"openvswitch: Fix log file permissions for fluentd access\n\nThe openvswitch log files are currently created with permissions\n0640 and ownership openvswitch:openvswitch. This prevents fluentd\nfrom reading these log files, as the fluentd user runs with the\nkolla group but is not a member of the openvswitch group.\n\nCloses-Bug: #1795609\n\nChange-Id: Id1019fff2eca58f76f690d2c823a2e4fbc27e350\nSigned-off-by: Norman Ziegner \u003cn.ziegner@hzdr.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db13b4d39bfc9cfa8a98417c05b2733a797bab64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db13b4d39bfc9cfa8a98417c05b2733a797bab64"}]},"branch":"refs/heads/master","description":"Rebase"},"6d393bca2ed349b7ff33977e6ebd774db893dacb":{"kind":"REWORK","_number":8,"created":"2026-02-16 11:54:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/962789/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/962789/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/962789/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/962789/8"}}},"commit":{"parents":[{"commit":"28501022bc17ca40fb6b8dba879d5df45bcab24e","subject":"Merge \"Stop installation of zake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28501022bc17ca40fb6b8dba879d5df45bcab24e"}]}],"author":{"name":"Norman Ziegner","email":"n.ziegner@hzdr.de","date":"2025-10-02 12:31:24.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-02-16 11:54:12.000000000","tz":0},"subject":"openvswitch: Fix log file permissions for fluentd access","message":"openvswitch: Fix log file permissions for fluentd access\n\nThe openvswitch log files are currently created with permissions\n0640 and ownership openvswitch:openvswitch. This prevents fluentd\nfrom reading these log files, as the fluentd user runs with the\nkolla group but is not a member of the openvswitch group.\n\nCloses-Bug: #1795609\n\nChange-Id: Id1019fff2eca58f76f690d2c823a2e4fbc27e350\nSigned-off-by: Norman Ziegner \u003cn.ziegner@hzdr.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d393bca2ed349b7ff33977e6ebd774db893dacb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d393bca2ed349b7ff33977e6ebd774db893dacb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
