)]}'
{"id":"openstack%2Fkolla~963258","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ic46b895775edf5e5fb2b637be49e2de1eb4adf36","project":"openstack/kolla","branch":"stable/2024.1","topic":"ironic-fix-stable/2024.1","attention_set":{},"removed_from_attention_set":{"30226":{"account":{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"},"last_update":"2025-11-17 16:29:31.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-17 16:29:31.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-29 13:33:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Ic46b895775edf5e5fb2b637be49e2de1eb4adf36","subject":"Fix permissions for ironic metrics","status":"MERGED","created":"2025-10-07 10:35:24.000000000","updated":"2025-11-17 16:31:17.000000000","submitted":"2025-11-17 16:29:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963258-ironic-fix-stable/2024.1","cherry_pick_of_change":940514,"cherry_pick_of_patch_set":14,"meta_rev_id":"445b98f5f6d2734f670308372601e2843fac125a","_number":963258,"virtual_id_number":963258,"owner":{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 16:29:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-17 11:54:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-11-17 11:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-11-17 11:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-17 14:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-17 14:43:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 10:35:24.000000000","updated_by":{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2025-10-07 10:45:34.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2025-10-07 11:49:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-07 16:55:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-09 05:53:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-17 08:55:48.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-11-17 14:43:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f63c0dcbea92612f72825a5e327f265bd4ca4abd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"},"date":"2025-10-07 10:35:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cf6cee61728ce25aa366020c64e1da2e45cec4a5","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2025-10-07 10:45:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"198cfed4f84f5615390155da9cf7d6e112f5ee64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 11:49:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b4e22b853f747ba8ca70c365e0709f2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/69035e4cc9e749d4b2c98b9ed4112c1b : SUCCESS in 25m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/22ba993a784647f7b8d79da637757cbb : FAILURE in 15m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5dc5655458be4ecc9c5afac9119a831f : SUCCESS in 11m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/35505f89243c44318f4d551c2499ba56 : SUCCESS in 28m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f15d3d8a1549415c923c123370a99ccb : SUCCESS in 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b44329fdd04919b443353461b7d3ea : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/173e069fffdf416c992602bc196e2aa3 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9e479cd4ec64335ad902ecf6887b22b : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf2394d135ec42ed94dc920a615fa803 : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20b44bd67d354ef98cef8304b0644a89 : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7605cc0bc4e41bf93802ac70d91fb05 : SUCCESS in 5m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/73b6d0c8be894dd8be10b97590b806ed : SUCCESS in 3m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71b71f4992e147d7a8808f51f248b7dc : SUCCESS in 19m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c80d61bbb8074cf697a12a62a8adfb16 : SUCCESS in 27m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f9b4e4184ff14a0ba2bd160508c77859 : SUCCESS in 49m 47s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a80cc5df64e240ffa37dcb7fd8c4ba62 : SUCCESS in 1h 04m 54s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ea4d9c18c41470a9354fd6876ff46b3 : SUCCESS in 17m 53s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8de69f386c0549cc9ed77d32e1d2a0db : SUCCESS in 17m 48s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/ac30d9999a134585aee1d0e55a384ffe : SUCCESS in 1h 02m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f23aa7a4c7f44b9ca60c5719382c025c : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3a2957da2dd883a84f918b15ded6e8b2fa35eb0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 12:42:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b471219359774d798ea953bab94d7c2f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/567c63b1671740a5b72aab5b95bb6f75 : SUCCESS in 28m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/59a68ebc6ac940dc81299ae69531c40c : FAILURE in 2h 04m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc88dca82d3f408b9464312ccd58e2c8 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f94dfa9a4218404fbb1d34386bec36fb : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f642ff583dee443a9b8453b48997ea2d : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c5dab8402ce3109b80fa20ad6c5bb9b73f455d3f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-07 16:55:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5a5da5f0566e73677431ed04408465b1e8c5f28","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-07 17:13:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8545e14b5e2e645e0f94ffa83542e038f3fe6d83","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 05:53:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fefdc64b64212202f5c050bd6d81d5d8f34db6b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-09 09:14:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"271d27dfb2315dd6d2db20ac306cd72b3681e786","author":{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"},"date":"2025-10-09 10:49:42.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c017f0ccd0e2e58f50896e8c35dad677dd3d5d18","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-29 09:37:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2859102820797b6eaaa8ef15f3fcf05b175012e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 09:37:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db74064bb7b3b3d82a01b64ebd089b6edaea4652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 10:58:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a5bba5b6bb046b1b8928c3e0530d6ed\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8e8b45bb0231477f9dd7ae535ef776a4 : SUCCESS in 20m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/54f193038faa4ccea5dde3dee7fa001b : SUCCESS in 30m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/950fb56cbfc44af49a2cb6695683bbc2 : SUCCESS in 1h 01m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52308b2c3db84c7fb1d009fc0fcd638d : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fea8043a7f9c4ac78236fd809927c303 : NODE_FAILURE Node(set) request c134f4deb241413e9954679efdfb7e68 failed in 0s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7547636d835c4f2d919f20f8efc6140d : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d3fed0e29846c1a480992353f2d168 : SUCCESS in 4m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/936962560dac4daeb1d70eae835b9c7c : SUCCESS in 2m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f7fefc9351c64f9a9338bcd7338e9b40 : SUCCESS in 16m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/afbcfe99a3954f3b99a3d751a705c56d : SUCCESS in 15m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0d350aec3484440386bd302c15e4baa5 : SUCCESS in 1h 09m 44s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b97bc1c5e0a44c14a93d49c769a16c34 : SUCCESS in 18m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f24a410ab80b4a1a9f1644efcbc0ac67 : SUCCESS in 23m 36s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/931434a2089846379436623a5f4b2457 : SUCCESS in 57m 22s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f4046f453226468eb4041eabc1bfc430 : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"af4bff00c2386be27d023879eaece04d0befda4e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-29 13:33:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81fcb1478099d3046022ce5c8aea97160221f86e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 15:05:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/516af34e899d4abe8ddd58200f5bcccf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e946ee7c7276407598f62347e058ca7f : SUCCESS in 24m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/071af8a760864c05a7ca2c7faa1e9da5 : POST_FAILURE in 16m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/59dafdd4952547859edeeb0848e7cae8 : SUCCESS in 28m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1e0b81ac46b449eca4ca00a2afc65e1b : SUCCESS in 27m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c3a5835a85734dea990c9d8b1b445e58 : SUCCESS in 1h 05m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0954b375ae5b4d75a1d499ef0c1b3456 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf35174ef59f4c149f076db17e025c5f : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21960198558b4e63be92f2e5624dc03c : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9987d986b9524bf1a3886efb0b0219e8 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/628662f9ba634044981c9dc514f98f6e : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d6b9d6f9ff49ddbe898a90cae3cebc : SUCCESS in 5m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8138624bc36f4168bb01e65e5f4c0e3e : SUCCESS in 2m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c3ac2678c3c54f3b885cb46604ce6f8e : SUCCESS in 23m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/95de914a607f4224b304062518733507 : SUCCESS in 27m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c14bc99e57554668a163ef036d822cf1 : SUCCESS in 1h 09m 43s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/0a45b9c74f4b4da493c8a957dff72981 : SUCCESS in 1h 22m 52s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dc72686f8304fad97c742d909a5f0aa : SUCCESS in 19m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9338b872d821463f9603c5ea778b9235 : SUCCESS in 22m 36s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/0fd10b94800449da95a35cd239988177 : SUCCESS in 1h 00m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d82389546e304541a7a180a1f9abf6ed : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a318d2e41efefb78faf39ceef92d9968aa9b2db1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 15:05:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1160f34f0210b72421fa780311314c55793f033","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 15:37:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/986a5532f53c41b39e5ec7300bf713d5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/80541945c74b43adbc85cfc8222c7d7b : SUCCESS in 23m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e1f6f87eb1cd4c80b19493d1387a0f3c : SUCCESS in 1h 28m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31ea3e2721dd43678cb2c0451ac1bd8e : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5108237c9bcf43babb0bf4d28c2caddf : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6be8ed925fd24446b2dda85cf1e24668 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a7a8c83b932b6fe270ee5c22e4346745de8c9294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 16:29:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebea0702ceac4174a7a6c77a92051895\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fedd8b3343424872b2799ca7de133648 : SUCCESS in 18m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/81c96c4c6f7449b392b2ca378854d19c : SUCCESS in 28m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/71ae4cdcee324594accb5a4527edeee0 : SUCCESS in 1h 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/706d21ea09d943208f45ba0a3779478f : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5847a8e4ee7644fd84c265804aa080a5 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76344f8d78f043c89a8612f3d61093af : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71857d719b5a4ee792f508f0ab4369d5 : SUCCESS in 6m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1002791b9865435abb93568e4053ed9e : SUCCESS in 2m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/993216968ebf4342a52617fe10dd1cd9 : SUCCESS in 22m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/942de096a78e4444be6d61fc76e1237c : SUCCESS in 26m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/aa528d7e67184483bacf196ee1eac7ff : SUCCESS in 1h 07m 59s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c8ffe0f18394aa59e9df55f1b6dc771 : SUCCESS in 21m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abc55c45e89f4cc491736bb254395c12 : NODE_FAILURE Node(set) request 313c1359acf249838e27cb2b0470e758 failed in 0s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/883a6007da024752a2800c1ac3dff1a7 : FAILURE in 1h 18m 33s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33dfa9e8c8564135998a036f97fa898c : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"db2d71f6ebf2728f6e514b8947536f00d45672e4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-17 08:55:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c363e17a2b34a66bff0e5f50a8a0d0e7f7c77141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 10:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6cf802c1c30423b8f372f4481ac31bb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9fdbce5bc1c044a980e2235a72b584e0 : FAILURE in 44m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/60d891796d9543d5ba82a6fa3af1db9e : FAILURE in 6m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9600c2f8e4954acd95d2483ae51be872 : FAILURE in 36m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d31299fa4a4d4910860c63643d65c562 : FAILURE in 27m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1ef49eec74f74b6aa6648e846ef0161d : SUCCESS in 1h 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f350907de9743338c0d24bffc5f4a96 : SUCCESS in 2m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e80ec3d8a8c4107afa44f59a40eab1b : FAILURE in 2m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d63f493b6c3941e68b4dfc309ca6a118 : SUCCESS in 2m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4464dd090f7840dfa9e8380cd1e8352a : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16d1f2051103450c90b4d4257bddd4d7 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92764933a2b84f46896b98b8869f7195 : SUCCESS in 4m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ff468f11227b4a9886d14b60ce3d00e6 : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/88647996b82d45e1911dec57b91cefa7 : FAILURE in 57m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cd5072c13dff4188aa490d91c84f8d0e : FAILURE in 22m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f3ee526e064d4477b5944d9f3e67081f : SUCCESS in 1h 19m 48s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/7fda6635d0cb4e98ad2b8d49e5de3399 : SUCCESS in 1h 26m 39s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/376389c18c42441999858de1a4c0bec3 : FAILURE in 46m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cabcab616f14626a1d399aa707f3c33 : FAILURE in 42m 05s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/b0b07d836d064b8e8bad1ae84a938ec5 : FAILURE in 1h 03m 16s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bb5139a7f4d04fb28355fff794c0b0c8 : FAILURE in 27m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f3688e23bb4a081366b96c7c4ec6538ee7c51aa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 10:48:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c4899ab37b48cb9613869419d2e526\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8b0deb21f55345e098bce15d665043ce : SUCCESS in 58m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/87b29b21f7074291ac2e2265a4b7768b : SUCCESS in 1h 50m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a06cf19035774f03baf9d62a60ed81c4 : FAILURE in 3m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e8e5afc52d1e435685ccad8a60a2a7eb : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/951937fdcf2e4f329d39af17ca971676 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3cdd4095f00b6e17fd7eb21d7e10d03988fb6825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-17 11:54:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_30226\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"}],"_revision_number":2},{"id":"7b2365ce0fcdf28cc797b51ac50676dd13b2910b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 13:56:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56c59782fef84e959e0edc15607795e2\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2ce2d857f3614f31a51b62748a9fec59 : SUCCESS in 1h 43m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/92645c44df74428ea1ddd4f87209b7bd : SUCCESS in 1h 58m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1cbc44a9cfd4694b28631e0da95c756 : FAILURE in 2m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/72db5ab134754d02943ad7b9e8887c09 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ca10b8cfb99045958028897b5b0c3d0c : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ebb3e17d617782596ca5c4124d13b2c40ef76c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:07:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2e24ea4dae94268af6857500e43b634\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/768ca7a13aec440ab7fec87e817d3ad2 : SUCCESS in 50m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/979179ce42854100af5354ef72ed89b0 : FAILURE in 6m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e360f315b5dc46ccbf22e35d133b901a : SUCCESS in 55m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/30d5bce0af6245c9ac346227e6fa3bfe : SUCCESS in 28m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ed4ca1c6d017406d822c5767c4a327c8 : SUCCESS in 1h 19m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f7922f106840fba159e2e7f2132fbd : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ee6a1db20245fcb7a15040d71dd990 : FAILURE in 1m 57s (non-voting)\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59971499dc8e44e4bfc2a9a2379238b2 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e9b3868c75d4fdd90d7cbfac02c1c5e : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7bc30510c14b03b5b36a684efb712f : SUCCESS in 4m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/87a41a2a40bc4277a48462f7b5409a19 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4d967eb3c544c5580ce0b9c16536e2d : SUCCESS in 3m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/465b627caf68481889a870c2b6b8bc04 : SUCCESS in 1h 09m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7b734673a0fc4bbfa22d3b0a21cd04d9 : SUCCESS in 26m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6a9f9ce60500438e88ad77f0ca8aca64 : SUCCESS in 1h 24m 31s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/22028f448c734e70bff96ed279c2a669 : SUCCESS in 2h 05m 24s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5270317898b848ecb05b1fd08651948e : SUCCESS in 35m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61245c11dddc49df8c226e3a5ac2d581 : SUCCESS in 35m 12s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/3f3496ce300e47d29197a540dd40adc3 : SUCCESS in 58m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b044816a60b842ab81102bf24373a41d : SUCCESS in 23m 44s","accounts_in_message":[],"_revision_number":2},{"id":"6ff2b5784ffd620498687dbd1d5472dba85c7db4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-17 14:43:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"967c3144e5ffb69e68fc9160bf200f9f943268bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:44:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5f2b433b540181e107976ab420b19631a5d91cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 16:29:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b77a59d087046e887cc70a9c50b6fca\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3a1e6d0fa2b844299851819d1bcef534 : SUCCESS in 47m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9eaf4ce9b8f04c25ae5af64362a559d9 : SUCCESS in 14m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4cf4c785271d4b97a22ffe4abaa945b3 : SUCCESS in 1h 00m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cf3da2e52be4cf5a355749c3792c0af : SUCCESS in 4m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1d7d5a8abe324908947e00b7a670f1e8 : SUCCESS in 2m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/72623c7776cb4cc59223732dccd807cb : SUCCESS in 40m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/25e32745050e49c3b80b3fed3e48e454 : SUCCESS in 28m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5aa0ca3fe9724baaa2785d3c72f1df2e : SUCCESS in 1h 25m 04s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7324b6dc46448dbb8e161705a043229 : SUCCESS in 38m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d165f05543bc4b2d924744f53fe60d75 : SUCCESS in 31m 32s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/8ce44d95fadc4fda908a69820c16c40e : SUCCESS in 46m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ea76117e6964c2eb41b2757c0b2ab26 : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":2},{"id":"de01c84bd5bab99c3d39ec39db43ac4978172d9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 16:29:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"445b98f5f6d2734f670308372601e2843fac125a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 16:31:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/260b6bccf5cb42879b39cd1a84e5d692\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e947a5d67c394ae88fc392e0c69dd388 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"329319dec38d3191d9031a897bfc55a19d91828b","revisions":{"f70da005c03ed419dffe8534c9ed56818607e513":{"kind":"REWORK","_number":1,"created":"2025-10-07 10:35:24.000000000","uploader":{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"},"ref":"refs/changes/58/963258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/963258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/963258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/963258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/963258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/963258/1"}}},"commit":{"parents":[{"commit":"07103ab0a0960e2a509fa349a25ad014ef675e38","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07103ab0a0960e2a509fa349a25ad014ef675e38"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","date":"2025-10-07 10:35:24.000000000","tz":0},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nSigned-off-by: Serhat Rıfat Demircan \u003cdemircan.serhat@gmail.com\u003e\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n(cherry picked from commit ed7c8399c4a2a43e9a4165db416b98f5be8d538c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f70da005c03ed419dffe8534c9ed56818607e513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f70da005c03ed419dffe8534c9ed56818607e513"}]},"branch":"refs/heads/stable/2024.1"},"329319dec38d3191d9031a897bfc55a19d91828b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-17 11:54:30.000000000","uploader":{"_account_id":30226,"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","username":"serhatrd"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/963258/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/963258/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/963258/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/963258/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/963258/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/963258/2"}}},"commit":{"parents":[{"commit":"34ca6e35c94ddba1e47722718f3fc81a1f03e28a","subject":"bifrost: Mark as unbuildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34ca6e35c94ddba1e47722718f3fc81a1f03e28a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Serhat Rıfat Demircan","email":"demircan.serhat@gmail.com","date":"2025-11-17 11:54:30.000000000","tz":0},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nSigned-off-by: Serhat Rıfat Demircan \u003cdemircan.serhat@gmail.com\u003e\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n(cherry picked from commit ed7c8399c4a2a43e9a4165db416b98f5be8d538c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/329319dec38d3191d9031a897bfc55a19d91828b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/329319dec38d3191d9031a897bfc55a19d91828b"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
