)]}'
{"id":"openstack%2Fkolla~963515","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I85ae37250ac4ef629b77ade04826a363e8047a91","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-09 15:13:14.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-10-09 13:49:04.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I85ae37250ac4ef629b77ade04826a363e8047a91","subject":"neutron: keepalived wrapper script does not rm containers","status":"MERGED","created":"2025-10-09 11:30:04.000000000","updated":"2025-10-09 15:41:37.000000000","submitted":"2025-10-09 15:13:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963515","cherry_pick_of_change":963492,"cherry_pick_of_patch_set":1,"meta_rev_id":"800cb8bbc03ff91f9230172bfe7d7c01f2a9aaff","_number":963515,"virtual_id_number":963515,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-09 15:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-09 13:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-10-09 13:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-09 13:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-09 13:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 13:49:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f3297cc2e33fe30ae6bddec8a9494711192ae811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 11:30:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cbb3b21d94141adfaf14104360c4430ef9284d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-09 11:30:33.000000000","message":"Patch Set 2: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":2},{"id":"172d8ad8d978575663c4c5e57ad43e0cf961b19b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 13:43:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af75fea6346349c79ef0e769e1ee3386\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4e406aa818e24f06adb3ded96f65ffe6 : SUCCESS in 23m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a22ab35468474adb834dc7af498c7640 : FAILURE in 33m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/13c2276e96bf4456a3682ce355dd1ea6 : SUCCESS in 22m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/50607f499b314dbaae933a69cc6e2baa : SUCCESS in 30m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/705fe27e8c9141708f66babad4591a9d : SUCCESS in 51m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0831378ff97a4387ac86713d1d523625 : SUCCESS in 1h 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b7f766440a49019b82e6684fc32463 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fafd612a6cf45e789222f186c7be6c6 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a87931c061864095b8fdfe815958ea62 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/320ee14dd6ba479db467949e05fa42c2 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e1cb0b1d64146149f02bf3fdfb081a0 : SUCCESS in 6m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b448ecb1ab3a4f0f81963250c436adb7 : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0e65314bdc174d2a9f8d0c0695cbd750 : SUCCESS in 22m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9dd06d22f3c1433e97950f899f9b1283 : SUCCESS in 1h 16m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0bb631f306744701a316327dd6c31294 : SUCCESS in 1h 13m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0b513b33134a43919671b19c5aa315dc : SUCCESS in 1h 49m 38s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9751b8d72a314706bf3e973788c624d9 : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/3bce446283f146cdb505bc42a57f6298 : FAILURE in 1h 28m 29s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/78abd412a4eb4f37b9b9ceac9b17335e : SUCCESS in 23m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/39407311251047acaf1f7dae2a0ac007 : SUCCESS in 24m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/54ed578a97e643198b98920a984dba79 : SUCCESS in 50m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9a1b7497e38843689c9cca99b72783be : SUCCESS in 1h 13m 52s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/dfa40e7a30a04e0e81007a25ed86914a : SUCCESS in 51m 15s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c390a7993a9a41a5b4b1549f08d99e74 : SUCCESS in 1h 18m 28s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2411251f82c1425cb95265675307822d : SUCCESS in 2h 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea712c16f30c014721714301ec69ab82298a736c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-09 13:48:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3ea43211526ccfa9f990b02bf44e278388336ed3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-09 13:49:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4228d3754d3c1145e6c88c5322edd0401f576bb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 13:49:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a984a2bdaaeb416c46c9f98d1e9973727c9d8606","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 15:13:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/085d934aef134fafba8ffefff6b8e8d2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/adb76559040344d8a791f95a3db6d5b9 : SUCCESS in 19m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/70dfe18468ea4103b2d954ab7a98157f : SUCCESS in 34m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b1ff78da2a414872a05c6d9e691ac55b : SUCCESS in 50m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2ef086468a9147f7864f36da84e9ea8c : SUCCESS in 1h 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82b82c78dc2468995b611a7d29ea668 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a66f02e010e4a208a4aca15133e4d80 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb8626d858b5471d996404ab0f3d253c : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25dbb382ae984138b83fab044a0e0e6f : SUCCESS in 6m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e16ccee716244eaa564bdccc136fa9c : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/aaa670e5e60c4d3d8732f599fdcb6146 : SUCCESS in 11m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5180e27cf6384c6788d39cf34ab92053 : SUCCESS in 42m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/46a7e0cdbba94414a84a4ea50e7147f6 : SUCCESS in 1h 17m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fb492c4fdbb944f9845c7d9a53b0eded : SUCCESS in 53m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/404c234b69974266a4b17a41d8e134be : SUCCESS in 21m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/09abb8ebe877487b940524af45601afb : SUCCESS in 30m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/665aff89dc8645d29a55732cd055c6bd : SUCCESS in 53m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0f37d9a5269c4b51b628821252c20f0a : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":2},{"id":"2287fc5a2b621a4fdfe0d890e60ea185d3185ed9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 15:13:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"74d2bad1ffbe14b57b5bc0b3c7ad5bab3a907c12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 15:14:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f25c1f3d55de4aa8a7ef139061c2c3d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151a80ec4be94f12b703f5f0632d7b52 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2},{"id":"800cb8bbc03ff91f9230172bfe7d7c01f2a9aaff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 15:41:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dccbcc6076f4f089e76f8926b238eff\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cb8f5dbdecc54ec0b8163178d4b8e457 : SUCCESS in 45m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02e712a721764ef1bc6e6272ffeadd14 : SUCCESS in 53m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ff8ce21f5b78471ab85adc01a2f2f565 : SUCCESS in 2h 06m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0ec46173210445ab7a6a8019340de43 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/600e1c7fda0a45879e1ff6e0a502b325 : SUCCESS in 4m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7b6e7d1f56f94555b1eda1de0d1c8e30 : SUCCESS in 51m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/050129ef83ae4572ae0854a1b16bec55 : SUCCESS in 36m 31s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"21952c928b6665caa15089d9ac9e73f6760faa90","revisions":{"dcbc5606a2bcf909b363e0aa65d9c67d711cc95e":{"kind":"REWORK","_number":1,"created":"2025-10-09 11:30:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/963515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/963515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/963515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/963515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/963515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/963515/1"}}},"commit":{"parents":[{"commit":"d26c62383a31ab45f47a45624668862b3d2473b5","subject":"Merge \"[stable] Use versioned Erlang 26 and 27 repo on aarch64\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d26c62383a31ab45f47a45624668862b3d2473b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 08:43:29.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 11:30:04.000000000","tz":0},"subject":"neutron: keepalived wrapper script does not rm containers","message":"neutron: keepalived wrapper script does not rm containers\n\nkeepalived wrapper script as the only one didn\u0027t have --rm\nin docker/podman invocation - so these didn\u0027t get deleted\nwhen stopped/crashed/etc.\n\nCloses-Bug: #2127170\n\nChange-Id: I85ae37250ac4ef629b77ade04826a363e8047a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 90d255daa32be1c80621eb94cba87a2bded570b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcbc5606a2bcf909b363e0aa65d9c67d711cc95e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcbc5606a2bcf909b363e0aa65d9c67d711cc95e"}]},"branch":"refs/heads/stable/2025.1"},"21952c928b6665caa15089d9ac9e73f6760faa90":{"kind":"NO_CHANGE","_number":2,"created":"2025-10-09 11:30:33.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/15/963515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/963515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/963515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/963515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/963515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/963515/2"}}},"commit":{"parents":[{"commit":"d26c62383a31ab45f47a45624668862b3d2473b5","subject":"Merge \"[stable] Use versioned Erlang 26 and 27 repo on aarch64\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d26c62383a31ab45f47a45624668862b3d2473b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-09 08:43:29.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-09 11:30:33.000000000","tz":0},"subject":"neutron: keepalived wrapper script does not rm containers","message":"neutron: keepalived wrapper script does not rm containers\n\nkeepalived wrapper script as the only one didn\u0027t have --rm\nin docker/podman invocation - so these didn\u0027t get deleted\nwhen stopped/crashed/etc.\n\nCloses-Bug: #2127170\n\nChange-Id: I85ae37250ac4ef629b77ade04826a363e8047a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 90d255daa32be1c80621eb94cba87a2bded570b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21952c928b6665caa15089d9ac9e73f6760faa90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21952c928b6665caa15089d9ac9e73f6760faa90"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
