)]}'
{"id":"openstack%2Fkolla~964793","triplet_id":"openstack%2Fkolla~master~Ib7f551d4a7c6ffe95786853d6899e704de4747b8","project":"openstack/kolla","branch":"master","topic":"kolla_trixie","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-12-17 14:28:27.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-20 06:13:12.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-12-19 20:16:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"2271":{"account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"last_update":"2025-12-20 07:24:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7f551d4a7c6ffe95786853d6899e704de4747b8","subject":"Switch to Debian Trixie","status":"MERGED","created":"2025-10-24 14:59:29.000000000","updated":"2025-12-20 07:26:08.000000000","submitted":"2025-12-20 07:24:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964793-kolla_trixie","meta_rev_id":"7c7dd398088a2d391d23f51122629b0a4b76b8e2","_number":964793,"virtual_id_number":964793,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-20 07:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-12-19 16:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-12-20 06:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-12-19 20:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-12-20 06:13:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-12-17 14:28:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-12-19 16:10:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-24 16:06:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-03 11:47:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-10 19:39:57.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"CC"},{"updated":"2025-12-11 20:29:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-12-17 14:28:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a718492bd621536a4cfa67b0c7c492e5b33f5c7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-24 14:59:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ca04ece6290d0d0e632a8cd8cf963b8db47b0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 16:06:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d393184056c437ea7c7cc38c376561c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d9234a0bddd04277b5a6d5ac47d0837c : SUCCESS in 10m 59s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c8852806dac144bbb4dbcb29e18a5946 : FAILURE in 3m 57s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/f462e187cf78470384ef6d26f3cddb9e : FAILURE in 6m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/665e1c765a4148b3836cc24e26f5d063 : SUCCESS in 11m 51s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/756b94482f4449b2b0d60e43508d7cdd : SUCCESS in 1h 05m 20s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d22a47377df4242a6cb9704fd363441 : SUCCESS in 20m 14s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8e0199b53a8b451ea9d03022c4a002b5 : SUCCESS in 23m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/59043ad0774c4b2384628eb1272b8f53 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9678c3195a445038c69c1457f698c9c : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ac27d93e35c422b9d80aa3372718ba5 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24557675bf184d9a8e1b47dc8499f708 : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b560bf0fbbe4ceba87c364971b8f254 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a8532e8cd643269138224a7b9f969f : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8fbc6cfe3107eedfd8da1371a95976b5800218d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-24 16:24:16.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"54f1178431d10aa8eefd035c9aab0b3e0240b84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 18:29:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5bdb4edbfda4bde8a9041689f57e21f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/11dfcd0f38b5485eb510152bf78fb8c7 : SUCCESS in 16m 48s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c7e6008b29244354a847250c4c0ddfd2 : FAILURE in 7m 33s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/62f83c3048df4efb9ff5310a5d889428 : FAILURE in 6m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d3e87b2fe31b40268ac12942776b13cf : SUCCESS in 17m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/49897912ce924857bb6209452aea3838 : SUCCESS in 51m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34ee943bb26846c5a0e772f2ef3a3255 : SUCCESS in 11m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d2b273c7280b4db2ba3c771dba7bef94 : SUCCESS in 15m 26s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/af5bf48cdfa54239b189eb065b1875ac : FAILURE in 9m 04s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/263efbfe9a074c7da55121ebd9c4a648 : FAILURE in 9m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/b51257e1d9984b5d99d5089e92b22cf5 : SUCCESS in 1h 36m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/90abf007f0e240f8a3f453c4bbd92110 : SUCCESS in 2h 03m 46s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/945b5a6b4b214e7d93dc22b9fe4477e4 : FAILURE in 9m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/6e5076bb674b4ee1b4ee3cbe7de588e3 : SUCCESS in 1h 09m 16s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/84fbaa18120742208c5ca85b322d2858 : FAILURE in 9m 29s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/e0503e95e4844d3692c4688751084a29 : SUCCESS in 26m 19s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/c307efa1c70d4d59aeb844adc0f215cd : FAILURE in 8m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/5501ef85d7284375b21559e6aafde4bf : SUCCESS in 29m 02s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/f2e2971475f2480dbbebb14f5d49c9e7 : FAILURE in 9m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/33cd5ca472e04448ad0cced0bc82c005 : FAILURE in 1h 00m 19s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/b99fcc4a68ce415eb11ba224bf681628 : FAILURE in 9m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/0d106e9c8527404399fd4a0db58d893a : SUCCESS in 1h 33m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c306849b144c4f6d986680e32da714c5 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac9a52a578a94c909e2e0733751eaa32 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3979afd90bee46aca9034666fd550b8f : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb410413af5b4af3afe12cb7257caac1 : SUCCESS in 6m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afed26cdfef1486bae28b28c15aa3c05 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/403f8cf2b716477d8d5e22691c02dfe5 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e99975cb34b1aa1e2ce245b1b2d7d1bb38e15fc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-24 18:51:32.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7e621beece9a6c3f75052f987e3c07a721823e34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 20:35:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c38227912d5a4d7bac7606d02376c3a7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/00e90ab757c246e3a54bc46a2cc7d166 : SUCCESS in 29m 15s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/1e51d45d62464123a3c9fe7a2d00b9fc : FAILURE in 4m 26s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/302d3674722948809e946e149de852f6 : SUCCESS in 27m 10s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/038498663cec4309a5c120e042365b6d : SUCCESS in 37m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/012256dd4b8348d19d62cce005f592c2 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/14a3373f642c41d4bc73e05eec873288 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6c38f7273ab94526bcc2bab592a9027e : SUCCESS in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6a0c7fae668d6534ed13addb34b82c52f4a31f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 20:56:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdf13872da1f45babbc8fe06a7a27762\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2fa0450eccfe4655b85d58c1d57f517a : SUCCESS in 20m 18s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/db56d96a737b4166907e3ad3e6f498c8 : FAILURE in 3m 38s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/d4add1f60b4b482982dcc522076e59f2 : FAILURE in 4m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/be22a0905e914d7b89a6c294ed2ff3cf : SUCCESS in 18m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/97e21d12d19b4daab41b51ba6f107ec4 : SUCCESS in 45m 14s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad13b5b9e13740f3bd8d6119fa9321ec : SUCCESS in 12m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/43a8150489a94072be603a8293f2bd67 : SUCCESS in 14m 15s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/9cfa49a325f642fe8f77d353e76ca3e9 : FAILURE in 8m 26s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/925a2d7b22db4fe19b97f0135758d495 : FAILURE in 11m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/ec1b2d3e3904482e9f087901e69453ea : SUCCESS in 1h 40m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7905c7478e8c4e00b7145af05ed9fb1e : SUCCESS in 2h 03m 30s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/7f0a9eb8ca014239a7b6c5f8c5e2e8fa : FAILURE in 9m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/ab3c9b803f0b4239934b40d42368ef7e : SUCCESS in 40m 12s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/f72a487e071a4cd0a89649691905b892 : FAILURE in 7m 28s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/456190699bbf47cb99641e3253aea579 : SUCCESS in 27m 31s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/dd64a21274794c40b5e37face3da47fb : FAILURE in 9m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/1dd92c8630cd4d7d94229fba739407b7 : SUCCESS in 49m 03s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/cd9cdb0ff3e04f66a762d9f03bacd6bf : FAILURE in 7m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/25a3d2aae21049d28258d790db0f2c09 : SUCCESS in 49m 00s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/1e46fd48e2cb485ea2623b33cb40aae8 : FAILURE in 7m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/6f5a04cb6d284c5b999d9e567472b048 : FAILURE in 1h 19m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e4d8da265fb94cdda16fba8a7328f6e8 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a35591d05845c7ad5638324f3a3ac9 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18d7c329b4a84bfa80740a0eef76c3f7 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d32ba5f0ae2747488fb9f778836ecc68 : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea2a17314361471ab27390f3a2897c8e : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e9eff8147764c63acd0313fb43d98a3 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"c505f7b8540a7d04a780d3311a935dd72aad603d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-25 04:48:25.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7e83d80234700ba898c986082acbba352d4c0de5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-25 05:44:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f37f1d6e27545808c5994f3b2694df4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/fd21aeca8a6b4b4381747728cca19c9a : FAILURE in 54m 16s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/82012a1ce0bd4dd99848221df9c0b779 : FAILURE in 17m 19s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4a8ad094a9574fb1ae36a927c3665e56 : FAILURE in 35m 40s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/fbfb171174ef4d1e89482e5017bb7681 : FAILURE in 48m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5ea592f864d4f25ae7840960fe091ef : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b932109ee0c6499fbce57a4986b46673 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b0064ba60e49429d990f4da3e31c10c9 : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5fff37cc35a5291bc05c9a3f3a5bb137867d53ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-25 05:59:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b5e58a7d4024d2aa666866aa2c1b3d3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/9f22a032599e49ac987800f23c42e94f : FAILURE in 13m 16s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c88ae4f3196740e08def5a8b98a57655 : FAILURE in 11m 04s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/5278b9d9a1cd4be0a74742c67bc54cbc : FAILURE in 13m 37s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/33856e11d3b34ba6b3f70415245d85a3 : FAILURE in 17m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b134a2a4c766444398319e1e73f5cf64 : FAILURE in 1h 05m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29ff45c877594402b4b8c0d2a70746a6 : FAILURE in 25m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/109da35320464771b53970b1188531c4 : FAILURE in 32m 12s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/888e9df140664aa9af2b12852dbccf50 : FAILURE in 29m 10s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2f7ab47607e74a3e81149f47bc3499e2 : FAILURE in 13m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/1e0c3212735d4cff9e10c9a227d9825b : FAILURE in 29m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9a2a16c709ef437c8b112fb8da20eac8 : FAILURE in 29m 18s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/2577c55ab26b4372abea79dce620a127 : FAILURE in 12m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/a19c5345b9a241f49d5bcdb41cec6cd6 : FAILURE in 17m 46s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/8474c71d0bb44b1d915223eb23c19a03 : FAILURE in 14m 48s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5676b76815ef40d885864f4bd8251543 : SUCCESS in 22m 40s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/d1e3e30469d0490199c6fb7f441e9cd7 : FAILURE in 11m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/b2290765f4c2431c9a266a3e19e11333 : FAILURE in 40m 05s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/25628b2f45e54e3cbaad3a7df0cca0ae : FAILURE in 15m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/55c369df8d874b25a9365982bc71a401 : FAILURE in 27m 52s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/f813b9c4d16c4a10a2b5b47a8c78351e : FAILURE in 13m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/375b8a7341a24f31ac079cda3504188f : FAILURE in 20m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ac2fa2f3a724754b1c0166333bef31b : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9565744573aa4d12aca3409f4f7c90b8 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a0c67caecd84595856ab2e978aa30c6 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5bbd40fd7444b8b8f1ca9e1f60023d8 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18dfffba8374489d9b87889777ebba2b : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61cb5095bb6a49b192ab4700c7bb2155 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":4},{"id":"281ade1b1d8c5b66412fda4b74652daedba21904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-26 20:09:22.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"733e4008f3f5a061ae2a8bf6e75ad15445faabd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-26 21:06:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/835d0ed68a9745c2bea81a0dfcd86c37\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/89ca6793dfc54ef8a89de9cd9f3e818e : FAILURE in 44m 17s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/679ef7986475436083b78a2610a02dcd : FAILURE in 7m 07s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/cbbd56b4cd70408581d128495d29c4f0 : FAILURE in 12m 31s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9215fb7b8e9948fe9df003879b56f4d4 : FAILURE in 44m 15s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f8b487318e034edca1d7a8bcca81a95c : FAILURE in 47m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1b638143c5f46ecb292ab8d87ae7c70 : FAILURE in 24m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ef4fdb926a349e3b6ed7acdba3d9471 : FAILURE in 32m 02s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/5049281bef9b4c6bba3db1fd322bf263 : FAILURE in 7m 58s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8a5d2bb1a42a467b9d079114bef1f19b : FAILURE in 7m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/484fd89c4ff14db7816a4a5ceb5bb489 : FAILURE in 25m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5cb11a7b07854e3faf1f65437d387338 : FAILURE in 26m 11s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/2d74efe534ee4018a11fa5c72cd1ae9a : FAILURE in 12m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/99885e7cdbf847ae9e5ea4f7805815b9 : FAILURE in 24m 06s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/29199379ffa54ced9330cbdb20753861 : FAILURE in 11m 01s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/0839324cde7a4d45bcb338bf319ddbfa : SUCCESS in 21m 58s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/9b3a4faff24545d586bd8296d8093e4f : FAILURE in 13m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/8c4588bcdb774d7aaf8c32cfde719a30 : FAILURE in 46m 52s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/eb30616870494c5eb6c7bea99d69e356 : FAILURE in 11m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/2c895561ec8b47cbb03b3c156aa368a9 : FAILURE in 19m 42s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/08adadc06f644ad1aa842cd5df2cb988 : FAILURE in 11m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/3bf99bb2259444e5a934a820aa8f1a5c : FAILURE in 18m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0fefedc3f92f49a3a6f90ee24ef41795 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95bd4b602be54ac9a8ba84559de942fd : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6d889af081b4e10ba9527a373aacad8 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e70d234ed7d4f7da4b7ede1f64aae78 : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71e60ecd2e78495bb2f34e8965bde76a : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4d74b42fbf4da0b4ba5877898f1782 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":5},{"id":"a32f79f51cb9cb1956d17239e3cadb589d284ac7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-26 21:14:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1104e7952d604b5b89c0c261061ecee5\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/323527aba14b428fb2d4ddbdba7545ed : FAILURE in 32m 30s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/956c7c1ab229460f8252cf830d4ba578 : FAILURE in 15m 44s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/43e3c8508b434d5ea43e9235c07a40fa : FAILURE in 52m 12s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2342f90510494d978646289c5781694a : FAILURE in 1h 02m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/22f34d1ef51b4e87899db75f67e12083 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/68605c27c9bf482baa781025d6c0983c : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bc7ea66774ee4d849df6581de0a8ed1b : SUCCESS in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ee47e380f41a29c566f7ec9c3a2f82b1207c6d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 05:33:50.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"9e71eeaf72f4bcd2928fffe86b76e99f77de33ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 06:40:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/226147d5bf3b4b869ec2ad5bd1b56b57\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b4ee7b271b294484800e04334786fe58 : FAILURE in 35m 06s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/eb953eefd79a4341aa5962e6aaac85f7 : FAILURE in 16m 56s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f82fe8d06de147b28c5d0cecc2c753d0 : FAILURE in 55m 09s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/22ca7409173b461cb8ae45cfcb63d211 : FAILURE in 1h 04m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ceacc397e1a644aba0ab1f01cc0db65e : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95e902cc10514f40ae2dedb228641efc : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0b18d05693b450b9e937297feae2ba5 : SUCCESS in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"14014c2fc75779c668df6586c77c219185acf3b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 06:54:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38f7cd543bf049dba17e876d3d756b59\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/aaed9a0d3c354b05a407ad6aa3dfab39 : FAILURE in 17m 44s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/164660221f0545709ee7fb3be8481d39 : FAILURE in 5m 10s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/67e8cec5e82e4db088f261f3f688715d : FAILURE in 9m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/81c48a353ea74e9bba5805da502fa9cb : FAILURE in 12m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/976e61257a134546be01f50c6094a361 : FAILURE in 1h 10m 32s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0f9b3eda31d4bf194eacfccb6096860 : FAILURE in 24m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd0898eb462c41029bac37234c7a8e03 : FAILURE in 44m 05s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/c43e8bb144514f75a138916b4a05b689 : FAILURE in 13m 02s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d08d11a11bec405b98021ab07c4eb42f : FAILURE in 11m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/02e133be45fd487c8af42857091796f6 : FAILURE in 26m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/6b027e908c5442a58af854b6cecfc8a3 : FAILURE in 28m 13s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/613101f5b70446cba4bec8437cc1e059 : FAILURE in 16m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/686d61015ece4cd98f733366491d4668 : FAILURE in 24m 36s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/8b4e5a5ec744447e8b94f447d348d87e : FAILURE in 13m 17s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f32f29aaacaa4c8890d26dd6d749e10b : FAILURE in 26m 13s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/8d674e79b38647018dd633543d01286e : FAILURE in 16m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/f4fee26c36ff4de492274d6b0c7d1fb3 : FAILURE in 23m 50s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/ba0e38a7cd5a4708a4c7c1320db26e38 : FAILURE in 14m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/c7ec6623d0724a67acb1921f011f904b : FAILURE in 19m 11s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/e16cdcb5f4ad4c23abbd72a513a8bbfe : FAILURE in 16m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/7104aa854e0546ebb350a56a844df24e : FAILURE in 27m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8e61847427914729b6164f24d7dbc498 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8059d174a63b4529a3f94367334b1c34 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b6bb5ada36c4ee39b58803c51932437 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d169ac628184ceab7ff66907962d893 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d8540975dc14e7f846f50ccceb4c632 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbfc2a712a974a2da94c9cf4c524109f : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":6},{"id":"9f5e279005939b6f1f6504730a141f692c6a1d7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:26:46.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"bafdfb765b0a4c44cc79c0e0789f3fb5c1b0f10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:27:28.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"e13c9c252a23b23b77a7307044fa2e0de19ea724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:56:25.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c04389f07c2848b8a037af521361fb26\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/a9c14d58da1b40e6bd7b814f2832479f : FAILURE in 10m 37s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/6a913e9bcaf040668e3525ffb593797a : FAILURE in 13m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7c1607a2bd694a7f86bca38ba654a9ac : SUCCESS in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61b560f677924dc099d4622b34427301 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3c327328f4548068577fa53d342ab42 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e5f9cb8e08541568acd15ae7a4815ca : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07fc3d047c57439c83ec6168e78556f9 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c71e30c5a88c4c15a79d925289f0ac6a : NODE_FAILURE Node(set) request 36e5fe76e2f641d7a2e24d38d15cde10 failed in 0s","accounts_in_message":[],"_revision_number":8},{"id":"a9cb359606c21a8ed87ee311ccf97ef2efc24789","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 19:10:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/931411db5d4845bb94b32bf901d2f15b\n\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/9029e68f8d644aa0a2d0a96830b15df7 : FAILURE in 15m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c68bedd335ea47249d231c5c551432f5 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10fc848f8356400ebf153aeb7dd5393f : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c3c45970cb145d1b5b8908a5e288694 : SUCCESS in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"424cfc77e3519c42d214cb2c05c35a6cfd5d57e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-05 15:07:30.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"f9df7d70de37d4f15de19d2fcb190010df3cb55c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 15:07:53.000000000","message":"Patch Set 9:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  The label \"debian-trixie-arm64-8GB\" was not found.\n\nThe problem appears in the the \"kolla-debian-trixie-arm64-8GB\" nodeset stanza:\n\n  nodeset:\n      name: kolla-debian-trixie-arm64-8GB\n      nodes:\n        - name: primary\n          label: debian-trixie-arm64-8GB\n\n  in \"openstack/kolla/zuul.d/nodesets.yaml@master\", line 20","accounts_in_message":[],"_revision_number":9},{"id":"796a62b32da9c413cd1e37af1bd29f9cff772f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 15:08:09.000000000","message":"Patch Set 9: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  The label \"debian-trixie-arm64-8GB\" was not found.\n\nThe problem appears in the the \"kolla-debian-trixie-arm64-8GB\" nodeset stanza:\n\n  nodeset:\n      name: kolla-debian-trixie-arm64-8GB\n      nodes:\n        - name: primary\n          label: debian-trixie-arm64-8GB\n\n  in \"openstack/kolla/zuul.d/nodesets.yaml@master\", line 20","accounts_in_message":[],"_revision_number":9},{"id":"d0c1dca910c476af61d2dfea6db758bc44b02108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-05 16:04:20.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"d01292becc372d05342c39a21fa617a305dd84cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 16:15:26.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  The label \"debian-trixie-arm64-8GB\" was not found.\n\nThe problem appears in the the \"kolla-debian-trixie-arm64-8GB\" nodeset stanza:\n\n  nodeset:\n      name: kolla-debian-trixie-arm64-8GB\n      nodes:\n        - name: primary\n          label: debian-trixie-arm64-8GB\n\n  in \"openstack/kolla/zuul.d/nodesets.yaml@master\", line 20","accounts_in_message":[],"_revision_number":10},{"id":"b7234a424ec71025d24b977a87b64cd27f0f8be8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 16:15:37.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  The label \"debian-trixie-arm64-8GB\" was not found.\n\nThe problem appears in the the \"kolla-debian-trixie-arm64-8GB\" nodeset stanza:\n\n  nodeset:\n      name: kolla-debian-trixie-arm64-8GB\n      nodes:\n        - name: primary\n          label: debian-trixie-arm64-8GB\n\n  in \"openstack/kolla/zuul.d/nodesets.yaml@master\", line 20","accounts_in_message":[],"_revision_number":10},{"id":"a46dec07408e56d8b9a363fa5f3db9d2182142bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 06:21:57.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"4ea2f57c82c9344a12a92d356cb4e8f4c3a02b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 06:24:58.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"8f1154dddef0df32338d9966c0025cc3b706bcff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 07:09:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d32848ef143d4d5b8b32720218b13d86\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ffc16019f4354a05b91960e4c91f5c9c : SUCCESS in 1m 46s\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/790d7c2f372b4078acc14877e87ab4c0 : FAILURE in 34m 50s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/752f377e20cd440ca40af2e5119b37c6 : FAILURE in 24m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/00e71651fa954ea38983c59ef5db93fb : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1fcf5743666444f9350eaca3aaa2846 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94dd7a99619a448eb4970c4bd96cd19d : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37e16a2b71e5469dba85ab166afff525 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c7281575cb4457a9eefa66f5f11600f : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a90d43a33dd40939a580c96b060712d : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":12},{"id":"0f6590a79ac0d8dc4c77ca439d6f01e9e694cda3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 08:34:16.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499b80d47ad24cb2ace13176f40a4b51\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/49a140ac6e594f4fa2afdac60a7ae7b2 : NODE_FAILURE Node(set) request a590a7e2b06f4ec7afa10feccf4bdd3c failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3697c0fe53734b23b27f4590812a1158 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0dd687b6a32f4c69935f753b2428f3ad : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/902d06f3b1a046c4b1bb6ae4b7557a26 : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"71daa28c3462c5c767b12cf3d1e163e2483c81b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 08:48:51.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"80299c559ae2572c86c54bb3cf6455fdafafe934","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 09:26:45.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89eea67a9b964e80a03da25d39d23ad0\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/afd15683f25b4a58b1a7e575abe1f4f2 : NODE_FAILURE Node(set) request dd1b1d19c4e0466288c64b0de515853a failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bf73722046e749d192f326bfaa98c60e : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f0aee218b1874ef89317f64d564cb28b : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b26650bbe2f940388e62b0b1366fb210 : SUCCESS in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ba6355f6427deb72217c7d827276202cf32b709f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 09:38:34.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ad7592ae95449bb5cfd02a1faad57e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b0c595ce95d44c33a44bbdc541be680c : SUCCESS in 1m 45s\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/787b6ea1bf94410fa91fbb9acbd4a04c : SUCCESS in 42m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/32f1e6393f1c490590a7d5d970b1b4d2 : SUCCESS in 19m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7579d0bf00374738a961822bc5a70818 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18766503e2224284937762eecc634844 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6463a1e72c1432e8252f7cc1694fdb6 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a686f292cf0c404c96ef1e8d3da33eb4 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2152ce974da74e1c8f6be34738d5d706 : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997192c174fe4c0eb68f1a2a40ec975b : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":13},{"id":"0d8dcb8c2e52f6f0672b9125e0514c0048d87b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 10:58:09.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"82f0c05a092e07069c73462d12d08001617d10c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 11:33:05.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"10f558734bd3a2bf4c799c8380d4272f6a2bd398","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 12:19:44.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5af34bd3ae3e4335ac091575e6ba8c95\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d9ed47e332764a648bbb78d7646aeaec : NODE_FAILURE Node(set) request afa4e4da542e4bd38b841311dbbcf05b failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42887292656a4240876140df04d8f0b3 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0cae852cbe0e45d6a2adabf849502093 : FAILURE in 6m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e81276d82b0a42f2bfb63bb83b9e2871 : FAILURE in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f24aadae7dacdef727daf48ea1eade3f566ed844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 12:28:05.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/654b3bc438144a909b4f95e6bc98bd6b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1fd0d86cf6204590bc2ff428a39a9405 : SUCCESS in 2m 06s\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2fec0cf2c30c4613a2a0f3a0391314b0 : SUCCESS in 50m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e72a31a047fc48f4a261a753218fdffd : SUCCESS in 29m 02s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7ab2b2c739a24297b9e450fa86d4a0ea : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccaa5fc5b0634cd1b5a3182d9f3df137 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14b2699aad1c4d9c99eb3981d7222a83 : FAILURE in 1m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ec3b73ea229445a9b31d9ae7ab0fdd5 : FAILURE in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48f0994bddd64dcdb372e6eb904ecbec : FAILURE in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4094fc2ea347474f83c8d99d07feaafa : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7101456a131a4f9582f270cd2c446dd6 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":15},{"id":"08b6d07290d2421734c0ea06445fe1fc5945bba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-05 08:47:39.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"c78fad33702a38b3f7fc72e685970ad9d06de49c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 09:41:27.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71c1a26c11ec403ba5df442f28f6e8a6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba43c34b19da47b6bc0829d5f131ca73 : SUCCESS in 5m 18s\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a01d13df7b0b4872ad436be3eaa58dd4 : SUCCESS in 43m 55s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/746e53d340144c548b66363b70dd22b2 : SUCCESS in 27m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f42e7aaa1c8941c1991cd8f6b5f67f89 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29321780614f4a689600657c700c447b : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e66f369a65a74ae5b9dbe8578e61baee : FAILURE in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc374fc10c584c50b91ee5bb5035bcf3 : FAILURE in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e68db484d8ed4526b88a25251bef859e : FAILURE in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b722fa8c800f4456a46c2300da69eb4f : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05bdb1a94e1a453eb933d1fe24ed4ab0 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":16},{"id":"95160a71d0d08afddfd5fa2c0a5dee6cbc275f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 09:58:22.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40d2cbf208bb46b48cfdbf78931eba58\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1ec0c951bdad448f95bbe00088f6482d : NODE_FAILURE Node(set) request ab7834eff373485f8e4fee78a9d0d1f2 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4d56ca19c2c4d2db6039bbea807f3cc : FAILURE in 3m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2f23933da6f40b59e732d416d8d2a1d : FAILURE in 5m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/df37feb22d454951b1fece6c077727dc : FAILURE in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5f31d91abb291d99ef33a9f4cdfe85249f1baace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-05 13:59:24.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"4af00f7c72419b5ea59cbb031b614dcf261e3a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 14:53:44.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0994dd058274b9a9c8d6a5121e09add\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3df0cb7ca0734c1eae722e7a322e9826 : SUCCESS in 3m 11s\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5e5aa2f81a8f47acab763fb1082c45d9 : SUCCESS in 48m 55s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/25d97290a699461b93f849efb6f713ef : SUCCESS in 29m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8c9ea20f32e84fa0b0cbdf3d132e88b3 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a816c6e8e6054c2195b901b13cd1bfb1 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89009c4a558e4fb79c0f5857a9c8a2d0 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/460721aa9d6c4aaa812d5593d031d44d : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af54d5837782401a8e2ef0b7af77f74d : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec8760d19b64122a3c1ffe86c3b503c : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e681f175ce947d89868712452198a63 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":17},{"id":"1c8857bd8580e75fe50148bbc08546ffa870f038","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:03:44.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d82909fdcd9243a0a12ee472448327f0\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7f1f84e2793d4ae48cc26c6eda9ee012 : SUCCESS in 2h 00m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cdbe10b528a2491289cc9af486436f48 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b6f2db152e246d887287ac1353a452a : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af540808b2b042d1b9566fa7b7c4377c : SUCCESS in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"45c3b04a6c9e2a10ea5b824b701f937e97788333","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-08 16:46:31.000000000","message":"Topic set to kolla_trixie","accounts_in_message":[],"_revision_number":17},{"id":"e3fd6edb1e1f38b28126b43b427bb6f51df7a1d3","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-12-10 19:39:57.000000000","message":"Patch Set 17:\n\n(5 comments)","accounts_in_message":[],"_revision_number":17},{"id":"3e0037c8a10e540938659b886d1b1267c975d759","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-11 20:29:35.000000000","message":"Patch Set 17:\n\n(10 comments)","accounts_in_message":[],"_revision_number":17},{"id":"529e3f7bbad799e3054c692199df9003f8cd016c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-11 20:29:50.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"01207e432ddfb91e11f94b03901b23f92986a5bd","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:17:34.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf861da49a254dc59222222e1dc5535d\n\n- kolla-build-debian-trixie-no-infra-wheels https://zuul.opendev.org/t/openstack/build/021bbf31e8f14e2ebea0a90f4934cd67 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":17},{"id":"a64448bd757ccd5c728bd4cbe03f6cfa3838af6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 14:41:03.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"869cb7c22abaaa134b03f58b326c2adce7ef06a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 14:41:21.000000000","message":"Patch Set 18:\n\n(7 comments)","accounts_in_message":[],"_revision_number":18},{"id":"9a76e749f76d28353205114452228028305207b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:41:22.000000000","message":"Patch Set 18:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,18","accounts_in_message":[],"_revision_number":18},{"id":"d391be4cb6f96e305bb41741df69fdf8de533bda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:41:24.000000000","message":"Patch Set 18: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,18","accounts_in_message":[],"_revision_number":18},{"id":"c826961989c6e2c825a2af2d409340f5bd40f3ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 14:43:21.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"d68c8c3df87425cd28fa55d0b86cbd95abd0d000","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:43:26.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,19","accounts_in_message":[],"_revision_number":19},{"id":"03c994e765f8ebc46d01d5ee7243a9a32dc1aedf","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 14:43:27.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":19},{"id":"c6405eba8aaa430fe6f175cd71620ad5e0dc4cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:43:28.000000000","message":"Patch Set 19: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,19","accounts_in_message":[],"_revision_number":19},{"id":"aca2cfa83d1484eee5d674b6204fbb315834b975","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 14:43:34.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"d0b1593652a375f2b21c193888237d0e203eddfa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 14:46:28.000000000","message":"Patch Set 20: Published edit on patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"34b799595b3e9a5c087a859178db342a52b44192","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:46:34.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,20","accounts_in_message":[],"_revision_number":20},{"id":"3ab1b39da331c84fddbf71ee6d778898f69a60fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:46:37.000000000","message":"Patch Set 20: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,20","accounts_in_message":[],"_revision_number":20},{"id":"ea8cacfde3d5e699cd7da7224e168863a867d34b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 14:53:19.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"776f407014121ac07a954f04968d4d80803a1e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:53:24.000000000","message":"Patch Set 21: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,21","accounts_in_message":[],"_revision_number":21},{"id":"23d32c34091f7aa4c4c7ee9150626898c7bc7922","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:53:26.000000000","message":"Patch Set 21:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,21","accounts_in_message":[],"_revision_number":21},{"id":"93775537bccad9288f24f03363eea6c640717f28","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 14:53:42.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"82f0b1afba9ba4c5d635c781d397df32ef4ab57e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:53:48.000000000","message":"Patch Set 22: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,22","accounts_in_message":[],"_revision_number":22},{"id":"23351b81919994072b17d44e5c6a71ea52410f54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:53:49.000000000","message":"Patch Set 22:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 964793,22","accounts_in_message":[],"_revision_number":22},{"id":"92695e5c0e8c3967f4eb08cd2499c77b3d2c2638","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:00:34.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nThe following files contain Git conflicts:\n* kolla/common/config.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"5198097d482dfba59a5a4bb885178584ba0ab8a7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:01:58.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"0146752cfcde5ee70740bb2fc5bb6670239817aa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:02:14.000000000","message":"Patch Set 24:\n\n(4 comments)","accounts_in_message":[],"_revision_number":24},{"id":"6412d6ffefd298e123677c36f7a1e71b875b2ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 15:12:53.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ade9c56cad14d2ea803737253742b60\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3f4fb26981e5499d8afa0b591c889a65 : FAILURE in 4m 37s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e979cd17373a48439abf89e2fe650e45 : FAILURE in 6m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b4e9c6131f264dc79fad770ae3ee5f9b : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/175f380c3dbe45f4b3f484026679f668 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0338dd466274b87b04ebcc011c6b4b7 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79506f9011654767b6ada30735f99f3e : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac05aa9ebbd04d1a98b74244658de32a : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65b3f6a3a3a4269b39f4e4362f2ebba : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3223bd00b414cc88dd728069926c628 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":24},{"id":"e5cfc37567c556b0dc1628329e9322a3841aa289","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:15:01.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/sources.list.debian\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"fbe0c3eb55874db7b13a59d1fab49b5e485c022b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:15:19.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"c584f7551302c4e703c4ac6d85f1819431cb32db","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:23:14.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"c236c17be0a822f162612ee97a18d6d2ab11bc9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 15:31:52.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09466454eedc4b27a22e7ace7541642c\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/53b2936d979d4517bc07a80ccb6ee5c7 : FAILURE in 6m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/751a3224316d4ed689f53958db310b0c : FAILURE in 4m 27s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ce1746fd4cac49ab90e8de73c5ec37ae : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55a669f97f940b6a0530c7f7962ba6b : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb9a2da5fd0b4749a7e768a620b114f0 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b299f60dd78e4de596cbf4d66adef179 : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0927a87e221d4ef0b8f51c4337740454 : SUCCESS in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c829bc8dd50f47bfb68c42568ae3a8ce : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9be6da3493d042a28f31288c20574ced : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":27},{"id":"24ce008b0b136be1e87b99368d3c4c583eb69b07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:44:05.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6818660bff4643e48e367a95c52efd70\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/797ddbac663d472a97060c7a6a74a748 : FAILURE in 6m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f90303e3f614b15bda9f9bd1cb2332c : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0eb58737b31547e2aa1053aff3974356 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f5af24278034d96897379860d75c39c : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"ca929054489a8a51939826e45cfa517bcb6195d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:24:18.000000000","message":"Patch Set 28: Patch Set 27 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"309dc590b1e2393269fb488f9daa4a9626c6b24d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:25:07.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"aec82e0b7b5108a08c2d654aaefeb08609194056","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:25:55.000000000","message":"Patch Set 30: Published edit on patch set 29.","accounts_in_message":[],"_revision_number":30},{"id":"85ac8be938aad1f8a873bd9440e373096fc70c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 18:40:48.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dce25e6783dc4290977ca49ccc0aa92d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c6893370c59749feb71a321dc9d1c342 : SUCCESS in 39m 54s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/253f82cc960c4c18831c722b422eb243 : SUCCESS in 54m 34s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2c83bcda7dcd452cbcd240c060376cf1 : SUCCESS in 23m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/fedcd437794d414aa9b3543475e27a3d : SUCCESS in 41m 03s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/819345d8ecb94165924c5f717e2a8a5c : SUCCESS in 39m 23s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8d0c5b8939e42d1aafa1e6ee57718b9 : SUCCESS in 33m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2038903fa2944f529318a9d13dec5040 : SUCCESS in 27m 12s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/59e157d0e49f4d99ae222d5d706eae05 : SUCCESS in 1h 46m 07s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d5f76bf6d6ac4cedab7df49a98378cf8 : FAILURE in 57m 54s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/6f98cf3ebe59496eb2ea2776f6f5c37d : SUCCESS in 1h 39m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1cf73ebf81f74449a1d3e78ad98d84bc : SUCCESS in 2h 11m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f86105805f6f45e3b48e8a6a24a11c91 : SUCCESS in 1h 45m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/86c9e6de415e4aa2bfbc49d95e77e28a : SUCCESS in 2h 10m 02s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/7982c6b729c44ab89817286389b1d26b : SUCCESS in 1h 02m 23s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/3c11073479734864bebb1833094f0f61 : SUCCESS in 1h 22m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/bc95454da63a492cb5b466287e22bffa : SUCCESS in 1h 12m 13s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/aefa2f17d90f4b64ac59de087d386713 : SUCCESS in 24m 38s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/2261ec3e7ddc43cba6af182c7835dbc7 : SUCCESS in 28m 59s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a3d6110ffe364cf9ab2a0d1b788861a4 : SUCCESS in 18m 03s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/828af28f706746539098417ab5cf9906 : FAILURE in 47m 54s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/fbe113aafe08493e857b387816d59d44 : SUCCESS in 31m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/dfb2de95797e4085b10dd0024a83b165 : SUCCESS in 1h 10m 56s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/7db584c5a9c742f68454711ad78dfef0 : SUCCESS in 49m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/6eb7e1215e6b4b2792da782e0fc4d5b2 : SUCCESS in 54m 08s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/406edc9e5f684f059b781b37596aee1f : FAILURE in 46m 19s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/98a900b9a7724c3580b0b89c3851f794 : SUCCESS in 1h 21m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/68386373a4c74a5e8e09f30f2fe6fc93 : FAILURE in 1h 19m 02s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b46178ce9e034f2d90e51d6d29fe198a : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c98d8779a34f18a80f2575572e0f8b : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13f8c99ed113439e85830320d5e8f641 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fb1c502233546f78f3ff85a2bd71d8a : SUCCESS in 4m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d89c311eab944e7985c3171dfb34a732 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8456d7a5f4fa496789f80e0c05065799 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6526dcf599804bb0adeb356d2312b143 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":30},{"id":"c1d605f5ecd802b4271f3b17c9e170003950ef0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 19:53:48.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5769b1d13f534adf9c4e0812c0214fb1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e93a0e865e014c3e8847e0eecfdb456e : SUCCESS in 51m 14s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3851d31685314be3a8f4bedd30966248 : SUCCESS in 2h 13m 53s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d1145385fac84ebcb032a954da114e2e : SUCCESS in 2h 12m 19s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/81cd6b1e687e40f189a5f70bdac869a3 : FAILURE in 1h 16m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/53c7e03ac942436d88cd53a66c322851 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b1c4fcb16164e5e9d5f045960619517 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4104bea6be26493a87b187fb542a6d6a : SUCCESS in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"f5f3a3d29e7d829c0ac0cdc05b6790afd07ec50b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 20:51:07.000000000","message":"Patch Set 31: Patch Set 30 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"cc2ac7fa85d70ecf843057bdf35e3f4df33bea6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 22:28:19.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d66a0ac5aab944598a1713ef6578c0b0\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7fc28032435b4ccb999e20c58c657c75 : SUCCESS in 1h 24m 14s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/4459b1cbc7dd4e418ad7209638b91611 : SUCCESS in 1h 01m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/555fb50986624230aca53b781c05b168 : SUCCESS in 1h 25m 05s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/38021e6b23aa43e8becd4abd37f4738d : SUCCESS in 59m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a49a54aed91245f18c299ccb429ca086 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/af7fe3312bc849519770f6dd61898927 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6fdf63481ef4bc3ab655ba43ea5cf85 : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"27cc17f369a476de7777e9900d0ee5a992d98045","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-12-15 23:01:45.000000000","message":"Patch Set 31:\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"5224cac1b335d5184a88310f391e4074d2a32912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 23:22:28.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5429a04e508a43068d24160202da8b75\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a2db08bf953c46f080599a3074607e2e : SUCCESS in 42m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/16c83018881049feb8c9420b828a4772 : SUCCESS in 49m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e6574fc62eac4a10848cef3ac330c85f : SUCCESS in 25m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c2df31e240604cdbb94d5b41eead5cf5 : SUCCESS in 38m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a8bf66f61af14a13b946e996ba5d69c8 : SUCCESS in 36m 59s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/444c74a8d3d740b6a5756b1be847f182 : SUCCESS in 40m 12s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/547f2021e92c47098d6e2c80adaffe5a : SUCCESS in 13m 59s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/ff8d21c5ee0c4e778cae1e99357b010e : FAILURE in 1h 39m 20s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/54d766701a894bd79f637bd26142d531 : SUCCESS in 2h 08m 12s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/51077dad9ec0495daaa6d9bceceee5cd : SUCCESS in 1h 41m 17s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/46d8df58576745d8bf6977a7ccefde9c : SUCCESS in 2h 28m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/ab65fef18d334a5bb5a9af35baf06c8a : SUCCESS in 1h 52m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/766522ed31944402920eea17be66badd : FAILURE in 30m 18s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/23d6f0506f004c9f944f17e464ae5981 : SUCCESS in 1h 27m 14s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/a5e3037be7ae40f0a820852f4b13cfad : SUCCESS in 1h 32m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/ac0ed41e82f148459b27f7f82c7ecee1 : SUCCESS in 1h 16m 32s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/fe4f832c98e74e6ea70123c9dcabeb2a : SUCCESS in 27m 28s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/09054ba4dffe4405b10b842da58c09a3 : SUCCESS in 30m 18s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/8afabcc1eb3f40c686820b8cefbad960 : SUCCESS in 16m 37s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/57c4ae71dc184de49d16705a6d1fad85 : FAILURE in 44m 04s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/a30ac34cb9d44845825d1ffbea59b541 : SUCCESS in 1h 01m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/b840543c83984f57af4f676ed95b1ba8 : SUCCESS in 30m 57s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/cf2840d406044dfaa0d7674b5db1b570 : SUCCESS in 46m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/9e18a0e8b33d444a9231f1104e1adc5e : SUCCESS in 1h 04m 06s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/6f9865935c23411e800561843f1d2420 : SUCCESS in 1h 21m 21s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/6acf0cc4ea304b389ade8f4d90178803 : SUCCESS in 1h 27m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2cdedc03b4674baab79a94289bee625b : SUCCESS in 1h 17m 55s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cbc8c084e4a346af8694a8f4055d76e2 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae84bc91c32d44c4a50246d4b53f56e1 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08beb9448a3f43d8a2e8ffb4be3bd228 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/866e1302b2e64c3080a5aa8537f72136 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f62f32a2e864476a9ea05fbd236c3f57 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de11421046c24a5194cc80d903d82649 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da85b86cdb084d209c08e774e8216a28 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":31},{"id":"63cd1cf644d7695e43adbe2bd190f74be393731b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-16 05:45:07.000000000","message":"Patch Set 31:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":31},{"id":"497a7aacc3b631d7050e9ffadbb0abd4c3ba9d32","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-16 10:03:08.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"ecc5ff1ea04bd25d0f71265ab864fd2ddb3bbdba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-16 10:03:12.000000000","message":"Patch Set 32: Patch Set 31 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"0ab3c58cd74b0c4a9ea944a175753f5214f7b7a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:05:39.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ce0261be4848c6a92ad3ceb779c779\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/fd11086d08aa433fb692ad8d545a37f5 : SUCCESS in 38m 28s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/439854d5715749528e2a0831b421989e : SUCCESS in 58m 11s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4f5905f867d04f5090a6be41abdbe662 : SUCCESS in 26m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/154adcfd4a964c75b77751d5c9b9ee3d : SUCCESS in 36m 58s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ac5b36ea4a24424b8e81cd863e21063d : SUCCESS in 34m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/847f56b300664d67bb76df7779fb05cd : SUCCESS in 33m 08s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb7ff3a0a4f64e9c8c71d509f03be85f : SUCCESS in 23m 44s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/9d3b88cabbd54433beb64154463209b2 : SUCCESS in 1h 19m 25s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2d2b57fa001a4818aae7bbaf3e3c38b0 : SUCCESS in 1h 52m 25s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/d079b0af4b5848889ba1769bfc43d828 : SUCCESS in 1h 25m 01s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/20676f253336404b8ea566fa3e292f2d : SUCCESS in 1h 59m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/ab6a17f0d4a04526a7ff0dab50d126cb : SUCCESS in 1h 29m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c239a892e1894f088f43a18a73cb2653 : SUCCESS in 1h 58m 23s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/43fdd2c4c6484b91a645d27cb13baceb : SUCCESS in 1h 02m 57s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/22ae17ed14bc4df0aed52faf2e235a9c : SUCCESS in 1h 08m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f221a8bf82c444db911efea9ce0f3402 : SUCCESS in 1h 10m 22s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/cc844bd60c7d4703bbe573158b9a75f0 : SUCCESS in 22m 25s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/dda635f2ea004e769ea9dca4f314e0aa : SUCCESS in 26m 20s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/7892fd1f79c2448993e142797c26b39b : SUCCESS in 28m 54s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/057c9418a87b4e748d9457d55d5cacc5 : FAILURE in 48m 50s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/84f71d1eb4a749c187480f51252f2f09 : SUCCESS in 49m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/d81c4968aef9448b94f566edb8291876 : SUCCESS in 1h 03m 26s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/c45d453b59e0496fbe300ff08ab58279 : SUCCESS in 48m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/2c1a883ae2a743768bc09242dffbfc11 : SUCCESS in 56m 35s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/5357a68b4d7547f3bb634329357a3c87 : SUCCESS in 1h 11m 40s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/a6c2409ca35d4e989e4009d1555a88b1 : SUCCESS in 1h 22m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/abde0d13d0ff41ce972e63be5924d447 : SUCCESS in 1h 19m 17s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b96139219d9d4ece95c20e9abc283958 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bba8ac4617a41bb847b04ff9ec7898d : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98518473329e44ddb84a752ba56b22b9 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d35831cb5fc44324ba4a3f4e95794b37 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0798838ec87c4088822877569040d76d : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8631870e20f34ae6b0aeb990821decc2 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3fbca479cf2469480bdb36bbf48b32b : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":32},{"id":"59fce9da551c7242fd277b4cba7aef97fcb426c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-16 12:15:23.000000000","message":"Patch Set 33: Published edit on patch set 32.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"c44a9e435233928d530fae90b769ea2f28408e41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 14:09:47.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2bbb266dc0e4138affc0d9c5cfbe8bb\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/221d6376b5bc48ad8bbacc052e3eb0f1 : SUCCESS in 1h 10m 58s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e10ebfd991ef47bcb0fda1e04653b5a9 : SUCCESS in 1h 44m 46s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/eecdbd698c9c44c1a978428e408ec7fd : SUCCESS in 1h 51m 19s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ad4cc7f318a3400ebb9983b7a79ae69c : SUCCESS in 56m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/339ae495242b43cdbefb3db175d35499 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cdd3a6404f394f049498e78dd0665173 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/56cec427342c4727b0960ff167b3032b : SUCCESS in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"8facd78f2842752610bd85ce259ed8937c2a8470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 14:20:34.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972ec4d7ea114fa9bbe9c418dcd3ffb4\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a9c66395bd634a25adf22c958c302a4a : SUCCESS in 46m 00s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/60e399a81ebf48779f397e0e3345c94f : SUCCESS in 51m 42s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d16d76a2058a433d821fa21e433a41a7 : SUCCESS in 24m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1a0959b6818046489c622b71c1a952d7 : SUCCESS in 40m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a8483dde593849cea44044b198f5211c : SUCCESS in 47m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39735cee957d4c17bdd8117c621feb0b : SUCCESS in 42m 54s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ab5e80daa494e0a8958385397ad6b80 : SUCCESS in 14m 04s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/66da544460d842d2b766d3bafbc58d16 : SUCCESS in 1h 16m 43s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f1f14316fbca447eaf9973d3a8f2f671 : FAILURE in 1h 23m 49s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/d448e7a8e26c455499c5d9345631d50a : FAILURE in 1h 17m 04s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d5087f43015f4ba6836b7d73d7f2b2e7 : FAILURE in 2h 02m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/168918e4c72a43c0bc48938c4a59aa63 : FAILURE in 1h 34m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/30b61a99b2c94f90bbb534509f9518ca : SUCCESS in 1h 53m 56s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/33832d20a5bf4ea19b0ddb0b836c6d8d : SUCCESS in 1h 03m 17s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/9e454957997445ba884e8a0d343ec074 : SUCCESS in 1h 08m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/187af592aca5496cb9e5e25bfa3deaf4 : SUCCESS in 1h 09m 10s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/55d66af3439d446390193e7f12b49f90 : SUCCESS in 25m 47s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/7154718db8364ff1a5f3c936e11ff9a7 : SUCCESS in 29m 09s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f78f43a83555464d8695b4b7f3fca9b8 : SUCCESS in 16m 25s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/e4e8f122436f40ad8d2fd8b09797be21 : FAILURE in 38m 47s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/5bf3a81d25a54e279e924ae4b79be6a7 : SUCCESS in 32m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/563e2d71ae634f21b4724dfc2dd16104 : SUCCESS in 52m 04s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/5b6527e15aee493babdc34309919e4f4 : SUCCESS in 1h 01m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/ed028cec3c5a4052b4a541f4e623200b : SUCCESS in 1h 04m 44s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/9b1e33c487994ddc91ed1ddd6b9d62df : SUCCESS in 1h 14m 51s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/90a50d8d14244b63b31f4ea2a2ada7d9 : SUCCESS in 1h 32m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/0d1abfc7e4114d68974771e98d3db1b8 : SUCCESS in 1h 29m 49s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d1db856d29284cd896b2a501c2644ab3 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dca2009729a4822922861e8df7e2499 : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17e6cfdaa03c4906a65f98832df8e7f1 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6fdf4593f94543e6ab6d31f2dd1d5a02 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/531f92b0e98c4dd688c2d5da9eacb75a : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad18fd623e74311a9ead7132b598281 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4bbd1118bd4faea7338b0dc5f4b8c3 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":33},{"id":"134417aebd408c62e364c85c8233c94c742703a7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-17 14:28:27.000000000","message":"Patch Set 33: Backport-Candidate-2 Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":33},{"id":"3aabe0430e2aaf51b9fe20f713d6e209b4a94988","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-19 16:10:25.000000000","message":"Patch Set 33: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":33},{"id":"848befa08f622a5c519d59aa8ba00b0c103bf320","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-19 20:16:18.000000000","message":"Patch Set 33: Code-Review+1","accounts_in_message":[],"_revision_number":33},{"id":"8873a2c3b6ffafde2cbf13a7fb205b965426ce2e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-20 06:13:12.000000000","message":"Patch Set 33: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":33},{"id":"146485119313b8dbf4cc36823ea3d393d632a70f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 06:13:50.000000000","message":"Patch Set 33: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":33},{"id":"7deae8a8af0e7588280f2a3242f49abc376bc54e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 07:24:34.000000000","message":"Patch Set 33: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad39ac9202714dd19217aaba286d555e\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0dcf3634424f4b0989d9c6e98292bcec : SUCCESS in 33m 15s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7e0a8a34f9a34b3687816dc20ec69fcc : SUCCESS in 25m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e9225a1e7d914f67884db6eeef385694 : SUCCESS in 40m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ec122157c5f44eb3add79dc59dae2338 : SUCCESS in 1h 05m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb7d8cdd57764f55bbb2a617f9a95d12 : SUCCESS in 35m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/443edd902fc446d3bdb84f08e91d4c61 : SUCCESS in 32m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1a7c8178afd5446f81ddc73fe117c783 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50892812084a4b78bf2dd8f812db03ce : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/124c91c4d9ce46d389e4365559d84141 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e463f84a7dc49f6b592733e612f6c8e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245da4f1f44341a3a63faa81a2c6ba88 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16df762a8f4f48c39245eb243aa9b8a8 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":33},{"id":"ceff31ef8df241810cd80a39d02df690b8bc166d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 07:24:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":33},{"id":"7c7dd398088a2d391d23f51122629b0a4b76b8e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 07:26:08.000000000","message":"Patch Set 33:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ba92979188249f8be28bee56d7ae3a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf2bc287340d4f4caa96fdb90da3b2e6 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5580bf54f7ab4804b9cfc77d47a6e234 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"cee27f6426f49b4242b0535d0e0dee6f8a44dbda","revisions":{"3ff27f824bf684266c21b2c1cff756c28093a834":{"kind":"REWORK","_number":1,"created":"2025-10-24 14:59:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/1"}}},"commit":{"parents":[{"commit":"2fdd26aa3823c12198f32405fbc3992d6be1333d","subject":"Merge \"neutron-metadata-agent: copy wrapper script for haproxy processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fdd26aa3823c12198f32405fbc3992d6be1333d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:59:26.000000000","tz":120},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ff27f824bf684266c21b2c1cff756c28093a834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ff27f824bf684266c21b2c1cff756c28093a834"}]},"branch":"refs/heads/master"},"1c97e9faa99f467f0e359d1794af2a90dc48c35d":{"kind":"REWORK","_number":2,"created":"2025-10-24 16:24:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/2"}}},"commit":{"parents":[{"commit":"2fdd26aa3823c12198f32405fbc3992d6be1333d","subject":"Merge \"neutron-metadata-agent: copy wrapper script for haproxy processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fdd26aa3823c12198f32405fbc3992d6be1333d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 16:24:16.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c97e9faa99f467f0e359d1794af2a90dc48c35d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c97e9faa99f467f0e359d1794af2a90dc48c35d"}]},"branch":"refs/heads/master"},"834e80aece39175ddb0b8c10e4e07c52517c31bc":{"kind":"REWORK","_number":3,"created":"2025-10-24 18:51:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/3"}}},"commit":{"parents":[{"commit":"2fdd26aa3823c12198f32405fbc3992d6be1333d","subject":"Merge \"neutron-metadata-agent: copy wrapper script for haproxy processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fdd26aa3823c12198f32405fbc3992d6be1333d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 18:51:32.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/834e80aece39175ddb0b8c10e4e07c52517c31bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/834e80aece39175ddb0b8c10e4e07c52517c31bc"}]},"branch":"refs/heads/master"},"aca6129d10c1bdd792edb222146cecd3069c849b":{"kind":"REWORK","_number":4,"created":"2025-10-25 04:48:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/4"}}},"commit":{"parents":[{"commit":"2fdd26aa3823c12198f32405fbc3992d6be1333d","subject":"Merge \"neutron-metadata-agent: copy wrapper script for haproxy processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fdd26aa3823c12198f32405fbc3992d6be1333d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-25 04:48:25.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aca6129d10c1bdd792edb222146cecd3069c849b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aca6129d10c1bdd792edb222146cecd3069c849b"}]},"branch":"refs/heads/master"},"b9344d5fa9257a839ea7eaa87b8fc422008bd7fe":{"kind":"REWORK","_number":5,"created":"2025-10-26 20:09:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/5"}}},"commit":{"parents":[{"commit":"2fdd26aa3823c12198f32405fbc3992d6be1333d","subject":"Merge \"neutron-metadata-agent: copy wrapper script for haproxy processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fdd26aa3823c12198f32405fbc3992d6be1333d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-26 20:09:22.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9344d5fa9257a839ea7eaa87b8fc422008bd7fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9344d5fa9257a839ea7eaa87b8fc422008bd7fe"}]},"branch":"refs/heads/master"},"40ca8ea7f17558157d06c99b4b0a263ef24b6ea7":{"kind":"REWORK","_number":6,"created":"2025-10-27 05:33:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/6"}}},"commit":{"parents":[{"commit":"2fdd26aa3823c12198f32405fbc3992d6be1333d","subject":"Merge \"neutron-metadata-agent: copy wrapper script for haproxy processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fdd26aa3823c12198f32405fbc3992d6be1333d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 05:33:50.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40ca8ea7f17558157d06c99b4b0a263ef24b6ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40ca8ea7f17558157d06c99b4b0a263ef24b6ea7"}]},"branch":"refs/heads/master"},"a46d2b26e03ee40571c884a68d5080e84d65996e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-27 17:26:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/7"}}},"commit":{"parents":[{"commit":"9bc8d4d0816ac89614a703170601046e57f54c57","subject":"horizon: pin pip to \u003c25.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8d4d0816ac89614a703170601046e57f54c57"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 17:26:46.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a46d2b26e03ee40571c884a68d5080e84d65996e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a46d2b26e03ee40571c884a68d5080e84d65996e"}]},"branch":"refs/heads/master","description":"Rebase"},"649987caf7a8adb702fd44bcd04ae00a88b50e6d":{"kind":"REWORK","_number":8,"created":"2025-10-27 17:27:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/8"}}},"commit":{"parents":[{"commit":"9bc8d4d0816ac89614a703170601046e57f54c57","subject":"horizon: pin pip to \u003c25.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8d4d0816ac89614a703170601046e57f54c57"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 17:27:28.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/649987caf7a8adb702fd44bcd04ae00a88b50e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/649987caf7a8adb702fd44bcd04ae00a88b50e6d"}]},"branch":"refs/heads/master"},"11ee4d079b8d5a24fb5689feff2190d0e6f9cad1":{"kind":"REWORK","_number":9,"created":"2025-11-05 15:07:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/9"}}},"commit":{"parents":[{"commit":"9bc8d4d0816ac89614a703170601046e57f54c57","subject":"horizon: pin pip to \u003c25.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8d4d0816ac89614a703170601046e57f54c57"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-05 15:07:30.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11ee4d079b8d5a24fb5689feff2190d0e6f9cad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11ee4d079b8d5a24fb5689feff2190d0e6f9cad1"}]},"branch":"refs/heads/master"},"65b2c7c3755591939719f7432753a42783db575e":{"kind":"REWORK","_number":10,"created":"2025-11-05 16:04:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/10"}}},"commit":{"parents":[{"commit":"9bc8d4d0816ac89614a703170601046e57f54c57","subject":"horizon: pin pip to \u003c25.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8d4d0816ac89614a703170601046e57f54c57"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-05 15:57:05.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65b2c7c3755591939719f7432753a42783db575e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65b2c7c3755591939719f7432753a42783db575e"}]},"branch":"refs/heads/master"},"f32c16b0eb58980aec555faa0e29dd65d3b2047f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-12-03 06:21:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/11"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 06:21:57.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f32c16b0eb58980aec555faa0e29dd65d3b2047f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f32c16b0eb58980aec555faa0e29dd65d3b2047f"}]},"branch":"refs/heads/master","description":"Rebase"},"abac8f158d51295bcbf1b5a5f95e4574939dcad4":{"kind":"REWORK","_number":12,"created":"2025-12-03 06:24:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/12"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 06:24:58.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abac8f158d51295bcbf1b5a5f95e4574939dcad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abac8f158d51295bcbf1b5a5f95e4574939dcad4"}]},"branch":"refs/heads/master"},"c09075d5785122e65e3df2e1237608eef80f175e":{"kind":"REWORK","_number":13,"created":"2025-12-03 08:48:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/13"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 08:48:51.000000000","tz":0},"subject":"WIP: Switch to Debian Trixie","message":"WIP: Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c09075d5785122e65e3df2e1237608eef80f175e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c09075d5785122e65e3df2e1237608eef80f175e"}]},"branch":"refs/heads/master"},"93ba58bc994525bed41de89ce281ee989b23b73e":{"kind":"REWORK","_number":14,"created":"2025-12-03 10:58:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/14"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 10:57:57.000000000","tz":60},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93ba58bc994525bed41de89ce281ee989b23b73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93ba58bc994525bed41de89ce281ee989b23b73e"}]},"branch":"refs/heads/master"},"c70243cdf1bfc0b0e160fe541d0f6fc42761b74d":{"kind":"REWORK","_number":15,"created":"2025-12-03 11:33:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/15"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 11:33:05.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c70243cdf1bfc0b0e160fe541d0f6fc42761b74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c70243cdf1bfc0b0e160fe541d0f6fc42761b74d"}]},"branch":"refs/heads/master"},"f3d4d378b6e4e4db7810105aa9f0b179bbcad6aa":{"kind":"REWORK","_number":16,"created":"2025-12-05 08:47:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/16"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-05 08:47:38.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3d4d378b6e4e4db7810105aa9f0b179bbcad6aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3d4d378b6e4e4db7810105aa9f0b179bbcad6aa"}]},"branch":"refs/heads/master"},"456e2333c5ceb959ac33412272230f5a4032e1e3":{"kind":"REWORK","_number":17,"created":"2025-12-05 13:59:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/17"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-05 13:59:24.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/456e2333c5ceb959ac33412272230f5a4032e1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/456e2333c5ceb959ac33412272230f5a4032e1e3"}]},"branch":"refs/heads/master"},"4f2f5506c41ea2d3b3947a2b909a06b949adcd83":{"kind":"REWORK","_number":18,"created":"2025-12-12 14:41:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/18"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 14:35:43.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f2f5506c41ea2d3b3947a2b909a06b949adcd83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f2f5506c41ea2d3b3947a2b909a06b949adcd83"}]},"branch":"refs/heads/master"},"e67b691ae707a5c9cd37c4bdad396a34efcff71f":{"kind":"REWORK","_number":19,"created":"2025-12-12 14:43:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/19"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 14:43:11.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e67b691ae707a5c9cd37c4bdad396a34efcff71f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e67b691ae707a5c9cd37c4bdad396a34efcff71f"}]},"branch":"refs/heads/master"},"ac01930ad7f5a27c7d4c6b9da8c49c989882d3d9":{"kind":"REWORK","_number":20,"created":"2025-12-12 14:46:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/20"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 14:46:27.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac01930ad7f5a27c7d4c6b9da8c49c989882d3d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac01930ad7f5a27c7d4c6b9da8c49c989882d3d9"}]},"branch":"refs/heads/master"},"9945cd88bc46d3b88a14a9ea15d2c0207c8321ae":{"kind":"REWORK","_number":21,"created":"2025-12-12 14:53:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/21"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 14:52:59.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9945cd88bc46d3b88a14a9ea15d2c0207c8321ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9945cd88bc46d3b88a14a9ea15d2c0207c8321ae"}]},"branch":"refs/heads/master"},"272cfcb57ad5eff94374b3d6c83eca64b66a56d6":{"kind":"REWORK","_number":22,"created":"2025-12-12 14:53:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/22"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 14:53:35.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/272cfcb57ad5eff94374b3d6c83eca64b66a56d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/272cfcb57ad5eff94374b3d6c83eca64b66a56d6"}]},"branch":"refs/heads/master"},"b128ed97b9bfab3029499bb02b9867d1ef282d04":{"kind":"REWORK","_number":23,"created":"2025-12-12 15:00:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/23"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:00:34.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b128ed97b9bfab3029499bb02b9867d1ef282d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b128ed97b9bfab3029499bb02b9867d1ef282d04"}]},"branch":"refs/heads/master","description":"Rebase"},"b581e654a0781c5387a55a9ddcf84717a3ef9769":{"kind":"REWORK","_number":24,"created":"2025-12-12 15:01:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/24"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:01:58.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b581e654a0781c5387a55a9ddcf84717a3ef9769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b581e654a0781c5387a55a9ddcf84717a3ef9769"}]},"branch":"refs/heads/master"},"84b8f33bbbef2314c0ee4dd721785a35aaccd65e":{"kind":"REWORK","_number":25,"created":"2025-12-12 15:15:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/25"}}},"commit":{"parents":[{"commit":"7787350e3e400d683f3e4c78d5cc202a7cc61fff","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7787350e3e400d683f3e4c78d5cc202a7cc61fff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:15:01.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84b8f33bbbef2314c0ee4dd721785a35aaccd65e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84b8f33bbbef2314c0ee4dd721785a35aaccd65e"}]},"branch":"refs/heads/master","description":"Rebase"},"f71cd89eec2bfccfd50d09c202248b67de34e1da":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-12-12 15:15:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/26"}}},"commit":{"parents":[{"commit":"fa281a2c16ce4aaca817eab9cad354a60c2b059c","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa281a2c16ce4aaca817eab9cad354a60c2b059c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:15:19.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f71cd89eec2bfccfd50d09c202248b67de34e1da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f71cd89eec2bfccfd50d09c202248b67de34e1da"}]},"branch":"refs/heads/master","description":"Rebase"},"77b8298f97d82fb4e175005784703ab47cb9ae68":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-12-12 15:23:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/27"}}},"commit":{"parents":[{"commit":"0f65bd828bffb113b35a0feb8d30abd8cae69746","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f65bd828bffb113b35a0feb8d30abd8cae69746"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:23:14.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77b8298f97d82fb4e175005784703ab47cb9ae68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77b8298f97d82fb4e175005784703ab47cb9ae68"}]},"branch":"refs/heads/master","description":"Rebase"},"dd53616d79fe84626380a4bd837b1fb8315a0375":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-12-15 16:24:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/28"}}},"commit":{"parents":[{"commit":"09a907808417a5d77c35b9c7616fc0ce8875baf3","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09a907808417a5d77c35b9c7616fc0ce8875baf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 16:24:18.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd53616d79fe84626380a4bd837b1fb8315a0375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd53616d79fe84626380a4bd837b1fb8315a0375"}]},"branch":"refs/heads/master","description":"Rebase"},"321e45a61dc8760c08aeb5192973f39e44aef09f":{"kind":"REWORK","_number":29,"created":"2025-12-15 16:25:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/29"}}},"commit":{"parents":[{"commit":"09a907808417a5d77c35b9c7616fc0ce8875baf3","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09a907808417a5d77c35b9c7616fc0ce8875baf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 16:24:53.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/321e45a61dc8760c08aeb5192973f39e44aef09f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/321e45a61dc8760c08aeb5192973f39e44aef09f"}]},"branch":"refs/heads/master"},"d72086e189ecfa93ac5129b30f2e97538de6e536":{"kind":"REWORK","_number":30,"created":"2025-12-15 16:25:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/30"}}},"commit":{"parents":[{"commit":"09a907808417a5d77c35b9c7616fc0ce8875baf3","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09a907808417a5d77c35b9c7616fc0ce8875baf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 16:25:53.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d72086e189ecfa93ac5129b30f2e97538de6e536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d72086e189ecfa93ac5129b30f2e97538de6e536"}]},"branch":"refs/heads/master"},"a6f472d89ccd571edb70f9553dc69b261dcbe5af":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2025-12-15 20:51:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/31"}}},"commit":{"parents":[{"commit":"807ed9744f35cd7ae8d6795ae6692bcf98da185c","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/807ed9744f35cd7ae8d6795ae6692bcf98da185c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 20:51:07.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6f472d89ccd571edb70f9553dc69b261dcbe5af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6f472d89ccd571edb70f9553dc69b261dcbe5af"}]},"branch":"refs/heads/master","description":"Rebase"},"578ba665e7b45c64c2eb566898ff6d0f064204e0":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2025-12-16 10:03:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/32"}}},"commit":{"parents":[{"commit":"a7ee5ccb2b551af5be80c0f082e503f6cbd54096","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ee5ccb2b551af5be80c0f082e503f6cbd54096"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-16 10:03:12.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/578ba665e7b45c64c2eb566898ff6d0f064204e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/578ba665e7b45c64c2eb566898ff6d0f064204e0"}]},"branch":"refs/heads/master","description":"Rebase"},"cee27f6426f49b4242b0535d0e0dee6f8a44dbda":{"kind":"REWORK","_number":33,"created":"2025-12-16 12:15:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/964793/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/964793/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/964793/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/964793/33"}}},"commit":{"parents":[{"commit":"a7ee5ccb2b551af5be80c0f082e503f6cbd54096","subject":"debian: Switch to deb822","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ee5ccb2b551af5be80c0f082e503f6cbd54096"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 14:58:36.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-16 12:15:22.000000000","tz":0},"subject":"Switch to Debian Trixie","message":"Switch to Debian Trixie\n\nChange-Id: Ib7f551d4a7c6ffe95786853d6899e704de4747b8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cee27f6426f49b4242b0535d0e0dee6f8a44dbda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cee27f6426f49b4242b0535d0e0dee6f8a44dbda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
