)]}'
{"id":"openstack%2Fkolla~964864","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I2cf22b874fcb885752bf3800c8ad8fa7a5eacff7","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-10-28 08:55:57.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I2cf22b874fcb885752bf3800c8ad8fa7a5eacff7","subject":"Remove container image of neutron-linuxbridge-agent","status":"MERGED","created":"2025-10-27 08:36:37.000000000","updated":"2025-10-28 10:22:17.000000000","submitted":"2025-10-28 10:21:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964864","cherry_pick_of_change":962279,"cherry_pick_of_patch_set":4,"meta_rev_id":"529affc866f7214ddb2abdafe0767cd02dd52c45","_number":964864,"virtual_id_number":964864,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 10:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-27 19:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-27 18:04:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-28 08:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-28 08:55:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 08:36:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-10-27 09:29:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 18:04:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-27 19:38:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"17cb68cd5cac2865c50c58cf264cdd8d0ba1603c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 08:36:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fe4575d07be3ba461d0860cde3e41e45ed3bc226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 09:29:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/503fe7891888497cace86774fa2b3293\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c54fe5c5cd05443797b85e42bf29d135 : SUCCESS in 21m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/93411ead72ff4099921194cbfe1f9f56 : FAILURE in 30m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a0dbe6d53b874710b434ef84bcf807a9 : FAILURE in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f1049fc6dc7a45ceafc228b391724120 : FAILURE in 41m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e00d7e6083ff4623927b86755bb80665 : FAILURE in 19m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/617d0c4056404c82af7bad06e9bdea1d : FAILURE in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d2bfa133664599af0327b2a7ae5a4a : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/672cf7d81fb447fa94d57d8bd0d82534 : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3823da3826cd4a9b979f2cb13239122b : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f5dd29e0bbc40169f8cff6b016c10a4 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4481f87a0fa54a0986fdd662b62d6109 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5955458b8d4b4b5890469bcf566b52ff : SUCCESS in 1m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/210cd1063fca4679a1a244ea5edd31d3 : SUCCESS in 2m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1ce6109ddfb4280be3a645a0b365358 : FAILURE in 39m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/78064673ea44497c8853588f0f2b6e44 : FAILURE in 42m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e07aaa741fd94dffaea5079c7617e802 : FAILURE in 44m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b5b759549f4e4a1b8feb54b80fc81d95 : FAILURE in 18m 57s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6d532180832d49b3b34e8404180b51b8 : FAILURE in 28m 14s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9fa5bdb14fa8425fa1d551a9f0a3f426 : FAILURE in 28m 19s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ddffd9705a7f45258ed85c0a53bffb8e : FAILURE in 23m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7be2748d395c4c559d176feccc8a90ce : FAILURE in 27m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1d431053c1484499b445a9af93468076 : FAILURE in 19m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/54d44f6b1dae41f9a5a71d1d8f9796d8 : FAILURE in 20m 04s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a1606d9bbaf346b68269f718a7dfa93c : FAILURE in 19m 08s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2ab4fa50023145649967b8688042bd44 : FAILURE in 21m 00s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/551394df8b624ea0901ac084e51f50b0 : FAILURE in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0158c7888bbbb21fadea110c1b23c7efff802ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 09:31:44.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0477cdf161f4e98025e418bb9262bd2433461bcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 11:04:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5958707e0a14fd38cbbec1080e5f1cc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/37e00edcfd22461eaff99e9b5c697012 : SUCCESS in 26m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/84ff44cc889d4893be0a5d230b22fa11 : NODE_FAILURE Node(set) request 7c8d2457c75f4085ba90879a65cc069f failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7999ce45ce3448b995c408876ae88dfe : FAILURE in 1h 00m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79dd98ddd7cb4e6386ea58ac813c3e62 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c08c3b0025214ec6a77c19217989ff14 : SUCCESS in 4m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3432b6d7120e4b5ba6bc217b537b3c59 : FAILURE in 40m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7a798fb814e84686b94f9f3cc4e9b3a3 : FAILURE in 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e5129170d5f0cf101fff2c048d01f876a6f50ca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 11:18:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed601ae8b2094c8084e734679f42ed93\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2c44e29a1b06422da208e47639830e4b : SUCCESS in 12m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c2d68354df074303b312b91c13cc0790 : FAILURE in 16m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/81863716f9b7434e9e6eb3297999787b : FAILURE in 37m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4bf266d26293440893a5c4a2bb7f318b : FAILURE in 26m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/daa746273aed440097785502fd586698 : FAILURE in 21m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c2f231f5ce0842e28a5981dda778932b : FAILURE in 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d32db2c77184046a0a335b8739fef9d : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45a2f0af84cd48598cfdd1e21e7a61ee : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/425a7496b1d84185bdc19209b719053c : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/528702535d0f4e15872b280c2a3b09a4 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5611737d914d38b9445b0baeabb6f8 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ca7a5dad4154c6eba5f33efdc801fb3 : SUCCESS in 3m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fbd69bac60f94e8187daa8a76586f523 : SUCCESS in 3m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/30c212c5595a4d4499ca1b6713f24e6f : FAILURE in 14m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a3601f6133424ea2b8a4bee64294cd88 : FAILURE in 1h 38m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/146a32db40974ce7960780fa4d413360 : FAILURE in 13m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e176bb02dad2476582db78edb03f0ed5 : FAILURE in 20m 27s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/73a1a634f4684596ae5209f9b349f8bc : FAILURE in 47m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c3575902a54d4a659946f9217b9892e8 : FAILURE in 26m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aca4f3574e2b4b68837423a9d8278e0f : FAILURE in 25m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7c0aa2cad724432d822692148ea2f051 : FAILURE in 28m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3c6b2c4c245f47b7b8d19f7a47e85a86 : FAILURE in 18m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b4b67e57b370441ab0fbaad88bd10fc6 : FAILURE in 22m 32s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ec2fb1d4afbc439cbcd4b3b99d1f6764 : FAILURE in 18m 54s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c7ce1e9fdfaf461bb82ee5aa17aaab61 : FAILURE in 19m 15s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4242900dace643cb944505f9a284fdd8 : FAILURE in 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4de767413fd56126c9dfc458a0b9edf2ab555244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:26:24.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_23084\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"_revision_number":3},{"id":"aba01b81f26216ddbb277f8a5600867f47cd5c30","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-27 18:04:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b0c817716e180d03086ff2e72426960d4ba93829","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 19:37:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e7590c4361471388073813cfd81e7e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5aaf9c2a6ceb4857b301ca6e680cfe4d : SUCCESS in 40m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4a534a8ca641457c8fdc59b4ba9d7d09 : SUCCESS in 37m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9dc0de3cd2664b0a8a9e16cba89fe254 : SUCCESS in 1h 46m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/451c654c0a9b4c82a19461b18c5a346e : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f1af1c1a6f0d4e3abaf0b1425edf5626 : SUCCESS in 6m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a7b0b2d16a834f12b48952d192951a00 : SUCCESS in 33m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ae07467ecd974f9a8932d02e83ea595c : SUCCESS in 48m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd4ec3a362f97c1998bfb7ffd6df12b885f47dd6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 19:38:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"63741a9b249a64da65bb48d7452507a3bd6f5dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 19:52:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b9c98bf55eb4e1b809e1603cf2f71b5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/79abb36122d049f895f648069f4c67d1 : SUCCESS in 12m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/156886a4dee145bfba181e43916eb365 : FAILURE in 30m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25bab4242d224ce8b9b65d1f709f9a9f : SUCCESS in 23m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0b0dcbb1b6f941a6a4df92391198a05f : SUCCESS in 25m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/634e7cb186264910a929b449ef1fe215 : SUCCESS in 59m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3e4b3f5fdaff40a3bd301c7a636274c7 : SUCCESS in 1h 08m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/055fa0c0a15247f593e5ba41895009de : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6e78775e23f4fbc8a6d98d78ace17db : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f66f4e87cd3847b5832ee0e533b1066e : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc18e9d34ca745918cece6a987d8707b : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17df419bc9624e9692fe6e2e2d1e3c7e : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f64864e36bfa436fa0424fda58d5c934 : SUCCESS in 3m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e477057ecce542a5a45d72125826ca10 : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2a0a0487badf49768ad2032ecfdf4d7d : SUCCESS in 20m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/81439e2c026f4be3aff65ef758a1f93c : SUCCESS in 1h 27m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4525d48db5924bd4b9fc8f67c8bc2548 : SUCCESS in 1h 13m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7de6f2ee4a4c4baebb329be51bc512bf : SUCCESS in 1h 37m 41s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/902d204b019b4bfcae7bcbf3a59f5090 : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/32dfba6021d74460be64be58a65f2155 : SUCCESS in 1h 41m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5760f317a73b483d93985af2c64dffcf : SUCCESS in 28m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1b544a475e3044ebaebde962c0d3bebd : SUCCESS in 32m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/98a936aa46b24a858641fe140d519fc5 : SUCCESS in 47m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ca043fef88ff4d0abe82df8185f4cf63 : SUCCESS in 1h 13m 51s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/76ebbc664079413aad67d3addeb42486 : SUCCESS in 49m 31s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/695d1b447b2d4daab5bf0dc4388b4703 : SUCCESS in 1h 15m 42s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/876bb4182ef7483b90a34ba202d2afa8 : SUCCESS in 2h 04m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e562823a5e0008b0d4b9a7d3256bf45477b5bbb7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-28 08:55:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cb2f396a7d580462adbdd252f4f52469af8d1523","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 08:56:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c0183b61784d3cbf7abf70ed36e67304fef8ba7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 10:21:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ddaee5dd3e549258d11754a33aeedd9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/caf3474208344e42b51cedec931bd3c6 : SUCCESS in 10m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a2d520e4908d402f8b3d6b14f95162e4 : SUCCESS in 26m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/15b02880923a4691b66ccfeab1ea3112 : SUCCESS in 52m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/33bbe1fb64c94725b56e5ecb520e2477 : SUCCESS in 1h 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3336988f26e4874b691c6fd84bc665e : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02e104b26b114757b9c1289f76eb3098 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7216bb92206647aea3fdfe9fa1999b90 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca8cbe6b21ea4fff97eed09a62964812 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bcf23ee310a4ac9901eceeec131c256 : SUCCESS in 4m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c149c785341544128f9c2d3644508a6d : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/85ea537109bc4c9cbf0c8fafba742cff : SUCCESS in 16m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c38b915e36d14daf9d50624fabaf8431 : SUCCESS in 1h 12m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e7b8f947794d47e19b1f1c4ea53268da : SUCCESS in 1h 08m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a35af6f336bd4b62891abbab23082c94 : SUCCESS in 1h 13m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0a578bdc46c7492aafbb817a8fb3debc : SUCCESS in 17m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/47f88bcdca6347ababcaf456bbe7da61 : SUCCESS in 32m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/af81e6a8945742efa53e3a700a9a52a2 : SUCCESS in 53m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5d7b008fea364e0c96a249135290c174 : SUCCESS in 1h 16m 50s","accounts_in_message":[],"_revision_number":3},{"id":"470e076ed548e714d20b5ec6a278a1ee90f501d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 10:21:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"529affc866f7214ddb2abdafe0767cd02dd52c45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 10:22:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b71d6e2a76a4db29a419c4a7e1abd07\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b87dc814e14a35be41881cbc725063 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93d5a856dcf143b88101462a9e9ea9ba : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e9efe07b3d3444665604ff8cbbaba2038104a561","revisions":{"5baa815f55f2e75a5c03b56a85f3e1c3522ed04e":{"kind":"REWORK","_number":1,"created":"2025-10-27 08:36:37.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/64/964864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/964864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/964864/1"}}},"commit":{"parents":[{"commit":"31de4b1fc30aff8eb67f92d930f383c41a0786bb","subject":"Merge \"doc: Fix Ubuntu Ceph version\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31de4b1fc30aff8eb67f92d930f383c41a0786bb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-25 13:30:03.000000000","tz":540},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-27 08:36:37.000000000","tz":0},"subject":"Remove container image of neutron-linuxbridge-agent","message":"Remove container image of neutron-linuxbridge-agent\n\nThe agent was removed from neutron during 2025.1 cycle. The remaining\nimplementation to deploy it is being removed from kolla-anible now.\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/962278\nChange-Id: I2cf22b874fcb885752bf3800c8ad8fa7a5eacff7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 3e12a642c2f4cd0600a2f9aa97ab9e24af3b7912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5baa815f55f2e75a5c03b56a85f3e1c3522ed04e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5baa815f55f2e75a5c03b56a85f3e1c3522ed04e"}]},"branch":"refs/heads/stable/2025.1"},"ef2c79864c4e2da86b82beb86960ce682c62720f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-10-27 09:31:44.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/64/964864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/964864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/964864/2"}}},"commit":{"parents":[{"commit":"31de4b1fc30aff8eb67f92d930f383c41a0786bb","subject":"Merge \"doc: Fix Ubuntu Ceph version\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31de4b1fc30aff8eb67f92d930f383c41a0786bb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-25 13:30:03.000000000","tz":540},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-27 09:31:44.000000000","tz":0},"subject":"Remove container image of neutron-linuxbridge-agent","message":"Remove container image of neutron-linuxbridge-agent\n\nThe agent was removed from neutron during 2025.1 cycle. The remaining\nimplementation to deploy it is being removed from kolla-anible now.\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/964863\nChange-Id: I2cf22b874fcb885752bf3800c8ad8fa7a5eacff7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 3e12a642c2f4cd0600a2f9aa97ab9e24af3b7912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef2c79864c4e2da86b82beb86960ce682c62720f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef2c79864c4e2da86b82beb86960ce682c62720f"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"e9efe07b3d3444665604ff8cbbaba2038104a561":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-27 17:26:24.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/964864/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/964864/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/964864/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/964864/3"}}},"commit":{"parents":[{"commit":"6f2a79debc926e4817f91b8afc8f9a2b6373e0b3","subject":"horizon: pin pip to \u003c25.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f2a79debc926e4817f91b8afc8f9a2b6373e0b3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-25 13:30:03.000000000","tz":540},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-27 17:26:24.000000000","tz":0},"subject":"Remove container image of neutron-linuxbridge-agent","message":"Remove container image of neutron-linuxbridge-agent\n\nThe agent was removed from neutron during 2025.1 cycle. The remaining\nimplementation to deploy it is being removed from kolla-anible now.\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/964863\nChange-Id: I2cf22b874fcb885752bf3800c8ad8fa7a5eacff7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 3e12a642c2f4cd0600a2f9aa97ab9e24af3b7912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9efe07b3d3444665604ff8cbbaba2038104a561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9efe07b3d3444665604ff8cbbaba2038104a561"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
