)]}'
{"id":"openstack%2Fkolla~964867","triplet_id":"openstack%2Fkolla~master~Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-27 10:22:18.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe","subject":"horizon: pin pip to \u003c25.3","status":"MERGED","created":"2025-10-27 08:44:45.000000000","updated":"2025-10-27 11:37:22.000000000","submitted":"2025-10-27 11:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964867","meta_rev_id":"0ae965b3e4be963c935ba7e5a381131217ba5383","_number":964867,"virtual_id_number":964867,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 11:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-27 10:22:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-10-27 09:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-10-27 10:22:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-10-27 09:40:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-27 09:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 09:59:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-10-27 10:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d54693776633267a310dcee362ac507265d5e3c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 08:44:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9609ebb1ecb323ae2b29855acff5ffeb2432bd13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 08:59:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"33d2fbaa3feb1a853e819ae2e4a59778013b8a3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 09:38:09.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"38f4ea9e3122dbca444d77b8047b8dd7c9f213c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 09:40:52.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"654d685e3a7a645d8c038de5ebbeb6c5a1c6d185","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 09:40:57.000000000","message":"Patch Set 3: Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"3c621d6d07843023cd3b4193d919b5ac81293658","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 09:59:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1fd776feb16fb9571d996a85f59d8777ad0831f1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 10:22:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb0014f404b14f9036b6ee27f2c2a710f5686b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 10:27:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beefb3ae967d4e9094b98ceffcccf506\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/395068440f3b4cb48955957ea6e4f1be : SUCCESS in 11m 36s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/d365f08319e540a7931c3defa900c062 : SUCCESS in 19m 38s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/115bba05eaba412e97f8af7842268c76 : SUCCESS in 28m 24s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a2ee3a43df4046dc8ebc3138f674310a : SUCCESS in 20m 36s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/809574363560468eb95afbfebfda3050 : SUCCESS in 49m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51a44010c72d48d48410f1e40d043dc2 : SUCCESS in 21m 03s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6807302fd063416bb8c0a62a0bdb9394 : SUCCESS in 24m 49s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e0c35eda140b4bca80e3e33e0e1dbefb : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b131f1af5ed44c4582b3ce59c86dc21b : SUCCESS in 1m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ad82cf60f434a5093eaa18bca776bb8 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53912f3269804c7285701a66a9c73d7d : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fc3087dd7e5401ca89b2c4e1dc00eb5 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0a213fc3c247d4a2e84ab5521c1e73 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":3},{"id":"67f6894d7e9fd43f26280d08c6eaba5cf3539420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 10:28:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"077c1d9bcbb7eb738ac34aa9955be936b85a0987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 10:47:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79412b065f62461b907fb46f671059db\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d6694abf6d1a49528a040eb3ac339a4e : NODE_FAILURE Node(set) request a74a4324faa44ba08781ff6c8d310411 failed in 0s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/1c58b5f0b0294249828d59457365ac5e : NODE_FAILURE Node(set) request 3cd15904265743bfb2a8f7f52e7f162a failed in 0s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/56c0637b972d44eb99430820d3041583 : SUCCESS in 36m 55s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b5022c8a6314434592e1233ad18d2c10 : SUCCESS in 42m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97e58ada87784c96ae53ebf12f6d8b49 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba008185605143ef90a4d195c0cc8bbf : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/065908948aee45c5abc3a86bffb0df32 : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a054005a0f77f126ee3145b5834be6f4a4ad9b0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 11:36:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76c53d7d30f474082d2ae437e0c710b\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/8eb4fb6a0df443f091b6ff35aabaec3f : SUCCESS in 21m 04s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/8050551305104c1fa635b369ae664182 : SUCCESS in 31m 01s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f5e5f359045141f1b41f785f6acc4766 : SUCCESS in 11m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/279eedac0d6141dd98e231407ede6fff : SUCCESS in 1h 04m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc00067e11fc4e76adb156b977100f2b : SUCCESS in 20m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a4dfda7d25a4b37a49ccdc8f85ce1a0 : SUCCESS in 37m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/11d90aec84dd44e8878ebf19c85998ca : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9751d6f102534325b713435e84700d2e : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1ab32e73a904561b7f561a817e8778f : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/751ea399a77e41b9bbf44fce601d8563 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843085b3614143878ca5900c99c02998 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":3},{"id":"989e4f603e262e9ec3a34ff5ce6667726fa15c17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 11:36:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0ae965b3e4be963c935ba7e5a381131217ba5383","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 11:37:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3dc0b96ec964e21b8decabb8037ae23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee744795fe34aa0813c43a22f7c1c61 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9bc8d4d0816ac89614a703170601046e57f54c57","revisions":{"91e1928b5d1c7e996bcece7291d8b9b1817948c0":{"kind":"REWORK","_number":1,"created":"2025-10-27 08:44:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/964867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/964867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/964867/1"}}},"commit":{"parents":[{"commit":"6bae37377c02af5b89bb2b7e06f8e3fc7fb10531","subject":"Merge \"Remove container image of neutron-linuxbridge-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bae37377c02af5b89bb2b7e06f8e3fc7fb10531"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:43.000000000","tz":60},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91e1928b5d1c7e996bcece7291d8b9b1817948c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91e1928b5d1c7e996bcece7291d8b9b1817948c0"}]},"branch":"refs/heads/master"},"45df9ca2fd943a5a7a114a338a6e42adfd36b7bb":{"kind":"REWORK","_number":2,"created":"2025-10-27 08:59:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/964867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/964867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/964867/2"}}},"commit":{"parents":[{"commit":"6bae37377c02af5b89bb2b7e06f8e3fc7fb10531","subject":"Merge \"Remove container image of neutron-linuxbridge-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bae37377c02af5b89bb2b7e06f8e3fc7fb10531"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:59:59.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45df9ca2fd943a5a7a114a338a6e42adfd36b7bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45df9ca2fd943a5a7a114a338a6e42adfd36b7bb"}]},"branch":"refs/heads/master"},"9bc8d4d0816ac89614a703170601046e57f54c57":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-10-27 09:38:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/964867/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/964867/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/964867/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/964867/3"}}},"commit":{"parents":[{"commit":"6bae37377c02af5b89bb2b7e06f8e3fc7fb10531","subject":"Merge \"Remove container image of neutron-linuxbridge-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bae37377c02af5b89bb2b7e06f8e3fc7fb10531"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 09:38:09.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nCloses-Bug: #2129916\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8d4d0816ac89614a703170601046e57f54c57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8d4d0816ac89614a703170601046e57f54c57"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
