)]}'
{"id":"openstack%2Fkolla~964873","triplet_id":"openstack%2Fkolla~stable%2F2025.1~Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-27 14:18:29.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-10-27 15:59:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe","subject":"horizon: pin pip to \u003c25.3","status":"MERGED","created":"2025-10-27 11:40:41.000000000","updated":"2025-10-27 16:00:08.000000000","submitted":"2025-10-27 15:59:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964873","cherry_pick_of_change":964867,"cherry_pick_of_patch_set":3,"meta_rev_id":"5c852b5f7f7cea4df854c79b04be1880d8fe27f3","_number":964873,"virtual_id_number":964873,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 15:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-27 14:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-27 13:58:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-10-27 14:01:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-10-27 14:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 11:40:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-10-27 13:20:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-27 13:44:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 13:58:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-27 14:18:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a815e1860c9f4ec94421102ecaba3a7b31a6c4a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 11:40:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"07290da4f8fff99a93ed69809a8ca7f13d960a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 11:40:49.000000000","message":"Patch Set 2: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":2},{"id":"48727055a22da785ce49370ed12d4779d808303b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 13:20:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0edf7e3838f4b0981c139c0222ab0f1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5fc82186857049aeba911c4cf523f16a : SUCCESS in 34m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/29f01e12608e4331a006375830899aaf : SUCCESS in 24m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/24dfe0c51dca442495ec4b1ec3470f35 : SUCCESS in 1h 37m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8d517b6c3ff4b6d9e0400acaa220c52 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d75d9e5587694f51a6bbd0cf5d864eba : SUCCESS in 4m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/21fd511713824c5eb7d101204ae747e0 : SUCCESS in 29m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/21a67edae82b488ea8a8e3b6f4fb4135 : SUCCESS in 28m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7c2d39bde8d57bb251fcd1ce83d12f65e9d38d8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 13:44:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f888256ab6ff48f29c2fec33a6e6ee96\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/737a68e08af34c2dbe9d1ff1744ade47 : SUCCESS in 18m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dae08d59825b46b4ad7e858695c754aa : FAILURE in 29m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c8d3c7dc6cd74f9bb45f58c312759393 : SUCCESS in 21m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/748f28018aa643039284bbe9d6ebe37b : SUCCESS in 28m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a8197f3a198c41a985ac6c8dc4317ff1 : SUCCESS in 51m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/33aea512836a4bf68985f114c5dd8b97 : SUCCESS in 1h 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f42fe9659c483c8d2997d352e040ed : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b52d43388d44839aaaf1b0d05aa5c8e : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eceda08ecd034a58ac2bac6ff9e8e612 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3599b6bac4c443d89f7982b434bf501d : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45c3f39ead7456a9130cf32c03d0667 : SUCCESS in 6m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/94ea7ca2577e4cb1b32d8bfd33b6ebcb : SUCCESS in 2m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/41ca58e738534e22a3cff56c659d1dfd : SUCCESS in 18m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2c92328b70ce4583ab75591eef41f0b9 : SUCCESS in 1h 36m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/33e8ccc17c304627bbcaafb049523873 : SUCCESS in 53m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/404bf31cb42c44eeb9e229e502cb6886 : SUCCESS in 1h 46m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/191f6ade07724c988dc76d1ae587dd99 : SUCCESS in 17m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f69505df20b749a985b6f5563bd93b17 : SUCCESS in 31m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/38f8519f18c84339a95401a2ec4fc339 : SUCCESS in 46m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7d69f71a5e1d43b7904ca33d6b69fb1d : SUCCESS in 1h 17m 32s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/d191d2289691450f983e0e1f26a50f6e : SUCCESS in 1h 58m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68f5bba0b3df66bbf9b358b0902c13db13ebbf01","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-27 13:58:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a83c2cad87f18e1116373acac03695b8ee01c416","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 14:01:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a02f0fe0c5762afaa4f160581d5f68dbde0f6ccd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 14:18:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a4f16a8852150b7fbab0995435f2945600d1908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 14:18:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"871bc5c9dd6f3ecbeeb9fc58f294d8e266ba7b70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:59:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d78af3525da14c90b935b48469eba26a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9e88629f57db44f4af0d73a1a15fc5bb : SUCCESS in 23m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9b5d70ba4d5549e4a29bf703d5dc6a0e : SUCCESS in 14m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/69a7406eab8b4f648e113c5e67fc6333 : SUCCESS in 53m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/62a951db90eb49918512e05a13457ab5 : SUCCESS in 1h 09m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f3ef8d75a84f6c8e0b24ea0a9ce5db : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecb4d2808e644b67a0262383e179c278 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b22ba919d75a40bda471bbfc627f7d41 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2828c0621e340479bb67c58ed90721c : SUCCESS in 6m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/414a1a6a5e834b44b76ed631fa0aef5f : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/458efa1510f04e72b582a4fb5f641580 : SUCCESS in 22m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/65166dd5c4994c46a15d7a7f16d1435c : SUCCESS in 1h 19m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a72ad3f742604797895e8773c05e792b : SUCCESS in 33m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cce669cdb63945dda273ac67e15a384c : SUCCESS in 1h 36m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aba919d0054042faa6b6b674ed3730df : SUCCESS in 25m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4827154c55784e1083f8c20eb7d8c51c : SUCCESS in 29m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/602dea6341d84679971efeac24066817 : SUCCESS in 47m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9d656c15e522411db1d328ca1b26efd5 : SUCCESS in 1h 13m 16s","accounts_in_message":[],"_revision_number":2},{"id":"59a501abbb255e191b54188c0add02e9f9abd1db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:59:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5c852b5f7f7cea4df854c79b04be1880d8fe27f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 16:00:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/922134d35a32486f80a7eef7191aba6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/866843accecd405dbeae205d934a65d4 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f2a79debc926e4817f91b8afc8f9a2b6373e0b3","revisions":{"101c2336c4d6600d5c95737fc65b5e09f8222be6":{"kind":"REWORK","_number":1,"created":"2025-10-27 11:40:41.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/73/964873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/964873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/964873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/964873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/964873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/964873/1"}}},"commit":{"parents":[{"commit":"31de4b1fc30aff8eb67f92d930f383c41a0786bb","subject":"Merge \"doc: Fix Ubuntu Ceph version\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31de4b1fc30aff8eb67f92d930f383c41a0786bb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-27 11:40:41.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nCloses-Bug: #2129916\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 9bc8d4d0816ac89614a703170601046e57f54c57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/101c2336c4d6600d5c95737fc65b5e09f8222be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/101c2336c4d6600d5c95737fc65b5e09f8222be6"}]},"branch":"refs/heads/stable/2025.1"},"6f2a79debc926e4817f91b8afc8f9a2b6373e0b3":{"kind":"NO_CHANGE","_number":2,"created":"2025-10-27 11:40:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/964873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/964873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/964873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/964873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/964873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/964873/2"}}},"commit":{"parents":[{"commit":"31de4b1fc30aff8eb67f92d930f383c41a0786bb","subject":"Merge \"doc: Fix Ubuntu Ceph version\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31de4b1fc30aff8eb67f92d930f383c41a0786bb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 11:40:49.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nCloses-Bug: #2129916\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 9bc8d4d0816ac89614a703170601046e57f54c57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f2a79debc926e4817f91b8afc8f9a2b6373e0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f2a79debc926e4817f91b8afc8f9a2b6373e0b3"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
