)]}'
{"id":"openstack%2Fkolla~964874","triplet_id":"openstack%2Fkolla~stable%2F2024.2~Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-27 15:59:13.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-27 15:59:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe","subject":"horizon: pin pip to \u003c25.3","status":"MERGED","created":"2025-10-27 11:42:14.000000000","updated":"2025-10-27 16:00:08.000000000","submitted":"2025-10-27 15:59:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964874","cherry_pick_of_change":964867,"cherry_pick_of_patch_set":3,"meta_rev_id":"c8b4507aba7178c10f4ca59a7a54ea3393f382ca","_number":964874,"virtual_id_number":964874,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 15:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-27 14:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-27 13:57:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-10-27 15:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"},{"value":2,"date":"2025-10-27 14:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"},{"value":1,"date":"2025-10-27 14:21:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 13:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 13:57:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-27 14:21:10.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-10-27 15:29:47.000000000","updated_by":{"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"},"reviewer":{"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"},"state":"REVIEWER"}],"messages":[{"id":"80e64a1158453b3f60f4306e35b0f1136ba96b8b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 11:42:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/horizon/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"428693d25b5d49e108a57a9c7866b99e11f7d136","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 11:42:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0e098bbe364e20561277761f0e7743f346d2275a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 11:43:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d11dc53a3947670a14e2e9137d826d8ab3927263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 13:16:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9db7caaea3347bfbd8407539959694b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/446cecfd98da42fc88da25bde2b0bfb4 : SUCCESS in 19m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/afbc92eadcd844db9245c35744d6a5ac : FAILURE in 33m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d37c6fe5f8774d35b0ab8935be2f4f60 : SUCCESS in 22m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2920054079cc4d0396ab9bbe6315459a : SUCCESS in 26m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d9e7729ece254c7aafb18f2d103776d9 : SUCCESS in 1h 03m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fa30ac281c8d4d29a05eeeffdcc8dad4 : SUCCESS in 46m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d614a7069c9c4d3b9d2a5a2beb5abe29 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a9845d96def4ef19341e431451c2a44 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e36f177a784442d69550654a2db26b1f : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09c0a54e0b824194b93c564015a4f0c6 : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b3de321d0b142d3a820ae81197ba967 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d0ddf116e744c23985eed1cec7fcc3e : SUCCESS in 4m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c70bc8b8706b41d5bbe5ec3728861acf : SUCCESS in 1m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6d53c2ac6032405ca5c76266f0f6e1ad : SUCCESS in 20m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/33aeddc4b48b4442bf271d47b611c825 : SUCCESS in 27m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/18bfe00d263f43e3b9c7bb9346d6d9d0 : SUCCESS in 59m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2535c33df9b44aa082b9b3cb63c79445 : SUCCESS in 1h 15m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b5fee3278b2e4a61bac1ef4985203268 : SUCCESS in 21m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7f31cbcf96654f529cf5be99f30469fa : SUCCESS in 27m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f76fd21398f54d55a853b7db669a8b27 : SUCCESS in 48m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/afaed0d8294f4d41b314fa9b785f6ff5 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":3},{"id":"01cc43234a7bd4c990cee118a1443c6e2d82011b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 13:52:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312f0a24e7f146fe8e0e55f50011050c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ff71703e17ec4b97b4e84e860dacf6de : SUCCESS in 30m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79f1a0b525a14c74a3b3cab41eb21586 : SUCCESS in 27m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ea69bf5af74c449890d069c0086f046a : SUCCESS in 1h 36m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c55b3c0c91ed40d8b27bc5d5017ce342 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6af4b1602b5f4db6b21bc7223d0023a8 : SUCCESS in 4m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/463a3b95a2c24827b460f5f09461291a : SUCCESS in 36m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e9eb33a3c7ff4db3abc1ad7a4750e119 : SUCCESS in 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6228a2177b6e4a345cef919139eae255b8e1ac58","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-27 13:57:46.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1483d8f03b88df9288e9fd5bf2d8ce794ce2e3b8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 14:02:13.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"c6d39cfdcf8f54116c572143c542527dcd297966","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 14:18:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"18a372fa420f9a2f14ea6e422984ba8af1411211","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 14:21:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b5e3be0a7fd01d724db666f9e9ce1b57af63b32b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 14:21:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1d91187dad8122cb73935bbedcb227834c5bf7e0","author":{"_account_id":37817,"name":"Alexey Kuchinskiy","display_name":"Alexey Kuchinskiy","email":"kuchinalser@gmail.com","username":"kuchinalser"},"date":"2025-10-27 15:29:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f7e3166c4a30569dd63dcbe9c2d6114d545095cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:59:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d867fba68a144b49b7981a273214529\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f3e0602331334b8b99d107fdd9bf3c6f : SUCCESS in 21m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b4e1bf7509c042c59012157dad0784a3 : SUCCESS in 28m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/393483cd93e64ec59e9038b79009e1a1 : SUCCESS in 1h 13m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5ab69fca3a424d938ae4b1519faf0ad1 : SUCCESS in 1h 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3387bd00cb84ae59da183fd9178ba52 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45d3232a535546a49eb5609e5dd98d2f : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/365fe2dfd2a540848311bdcdecb290b4 : SUCCESS in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864c5277e700409dad0a7bc548120c83 : SUCCESS in 4m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/db780389882a42c08128c3c7c5163ccb : SUCCESS in 2m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/228d819a8be645b585e9e5a3eadf385f : SUCCESS in 21m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4eb3af6dd3854600af591be8c7815aa2 : SUCCESS in 28m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d2c00fa45c184381bd15e0b257bc018b : SUCCESS in 1h 02m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/171735e45f5a439683e752b36ec015b8 : SUCCESS in 1h 21m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f1b2ebd9e90b410295f0944e09dce218 : SUCCESS in 20m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/14fff5a9dd004183bc3c40241a3e0a02 : SUCCESS in 34m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e8423ffcd5c94fd882db10dd640cf740 : SUCCESS in 1h 01m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/71cea8b7315f4d11ba25166d618feed3 : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":3},{"id":"3ad4a350efc7eb1059363738cb63ce8f65d76c5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:59:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c8b4507aba7178c10f4ca59a7a54ea3393f382ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 16:00:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83bf0a0fe4424d9a9bdcf8ee7541201e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6345944b960744299d40e7ff7b438c90 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d7e859067daaab5d0383bf8e4ad4ee4c9b910a55","revisions":{"2d7f50c2d46e9f55c72a17dffe1d315b9b5a3704":{"kind":"REWORK","_number":1,"created":"2025-10-27 11:42:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/964874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/964874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/964874/1"}}},"commit":{"parents":[{"commit":"dd08eb3d0a0145069f02a097c270d684a36fb60d","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd08eb3d0a0145069f02a097c270d684a36fb60d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 11:42:14.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nCloses-Bug: #2129916\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 9bc8d4d0816ac89614a703170601046e57f54c57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d7f50c2d46e9f55c72a17dffe1d315b9b5a3704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d7f50c2d46e9f55c72a17dffe1d315b9b5a3704"}]},"branch":"refs/heads/stable/2024.2"},"741c3462474010fe1ce245c662e8c0e01ac36987":{"kind":"REWORK","_number":2,"created":"2025-10-27 11:42:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/964874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/964874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/964874/2"}}},"commit":{"parents":[{"commit":"dd08eb3d0a0145069f02a097c270d684a36fb60d","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd08eb3d0a0145069f02a097c270d684a36fb60d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 11:42:35.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nCloses-Bug: #2129916\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 9bc8d4d0816ac89614a703170601046e57f54c57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/741c3462474010fe1ce245c662e8c0e01ac36987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/741c3462474010fe1ce245c662e8c0e01ac36987"}]},"branch":"refs/heads/stable/2024.2"},"d7e859067daaab5d0383bf8e4ad4ee4c9b910a55":{"kind":"REWORK","_number":3,"created":"2025-10-27 11:43:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/964874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/964874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/964874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/964874/3"}}},"commit":{"parents":[{"commit":"dd08eb3d0a0145069f02a097c270d684a36fb60d","subject":"doc: Fix Ubuntu Ceph version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd08eb3d0a0145069f02a097c270d684a36fb60d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 11:43:22.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nCloses-Bug: #2129916\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 9bc8d4d0816ac89614a703170601046e57f54c57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7e859067daaab5d0383bf8e4ad4ee4c9b910a55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7e859067daaab5d0383bf8e4ad4ee4c9b910a55"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
