)]}'
{"id":"openstack%2Fkolla~964875","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-27 15:59:17.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-27 15:59:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe","subject":"horizon: pin pip to \u003c25.3","status":"MERGED","created":"2025-10-27 11:42:57.000000000","updated":"2025-10-27 16:00:50.000000000","submitted":"2025-10-27 15:59:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964875","cherry_pick_of_change":964874,"cherry_pick_of_patch_set":3,"meta_rev_id":"f0f1e93e8c3c9ff78fefaa84e73cbd27ca5785d8","_number":964875,"virtual_id_number":964875,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 15:59:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-27 14:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-27 13:57:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-10-27 14:21:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-10-27 14:21:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 12:48:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 13:57:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-27 14:21:15.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"058aafd9a31c3a3eb2d0282a4382ea8296ef970c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 11:42:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"8cca4402f7d059e2368d5200a38bcb5aa254cf70","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 11:43:31.000000000","message":"Patch Set 2: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":2},{"id":"9a981a117505c7a88f9c0e1eebb1eba2ffbd1b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 12:48:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1587a28c056248209e485c8c657194e7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0f0003031acb4993acee904af8a13799 : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b1b539d54fa747bcac9da92649936dc9 : FAILURE in 16m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a8937c07a4eb44a9a47758c263b1017f : SUCCESS in 10m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6e2269367b5740c586d0eeb83f44281b : SUCCESS in 26m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fac6cadb7f5b4b44939e1388db169a80 : SUCCESS in 53m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d0d63a18214535803310721aa1362e : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e24adbac4dce4045a4981df053e69170 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80747194a7474ef68363b82de8cbcaad : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f73038e0f864c0283823aa0cf080fe6 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c521f1bad7e242a6978880640a8b62f6 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db799e41ff734cb1871a636ed6f215bb : SUCCESS in 4m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/abb51c8987964f5aba8d239fb4376bbf : SUCCESS in 2m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/03d1033bc1974e83a83e89d7b1f3f977 : SUCCESS in 22m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/24373710372d41b48dfabc2e9aceae43 : SUCCESS in 27m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/32234e910ca3474f960cf6b79bfdc25e : SUCCESS in 50m 45s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b12bef02777c4aa5944d91a821dd440c : SUCCESS in 17m 53s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b60064b1f084beea417792191e1d09b : SUCCESS in 24m 39s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/af65e06ba7ea43dcbf533d3daa66f1ee : SUCCESS in 1h 00m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/602182a897c44b75a5f75fc94cd59618 : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":2},{"id":"789b2e6f057ee6ec10492be6b3f3b3b06f3d0222","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 13:54:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c7f92c06ed41289c9513cf0ffac582\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ea8f09d74be943cfa605995c67f2b594 : SUCCESS in 38m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/03236bb104bf48f3ba216c221b922a47 : SUCCESS in 1h 33m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b5528e6a56c47c5815ccfc5607f51f5 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/060597f9d0a246e58f24dec7ef448b09 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc9d40bfc1db4fc9af1b3fb88abd65cb : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"17c760a3a763640fcce933b18b6ab3b7f6844b59","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-27 13:57:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"381aa17557b8923c150305f73c688cbc828bb118","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 14:02:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"28a0873e76b26df616a48d18d9e8286ca7ad7d5a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 14:18:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"88696a4869f9bd05f1886286728e63f33b4e3c5a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 14:21:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3cdb89520b54567ed325b772993e345b20a42b85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 14:21:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac8d2e1a4796ea9f722d13febe3657e69168b7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:59:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d00d0de3384c2c992fb6cf4b2069aa\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bc535c8c68884dcd9d9d8ac58a7ab117 : SUCCESS in 19m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d70aad3c36d74792847780eae900a469 : SUCCESS in 14m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8fbc4184bc9049f9a131c04abfca3656 : SUCCESS in 1h 19m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65d97fee6b614f2bb163ce14b525cfef : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e4ddb4292e64a2f9b0a1903284fe6e3 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ae09a12c8674bb8aa28df09bce1caaf : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b606b866652544e7a0a8b2baa669ded7 : SUCCESS in 4m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/770ff60f0ebd40dc88486e5e6c8e8a6a : SUCCESS in 2m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3c4801b2c2dc4606b9d7697348fa8762 : SUCCESS in 21m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7fd6f39d1d82486f9065a57bff9eadd1 : SUCCESS in 20m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/86661ef76dc84b0991412a488724ad6f : SUCCESS in 1h 08m 06s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/318de88bd3544c2499efdec493db3928 : SUCCESS in 19m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20f07ec3a4904120a9ff09bfc8528a38 : SUCCESS in 10m 43s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/13471baeb880407c97c55b4907a1726c : SUCCESS in 59m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34e751b2db174fd78736ea54961a3d10 : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":2},{"id":"749464567436c1306cdcf663703e5dcd3b7280a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:59:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f0f1e93e8c3c9ff78fefaa84e73cbd27ca5785d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 16:00:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f575fee03284adfa62d0a39920f4968\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e2cfa547404fb99a1d862bbbe5c248 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"becd6081d8fc0d72ccdc57f70b57fd29e179f73d","revisions":{"2cfb1fed7a5a537b46c5a3b71d2621d939b627ef":{"kind":"REWORK","_number":1,"created":"2025-10-27 11:42:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/964875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/964875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/964875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/964875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/964875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/964875/1"}}},"commit":{"parents":[{"commit":"07103ab0a0960e2a509fa349a25ad014ef675e38","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07103ab0a0960e2a509fa349a25ad014ef675e38"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 11:42:57.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nCloses-Bug: #2129916\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 9bc8d4d0816ac89614a703170601046e57f54c57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cfb1fed7a5a537b46c5a3b71d2621d939b627ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cfb1fed7a5a537b46c5a3b71d2621d939b627ef"}]},"branch":"refs/heads/stable/2024.1"},"becd6081d8fc0d72ccdc57f70b57fd29e179f73d":{"kind":"REWORK","_number":2,"created":"2025-10-27 11:43:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/964875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/964875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/964875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/964875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/964875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/964875/2"}}},"commit":{"parents":[{"commit":"07103ab0a0960e2a509fa349a25ad014ef675e38","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07103ab0a0960e2a509fa349a25ad014ef675e38"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 08:44:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-27 11:43:31.000000000","tz":0},"subject":"horizon: pin pip to \u003c25.3","message":"horizon: pin pip to \u003c25.3\n\nCloses-Bug: #2129916\n\nChange-Id: Ib1b40a65cb2667b53c72f00d16bfe280bcf24bfe\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 9bc8d4d0816ac89614a703170601046e57f54c57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/becd6081d8fc0d72ccdc57f70b57fd29e179f73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/becd6081d8fc0d72ccdc57f70b57fd29e179f73d"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
