)]}'
{"id":"openstack%2Fkolla~965458","triplet_id":"openstack%2Fkolla~master~I395cfdf22cd87a43294dbae561a8872cb0b48e1b","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-11-04 12:52:35.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-04 14:33:28.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-04 14:33:28.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-04 13:24:50.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I395cfdf22cd87a43294dbae561a8872cb0b48e1b","subject":"Bump versions of packages in sources.py","status":"MERGED","created":"2025-10-31 10:08:21.000000000","updated":"2025-11-04 14:34:20.000000000","submitted":"2025-11-04 14:33:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965458","meta_rev_id":"fed4276fa65f5178f20ee364b0f1a40f2962e684","_number":965458,"virtual_id_number":965458,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-04 14:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-11-04 12:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-11-04 13:24:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-11-04 13:24:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-11-04 12:52:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 10:18:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-10-31 11:30:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-31 13:34:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-03 06:16:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"bfaa365bf52d50a6074c35813172e7e02ce3c4e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-31 10:08:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41ffe046134b4390a2d526d52ba9a3bb653b0bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-31 10:15:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c0c367b6bad98bf813bf277a5b4e98db3ee009ce","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-31 10:18:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b2420f35c90bcede4c7a6dfcbf27725571941963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 11:30:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24c41852714a48a286992bed5e5c1019\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/aeba7d8c9fdf467ba406b081f4d39915 : FAILURE in 15m 32s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/a322f2b6689847f4b61231464941b666 : FAILURE in 21m 14s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/fd48561306be4095b775505228e67546 : FAILURE in 25m 02s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/89c10476bd4f463f99d11836333dc699 : FAILURE in 22m 48s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2aa3bf4b45054dbf8bd815f4dc51d7bd : FAILURE in 1h 07m 21s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bed54bc046ca4c5fba68cb5cfb302dd8 : FAILURE in 21m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e659651e2f9b49b1a30ed93780c64167 : FAILURE in 12m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2ff1386981934de4ba3a333147462cad : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/829592ffd1b74b4294183d366b796b48 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb11087481ba49a28428b734c203e866 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faa27db1479346c1936518d2485fad4e : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8ca0b60212f446fa40b6afa3ee34eb3 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9601d9d8afa14fedaac356a9c2796dee : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5081c19b34c4405ca60c9100fabba3f3 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"fce8a45a6409a8b6fc6547220023f87b4cc96ddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-31 11:33:44.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a1bc4d2e5684be55117b33b86e84e6bfc0414609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 12:19:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d9d74040b30491f9976899d201624ce\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a5a72f52e1444336a7acaa429a5070c6 : SUCCESS in 11m 33s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/4800836e3b17499e986903b862ad1c0b : SUCCESS in 19m 21s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c3b3aedb821f4ce8a990634963073d4b : SUCCESS in 27m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8ac19a530dde4d49991e7a784ccca87f : SUCCESS in 22m 21s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4f05e26f7f09448aa24271dfd970c239 : SUCCESS in 35m 59s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da36b0f80a584a36b96cfc66b32b65ca : SUCCESS in 10m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1803697cf2444c5d98853b92bb097165 : SUCCESS in 13m 02s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/394de532dbe743208f14e799dc4ad154 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60db8ed43b2f4885a16fe8209f3f4bf3 : SUCCESS in 1m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45b8e6e37d5c4218aed0c4faa1266639 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eaa3e689060c46668754aab19489581d : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2593526935064168a9a9e6e7d89b99ec : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287aa3316f01481bad7bf1b5966cad48 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c622caef40bb420cbded79a2c5a993f3 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6392e103c70f66d05c3ef4ad81cb1d3c2a02ba2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 12:58:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1da64eb9f7264345a8cd85ab323de02e\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b7211e5466824760b561cc8b6de7e442 : SUCCESS in 29m 22s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/e29917566c0844b8bc8a3a2294528fb0 : FAILURE in 39m 48s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4a185b3e7014436497a504840ea86ddd : SUCCESS in 55m 22s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/34daf438d54f488bbfcc14a468a0c7cd : SUCCESS in 59m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39197c7208e74e1cadfa9dfa49fd79a2 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2b35ab7ddef468886ac9d7337cac3ad : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a85f69adc2b94b779215890f7bba9912 : SUCCESS in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92ea49da3b635df69bfacd0f9243b33c665b43ae","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-31 13:34:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e65c4aa2b76795a86122eb8aa7ef92afdf9e96f9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-03 06:16:58.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"31815ce9db5ccf53e45e7d3d667b71e8e165ec2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-03 06:37:55.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5f420ad76397fd7cd2b39d7c9642856ee43e9fd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-03 06:38:43.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"3a65295af1e45e1582abfe95a091b5712f234806","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 07:16:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71cb1830df84481d98511364d04e99e7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2459fb07d00b483593bfb6d83d077a71 : SUCCESS in 36m 39s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/beefd089d9374f958491b0fbda5564df : FAILURE in 27m 31s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/657a908802434cb3959c4642127fc231 : SUCCESS in 35m 19s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/87159720f1704c88a1bf69ec1dc7ec68 : SUCCESS in 29m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c7ebe795dd114188873995f08112ed78 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31914c4615fd47caa6a8253496b9edd5 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b7c78b225da4ed8950eabe02dd78929 : SUCCESS in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"86e2ee3e65071037d9613cbc277cafb7c20b4757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 07:48:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bf10769422e4c658c941283f9f3b59e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/33b469945e3848dbb660ddafcc3188e7 : SUCCESS in 14m 50s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/6f23e826420548219fd4e8068e046d53 : SUCCESS in 20m 29s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/925d65b46aff436a80eb9230948be3e1 : SUCCESS in 30m 41s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3a33550ec9434ebea92ba51ac286a1ff : SUCCESS in 25m 03s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/43970f4096b547469dd2c40406e0b4e8 : SUCCESS in 1h 01m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5597e1cf2d30416fbca7eb6e527dd1b6 : SUCCESS in 14m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ebe04f3a18e40d0b95d4f3d1ce85319 : SUCCESS in 22m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a6dbd79c58d34203b32420460472ce19 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2631d79965b4d22afcbef1104be6e14 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2086039ac7954590af1c99c1cb582167 : SUCCESS in 1m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ae7fc519fbb4c34921122d82282c6f3 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/533d9dd01c124057bf0d596e51e9675a : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43403513dc74d9490c291e46c09316c : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44892a779eb145fa93b054b81a3debdb : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":5},{"id":"7cf5908fd99ff67f799981243d259b76ecf1a53c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-04 10:46:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2fbbb02d81e5b4c2e69bf25b8cc7c9fc811a9221","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-04 12:52:35.000000000","message":"Patch Set 5: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"001704c8d2dd096bcf8b345881dc97786181a641","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-04 13:24:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1c3da288d07d7e445ef2688ccd4baae9476398bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 13:25:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"089cfcfae4dadd1660df5f0036aafaa0a9f2044f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 14:33:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16156c84c43f43eca306f1fe044665bd\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/f7d2f330828b48b8a6671da5179ff9c4 : SUCCESS in 21m 36s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/06d9e743a4554de59043c1410e5c6eb1 : SUCCESS in 27m 47s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3f212bb8f7094bc4a9242d2e04de547e : SUCCESS in 20m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8ea25fd871a34774aca9e1b2d0b9299f : SUCCESS in 56m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7d567e538e246e199b391b543f89137 : SUCCESS in 21m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2d7af3f42ceb40c3b1384d91fc337024 : SUCCESS in 26m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b71779d9b53a403abd63a9977de26bd8 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c7b5ebc8ba4ce59ec069873835abda : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f54a0b0127e44fcbd66d3a3c2362904 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32b4cc3c285b4a70a557b786cca07157 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d770b409f33e436bbc4ca1eebc1b3436 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5232ac0a14034ceab51c2a5c46d244d1 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":5},{"id":"8ac0fc8397f08145b0aee07fe63fd02eb63ae1a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 14:33:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fed4276fa65f5178f20ee364b0f1a40f2962e684","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 14:34:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc2f60ca854421ca14ea143954e1ecf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598ad6d7a3984f9485410ef0d959c522 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bcc2f3fc1c04f79bd87ca65c1286e4a : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5ad545cc725e1f072784fe4973ed6428963ea0fc","revisions":{"2e978cd361845674867730b550905f58310752ba":{"kind":"REWORK","_number":1,"created":"2025-10-31 10:08:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/965458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/965458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/965458/1"}}},"commit":{"parents":[{"commit":"40bcc4297802dd2f765d61496fb1ef814ff0b8c4","subject":"CI: Fix publish secret for arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40bcc4297802dd2f765d61496fb1ef814ff0b8c4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-31 10:08:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-31 10:08:19.000000000","tz":60},"subject":"Bump versions of packages in sources.py","message":"Bump versions of packages in sources.py\n\nChange-Id: I395cfdf22cd87a43294dbae561a8872cb0b48e1b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e978cd361845674867730b550905f58310752ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e978cd361845674867730b550905f58310752ba"}]},"branch":"refs/heads/master"},"92310ea4f588c6fad546e0cef655a07979989a0b":{"kind":"REWORK","_number":2,"created":"2025-10-31 10:15:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/965458/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/965458/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/965458/2"}}},"commit":{"parents":[{"commit":"40bcc4297802dd2f765d61496fb1ef814ff0b8c4","subject":"CI: Fix publish secret for arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40bcc4297802dd2f765d61496fb1ef814ff0b8c4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-31 10:08:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-31 10:14:58.000000000","tz":60},"subject":"Bump versions of packages in sources.py","message":"Bump versions of packages in sources.py\n\nChange-Id: I395cfdf22cd87a43294dbae561a8872cb0b48e1b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92310ea4f588c6fad546e0cef655a07979989a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92310ea4f588c6fad546e0cef655a07979989a0b"}]},"branch":"refs/heads/master"},"e1d312764df8ffbf6b5023f56c97f445d2106557":{"kind":"REWORK","_number":3,"created":"2025-10-31 11:33:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/965458/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/965458/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/965458/3"}}},"commit":{"parents":[{"commit":"40bcc4297802dd2f765d61496fb1ef814ff0b8c4","subject":"CI: Fix publish secret for arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40bcc4297802dd2f765d61496fb1ef814ff0b8c4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-31 10:08:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-31 11:33:44.000000000","tz":0},"subject":"Bump versions of packages in sources.py","message":"Bump versions of packages in sources.py\n\nChange-Id: I395cfdf22cd87a43294dbae561a8872cb0b48e1b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1d312764df8ffbf6b5023f56c97f445d2106557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1d312764df8ffbf6b5023f56c97f445d2106557"}]},"branch":"refs/heads/master"},"bab4458c59405b4cf42e6d6936fc6911df7b9061":{"kind":"REWORK","_number":4,"created":"2025-11-03 06:37:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/965458/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/965458/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/965458/4"}}},"commit":{"parents":[{"commit":"40bcc4297802dd2f765d61496fb1ef814ff0b8c4","subject":"CI: Fix publish secret for arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40bcc4297802dd2f765d61496fb1ef814ff0b8c4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-31 10:08:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-03 06:37:55.000000000","tz":0},"subject":"Bump versions of packages in sources.py","message":"Bump versions of packages in sources.py\n\nChange-Id: I395cfdf22cd87a43294dbae561a8872cb0b48e1b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bab4458c59405b4cf42e6d6936fc6911df7b9061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bab4458c59405b4cf42e6d6936fc6911df7b9061"}]},"branch":"refs/heads/master"},"5ad545cc725e1f072784fe4973ed6428963ea0fc":{"kind":"REWORK","_number":5,"created":"2025-11-03 06:38:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/965458/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/965458/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/965458/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/965458/5"}}},"commit":{"parents":[{"commit":"40bcc4297802dd2f765d61496fb1ef814ff0b8c4","subject":"CI: Fix publish secret for arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40bcc4297802dd2f765d61496fb1ef814ff0b8c4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-31 10:08:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-03 06:38:41.000000000","tz":0},"subject":"Bump versions of packages in sources.py","message":"Bump versions of packages in sources.py\n\nChange-Id: I395cfdf22cd87a43294dbae561a8872cb0b48e1b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ad545cc725e1f072784fe4973ed6428963ea0fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ad545cc725e1f072784fe4973ed6428963ea0fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
