)]}'
{"id":"openstack%2Fkolla~966017","triplet_id":"openstack%2Fkolla~master~Id6684e4356f68e4348c0f8c1c0c1e5397553397e","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-18 13:44:08.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-18 14:40:00.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-03-25 12:05:55.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-03-25 12:05:55.000000000","reason":"Change was submitted"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2026-03-25 12:05:55.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-25 12:05:55.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-03-25 12:05:55.000000000","reason":"Change was submitted"},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2026-03-25 12:05:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id6684e4356f68e4348c0f8c1c0c1e5397553397e","subject":"rabbitmq: Update to 4.2","status":"MERGED","created":"2025-11-04 09:59:19.000000000","updated":"2026-03-25 12:06:56.000000000","submitted":"2026-03-25 12:05:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966017","meta_rev_id":"1d29625acf7bd091798eaf830cf011732f0484b5","_number":966017,"virtual_id_number":966017,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-25 12:05:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-18 14:40:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-03-17 09:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-03-17 09:33:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2026-03-17 09:33:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":1,"date":"2026-03-17 09:33:08.000000000","_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-03-18 13:44:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-04 10:32:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-04 11:38:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-04 13:01:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-25 10:16:59.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-11-25 15:41:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-10 11:33:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-02-10 14:31:29.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2026-02-11 08:10:05.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-03-17 09:32:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"2bbd8ab487a217a6f5e6b8b69be87ea9aa10af53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-11-04 09:59:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fa3f0115e106b06967bad55a35c92bc6529fe24","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-04 10:32:16.000000000","message":"Patch Set 1: Review-Priority-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d94e83d09805f01c66134abeda2a49306709f07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 11:38:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a6ba2775ce24fd18b43d8fcbed26b50\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/eaa9127ba3be4b03a1dafd6909d84aa7 : SUCCESS in 45m 03s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/cbf840309c3f4f4ebff238b95b2651bb : FAILURE in 37m 52s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/3deed15e21a745ba88f59b497bebd16a : SUCCESS in 38m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/c8bb56f4b153411fba8212ad5207499c : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8875a6037f5542e0982b9105b7f941ec : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0da4650dd37b4ccfbe1a03c062e2fa5d : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bdb3caf66bb040aab47c7e4007123d0c : SUCCESS in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf77be114cb598b4ce749c484118b2efa7559787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 13:01:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b159c5a858a143d589f8577491183de1\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/5091e426515d444d8a3f2b7a611633df : SUCCESS in 23m 12s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/257d3abef5444987957889b1532a9127 : SUCCESS in 22m 05s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/479dbd092f85460f896547bef5c352a6 : SUCCESS in 29m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b3ad91dcf2f0425a87837b9909a15dd1 : SUCCESS in 14m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b5f518b1b9ce48ffb764156b34e066d0 : SUCCESS in 40m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a1f376b20ee4fd79364df5b83d10dc4 : SUCCESS in 20m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd1ae469ca90456abee7987dcc0002e7 : SUCCESS in 25m 36s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/780e8a1fae77443281adc0c6ded4d65a : SUCCESS in 1h 27m 37s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4cd58465deca44dbadb2fe570ec1debf : FAILURE in 3h 00m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d1855ccf94c1496195bec819e1528ac0 : SUCCESS in 1h 32m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8d51965205a54829a4082134cd251151 : FAILURE in 1h 04m 15s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/98ab8e2478c646eb8fb77dbc5316d6d0 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd30b4b2e314ac882a7da03647975e0 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0efe4c02fee64ee79b33dbb2f82e4447 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4171bf132fca40b38bc514bd08bdab62 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/099a189a9f3a4d6ba282059cd0180827 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07969f299c8b4b71ab79d2dea59a6541 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d9a897019f64312acf02c6cab14a6c6 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bed150dddf917d820b3b4991f57fe6fa5ac05730","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-25 10:16:59.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba5b0c0ea89a3ea6a93b9eb8419296d620959e29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-10 11:32:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_37306\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"_revision_number":2},{"id":"cff64a6b3838f2b184fc9fcc44b8edd3ad1896a2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-10 11:33:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7bde4eb28b7415152baf39a263bd4dfd170c09fb","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-02-10 11:47:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d2b4085581274bcfe9ecefae8f05c8dfef8b4ea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-10 12:05:30.000000000","message":"Patch Set 2: -Review-Priority","accounts_in_message":[],"_revision_number":2},{"id":"d18f7cc3c55ce1042e3af254ad4d10ff2d1bee72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 13:28:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e58cdfedf763489f9e3511c619a590ee\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e188282ba3a4483bab867844a331c982 : SUCCESS in 1h 52m 41s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/ca65233d094e4da79eee8eab64c5cb9f : SUCCESS in 1h 11m 13s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b2ff05c599194729b77aea7345395982 : SUCCESS in 1h 53m 43s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f9f278396b8c40428330da67c8522934 : FAILURE in 56m 03s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ea6c6ee6a2bc41e0ba6ab07725cd1071 : FAILURE in 1h 53m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6b19b358c7841d684a0d5f6ab2790d3 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2fb29d8547c34cdc9a7588fbdb1e1029 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ad33a89644b439b923fc279179a17fc : RETRY_LIMIT in 2m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"830068b6b9037a8281a34534a558730f1da164bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 13:52:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c25b53b6187140fc9b4a29c91eeab055\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7d51e4891fbd47aab29672b2b0e957de : SUCCESS in 49m 57s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/bc0cb05dcae74cc39417ea61546d0350 : SUCCESS in 46m 04s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ce891126a6ce4c7d81f1d9b7773f566c : SUCCESS in 11m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0b6645b2e4274f82ae3c4130d61800ce : SUCCESS in 42m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/62ff65a5ce754579889756350f48aa06 : SUCCESS in 35m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5baa071155e94f24b523462e4d8f3d64 : SUCCESS in 1h 00m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45d30024c18d41f6af8ef9f00556e0ce : SUCCESS in 29m 51s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/8a74de03087c4228aae907dc7d1ba1a0 : FAILURE in 51m 01s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/cc1ab4b42a5841e09576b69a712a40dd : SUCCESS in 54m 50s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/77289070264f4816981086a151f4154d : SUCCESS in 1h 21m 57s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ba31fb4042804a318158a159a9d1dc3c : SUCCESS in 1h 00m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ad13c7595ea84b40af2fc6788444432f : SUCCESS in 1h 31m 52s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef250a0427f34bdcb07e1da20056acd4 : SUCCESS in 58m 13s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a911a2bf057247308fc13fe8a9cada93 : SUCCESS in 1h 30m 20s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9fcaba6d457c43099e9f9df0331cfa34 : SUCCESS in 1h 37m 06s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/93e125d0567f4261a56ee305a60cfc7c : SUCCESS in 2h 18m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/bee308b89e8c4d04852decdcd78ac24b : SUCCESS in 1h 28m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/18bea3aa9ef24d08bbe20fe1569c64ca : FAILURE in 1h 34m 24s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8bd4bd1a432a4759a828e1e0d66f7680 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bf4464ca00047c28445f61e80e1e7a4 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f84d9a9945c442ba922523dc719cdc70 : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f84aa19bf1444a33a721988cbce0ecc7 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3fcc5824bf4498e90afffb86ae1faa5 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66dfef697bc448cbbff2a72b1d5a31aa : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1962f3f51f9544188db80cd21d7115d4 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"981e019c685a1daba85e09a41c49978abf5f6c0f","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-02-10 14:31:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c67cc19b7182c297e1e713e2853d98255994e8e5","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-11 08:10:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"21b97e250e31c07066cedc5754485120ff80844c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-11 09:13:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ad73fc26ce67e88e116dff09eb91d22d5dd535e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-12 13:48:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_37306\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"_revision_number":3},{"id":"17c8919c3a1ca2905c4537ece6eac6a80475737b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-12 13:49:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cc215c7f4c21fe15950942a07c5a8e2e38910c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 16:05:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb1a7ae497eb430bb1e2cb54d5809652\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/938c95607bc940929ec531261cf1c8af : SUCCESS in 43m 23s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4328ac22f7884c0c8895adc7dd250c4f : SUCCESS in 34m 25s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a7f6c346f20e476b8a4e7deb5b1e13e1 : SUCCESS in 28m 30s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9cf39312ab6840e799c043ead19a3445 : SUCCESS in 38m 31s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/935fbcc89cb0429bb2e2af60c03e8335 : SUCCESS in 38m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c28f163418d49679e7fc13f3785164a : SUCCESS in 34m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0353ecc642f849ddad8e3fcbfbfac32e : SUCCESS in 32m 15s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/cd3082cef1784320b8651fbc99b1c481 : FAILURE in 59m 41s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/938bb9ce9e5d4f798385b7fbbb5203cd : SUCCESS in 1h 06m 50s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/6f30c1fe1ca84737ad02ec70492e3dc0 : SUCCESS in 1h 24m 53s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/260812c5205347349091e84e033fc639 : SUCCESS in 1h 10m 44s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d0849a835fc44647b48a84aa4d5496f5 : FAILURE in 48m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d5fd4d3196d4f36a862421e4e0fa8cb : SUCCESS in 1h 05m 09s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c24d8dec98ed476da42d88bc7cc164f1 : SUCCESS in 1h 20m 20s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e7e72c9680b5449ea8d76f6e64c30811 : SUCCESS in 1h 30m 24s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ca0627c136c5407da0148eb0df8ac2b2 : SUCCESS in 2h 09m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/4cc85e79ac3745238aded8ea0284f5c8 : SUCCESS in 1h 31m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/70ab9114cd6344b8beaa704389876f02 : SUCCESS in 2h 10m 16s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/94f1f337d7564e2ab12be82426ecabf9 : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91864e1258eb49a8948c46dcbb3e2fca : SUCCESS in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3d86a0339004cbf8a8bf4cc551d2839 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cece2a8ae5844cbb98a3afeab96b639 : SUCCESS in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/40f90b89aa25420486f6af31060f4bb8 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26bb29781fbd449594aa68fffdc6e251 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce08c6520f964ecba1d4a07192441f5e : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d180ec5f225e46a7999561fe152c7250 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c32beb10020d9174ef0950301c5b7d7802ba8db6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-12 17:37:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bf78bb1a29d5f2dc6e20fb520b7e615e37df50c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 17:37:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e0959c8239b4a6388b4a9740004cdca\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1a6fa423594f40ad83855eaaad2ea3c9 : SUCCESS in 2h 36m 46s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1a40b76bddff46208cb200a496f6fbaf : SUCCESS in 2h 50m 31s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/cd87d12565ad4b4c95a38e144be13ff6 : SUCCESS in 2h 35m 37s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8bfa36e05eed4d06b50459ee0564a50b : SUCCESS in 2h 34m 40s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/1eece5c7da954c3a984f837096616c8c : FAILURE in 2h 13m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb3a6365aa0b4730b50fe901603a7077 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0d3089220134c81bee84749cf1fa9c4 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a50c057f3bc24cf997fb2cb9d32f0282 : SUCCESS in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"015913af35fa8a00285834bbfe8dcbd402352796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 20:02:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b00b15a94b94cbd835cc88a7526b682\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0a2884fd15814aebb24c85568d298354 : SUCCESS in 38m 23s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d6e899cf1a3a41b092b898147c9f70ba : SUCCESS in 45m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/228900cd36ae4d1c83e7682b6e80484b : SUCCESS in 27m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9a3807b289c54fdc933c406560a306d2 : SUCCESS in 41m 31s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/107eca706d014d49ab535d54171093c9 : SUCCESS in 1h 08m 08s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14c28ddbe923486f85fc80877495fa3b : SUCCESS in 54m 28s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a287d42379af443f897277790f1b5870 : SUCCESS in 26m 54s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/b2a170b8121d43a19f96157fe6a1b246 : FAILURE in 46m 39s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/037dac22139344e6a5a74eff204f52e6 : SUCCESS in 56m 29s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/55f6adaeceae4a5186cc5b51c47c1b12 : SUCCESS in 1h 13m 08s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c0290446726947a9a0e916ab509cfcb6 : SUCCESS in 1h 03m 50s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fbb59e8092544da8b8c2667e4d8a7ed5 : SUCCESS in 1h 22m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0d28cc3095f4706a1e8e231607f38c6 : SUCCESS in 57m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3d851705d4a84389841edb61beb2e96e : SUCCESS in 1h 21m 09s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/3c4b073170fe4e4aaf62f1c9f9ab3f23 : SUCCESS in 1h 36m 54s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7696013070274e7ca3aba8e9584d0ced : SUCCESS in 2h 13m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/c495f3bb08b546ae9d8649c791be6914 : SUCCESS in 1h 26m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4eccff7f90d04137a1401351815277ba : SUCCESS in 2h 11m 06s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c5d8337e2d6e46e19caab2bb62e3de6c : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52506138664a4a3d9c49f6262cf8b773 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcdc19ac1e9d482993edba9823562c04 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa8d47a0cb97450f8ec0a1e4375d97d5 : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59bb0e2746324a8e8e2fa9e7437a0f16 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1e7106ad21d4f429f1a9e895437e3f8 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd6b0d7dc8441edb632449b2fb58d6c : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8df1c6a8ded84a92b7601eda426315e0 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"34764ae7d5d9028ae9dba8ebaa5328676898ebcf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-12 20:20:12.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"7cf9144242feb5997906987e24a6a656f8ca6968","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-17 09:32:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9008875cc9bd5e2840d5c55a1db4770527fb666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-17 09:33:08.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5953c1662a91fc37034df77e64e50783b717b02e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-17 09:33:37.000000000","message":"Patch Set 4: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"68704d74ff4304534b88db9e431b2f2b8fab6613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 11:41:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97c37c1c979545998557213b32f56c7e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/50cb61f2ecef4cf593a2e9f8c73eca85 : SUCCESS in 48m 25s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/1fd725f678cc44818a4115d6cc2fb34b : SUCCESS in 41m 32s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/453ab6f2416445629219b361e2d3b984 : SUCCESS in 27m 30s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a816f68aaf344e1db57d1fa9d9aa9f58 : SUCCESS in 51m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a9d08c24acc54e9ebfa0c7bf43f244b2 : SUCCESS in 43m 31s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5b3dba0ee38426f887f9766cf762204 : SUCCESS in 52m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7c025d2c9cff496e86cc61d522229f08 : SUCCESS in 27m 57s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2eb7d0b287694772975127323bc1e99c : FAILURE in 49m 35s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e75eff22d677441a8d8e65a351812ba0 : SUCCESS in 57m 38s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/95e5b7705dbc48b281787e17f88e1817 : SUCCESS in 1h 13m 25s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/57cae025cb9447948a8999ca2288f550 : SUCCESS in 1h 02m 24s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/97cfb6032d04487696cf21159d8a1d39 : SUCCESS in 1h 23m 48s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0e70be5b75247e791e8b988f886884e : FAILURE in 1h 30m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/593e9d7375a0469b927039512538bfc3 : SUCCESS in 1h 22m 48s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e8a679ec50434c8e9c8c11b7b6e89ed8 : SUCCESS in 1h 35m 02s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ab4ddc9b41454a14afe2d0696b076519 : SUCCESS in 2h 06m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/1060c951a2974995b1a4a479d3c06148 : SUCCESS in 1h 32m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/31f0abda08b04d5f949399b71c02deb8 : SUCCESS in 2h 05m 00s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe2b523a18d2459a961e1e8ff24d3109 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d1c0b514774165ac274ac201875d05 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d388aaf056c46798481c58f282e580b : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/779be48877a644c4abea6ba0ca09b70a : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbf59f49738a46e3983cc2207bec16f2 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f65e76a180d4408b6602f44575703ec : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99d290328664407bbfef28e4b18c250 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31ac484a565241028628bbf8b52befef : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":4},{"id":"5ba8504f607a64ef29dde1ce79ec5f8cac3b6502","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 12:44:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1763d94f9ed4574af2af18bbdb35579\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/725d063feb374001a709306da5337619 : SUCCESS in 57m 33s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e49ec17a8e6d44b382728c692c2396be : SUCCESS in 1h 10m 34s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0e4e2bf2032544439f5a59d0d464f3af : SUCCESS in 2h 25m 29s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/3d896fed5d9e49439419ca8f6fef5e93 : SUCCESS in 1h 02m 15s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/9c2a2933c0f64bf18915cf4c464b49a0 : FAILURE in 1h 09m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f6cfe55256034cbd983b203c51720931 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3b9fe8bf562f40d2a0d4b4785d3efd15 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/65634b6d947343d18d2766d31ab0cd7d : SUCCESS in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3f1daf79657a78926927821839d7a01639ccdf1c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-18 13:44:08.000000000","message":"Patch Set 4: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e75222d4c2512494821ecdff03d75a0536facbab","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-18 14:40:00.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"23a6ea40fa075be683ea6c7b6781258e02b107a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:03:43.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d1d76ceede74043a82638f352929254\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a1b2770acce24801904f0551e805c006 : SUCCESS in 39m 41s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/964b325a64c0483384a101f19c74cf29 : SUCCESS in 41m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7676abdfcf41499f9e2c452e80224070 : SUCCESS in 14m 47s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/13f52c9fcd3443eb8a4d3001cebf1628 : SUCCESS in 39m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f4f69deef07b414aa1d454ad34f630b3 : SUCCESS in 38m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aa40d997ccb48909a31c8cec7965236 : SUCCESS in 35m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/47d7bd81cf48468389740d05cabbac8c : SUCCESS in 30m 44s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/de57a7499e534f74864c4503cf2d654e : FAILURE in 1h 04m 09s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ae04963c208048d189eec75bae285db1 : SUCCESS in 54m 00s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/0d9fc5d3fa8948dc9137e73cfeff68df : FAILURE in 1h 22m 47s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/5ec8e77e191f4c25ad34ca7d58cf82ab : SUCCESS in 1h 01m 06s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7e20dc2bb03546d9a3393829a2383565 : SUCCESS in 1h 25m 16s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18fa48939aad40c1a9ede6536a08d27d : SUCCESS in 1h 00m 55s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7c2f108af71b47df918c70ffbe3af44d : SUCCESS in 1h 23m 03s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/0f3b65212dbd4053b2228e42f79463ca : SUCCESS in 1h 41m 29s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a6908a18836044a191d3b0af3a0c8824 : SUCCESS in 2h 02m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/413be77b8ef240b992bb0005e193e54d : SUCCESS in 1h 41m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/febed8e2996b4cd18576fdbccfe116c6 : SUCCESS in 2h 09m 37s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/90dca60388a848bfb5325f789de59152 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272a79c1d39343f295a0d1262b1234e2 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a75d3740bf204dc6ac86d5ebe1bcef9d : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85a9d09dcb1d474cb6256ca18b5d6bbe : SUCCESS in 5m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a08e1a4f8c041418094086fb52414fd : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5379f7c6cd0044c3814099d84103d0cc : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fcc377234e04cdd87dd0cb2130e5f7d : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29e5d7964dbd409d9b414038cf43b0b7 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":4},{"id":"2cd6aa63b3020f61a2f141cb82e2d06a0098b7ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:36:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36604b1b59ea4d2d9b6aabbb9d01153f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/677ba0f19b0447f8b6e0d43e21ece069 : SUCCESS in 53m 41s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3c19467a11cf42f08056373a5c411787 : SUCCESS in 2h 14m 27s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b528807ef46f4e5f8c507a427826f7c2 : FAILURE in 1h 42m 52s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/52cc1e2fc7cc412d9f247a612f031f23 : SUCCESS in 1h 20m 35s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/9b78921c78cb4d80b8426b0895505c9c : FAILURE in 1h 47m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/efe29bcae642427f8d675eeb72b0b0b4 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52a63138b4154554adb3e1b4893b7eea : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d409f195cf37405586ef7d913523f70f : SUCCESS in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"845da2193aed99a77a214489c0ed1efd626f049b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-24 15:32:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"308a13dfe915143043f5bdf46bcf604dd8f34d05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 18:04:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4af26cf75bf2436685e9bbc8d65e9b1a\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d33ad00000bb4dbaa54688f753887c3c : SUCCESS in 58m 31s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/c25d661c7e9a45009c4f8f4c2c4a0d72 : SUCCESS in 51m 39s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8070dffb39e7488c81ff4319c4544d65 : SUCCESS in 46m 32s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/5f4fee7458a3441c93f1f4c1ab95f7f0 : SUCCESS in 1h 15m 58s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/def37d90ed25475abc093d83c95e7ef5 : FAILURE in 1h 09m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e75ce1fe068f4f8f83fa7347d68e6b5a : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0ec7446d39034720a4557fe477cfc497 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/019d19c7ae80457a9b56976cf5d6aa7b : SUCCESS in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c4c1c4b7a2da0055bc3adfc81476aafc7af652d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 18:27:26.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5506178f787741e18f9a6d628f022b7b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8f33c7e7c728499c8c5d33dd57e9efd9 : SUCCESS in 44m 38s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f1ebac6d5cf5406da1ec7867d77f865d : SUCCESS in 42m 37s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/197cee0dfbd8404fb19d3d8ea6567468 : SUCCESS in 28m 08s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/923945e1739c4ea8b9f868482c8aed7f : SUCCESS in 44m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/37a5e822ea8c4542a04bfc776271e311 : SUCCESS in 50m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/220a6ea49f8d4c32aeb3e917e4c139a7 : SUCCESS in 53m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/56f3181df40a4e22aa0c72ca8e7e9f34 : SUCCESS in 34m 34s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/f48be48825294775ad465aa2101e019c : SUCCESS in 1h 10m 49s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8f51d651a04b4466bbc334661ab5fef7 : SUCCESS in 1h 11m 21s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/5f9534649fd0440f8960d1b078f29f03 : SUCCESS in 1h 33m 27s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/9960cd0d079a42e8bfbaf9eb58e26375 : SUCCESS in 1h 21m 11s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c3ad4dec727844cea3440c5ad6c250c1 : SUCCESS in 1h 47m 58s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8f12574f80e4a0c9d5263479f3a38f8 : FAILURE in 2h 27m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d7bec8a5286943ccbf5450628d569037 : SUCCESS in 1h 42m 59s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/d1273890163544dab6d121e79a3db9a4 : SUCCESS in 1h 52m 37s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f405bab7935e4c019cc35ed3d799c858 : FAILURE in 1h 47m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d35ab010af044f49b66bba31600aa1dc : SUCCESS in 1h 50m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/66f11bc029884cfe8bc5493622a42b2f : SUCCESS in 2h 30m 08s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1f6f9fceb74d4241afcda8c47d6113dc : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c5741b72c74db5b4751deeec2a34a0 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4b3961ff2e64242b88c09e32feaf8dd : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58d57f5f04c84254b2459dfd724726b2 : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f3a851b6384c4c94b1e41b8aa447885e : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7fa8a755cce405db7be91dc76b9fac7 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51b319134b06491a82b58f57c2d325fc : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b1ac78184674992a2fbdd3f3e096bf4 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":4},{"id":"e3f85122dce3e6bab525823e9238d2edb9e71b06","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 05:34:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"17eecc10d77888d029f50678715ac34cfad3d935","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 07:04:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6da0165515f147a1b5f388567b8f1a11\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7afa8d4abc4c4981b69eb1111de53329 : SUCCESS in 1h 10m 46s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f7a134a6dd564acebd05f4988e210b0a : SUCCESS in 1h 13m 09s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/13264e11144a4b778d84c27143b99a60 : SUCCESS in 1h 19m 10s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1ec2999eb89d49e79a1d59773ab9e558 : SUCCESS in 1h 11m 32s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a6d9f7b7af2a4fec85ee03e79c8cd874 : FAILURE in 1h 27m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38c4c979e8dc4c2d95d0efec5b7d38d9 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/53e35b097dae4080a6137294d21f5212 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dcf4b0bcbef14eadab239ad34687ea08 : SUCCESS in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4cb34277bf7fd19d73f28b06ce104e4b74bac90e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 08:20:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6ec53daf0c435bb3eeaaa445e109f2\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4286cfa4e7cf42db875757cb057ffb8b : SUCCESS in 47m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/cee96c78cd6d44b187d6a0ca482f7cf3 : SUCCESS in 38m 51s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/521c5a0efe18415ca466de13cb2a0c43 : SUCCESS in 26m 08s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/41e7f85554454d568d1f4d21a9b45999 : SUCCESS in 40m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4e8ba2dcaaa54e5399768b816a056b20 : SUCCESS in 1h 04m 19s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d84eb2002d1a4b81b553c5aee4dc8561 : SUCCESS in 50m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb8b4c087c6f4a8a89939366e7a365cf : SUCCESS in 29m 19s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/961a024638c7460f8e7634e1566e64bc : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/74e23eead09e4802911ffc93d2c5400e : SUCCESS in 1h 11m 15s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/7cc55985750f4264ad2d6b87daf35e59 : SUCCESS in 1h 43m 26s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/62c3c7730cde48589f54bc4743e3e35b : SUCCESS in 1h 18m 31s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ea0c7c1fd4094942a74a133d104d734e : SUCCESS in 1h 55m 07s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d7cc2c0391b411dbc81e7ceea2933e7 : SUCCESS in 1h 19m 48s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9d0e0936381142968ecea65eca445253 : SUCCESS in 1h 56m 42s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/f6efda1296bc44e8ba2fd3691ae70a37 : SUCCESS in 1h 57m 00s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a1d3019e73344f18b41adc1e5c8993b2 : SUCCESS in 2h 43m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/efbe6db2970d4f5a9e39aea954c8bd9f : SUCCESS in 2h 02m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/adf42984662d4219b469df77daad0ff6 : SUCCESS in 2h 43m 08s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9ff7c5b519d64e0ea99efb49a4c1e7fe : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12488a798353498381ef333b91371156 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79704fcb24cb44f5bd67a6e133dc2eb7 : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35fc2666eac94fc39e95125d80d35f26 : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9895733a6d324a35bacca19c42d3314e : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0853c6611644426d842fd954ab915b3e : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa52a68305ed42d2bfb85c75ea6e0472 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ab6738f88d74559a19d702c91340f5b : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":4},{"id":"1b0542e28944dd967403d1ac3c23597124fa00ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 08:20:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3d3a1126527d07f954fe803f94f4334429d37626","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:05:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1673183c56c84a1ba1123b937ea5c6fa\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7042c2a7c1f24ca6a8da9673d473db76 : SUCCESS in 53m 56s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f6f9398a1ad44ff0adc8850fd8512c86 : SUCCESS in 28m 20s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e905c164c8ba48398cf196e58f81d417 : SUCCESS in 49m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/79fc37b304754e0f83c60f0f5ae4ecaa : SUCCESS in 41m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53a029976dc6457fa22b0f322670bc71 : SUCCESS in 49m 18s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ec78e0de225e48a3844bf79dd0d9a3a6 : SUCCESS in 26m 13s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3c7d486eaa3d4cea869668535085c3c9 : SUCCESS in 1h 29m 44s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b5dcf57949544cbb865873c323271824 : SUCCESS in 1h 35m 14s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/bdb39c7a651f4e34abb4a2ba13516e74 : SUCCESS in 1h 35m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/72ace757897f46dbbe8e757461237232 : SUCCESS in 2h 08m 28s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4027a7db2999458ca3df7b58030a818c : SUCCESS in 1h 43m 27s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/850a5e747292413490af2d8c261f4a15 : SUCCESS in 2h 24m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b80c6499614420ea43eb0c989be65d8 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc0666d826c433fac6703ce5456cee8 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/405bdf86c48445ee8b5caa97d0911fe1 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae20428cbf044162bfe108689e95f86b : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3472abd295a9474ca183e8c56f4de955 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aae80af8a034955a0b49bb8e30cb917 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":4},{"id":"bbf017a6c98365c073d2c8730b5a3323de706f1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:05:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1d29625acf7bd091798eaf830cf011732f0484b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:06:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/850bd9b0741845148abfc6870a849572\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f36cd4934a4483ba11b9b593555f3fd : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2863732e7f5e42d49bbcd7b9f69938ef : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1b14eb8df7f1a34edd41e945df33ffb8f1c56b79","revisions":{"f6d38b24c003e50e3403024950a061ae20ab7d86":{"kind":"REWORK","_number":1,"created":"2025-11-04 09:59:19.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/17/966017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/966017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/966017/1"}}},"commit":{"parents":[{"commit":"40bcc4297802dd2f765d61496fb1ef814ff0b8c4","subject":"CI: Fix publish secret for arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40bcc4297802dd2f765d61496fb1ef814ff0b8c4"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 09:55:19.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 09:58:54.000000000","tz":60},"subject":"rabbitmq: Update to 4.2","message":"rabbitmq: Update to 4.2\n\nChange-Id: Id6684e4356f68e4348c0f8c1c0c1e5397553397e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6d38b24c003e50e3403024950a061ae20ab7d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6d38b24c003e50e3403024950a061ae20ab7d86"}]},"branch":"refs/heads/master"},"76bce398ee3730f49c1020e38f4b5795e2e89381":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-10 11:32:45.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/966017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/966017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/966017/2"}}},"commit":{"parents":[{"commit":"a198d189d05e920a157efc225a8a370b548af5a1","subject":"Merge \"Remove special handling for python3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a198d189d05e920a157efc225a8a370b548af5a1"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 09:55:19.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-02-10 11:32:45.000000000","tz":0},"subject":"rabbitmq: Update to 4.2","message":"rabbitmq: Update to 4.2\n\nChange-Id: Id6684e4356f68e4348c0f8c1c0c1e5397553397e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76bce398ee3730f49c1020e38f4b5795e2e89381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76bce398ee3730f49c1020e38f4b5795e2e89381"}]},"branch":"refs/heads/master","description":"Rebase"},"4ba88f775440aa22f0c104eb454ac941a1515acc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-12 13:48:59.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/966017/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/966017/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/966017/3"}}},"commit":{"parents":[{"commit":"761b0c0beb564d81fcbd420099ea7512e83bb789","subject":"Fix MariaDB version number","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/761b0c0beb564d81fcbd420099ea7512e83bb789"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 09:55:19.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-12 13:48:59.000000000","tz":0},"subject":"rabbitmq: Update to 4.2","message":"rabbitmq: Update to 4.2\n\nChange-Id: Id6684e4356f68e4348c0f8c1c0c1e5397553397e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ba88f775440aa22f0c104eb454ac941a1515acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ba88f775440aa22f0c104eb454ac941a1515acc"}]},"branch":"refs/heads/master","description":"Rebase"},"1b14eb8df7f1a34edd41e945df33ffb8f1c56b79":{"kind":"REWORK","_number":4,"created":"2026-03-17 09:33:08.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/17/966017/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/966017/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/966017/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/966017/4"}}},"commit":{"parents":[{"commit":"761b0c0beb564d81fcbd420099ea7512e83bb789","subject":"Fix MariaDB version number","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/761b0c0beb564d81fcbd420099ea7512e83bb789"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-11-04 09:55:19.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-03-17 09:33:01.000000000","tz":60},"subject":"rabbitmq: Update to 4.2","message":"rabbitmq: Update to 4.2\n\nChange-Id: Id6684e4356f68e4348c0f8c1c0c1e5397553397e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b14eb8df7f1a34edd41e945df33ffb8f1c56b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b14eb8df7f1a34edd41e945df33ffb8f1c56b79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
