)]}'
{"id":"openstack%2Fkolla~967299","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I46690597b3a6536cee14e5a76ba963b5fa9dfd34","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-17 11:33:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-17 10:40:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I46690597b3a6536cee14e5a76ba963b5fa9dfd34","subject":"bifrost: Mark as unbuildable","status":"MERGED","created":"2025-11-17 09:01:14.000000000","updated":"2025-11-17 13:14:29.000000000","submitted":"2025-11-17 13:12:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967299","meta_rev_id":"c028e02fcde68a52987e60e348d73ce3f78f3790","_number":967299,"virtual_id_number":967299,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 13:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-17 11:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-11-17 10:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-17 10:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-17 11:33:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-17 09:13:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-17 09:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-17 09:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-17 10:49:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"a9a0f62bebc8d5e0c05896bf28d15df3e6bedd8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-17 09:01:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e07942bd154e96dd4871ee84d915e25b23a53d2e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-17 09:13:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1eae34026e4285ffaa6c9b9ba3b694ab23720d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-17 09:22:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"17c464fa70b3fb98f5d2904796691cb7075eaf2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 09:23:00.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.1.  The\nproblem was:\n\n  expected str @ data[\u0027templates\u0027][0]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs:\n            voting: false\n        - openstack-python3-jobs-arm64\n        - openstack-cover-jobs\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/kolla/.zuul.d/project.yaml@stable/2024.1\", line 2","accounts_in_message":[],"_revision_number":2},{"id":"9352006d3bca173af61bebf3cbd14f6570df1b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 09:23:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.1.  The\nproblem was:\n\n  expected str @ data[\u0027templates\u0027][0]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs:\n            voting: false\n        - openstack-python3-jobs-arm64\n        - openstack-cover-jobs\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/kolla/.zuul.d/project.yaml@stable/2024.1\", line 2","accounts_in_message":[],"_revision_number":2},{"id":"6c5ba2e66e8e27267a292750c17d087f8afc30e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-17 10:09:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"44dcf5b919b1274a27b1cf98baeff9e8dae59595","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 10:09:40.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.1.  The\nproblem was:\n\n  Project template openstack-python3-jobs is already defined\n\nThe problem appears in the the \"openstack-python3-jobs\" project-template stanza:\n\n  project-template:\n      name: openstack-python3-jobs\n      # NOTE(gmann): We are using this generic template for python version\n      # testing. When a new development cycle starts, we need to do two things:\n      # 1. Update this template for the python version jobs as per the\n      #    new development cycle tetsing runtime defined by TC\n      #    https://governance.openstack.org/tc/reference/runtimes/\n      #    Use the \u0027branches\u0027 variant to run the applicable python version\n      #    jobs.\n  ...\n\n  in \"openstack/kolla/.zuul.d/python3-jobs.yaml@stable/2024.1\", line 2","accounts_in_message":[],"_revision_number":3},{"id":"69c071c8c8f5bd4c52bb0b5fe026a548a765e451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 10:09:57.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.1.  The\nproblem was:\n\n  Project template openstack-python3-jobs is already defined\n\nThe problem appears in the the \"openstack-python3-jobs\" project-template stanza:\n\n  project-template:\n      name: openstack-python3-jobs\n      # NOTE(gmann): We are using this generic template for python version\n      # testing. When a new development cycle starts, we need to do two things:\n      # 1. Update this template for the python version jobs as per the\n      #    new development cycle tetsing runtime defined by TC\n      #    https://governance.openstack.org/tc/reference/runtimes/\n      #    Use the \u0027branches\u0027 variant to run the applicable python version\n      #    jobs.\n  ...\n\n  in \"openstack/kolla/.zuul.d/python3-jobs.yaml@stable/2024.1\", line 2","accounts_in_message":[],"_revision_number":3},{"id":"32a64b8d1f450bbd79de6d314605b203d245b1f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-17 10:10:14.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"31e17392e80336cb1a52a2aa39efb7ba663e34fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 10:10:54.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.1.  The\nproblem was:\n\n  Project template openstack-python3-jobs is already defined\n\nThe problem appears in the the \"openstack-python3-jobs\" project-template stanza:\n\n  project-template:\n      name: openstack-python3-jobs\n      # NOTE(gmann): We are using this generic template for python version\n      # testing. When a new development cycle starts, we need to do two things:\n      # 1. Update this template for the python version jobs as per the\n      #    new development cycle tetsing runtime defined by TC\n      #    https://governance.openstack.org/tc/reference/runtimes/\n      #    Use the \u0027branches\u0027 variant to run the applicable python version\n      #    jobs.\n  ...\n\n  in \"openstack/kolla/.zuul.d/python3-jobs.yaml@stable/2024.1\", line 2","accounts_in_message":[],"_revision_number":4},{"id":"485dd4de13693314efebd51ac59b54ef28f71cd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 10:11:05.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.1.  The\nproblem was:\n\n  Project template openstack-python3-jobs is already defined\n\nThe problem appears in the the \"openstack-python3-jobs\" project-template stanza:\n\n  project-template:\n      name: openstack-python3-jobs\n      # NOTE(gmann): We are using this generic template for python version\n      # testing. When a new development cycle starts, we need to do two things:\n      # 1. Update this template for the python version jobs as per the\n      #    new development cycle tetsing runtime defined by TC\n      #    https://governance.openstack.org/tc/reference/runtimes/\n      #    Use the \u0027branches\u0027 variant to run the applicable python version\n      #    jobs.\n  ...\n\n  in \"openstack/kolla/.zuul.d/python3-jobs.yaml@stable/2024.1\", line 2","accounts_in_message":[],"_revision_number":4},{"id":"b9dc8f9b5778c130ad3b728810a8912c9d0779b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-17 10:15:03.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f12e02268595b5c660bbae0c761988298d879fbf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-17 10:40:25.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"793eb17dab3635ecec99b8064ed86d8194ec9340","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-17 10:49:35.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b1f09693e82adba1596abed578063494ac298a90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 11:29:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0bac8c89f0449ec8b724aeb7051f3b2\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/132ba813cca84a1a9cd32a0db7cd07a5 : SUCCESS in 1h 09m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ff3c2c0a72d4d1fb685bc8219792f6e : FAILURE in 13m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f23b0584518143f0a9c9edb16c08e1af : FAILURE in 2m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e86c5a2d2a0547c29922f7b1403ba77c : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c1c12af71d5e46a3ac596e57eb91a869 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":5},{"id":"6e81531442821d3a7e6b47e0cd141ed06b888c16","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-17 11:33:30.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"090b23a9316526e3b84674d7bd8e113933799299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 11:41:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd4180978394b518d379f7368eeef4c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bf8acc67843545db85a9da7607e17935 : SUCCESS in 48m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ceb9d98ce0e14fbba0db89354c3b1043 : FAILURE in 8m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/77994f548bbe4b85832c05530e6760ff : SUCCESS in 32m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3b1e682b2349449c87c174185a1a1cd2 : SUCCESS in 13m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87cb19a3da3748d6be7dbd48fe01905d : SUCCESS in 1h 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936ff153b98e477ab1099fea8447604c : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b43526c97c84855b034c7208b26d364 : FAILURE in 2m 24s (non-voting)\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7974fc66cf54489dbff94fe8649c1fca : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40dda8b9332640c4bfa28212c95cd7f8 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164aafda708d440f8dfd283b35e573f1 : SUCCESS in 2m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/95bff447732c459eb35b1a63074b91c9 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ca1c80f033d44fbb8b7030a4c161f23 : SUCCESS in 3m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc4092134e6b4923960d699094ad0f03 : SUCCESS in 56m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/74329af1d60c4ec48b639a9143c0ea68 : SUCCESS in 23m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6cf9edb8cae94c1e9af5bf1116c5904e : SUCCESS in 1h 03m 17s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/874e9053fc32453f94ac87e41ac8dd44 : SUCCESS in 39m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6b44136d501401294977fb5071e4248 : SUCCESS in 38m 56s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/d3c697d77f5e4c97b0d145124fe73286 : SUCCESS in 46m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d92fd36e7e94ef6b2f732ac964e260e : SUCCESS in 25m 58s","accounts_in_message":[],"_revision_number":5},{"id":"ac5d38a796b6fbed157792ad43cd62c3bf7ac03b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 11:41:48.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7c80b190967c3c4bf3c0e5a84c90521432283d6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 13:12:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c06966742294793a7642ee20ab1f70c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d71e86ae9b5140ee853bfc818371ce7c : SUCCESS in 33m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cbadf034aa4f43ed9201d72057bcd985 : SUCCESS in 15m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e25c8a5b396d4b5d96dff5823b1b2236 : SUCCESS in 1h 16m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a5b133a41544c1a44699d983141afa : SUCCESS in 6m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eb31e38c15434a1a98674b2e85b33ea5 : SUCCESS in 2m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7dd9f9f490b94a3e8c16c9a49ce898f5 : SUCCESS in 53m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/784ba68483e544d393e5f7af70631d5c : SUCCESS in 33m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9847520c4c3a44b9964022468fac5f1f : SUCCESS in 1h 25m 08s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6e559f028644dd49c06d1a63667e18e : SUCCESS in 38m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c868020444b4755a1caded47d8a077c : SUCCESS in 34m 44s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/0348fa11afce4d4ebe0164f575d27cee : SUCCESS in 43m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62eafdabb08c45c28183d08764159f7c : SUCCESS in 12m 03s","accounts_in_message":[],"_revision_number":5},{"id":"f2978c1abca3df09681f8eafea673db917712deb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 13:12:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c028e02fcde68a52987e60e348d73ce3f78f3790","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 13:14:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b153852c4743d1bbcbcc6449e85784\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063508c0c20c47778420677a03f36731 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"34ca6e35c94ddba1e47722718f3fc81a1f03e28a","revisions":{"995585efc6dccd721cd913ba24790d73f9e45589":{"kind":"REWORK","_number":1,"created":"2025-11-17 09:01:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/967299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/967299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/967299/1"}}},"commit":{"parents":[{"commit":"2c31403936a07404ae3fcc0661c874c6b8c59758","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c31403936a07404ae3fcc0661c874c6b8c59758"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 09:00:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 09:01:11.000000000","tz":60},"subject":"bifrost: Mark as unbuildable","message":"bifrost: Mark as unbuildable\n\nIt\u0027s failing since switch to unmaintained\n\nChange-Id: I46690597b3a6536cee14e5a76ba963b5fa9dfd34\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/995585efc6dccd721cd913ba24790d73f9e45589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/995585efc6dccd721cd913ba24790d73f9e45589"}]},"branch":"refs/heads/stable/2024.1"},"a1bdd9ba61fe0c71c612e40528d9cf921fd5f55f":{"kind":"REWORK","_number":2,"created":"2025-11-17 09:22:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/967299/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/967299/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/967299/2"}}},"commit":{"parents":[{"commit":"2c31403936a07404ae3fcc0661c874c6b8c59758","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c31403936a07404ae3fcc0661c874c6b8c59758"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 09:00:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 09:22:25.000000000","tz":60},"subject":"bifrost: Mark as unbuildable","message":"bifrost: Mark as unbuildable\n\nIt\u0027s failing since switch to unmaintained\n\nChange-Id: I46690597b3a6536cee14e5a76ba963b5fa9dfd34\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1bdd9ba61fe0c71c612e40528d9cf921fd5f55f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1bdd9ba61fe0c71c612e40528d9cf921fd5f55f"}]},"branch":"refs/heads/stable/2024.1"},"95bb22d906756c929630431e59997287b523132f":{"kind":"REWORK","_number":3,"created":"2025-11-17 10:09:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/967299/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/967299/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/967299/3"}}},"commit":{"parents":[{"commit":"2c31403936a07404ae3fcc0661c874c6b8c59758","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c31403936a07404ae3fcc0661c874c6b8c59758"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 09:00:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 10:09:03.000000000","tz":60},"subject":"bifrost: Mark as unbuildable","message":"bifrost: Mark as unbuildable\n\nIt\u0027s failing since switch to unmaintained\n\nChange-Id: I46690597b3a6536cee14e5a76ba963b5fa9dfd34\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95bb22d906756c929630431e59997287b523132f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95bb22d906756c929630431e59997287b523132f"}]},"branch":"refs/heads/stable/2024.1"},"0a71a10fd37c96b128dd0085c2d2f258296605b8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-11-17 10:10:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/967299/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/967299/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/967299/4"}}},"commit":{"parents":[{"commit":"2c31403936a07404ae3fcc0661c874c6b8c59758","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c31403936a07404ae3fcc0661c874c6b8c59758"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 09:00:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 10:09:14.000000000","tz":60},"subject":"bifrost: Mark as unbuildable","message":"bifrost: Mark as unbuildable\n\nIt\u0027s failing since switch to unmaintained\n\nMarking py38 job as non voting, since it\u0027s now checking out master.\nIt can be marked again as voting when we branch to unmaintained/2024.1\n\nChange-Id: I46690597b3a6536cee14e5a76ba963b5fa9dfd34\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a71a10fd37c96b128dd0085c2d2f258296605b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a71a10fd37c96b128dd0085c2d2f258296605b8"}]},"branch":"refs/heads/stable/2024.1"},"34ca6e35c94ddba1e47722718f3fc81a1f03e28a":{"kind":"REWORK","_number":5,"created":"2025-11-17 10:15:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/967299/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/967299/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/967299/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/967299/5"}}},"commit":{"parents":[{"commit":"2c31403936a07404ae3fcc0661c874c6b8c59758","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c31403936a07404ae3fcc0661c874c6b8c59758"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 09:00:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-17 10:15:02.000000000","tz":0},"subject":"bifrost: Mark as unbuildable","message":"bifrost: Mark as unbuildable\n\nIt\u0027s failing since switch to unmaintained\n\nMarking py38 job as non voting, since it\u0027s now checking out master.\nIt can be marked again as voting when we branch to unmaintained/2024.1\n\nChange-Id: I46690597b3a6536cee14e5a76ba963b5fa9dfd34\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34ca6e35c94ddba1e47722718f3fc81a1f03e28a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34ca6e35c94ddba1e47722718f3fc81a1f03e28a"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
