)]}'
{"id":"openstack%2Fkolla~967475","triplet_id":"openstack%2Fkolla~master~I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf","project":"openstack/kolla","branch":"master","topic":"update-debian-package-source","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-19 04:48:55.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-19 07:09:12.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-19 07:09:12.000000000","reason":"Change was submitted"},"2271":{"account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"last_update":"2025-11-19 07:09:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf","subject":"Transition to the new rabbitmq Debian package archives.","status":"MERGED","created":"2025-11-18 05:02:54.000000000","updated":"2025-11-19 07:09:59.000000000","submitted":"2025-11-19 07:09:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967475-update-debian-package-source","meta_rev_id":"79e02b6b25f5428102702f125f5b346eb3f2f8b7","_number":967475,"virtual_id_number":967475,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-19 07:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-19 04:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-19 06:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-19 06:14:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-19 04:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 06:06:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-18 07:41:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 07:51:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-11-18 07:51:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-18 10:38:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"825fe1fa4f21d3321c322956126d7b9b644828a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-11-18 05:02:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9e3bee121a22448040fc77ad17d686254264957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 07:41:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c1bb8e0e354af4ae8f050a3f83e3a3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ace2833c8c7a43a2b51ebf20618a7306 : SUCCESS in 42m 15s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/646adf3013ce423ba2d44f5bcf51ade1 : SUCCESS in 42m 47s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/678b26d338f7485798f81bee60f0de18 : SUCCESS in 18m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e58d296e3c7e415780594d9c22a0e013 : SUCCESS in 37m 21s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/769f4570dad94fdf9267ea3a74eb4bf0 : SUCCESS in 46m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c374a5a73194a5ebdaa44e6da654afa : SUCCESS in 39m 38s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1bca109a0882413e9be3393af0983467 : SUCCESS in 29m 28s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/e058b4073a754f1aad2f14c62ba74c62 : SUCCESS in 2h 04m 52s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/357b2411184343f995774db751d33cb4 : SUCCESS in 1h 21m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f1beed88000e4068ae3965369084d29e : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0246924dbcfc44f39bc4a1e374991751 : SUCCESS in 2h 34m 05s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/0a12ae773a9e4fa28dd2c9fd3ae6552d : SUCCESS in 1h 27m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/5090250465764149882eb962bedbcc93 : SUCCESS in 43m 40s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/d0196486ff444a03a3548967b9247103 : SUCCESS in 26m 26s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/c963e3c312de413c9b78f9c1ec9ec536 : SUCCESS in 26m 41s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/db4107b0722145f1b39d0f2c210507e2 : SUCCESS in 51m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/74fb5d0eadd24b3ba274ed4e14e7f3ce : SUCCESS in 53m 31s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/806a68e05a2441daabd81b7dca484d51 : SUCCESS in 46m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/a350d788244e4338819e35b0d2127e2f : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/ea25dcc210e84f82b409af10e8cc98ca : SUCCESS in 1h 08m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/1963e7556a3b417b81ba185aee30d34d : SUCCESS in 1h 18m 29s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2433143d1cbf474c89455a9b18c6527e : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8451c41afa2c4f1485c4d35145e9abd7 : SUCCESS in 2m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc71a9dc6afb4209944ca3b1bffd678d : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1bc5876a70e40489928c9933f8c518a : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c01dfa3cc93e45af9d1a7e9535946580 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5586706b93494b1986970c69d6fe0a73 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6157654ee23e1272130cf50f244b69e4e4892c75","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 07:51:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69da01fe5b7aa2244392163c6be9efecd0031947","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 07:51:14.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"797de2d09aaa814aa1fb4260ef0f8f74fc3cbd52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 08:02:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ccf2e576d094720886e97f9bee1df7b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/0dd931812d9f4e47b935546ab6abf885 : SUCCESS in 2h 57m 01s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/a72ede2e92584976a99efcf88da8befd : SUCCESS in 2h 34m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1cbbd5843b2542ec9777879d745d328c : SUCCESS in 2h 46m 26s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/666424a4668847e1adef1e1385783a03 : SUCCESS in 2h 55m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35ed029334354be48100def89f79d98c : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c656709a87649c5b0c597204ac7997f : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85bbb56413b848c79c700e19efde7d70 : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95ce934384d566d12f8a2084e7f116b2903c3502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-11-18 08:51:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a9f92db4165d6fab3c675777f65d8df542ee41de","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-11-18 08:52:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"72f9e6c54256dd9e67210d894294adacc682f2eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 09:04:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b52f3983695d2f77dab37b0c83fe515aae32b9c8","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-11-18 09:12:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99b15469c789b001ea1ee9c6772ff5c5c49abb51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-11-18 09:35:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"36b793578a80243269ff4d661220b05b7b4bb0c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 09:36:34.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ef5d071adbfc2f339bf07abe77be7b8aca81c8da","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 10:01:11.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"799d6298c296b0e58a26341e51c02f61f538b80f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 10:01:15.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"ea8f4ff25cf663893815e5f250f7a3de585bd5a4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-18 10:02:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6282222991b7a78a7c780de6ffca086b0ef39268","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-18 10:38:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8cd067f95a612fc96a9f49d6ef8fed3bd30ec9f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 12:05:23.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"0e658dfc8ba651caee096e55a071419583006ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 12:54:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad2d0ac69739441e8449f6f33b763ddf\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/11a59c835a734684bb31b150154bf1a6 : SUCCESS in 42m 42s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c871726622f74792a9565b826476d775 : SUCCESS in 42m 42s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/e094726614e54404a05de2ad1a7abd7b : SUCCESS in 20m 32s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3a776c6285074acdbde6b48ef75dbd91 : SUCCESS in 42m 01s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/339b61b2eceb41059b593216079feeda : SUCCESS in 39m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0bf48048fb0407ab7ae52daff1513b0 : SUCCESS in 36m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c4c5d317d78947a190043305b403ba89 : SUCCESS in 21m 46s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/e34f04c007e5446f8817a6eb5ec79e20 : FAILURE in 5m 12s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/aa38f222522849debb554065ffa27312 : FAILURE in 8m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/b70a380d4ca046d39fb22ea71b128c13 : FAILURE in 7m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8d18a121ed55473082f3f7df74be8651 : FAILURE in 8m 22s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/a684693bfb15439c98fee5c82f660037 : FAILURE in 4m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/5ae0b5d1199f40259426faecb8163407 : FAILURE in 6m 34s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/6b5adf231c994d73acbe94177cfcfcbd : FAILURE in 4m 22s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/262195e9edf74f28879203036298edcd : FAILURE in 3m 31s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/84ec529707b44e62b543b5eb71a48ec4 : FAILURE in 4m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/7b66ca65ba254a0c9ce91eb7d3d548be : FAILURE in 4m 15s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/7f1d7abac2874799b5b331513e739cc6 : FAILURE in 4m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/3fdbf3eea31a4dcd8ec5ddfa80fda0e9 : FAILURE in 4m 11s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/7ddc1051fe264db38523b39dfe37782a : FAILURE in 6m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/e19f3e59a9b54577b8a3a460eb430f9b : FAILURE in 6m 41s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e078fc43c5484fe09fe4d583c93f9369 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04adea738c0740e681685a41ad8b98a9 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d45caf8394974620b5640ece628a86dd : SUCCESS in 1m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aedde96f21054f39920147137e1532a7 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ce1df4097ae4fb9b622016173dbc5ff : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe9ad15c312542b4b8910747e54dfa4f : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cd47b92c4824c5f83e05800ae05f937 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":5},{"id":"97c330a74d1b4bd8f030cf009f11be97e6895755","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 12:59:11.000000000","message":"Patch Set 5: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2fe4c0be2ac98c2f6f9a1b388d853d66e535192d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-18 13:05:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"670a7dc3777f8e7d1abdc109a8912b10c2b9b9a9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-18 13:31:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"febe21ad97c2dddd786b04a5c79fa20b687711ab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 13:33:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5a4cff475d5b301d4423afa7e9d8b67e7014fb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 13:57:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f0280bca6bc4793a04bf9059a14be95\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b8bb822c88df4dd8aebc44da67612c8f : FAILURE in 42m 31s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/4afb06f82a064000b8f4822827a4972f : FAILURE in 5m 54s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/ff58f6b7cb964f189be71958cc335254 : FAILURE in 7m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b5a4e9d0fae7438f9501035ca7e86bbf : FAILURE in 39m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5540551c264847d9b21e5b16c9481977 : FAILURE in 53m 11s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdebacfdc2304228a1c956ca935ae187 : FAILURE in 7m 58s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/49bd0745119b46ebad69112bd5d4bfac : FAILURE in 7m 23s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/c30b743daafc467886d4f7a56d021b62 : FAILURE in 5m 34s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/84980136149148309fa167a784ef8059 : FAILURE in 7m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d7a5854eaf1940f597cbc327f76fe53c : FAILURE in 8m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ef194fa0327e42678d51cddd00bafc22 : FAILURE in 7m 28s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/0a0684b63a6744ea9c39b6fc7705bc72 : FAILURE in 6m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/c3d5e5e734e4431c9138a109b3027133 : FAILURE in 6m 27s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/55cf58d4d28d493695d6214b3d7d1d05 : FAILURE in 5m 27s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b3ad71857f1c45b4a5951a068fde90b3 : FAILURE in 5m 23s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/4d8fe2ba9d664a9f96eda2b80404c130 : FAILURE in 5m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/229a6dcdbe58445bb41e279094cb7e01 : FAILURE in 4m 12s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/17f4be4fd92348e082f9a820804d5afb : FAILURE in 5m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/f85518a582d4408b83de2afa271f8db0 : FAILURE in 4m 14s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/b953dda61a6c464aa26551a992912b9e : FAILURE in 6m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/345a721437b34a1f9a9fa122e230e3a0 : FAILURE in 5m 04s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa07d0eb25c64516b1bc32af2f25370c : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25fdb229fb694ec09c6143ee705c59f1 : FAILURE in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7700c61614fa410caf5f8fc702779b75 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a88890e97562408dbd2972848822c0f7 : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e31cd296d7474d7e85dfcdaaf917f204 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4313dee1f71745f3b7763118a27bf0d4 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39a28c64acca442abcde002d33fe7d39 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":5},{"id":"412885d69a1621de6ed355e3aaa038b532c30e81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 14:49:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a656e2219c1445b385b835721498c5d7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/ddfaae881aca4f598e317fab74cfb146 : SUCCESS in 1h 07m 33s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/0c737bccaaa24a7299f159d4bdf40d09 : FAILURE in 1h 10m 22s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/02430aa05879471cbf1cf647fc061470 : FAILURE in 2h 22m 19s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/c0b8e7bb218046cfbe9a185ae7e51c24 : FAILURE in 2h 41m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1010e4fb8ec34ec49945fcb4b91f1fe4 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea1cfd3e828a4ef780544c40e05316d1 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d0c92985b6b4629a2c252073a4e7626 : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cabaa6f58774d851c9085872d98dd460f64217d8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 14:50:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"095928a10d7373c9f94b94d391cb68be83ff3ca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 16:05:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d25801d8c2b6459495ea85ce12197f4c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/916aa74b8fae41d98e98ce58bd258e13 : FAILURE in 39m 16s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/a5fd6268f5cb4cd990c26579bfe325d2 : SUCCESS in 52m 13s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/d80a52ede8a74b71a1c1427d31115e58 : FAILURE in 40m 30s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/18d0f7e2dcbb487e94c9525aada2e232 : FAILURE in 37m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5529ee88daf142929bfc7e4c5a80d4b2 : SUCCESS in 51m 24s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e873b17bcbdf4b97af3de7e947c071e0 : FAILURE in 33m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/03b49aeeb86a459c98c9f47487fa25cb : SUCCESS in 38m 29s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/4962820781a544019e346d3e8d036c50 : FAILURE in 7m 21s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/995264280df44c6f86789501f3199ff4 : FAILURE in 6m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/bf49eab3fc0b43e09d04e51a7d4ee53a : FAILURE in 7m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/601aa2f616074b098098ba483ab74207 : FAILURE in 7m 11s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/98d22071a0834b9b9cf4e33369d84903 : FAILURE in 7m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/68705ef60c2d4f879dfbe23b99f3f843 : FAILURE in 6m 30s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/04d1d4a5dc0c4bbba327595ead2d87f2 : FAILURE in 6m 09s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/7eb55c8dde31498283291d05bf2e3965 : FAILURE in 4m 57s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/25373426223143ebbed1ad8e8e5fddff : FAILURE in 4m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/26f274782a51408e91cd11fa59297408 : FAILURE in 3m 59s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/07f53d6e01c84a5aa70b92c8741e1291 : FAILURE in 4m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/327790aecb3142d7a4ea317f5f226243 : FAILURE in 4m 41s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/ce4d072503ec4ab8a823c0fcbbf4caae : FAILURE in 4m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/68f650ddde7d4357a90ed72254d87eab : FAILURE in 5m 49s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7363ffa7c9314c27935867777a3d7dc2 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24f8560df64433fbe3eab9f65bba6a7 : FAILURE in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f035944fcfc4ec5aa60ae3648da9ef6 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e09100e5958a44589e6f6069c1379a5c : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56a86ef2481042a2af9a07d396851d72 : SUCCESS in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a24a580abd4e238fb913f528291c6e : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9094aea55b144b89afe5c8c328114c64 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":5},{"id":"8f4fd611db5258c1873db24dcadf711c4bd4f17c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 16:12:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f94f0ce5561a16efbb97190c2389ea2f032ec3f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 17:57:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0a666bd75948edb50511bfc0fb8b6d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/953ab06e49514ca791e03e513f7ea10b : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/595b94f9b54e460ba024851777fba13f : SUCCESS in 2h 45m 57s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/5accf6f6e778474b8748e098ca178549 : FAILURE in 1h 03m 13s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b42dedad571b4e5bb099a00fe61d5055 : TIMED_OUT in 3h 03m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85aa8631622740239bf9bfb68c9233ff : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be89688cf07b48899cf2a07a6f12b426 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89963643cdb54e4c9e5bb48301115b6d : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dcba7b61ceedef671d8a4a07547d98dd26351217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 18:21:25.000000000","message":"Patch Set 6: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"da2042dc79c68db14bc7ab8efe53a699311ba963","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-11-18 18:27:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8db2bccbb0b43e2e05f93f39de6a917cbf62d0e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:32:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b358b1381f784019b2472f4f3612dfce\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a1e15fe3df5341c8b11577e1d8b0c3e7 : SUCCESS in 2h 07m 43s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/d4edfaffa1a54149910bee92ca504f59 : SUCCESS in 1h 43m 55s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0bc33d0c84c24cdfa96813cd64c6b276 : SUCCESS in 2h 01m 37s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2bca170e0ec64eb182feeb31c2669ef9 : SUCCESS in 51m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc94c86e1bfa496a916796042b41c998 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29d13d3ad10744919ad7b117b428fe95 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8002609380e1405ab8123896c50d0d71 : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7e540ae523c094db558b4c2640650ebe669a2424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:44:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6619f239d64e4c9c9bb2b3e970f8bd14\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2e5e4a4d090548c39369d2512f356d7a : SUCCESS in 42m 01s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/46ba09d014b94305b626aaa991e35b7f : SUCCESS in 48m 12s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/a434937a2168422bb02cf99dcb1d73b2 : SUCCESS in 30m 33s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/09f0e49537e64d0e820b9bcfdce9eae7 : SUCCESS in 41m 43s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b1a8ca940eee48c1b089347957252874 : SUCCESS in 1h 04m 24s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9721e761f41a414e8c710b236744b2a0 : SUCCESS in 41m 09s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a2fb2e79f9404ff4bc1cb4685dfabb2a : SUCCESS in 16m 06s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/9ecfde2dbedb452685a00690cf8b8f98 : SUCCESS in 1h 27m 06s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1547a20e5ca04271865aab3f70f4b47f : FAILURE in 1h 13m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/bf919733b0024085a4cbbb9e2ea3c323 : SUCCESS in 1h 46m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dddc711ea62d4583a6fb084368573e6c : SUCCESS in 2h 21m 43s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/3c6d12f19d2b4cfb9a1e9d925d3ad519 : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/20316fba32304fe9a234a2a2b0c018e9 : SUCCESS in 1h 12m 07s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/e69fc14e39b54bfe9324c7397a86b751 : SUCCESS in 34m 09s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f634ac310146424da06695f6533bfaeb : SUCCESS in 29m 03s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/7ef0cce3c76b4eeaa87db27a639c6237 : SUCCESS in 1h 00m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/90c712049088409fb7ec5cc18ed1f627 : SUCCESS in 31m 48s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/3b53d8f84a714b0c972d85456baefd97 : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/82d7302e20dc4beeabeb89ee35d8a601 : SUCCESS in 59m 30s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/c2306125cd154dcbbe76556a5dadf811 : SUCCESS in 1h 12m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/3554036718fd41f8830aa3322ac9b79c : FAILURE in 56m 27s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/585b3dcfeb0c46f7a2feaa319b46eebe : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1ba1a73f8eb4b48bc19af8cd1007d61 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8246c8fb023f4b25a36ab026caaa74bd : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/890d4ef4cb35402282d67a95a2a2805a : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbf9be4c0ba247e0a54233d458f648d5 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7f6a58818f34717aff664a8a6d1d401 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0731171ca7264349a73497803591182a : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":6},{"id":"078abfe4ff8e65bb346ed5c3478cc945449d49b7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 04:48:55.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8f63fc18c9f0ccc6aa3e93170c6781eb63f9874f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 06:14:27.000000000","message":"Patch Set 6: Backport-Candidate+2 Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"417951538debe1054f1d5e3510a26c9f17c5e678","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 06:14:30.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"85d370d3d76120f6aa8a30c263992abed63797db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 06:14:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"72dd4e1a9f18fd89936585f5387b337aff12d3dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 07:09:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7facbe607d2d4044bf6ca3e989ee068d\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/026d32db4ff7441d932d69d0a5250b77 : SUCCESS in 49m 49s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/328e527dde5c4bbf89153bea0c83de21 : SUCCESS in 32m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/046364d3ab0a4c259c10c8255daed47c : SUCCESS in 41m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0139c644ea474db58789d53fcb6102bd : SUCCESS in 38m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d826203caebf4fe981813ec80b7f864a : SUCCESS in 35m 25s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/52523cd3b39c4715826b7bda99b9e5c7 : SUCCESS in 28m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0537a745160845edb15eee68e9e214b4 : SUCCESS in 1m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79cb2b6d26f14c83b9c9b554a8d73cc2 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0544ff2b9c024b08b34d9ed1d765e055 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1239698be8284d30a6646e4789885556 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10910510da76459b93582dd21007ce6f : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32ae867159dd441792d1361eb33020d8 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":6},{"id":"6d40f0e3c327eb25bf344ce1c324b550257f9842","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 07:09:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"79e02b6b25f5428102702f125f5b346eb3f2f8b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 07:09:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2c8dc028cfe4a0c8c7a650567021e97\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6246232e740b4f52aeb10a8413756d95 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2419d5009f84c22abbea78611fa40b1 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0fc73a644bdb7e4f583ba5ebf7ee3fcd60fb8ffc","revisions":{"edc396d39f35817ca6cc807c0cd1f6f9da5e8966":{"kind":"REWORK","_number":1,"created":"2025-11-18 05:02:54.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/75/967475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/967475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/967475/1"}}},"commit":{"parents":[{"commit":"a03766b07f753e233b9b142f2d0168bacbc23eb8","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a03766b07f753e233b9b142f2d0168bacbc23eb8"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 05:00:10.000000000","tz":660},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 05:02:26.000000000","tz":660},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one.\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/edc396d39f35817ca6cc807c0cd1f6f9da5e8966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/edc396d39f35817ca6cc807c0cd1f6f9da5e8966"}]},"branch":"refs/heads/master"},"2dbfe0157b5a464c506e512c61ca30b6b703bf79":{"kind":"REWORK","_number":2,"created":"2025-11-18 08:51:27.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/75/967475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/967475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/967475/2"}}},"commit":{"parents":[{"commit":"a03766b07f753e233b9b142f2d0168bacbc23eb8","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a03766b07f753e233b9b142f2d0168bacbc23eb8"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:45:29.000000000","tz":660},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:51:24.000000000","tz":660},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one.\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2dbfe0157b5a464c506e512c61ca30b6b703bf79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2dbfe0157b5a464c506e512c61ca30b6b703bf79"}]},"branch":"refs/heads/master"},"7a8e43b55ad9b60ef5100d9a186d419ce2b29918":{"kind":"REWORK","_number":3,"created":"2025-11-18 09:35:15.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/75/967475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/967475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/967475/3"}}},"commit":{"parents":[{"commit":"a03766b07f753e233b9b142f2d0168bacbc23eb8","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a03766b07f753e233b9b142f2d0168bacbc23eb8"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:45:29.000000000","tz":660},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 09:35:03.000000000","tz":660},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one.\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a8e43b55ad9b60ef5100d9a186d419ce2b29918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a8e43b55ad9b60ef5100d9a186d419ce2b29918"}]},"branch":"refs/heads/master"},"991346a727204a5207d6d63ce1a5ce34bee8e893":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-11-18 09:36:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/967475/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/967475/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/967475/4"}}},"commit":{"parents":[{"commit":"a03766b07f753e233b9b142f2d0168bacbc23eb8","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a03766b07f753e233b9b142f2d0168bacbc23eb8"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:45:29.000000000","tz":660},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 09:36:33.000000000","tz":0},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one.\n\nCloses-Bug: #2131736\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/991346a727204a5207d6d63ce1a5ce34bee8e893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/991346a727204a5207d6d63ce1a5ce34bee8e893"}]},"branch":"refs/heads/master","description":"Edit commit message"},"27d2f0d72a117cd0cb5b6abc97b3334fea9367a9":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-11-18 12:05:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/967475/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/967475/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/967475/5"}}},"commit":{"parents":[{"commit":"a03766b07f753e233b9b142f2d0168bacbc23eb8","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a03766b07f753e233b9b142f2d0168bacbc23eb8"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:45:29.000000000","tz":660},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 12:05:23.000000000","tz":0},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one\n\nCloses-Bug: #2131736\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27d2f0d72a117cd0cb5b6abc97b3334fea9367a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27d2f0d72a117cd0cb5b6abc97b3334fea9367a9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0fc73a644bdb7e4f583ba5ebf7ee3fcd60fb8ffc":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-11-18 18:21:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/967475/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/967475/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/967475/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/967475/6"}}},"commit":{"parents":[{"commit":"a03766b07f753e233b9b142f2d0168bacbc23eb8","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a03766b07f753e233b9b142f2d0168bacbc23eb8"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:45:29.000000000","tz":660},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 18:21:25.000000000","tz":0},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one.\n\nCloses-Bug: #2131736\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fc73a644bdb7e4f583ba5ebf7ee3fcd60fb8ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fc73a644bdb7e4f583ba5ebf7ee3fcd60fb8ffc"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
