)]}'
{"id":"openstack%2Fkolla~967507","triplet_id":"openstack%2Fkolla~stable%2F2025.1~Idd809f5efd2d308c1565d7635e67af0a431d18f7","project":"openstack/kolla","branch":"stable/2025.1","topic":"kolla-2025_1_el_10","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2025-12-04 14:08:30.000000000","reason":"\u003cGERRIT_ACCOUNT_35264\u003e replied on the change","reason_account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-05 08:37:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd809f5efd2d308c1565d7635e67af0a431d18f7","subject":"Add support for Rocky Linux 10","status":"MERGED","created":"2025-11-18 11:21:06.000000000","updated":"2025-12-05 08:39:02.000000000","submitted":"2025-12-05 08:37:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967507-kolla-2025_1_el_10","cherry_pick_of_change":956554,"cherry_pick_of_patch_set":67,"meta_rev_id":"6ec0bdba2d09a1879428fb7bd58695eee43660f2","_number":967507,"virtual_id_number":967507,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 08:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-12-05 06:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-12-05 08:22:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-05 06:13:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 11:21:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-18 11:21:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 13:28:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-12-04 13:53:42.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2025-12-05 08:22:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"56c7d82d1abd9e207dc53dce71c7deb6a5ada93f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 11:21:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/base/ci-rocky.repo\n* kolla/image/unbuildable.py\n* tests/templates/template_overrides.j2\n* zuul.d/nodesets.yaml\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"f4685f3004219bc90dff6c44fb0959bbcb6c2098","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:21:28.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/kolla/zuul.d/nodesets.yaml@stable/2025.1\", line 1, column 48\nWarning:\n  Configuration in openstack/kolla/.zuul.d/base.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/centos.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/debian.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/project.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/tox.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/ubuntu.yaml@stable/2025.1 ignored because project-branch is already configured","accounts_in_message":[],"_revision_number":1},{"id":"7a14eeef8cba98d567c568ba684f26957a74c912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:21:40.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/kolla/zuul.d/nodesets.yaml@stable/2025.1\", line 1, column 48\nWarning:\n  Configuration in openstack/kolla/.zuul.d/base.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/centos.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/debian.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/project.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/tox.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/ubuntu.yaml@stable/2025.1 ignored because project-branch is already configured","accounts_in_message":[],"_revision_number":1},{"id":"a34ccdfcaf4054617b9170def25b199f3d8e8bab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 11:23:02.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2bf5ee5fe5c0728e73f72ce9f25f484351fdeeed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:23:21.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/kolla/zuul.d/nodesets.yaml@stable/2025.1\", line 1, column 48\nWarning:\n  Configuration in openstack/kolla/.zuul.d/base.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/centos.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/debian.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/project.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/tox.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/ubuntu.yaml@stable/2025.1 ignored because project-branch is already configured","accounts_in_message":[],"_revision_number":2},{"id":"c5f3c038ff15f6eaecb0f0b1f020f8d1e98ce235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:23:55.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/kolla/zuul.d/nodesets.yaml@stable/2025.1\", line 1, column 48\nWarning:\n  Configuration in openstack/kolla/.zuul.d/base.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/centos.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/debian.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/project.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/tox.yaml@stable/2025.1 ignored because project-branch is already configured\n  Configuration in openstack/kolla/.zuul.d/ubuntu.yaml@stable/2025.1 ignored because project-branch is already configured","accounts_in_message":[],"_revision_number":2},{"id":"bfab3b0138d2e8701c85f7f42f46b0788a551b16","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 11:24:51.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2f7707427166061f9cc9cd784f73da8925d678a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 11:29:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8ab24358af100c455f042bf655fe97c2a7baec45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:29:38.000000000","message":"Patch Set 4:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  while parsing a block mapping\n    in \"openstack/kolla/.zuul.d/base.yaml@stable/2025.1\", line 84, column 3\n  did not find expected key\n    in \"openstack/kolla/.zuul.d/base.yaml@stable/2025.1\", line 86, column 4","accounts_in_message":[],"_revision_number":4},{"id":"cb84c186ca4d2f1782056118f3d24565385ad5a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:29:51.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  while parsing a block mapping\n    in \"openstack/kolla/.zuul.d/base.yaml@stable/2025.1\", line 84, column 3\n  did not find expected key\n    in \"openstack/kolla/.zuul.d/base.yaml@stable/2025.1\", line 86, column 4","accounts_in_message":[],"_revision_number":4},{"id":"3da16e9e708a2d0a052e1b571899b118a1ff85aa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 11:53:24.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7cfd9e8cb8bc3b33e97218784c324985134d5abe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:53:42.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  The label \"rockylinux-10\" was not found.\n\nThe problem appears in the the \"kolla-rockylinux-10\" nodeset stanza:\n\n  nodeset:\n      name: kolla-rockylinux-10\n      nodes:\n        - name: primary\n          label: rockylinux-10\n\n  in \"openstack/kolla/.zuul.d/base.yaml@stable/2025.1\", line 84","accounts_in_message":[],"_revision_number":5},{"id":"1d7af2820172cf9ad2d69809a9714970cdeb193c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:53:55.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  The label \"rockylinux-10\" was not found.\n\nThe problem appears in the the \"kolla-rockylinux-10\" nodeset stanza:\n\n  nodeset:\n      name: kolla-rockylinux-10\n      nodes:\n        - name: primary\n          label: rockylinux-10\n\n  in \"openstack/kolla/.zuul.d/base.yaml@stable/2025.1\", line 84","accounts_in_message":[],"_revision_number":5},{"id":"a61357caae87934ee7fb3d6fbbb3bd132efe3892","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 11:55:45.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"9fdf3860214f283648e00d69ca246833fa821963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:56:02.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  Job kolla-ansible-rocky10 not defined\n\nThe problem appears in the the \"opendev.org/openstack/kolla\" project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-build-rocky9-podman\n          - kolla-ansible-rocky9\n          - kolla-ansible-rocky9-upgrade\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-rocky9-cephadm:\n  ...\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1\", line 2","accounts_in_message":[],"_revision_number":6},{"id":"6ffd158a37b6b3a3a1b2560b057ea74b3137d65a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:56:16.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  Job kolla-ansible-rocky10 not defined\n\nThe problem appears in the the \"opendev.org/openstack/kolla\" project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-build-rocky9-podman\n          - kolla-ansible-rocky9\n          - kolla-ansible-rocky9-upgrade\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-rocky9-cephadm:\n  ...\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1\", line 2","accounts_in_message":[],"_revision_number":6},{"id":"96b86f10f4f8188b58e49d96b9dd87aef2538850","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 11:57:06.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"10b2012ea61afb573ff0072790f91ede2d2f5966","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 11:57:22.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"0142a857ffe29f23e5d910232b4342409502c534","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:57:26.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  Job kolla-ansible-rocky10 not defined\n\nThe problem appears in the the \"opendev.org/openstack/kolla\" project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-build-rocky9-podman\n          - kolla-ansible-rocky9\n          - kolla-ansible-rocky9-upgrade\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-rocky9-cephadm:\n  ...\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1\", line 2","accounts_in_message":[],"_revision_number":7},{"id":"6476160bb3760aff031ae2af2194411849cd3aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:57:43.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  Job kolla-ansible-rocky10 not defined\n\nThe problem appears in the the \"opendev.org/openstack/kolla\" project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-build-rocky9-podman\n          - kolla-ansible-rocky9\n          - kolla-ansible-rocky9-upgrade\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-rocky9-cephadm:\n  ...\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1\", line 2","accounts_in_message":[],"_revision_number":8},{"id":"dcfeb366084cf2f8857ff9309825ef94cdbdc494","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:57:57.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2025.1.  The\nproblem was:\n\n  Job kolla-ansible-rocky10 not defined\n\nThe problem appears in the the \"opendev.org/openstack/kolla\" project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-build-rocky9-podman\n          - kolla-ansible-rocky9\n          - kolla-ansible-rocky9-upgrade\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-rocky9-cephadm:\n  ...\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2025.1\", line 2","accounts_in_message":[],"_revision_number":8},{"id":"fc4826183c9764a2ba823351c9def1b7a2318f9b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 12:00:36.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d91ea9bff85a8a00f45fa9a3c8112f56df3d3df8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 12:13:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c53f4e4111843a38405ed25c2a9181a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/385b0d45e1484144938bec40f4727590 : FAILURE in 9m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/82509cfd6e54479cb2c3d11b29f56aa6 : FAILURE in 5m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/023991d8bef14a1c843bcf326807ff02 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a3799a2ec8b4bdca6e156ccde2de084 : FAILURE in 4m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/074f05a98c984925a617592b77c43636 : FAILURE in 5m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a633b70aa8434c0ba41db47e821c9569 : FAILURE in 8m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/46b6fc711f9d4754899307472f12c746 : FAILURE in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"47e4af8aa3704fbca8c9820f037e2e77e8b95843","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 12:16:27.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"e440e22b8900f3c351984b17399b842748ad3900","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 12:26:07.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"dcb16e98be9c7d967926bd9ece54bf39e82a0143","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 12:37:30.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"06bf725a158b138575255e29b475956b6f3040a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 13:35:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d5b4889b05443dc8b1cacb0a443bbed\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c699fa808ee04c80902ce86016b6cd4d : FAILURE in 42m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4a257d020f2641e2856f018cce7c9d2c : FAILURE in 4m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/177d6e8a0c504d5bb3cd54d6e3f5ef6f : FAILURE in 7m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f1e83d135e1b4afb8ec9644e9c9b9583 : FAILURE in 7m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f828768c03e743acba1ff5f7f0806119 : FAILURE in 4m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/734bdefbfff84974b752bed10e409d4e : FAILURE in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f90a347e4649c4b861a18740e3856b : FAILURE in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90136d325f142fc8747733f80321d3d : SUCCESS in 2m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddfa86028c7a406f9a70bde13cae6114 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e82b4631ddf4a68a04d040b829d244b : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea91005581534f93b1264df204c1c64f : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87bd68d21c5942ec8638bf567c7499d8 : SUCCESS in 3m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b3b0cd7924345c19304da285a6a1506 : SUCCESS in 2m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/78775e48b63143ea997a312502463911 : FAILURE in 45m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/321380254e994763b617f2482c6efa84 : FAILURE in 56m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/97940a531e20497ebbbb9eabf7d973b1 : FAILURE in 5m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0454a8035d8b410c94894b3cefb939f0 : FAILURE in 4m 48s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/3ee7cfd16e704c128676d970e4ef4f7d : FAILURE in 31m 33s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/aa90427756fe4d88a6bc3b3b7cdc13cd : FAILURE in 5m 37s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6634ca81c55b4492a35ed58507a05a8e : FAILURE in 4m 41s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/bf0df05543ec40d8974b965ff107d6c6 : FAILURE in 15m 33s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/84d5f159d3ff4aa6980d5adf2c7108ef : FAILURE in 6m 09s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/914b54c44753414c9d58b7f4218b1333 : FAILURE in 4m 45s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2122b9ba073a43728c3e8e76932b35d5 : FAILURE in 29m 49s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ba046be4238b4d569b36ddecd918a157 : FAILURE in 4m 30s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/51dae5cc743140f8a25eb00021d81d61 : FAILURE in 36m 54s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/b0d82cb2860943348b72ea5b60606c6e : FAILURE in 46m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8a744cba6c9c4dab95664d1e9c77d9fc : FAILURE in 7m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fc908d07a3fd40a8bb2f0df6f5de7f36 : FAILURE in 6m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5b21f35abda847c88ca893858b097f11 : FAILURE in 3m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/31ee59081b3b47dab1599f3fe5d6d2d8 : FAILURE in 3m 51s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ba292930bd05421eb0da146f64c37525 : FAILURE in 4m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/194d26111fda483c91e3636a6f1d69e1 : FAILURE in 4m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/45875d4bc0d0412d9a044510b668274d : FAILURE in 4m 15s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3ec20220d89e4fcfab3ff51ff65b4242 : FAILURE in 3m 44s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/1baa70e9e79f42a988e133f054ae6e48 : FAILURE in 5m 19s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b4c35aef4a244d75849cc2cd12630fda : FAILURE in 4m 38s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1e1e577b97e84d9aa70d05334d34fb2b : FAILURE in 4m 46s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fade67093b7a4a9db1a38a780cd655e8 : FAILURE in 2m 46s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/82caebf64d9d4727af52ec382f12e338 : FAILURE in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2a80e3b9ed593b239e402654a4c3ce0ac4df7eaf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 13:46:39.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"52776c88e3efca4ada3b8577db8a45be23cf86e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-18 14:04:49.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9672d350e779534b7f62ff9f071062dfacd3a008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 15:21:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abea5198392544b59f083319c85f688c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/33d5b7befe2f44438b83a4cf317235d6 : FAILURE in 40m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/950a46b10a1b4021b8647a5e231e02c6 : FAILURE in 4m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/16621b67ef80409892085b4149fda44c : FAILURE in 49m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aee3b6218d1946778e91103e5fed4af7 : FAILURE in 28m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/414f9fb3fd6f4bcca11e4a8a42c4f75d : FAILURE in 4m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5ddd3fe21a1c40c0b32a457a3c7faa60 : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1901fae40db848b99ec38b4e0a6195ef : FAILURE in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/784dbfca7bd34806a0d1c59b31fa9a41 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64efea0e502642e189644cfe0284076c : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b29721dbe66e4ad7a9ede0cc47547024 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc4785b888146428c7d28408a4e79b8 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac5972aef4604b6bb9109ce4755082aa : SUCCESS in 4m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a4c00f13b9724ccd836d43c5a80f8548 : SUCCESS in 3m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/424d65d5aeb341c094f30d213ab52527 : FAILURE in 55m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/998091b6d2704ab6ba5473f55489b86b : FAILURE in 1h 10m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/121c0c45c61744d19369e7bb88b78ad3 : FAILURE in 5m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5619550d5c6e4707bc8a42ac0ee6a299 : FAILURE in 4m 29s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/bb2fe0ccde664fae9beb747d1dd9c9c1 : FAILURE in 5m 06s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4caf5ba1d5ba420f88f7cd6529ac0e3a : FAILURE in 5m 38s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5f5ff54aaecd4097bc480caea0782a09 : FAILURE in 5m 17s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/551e001493104a5d815f3ce3fd8bf7f0 : FAILURE in 5m 19s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/aa84c16141b74e5caac14861617404fe : FAILURE in 7m 53s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6151ee0041ab49a581d22475c286e884 : FAILURE in 4m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0558bc35bad84eb2a44e259cd23ad432 : FAILURE in 6m 26s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/763cd06289814bbd98ac881ae048f6d2 : FAILURE in 4m 34s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/65016f9500d54c1caedc2bc04ec55271 : FAILURE in 41m 08s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/3e6d76bf106f4e25a0ec4c84233acabe : FAILURE in 41m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/37d21533843945f8801c464bd8426c70 : FAILURE in 47m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/90932218e43c451e9575b310c681046e : FAILURE in 29m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e9f061e900f54beb8b70b2ff42621f89 : FAILURE in 4m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/91d36fb311f5451b91681a4ba352ede1 : FAILURE in 4m 22s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/20d5d64214494f19b246320c99ca48b1 : FAILURE in 4m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bd09836e35764c5c859f7df61c2b493d : FAILURE in 6m 00s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9f38cfccdfff48b29a5420317412702a : FAILURE in 4m 37s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ef623acebabf4c5fa4e1f2e29a7c69a6 : FAILURE in 3m 16s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0abb5416213f4a398bfbd845d7a3d55a : FAILURE in 5m 16s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b95810db427b47d3ae44c029bc5f9a44 : FAILURE in 3m 56s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/707146e30e7546f68466c0b8fbbceb13 : FAILURE in 4m 29s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bfccd3c6c18e4aaab1e44f6b4fd56f45 : FAILURE in 3m 27s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2cad054f4dc94609a6a248a34268b85b : FAILURE in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e06f6bc6d1fcd83164a2dbe22846713856af2f3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 17:03:38.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0952704977bf450a98f1c6e39841a87b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d4d9db1973564b059548e4434f112a28 : FAILURE in 2h 44m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac4456d6eec942398a37c2e5bce65b11 : FAILURE in 2h 30m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0965412e03ef48feb7496e1819f2a4bf : FAILURE in 5m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6da3ae9e66f547e7be897791f6ee8726 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ed3b60ae90eb4abcbe84b94b575aa36e : SUCCESS in 5m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/619627cc02214d2d8ca12a6d00f3f5dd : FAILURE in 1h 43m 24s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/75cffa38584b45e8aed8f3acf2a81b65 : FAILURE in 2h 47m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c6e13814ed8f45bd9613f775e2e165ec : FAILURE in 2h 42m 17s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"296771892921f02ce67ddf8843bcbf10d3f2acbb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-24 16:09:58.000000000","message":"Patch Set 15: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"c4c55c55d0db7988b2f76885630da2431bc03698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 19:00:01.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e899f9858364fa2b0de4ff2f8bc2367\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/28c32c8fb0414ae8afabf291420884d5 : SUCCESS in 44m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/48a6a9c362f54070a8ae15084c8e2882 : FAILURE in 34m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/92f1646a20d541b4adcaad3527225e30 : SUCCESS in 44m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ad1666f559bc45ef8e9a6494699c5c6c : SUCCESS in 27m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0ed0d292683e4f96b75e680ec1346458 : SUCCESS in 1h 00m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1a6d917d04664ff980127a0713913011 : SUCCESS in 1h 44m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a39d002ef44df993c866fd660b0858 : FAILURE in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abff3a43cd334a7baba32cf0e33b1ffa : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2385cfa39c374347a2606a55c43f79e2 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37413a7d45694da8bd2c8ace17dc1269 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982171eccc3844179873d007668c270f : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9e09463a8b1446a94329262159eeb75 : SUCCESS in 2m 59s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4794ab3bb3464e5891f3cd40c01fa45e : SUCCESS in 3m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/87de0369f56046298a2447def96a994e : SUCCESS in 59m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2b43780c1a1d4daca5078e9774b44a18 : SUCCESS in 42m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/446e10f1777048b78b6a042c7a2f7eeb : SUCCESS in 1h 15m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/44536cc6011e47af85f039868b1f90bf : SUCCESS in 1h 54m 26s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/2f5c53b8176d4c378cb712f5f472d2af : SUCCESS in 2h 32m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/67a9284aad784be9ab8fcd44db85a079 : FAILURE in 2h 08m 54s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3a62fe81452941a39f5fdee9af182aa3 : FAILURE in 1h 12m 10s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e1137829026e4fb7a3801a8472f1b651 : SUCCESS in 28m 26s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4292e75b49d3467dbf36a405aa2c53bc : SUCCESS in 45m 29s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9246e31fe9074132ae719ee170854d12 : SUCCESS in 52m 10s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/6de449ee609d4193946d89ea4c0f574b : SUCCESS in 2h 05m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bf8dfe8a0cfb427e9bd56a6d2f6eeffd : SUCCESS in 41m 44s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/d750c8338f6f4ed4aebbd4357e031062 : SUCCESS in 46m 47s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/80412f7b64294f4e94e9659e3ce8ac5d : SUCCESS in 1h 05m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/459138e8f0a240a0ba923408ffc1a624 : SUCCESS in 44m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f49dc265ef974ebcb15cb96a7800d7a9 : SUCCESS in 24m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d957351ac9e04839bc81d0b1063e207a : SUCCESS in 1h 12m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/91e7167dcea64c62a9c1c8ef62468c38 : SUCCESS in 1h 40m 50s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/765ae9709de64a07b1f6c187c8871e2d : FAILURE in 54m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/558ff4e6c9404b0b990d326f09502e7c : SUCCESS in 2h 44m 24s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6afe238d0f8046ddae37b315451041ac : FAILURE in 1h 20m 02s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a3185832b5d8428da8cc9135c5d46cce : SUCCESS in 27m 21s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/97fbd1d0e9c44a688d05c4707db31a5c : SUCCESS in 1h 07m 06s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/47172e1fbfce484e8fb0d2efa395b6c6 : SUCCESS in 1h 20m 49s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1b93c813b33140c6888220a1cc42427c : SUCCESS in 2h 14m 50s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ad7344891b3747c3a31eb683a2301095 : SUCCESS in 26m 52s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/16f7b4ec1bdb41b2b85ddcb56cb6d0e5 : FAILURE in 1h 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"9f9961ac9300852512c2e8b4cb59b51413f0cd3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 21:13:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8ab7f0d7cd494795a7a081a140a342\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dfedf3ea141e4cbca087079ebb24b834 : TIMED_OUT in 3h 05m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1d726af4f9c74828b92472880ddc6d4b : TIMED_OUT in 3h 03m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3f83a5450943468eb9807d0416af90b9 : TIMED_OUT in 3h 02m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/012163115a384e5d82272914c7bbe45e : NODE_FAILURE Node(set) request 5d9d40de88664f7289cc89b692d793c0 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/54dfaa2420d740bdb98401668ffaec4f : NODE_FAILURE Node(set) request c77c38c713694166b69f6d75e2848344 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/01fda5994c244596bafd574482e0afbb : NODE_FAILURE Node(set) request 6052cc3b0fee4376a2089fe500b94368 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/dfa2083f666e4e798f5e0ef0a3902377 : TIMED_OUT in 3h 03m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/75c4964135c646d893efc00a9280eb65 : SUCCESS in 1h 08m 40s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f13d7b519aa37c8967f3a46b0d09ab4fb4b59326","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-27 11:40:46.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"1e1f96662f7df1d30942f93c78ca97bad8a127dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-27 12:29:54.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n* docker/base/proxysql.repo\n* docker/base/td.repo\n","accounts_in_message":[],"_revision_number":17},{"id":"25b258d5da3e0c20003d931d0ecef3bedcfa3e18","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-27 12:31:10.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"2e5520f7b776157465f050ccff11ec9b5214d893","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-27 12:33:57.000000000","message":"Topic kolla_rockylinux_10-stable/2025.1 removed","accounts_in_message":[],"_revision_number":18},{"id":"252631b80fab176096fefa6b722238f7646d0e46","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-27 12:34:04.000000000","message":"Topic set to kolla-2025_1_el_10","accounts_in_message":[],"_revision_number":18},{"id":"c14814de8aa3cce07e4d101356ba021a3111afe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 12:37:13.000000000","message":"Patch Set 18: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla/+/968636 is needed.","accounts_in_message":[],"_revision_number":18},{"id":"d4d3503afb065ffbc11a8fce84c1262db3bbed77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 12:37:35.000000000","message":"Patch Set 18:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla/+/968636 is needed.","accounts_in_message":[],"_revision_number":18},{"id":"cf08e839642c1102547ff576ffb09ffbe4f8e301","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-01 08:13:58.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nThe following files contain Git conflicts:\n* kolla/template/methods.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"c7e2bc73366ba9f9813e7dcbd266ed94fdf9ecfe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-01 08:15:13.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"40bcd954f7cb65d39b95f65ba48bd3dca1e8ee4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 08:38:35.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a31a598a0d9432cb15d2cb523ee3d89\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/13c466c5f4e847d0b9bd4224638921c7 : FAILURE in 4m 36s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5db60c9227fa4b6f97cfde2d7be6196a : FAILURE in 2m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4dba67700cb047f1a2e23e8a6de3817f : FAILURE in 8m 33s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/414b9b9d4f1b4cbf9b9144046d59686a : FAILURE in 7m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/57d7e2a4b01d4af791369300681909dd : FAILURE in 4m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ae7530122d3542ed9b992b53c308a2ea : FAILURE in 2m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/261f921cedb84581b3e3c73a458f5cb8 : FAILURE in 5m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/29be0f758a40468dafb693246e8f1a71 : FAILURE in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b680eacc8a58493e9cbbeb8646f64c13 : FAILURE in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e121f79d8df74b65813644106eb8fe96 : FAILURE in 1m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d14ca8d2a3a44d580a8a69aa9ce197a : FAILURE in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6795c3111892419e844d0dee0a11fc4b : FAILURE in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0974efedd53941e694250869d781c8d2 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a6f0a5b23df47e2b2f46ba1c2c2cad5 : SUCCESS in 3m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/54a1d8bcfd8c4f5bb493e95266a8a42e : FAILURE in 3m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/783f9ee2522743478151fcfd725f17d2 : FAILURE in 4m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0b10f59bd1ec49eb9e0b6ac8fa677ee7 : FAILURE in 4m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bd342295934a44db85f2e8f47167c58a : FAILURE in 10m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da67b2f8dfa145f4b395031519b6a626 : FAILURE in 8m 24s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5b850aadd74f4a809b90391412814fb9 : FAILURE in 9m 34s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/727f1087cd2349b3b00cd00f41f0d6d1 : FAILURE in 9m 11s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a51dded29938420a8dcd497c3ae87541 : FAILURE in 9m 11s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/74ea0975648c4f75a2320165ab21de51 : FAILURE in 8m 54s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/2a4cd72e97d7415286c66c796c05ecd2 : FAILURE in 9m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9f0b7d0566c8446da1fefa5905e910ab : FAILURE in 10m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/11a6c83c186c4685b261faae289722ae : FAILURE in 9m 12s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ace9b9cea58d4436ba6e742f1be84d3c : FAILURE in 7m 57s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/8ed0626906a94be69687221ba8801a31 : FAILURE in 5m 54s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/b7853ad01a5e4dc0a97c27a7773f8125 : FAILURE in 5m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e8a0ecc6c87e42b5a987af6ebebf24b2 : FAILURE in 3m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6204031f57914c2988bbd91c9e4a936c : FAILURE in 3m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/10b6bfca52b04635aaae4dfe36d724fe : FAILURE in 5m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e825494caee6495fa3141167d6a4809c : FAILURE in 6m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/543c1c97d5044027a30b2bab0afa6542 : FAILURE in 7m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8ea718ae13e44f25a9b63b5b85d50598 : FAILURE in 7m 39s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/13f45fbdea8348cf878f865d8fe5fc5f : FAILURE in 9m 59s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/34546074f50648058bad1f4e15765355 : FAILURE in 7m 33s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/dab87673f79446709cee2b16d71f45e9 : FAILURE in 8m 01s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1bd668c65d83430b90a2c8970f33f0cc : FAILURE in 6m 09s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/10b890c499bb4eb8969c29721d2b7ae7 : FAILURE in 6m 49s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/15fcac37be8544e6aa612a7d0543f72c : FAILURE in 6m 25s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/dafef7522c42443ba7ca167c666c0852 : FAILURE in 8m 22s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/keystone/keystone-ssh/extend_start.sh","accounts_in_message":[],"_revision_number":20},{"id":"0284d6288df964d8a0d51be91f1b250f53f0ed08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 09:27:33.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24aa72982afd41399d130435d172558a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d81079e2755d4631ad57c3181a179e48 : FAILURE in 5m 42s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/4fb71024095446a4bf7eb3c7018318b6 : FAILURE in 5m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/711c4331efdd4066914d3779544e1a8f : FAILURE in 4m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3f424b3a02384245b2cbfe230748ce25 : FAILURE in 20m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69e822de0bc44cfcac6fa9c2c4c096af : FAILURE in 3m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10e95a17c80c4ab1b9aafa4cfd6726f1 : FAILURE in 8m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1526077963814034bb48a6e43353b5fc : FAILURE in 6m 02s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/a484d864a7c44d67970ff420b73d70b8 : FAILURE in 16m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2a98ca158c604b798bff1fbbbfedf623 : FAILURE in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"c2e0f148b9c8d92a13ddb56f38a4ad4f566a5c2c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 12:53:37.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n* kolla/image/kolla_worker.py\n* kolla/template/methods.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"396746dd889b7c6541adb056e851f51e5c3119cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 12:54:20.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"3b01de28ac6e2ce9bfc65179d0d8f18da7e2e9f4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 12:54:38.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"a745dff7134046c52d71251875c3aa4fff12d66f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 12:55:28.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"114113687c84eee9206e0c05636c57c87885b8f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 15:56:08.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n","accounts_in_message":[],"_revision_number":25},{"id":"14856673e2e6b26a80f939b2227ad1165c12df46","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 15:59:52.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"7f50c146fb3c1d8f566618a96bc405f8cd8602c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 16:00:42.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"734308b3bb2279a604aecbc34e60525ff22e7992","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 16:01:27.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"6dc289c91f472c5a7fab428791206de912b948be","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 16:18:25.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"2d0b7ad612e3877b8926be6806486ce28cee772d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 17:24:13.000000000","message":"Patch Set 30: Patch Set 29 was rebased","accounts_in_message":[],"_revision_number":30},{"id":"4d1ed6e23c389d6eae995726b86c7c35b555938d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 21:35:50.000000000","message":"Patch Set 30: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbdd2511d1074dfd91aa9bbb649cb678\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d006b3bc3dc74112982c92c6a605fe6f : SUCCESS in 1h 04m 24s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/891215f93e9a4835b9e35ac8dd3bd770 : FAILURE in 4m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bdbd9b5b5b6240989990499e4bb4a357 : SUCCESS in 1h 05m 03s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/8d32a4f835d04b84bba500fe30f35ae7 : SUCCESS in 47m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/20303b6bd83b4ee294cc1f37a6b92aee : SUCCESS in 44m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c8c4c575074e43c09fcc96cfbb563861 : SUCCESS in 28m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f9289d87e85149449f89f527b8a6915e : SUCCESS in 58m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/390a7b2a2f4c49ae94ac8c6309b60648 : SUCCESS in 1h 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e405a2951dbc4663b72e792e7ee0e798 : FAILURE in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3812fa2785b5472a804359242c555258 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a233d93bdc474557baae502b0e49f329 : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a50dcb966fd4ed0bc905da6ae648daf : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f47a28b67454453187fbff984baca603 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08602d38aae74a1badb50dc4ccfb238e : SUCCESS in 3m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/78c35f18d2e54522ba69eded0b30b0a1 : SUCCESS in 1m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/83139c2c8e3b4befb89193c4c456ab79 : SUCCESS in 53m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3cc127ff0aee450d9ca874997ed2be53 : SUCCESS in 1h 23m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/82aa6e4f901b4c86bc59d9e9a147d6fb : SUCCESS in 1h 26m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7b51c7ed87b943d7a1a6670d7b9819d3 : SUCCESS in 2h 05m 00s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4f5815dafe5049eaa7b9affe67cc18b3 : FAILURE in 2h 06m 35s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a9cb882741284841a9eac2285b3d9b3a : FAILURE in 1h 16m 41s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e71b240406554b2ab4793b1a98e37983 : FAILURE in 1h 21m 07s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/98d94fa6320349f49e34811ebc8048b5 : SUCCESS in 32m 39s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f426f364b9104c54b06f991c45408b20 : SUCCESS in 1h 21m 00s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/63bffb81ce514beb9df324ab2d2d6544 : SUCCESS in 1h 21m 41s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5ae6b9e7432a4a009b650653b608bfd3 : SUCCESS in 1h 57m 59s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dc3d8c0aeb6549c7b3729762b6da2826 : SUCCESS in 37m 24s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/0529cfbab69b406ba68e91cfbdc508e8 : FAILURE in 4m 34s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/b1ffcd21aefa443cac603045f2ad6ba4 : FAILURE in 4m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9d8bad8e49224789a38d553bc46f32df : SUCCESS in 36m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0d221391c90b4198ab68011e21c3fdc9 : SUCCESS in 14m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/96b4ad39dde940438161dee484bac067 : SUCCESS in 1h 04m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/58c4dbc5e88a4fe399eab5eb7af611f6 : SUCCESS in 1h 37m 35s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c719034920cd4a6dab170ff5d9e878c5 : SUCCESS in 1h 53m 51s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/88f71fbeeffa42c591588695a35463d7 : SUCCESS in 2h 28m 47s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/170c1c5bb21048d6873163e057580ec7 : FAILURE in 1h 13m 56s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/0c5b59e01daa4caea83592b0df48f16f : SUCCESS in 17m 03s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f4e5d3082b144ab4a54621e6665b85b4 : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f739e778a80a4b2492033a70714a1ad5 : SUCCESS in 1h 12m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a29cbd528f324d609433c786d4f9c920 : SUCCESS in 1h 37m 14s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1134dea6cfe34fe5a4ec862375ca77aa : SUCCESS in 34m 07s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/6a83589a10174d1c8e6c392e7d370119 : FAILURE in 1h 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"d6642ab94390ad849113f257016233e68aad6180","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 22:08:47.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22f9e53ba652454ca71614f9b6332af5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f68a484a1f4841f3bc3ec9757a432873 : SUCCESS in 57m 29s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/868d7c642cef4eddb6f0c291a8fad390 : FAILURE in 10m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/96c4fc23a1d04f9bb46e01d0e9737e1a : SUCCESS in 50m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/78885cc7551446b8a9ea0bd41c390809 : SUCCESS in 1h 24m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d7c061241df4ee5b7fb1c83159c1755 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b23addc3eda4ac19cf45887491a9ffc : SUCCESS in 4m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dd5a46b8374b4e7f84102a6cff850120 : SUCCESS in 1h 33m 33s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/cfa3087478da4fae8adb777588b17039 : FAILURE in 10m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4dce442f1bcd4f61ac604ec4a94827a5 : SUCCESS in 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"250cc3d0b59e2769faf91c1a8b7f8ba41bd844d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 10:54:35.000000000","message":"Patch Set 31: Patch Set 30 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"16b3602445acfd3cf2a6e62ca159f500987f1026","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 10:55:06.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"cfdaaae310a1aad627556433dda7db9d943047f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 10:56:43.000000000","message":"Patch Set 33: Patch Set 32 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n","accounts_in_message":[],"_revision_number":33},{"id":"b31cbfa664fa044c8b45cd347da6645e086eb800","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 10:57:21.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"8ad72f5ab6880909ebcadbc1b08f82bd05e02e6c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 11:17:40.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"cbc36d6ffa6c552de67e04d3dfbe249191974e3a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 11:26:50.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"9bd23e87c9ac04950fdd17f8863fa13215d61ae8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 11:51:19.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"db302fa74f5b43cad5a889470c5225e16ca500c5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 12:03:46.000000000","message":"Patch Set 38: Patch Set 37 was rebased","accounts_in_message":[],"_revision_number":38},{"id":"c207a59b7d0e4e559565ecb3bb1baa592b135ae9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 12:05:07.000000000","message":"Patch Set 39: Published edit on patch set 38.","accounts_in_message":[],"_revision_number":39},{"id":"8f07f15d591e05f79f72933a53dde385bb882cf9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 12:10:34.000000000","message":"Patch Set 40: Patch Set 39 was rebased","accounts_in_message":[],"_revision_number":40},{"id":"a049288e3a3defb24128360f054454b91c6fc139","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 12:49:54.000000000","message":"Patch Set 41: Published edit on patch set 40.","accounts_in_message":[],"_revision_number":41},{"id":"6465dcc0bcb0eeadefe9aa599c6765081c48c462","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 13:24:31.000000000","message":"Patch Set 42: Published edit on patch set 41.","accounts_in_message":[],"_revision_number":42},{"id":"6986cbd59f60af0241aee9309a645d7b854447f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 13:49:06.000000000","message":"Patch Set 43: Published edit on patch set 42.","accounts_in_message":[],"_revision_number":43},{"id":"42b2ecd90b9e21cdc16e470fd9f0d0eccd3f4c07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 14:07:48.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed7a4d5a83024448869be8c03a9612b4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a01982c9883e4db5acea0b46cf8a4914 : FAILURE in 4m 55s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/2f50b4acf974410d8bb1997412f1978e : FAILURE in 4m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6ad7e03b41bd4a5fb23e92d886da6dc4 : FAILURE in 3m 32s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/86c48e78561a48559035c5a73c6af0f0 : FAILURE in 2m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0dd401239f3840e198eea6e7e2beeea0 : FAILURE in 3m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eefba120d72544f8a8e1f9cd14b6674e : FAILURE in 1m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8d2f92b3e43d44059ece2d79e2e8705a : FAILURE in 2m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/76c00513c62f40c2858eec5b0dc3f7b9 : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2c23ea5d084fd2a87d34ccaf7d208b : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20792557fe0f48a096daa8bd69e568fa : SUCCESS in 2m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5317d9fc7ace4b02b6fccda6c783e01d : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f404d80ff0347528132056a7dc13c20 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69491c965e94d9c971375181ae066fd : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/797514b5a69f4d1aa8b64187f25ed797 : SUCCESS in 2m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f9fa41fa4d934f33ba634d675629d4fc : SUCCESS in 2m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e6a8039654264fcb815223c2cd160d30 : FAILURE in 6m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/df9dac4792bf4c0288a1109e38ae1027 : FAILURE in 5m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c6707c1ec30246c4ad628719b5446eb4 : FAILURE in 4m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/06dc4420636c45b1ad819107713707b6 : FAILURE in 3m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0b0cf9780de04ce8b965d8d07b4041f3 : FAILURE in 5m 41s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/488dde962baa43d49214fe3e24aa2881 : FAILURE in 3m 03s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9c9cb06d4193491bb0d0a65ddbffc08c : FAILURE in 3m 10s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/0f4645c851404555963cb4966d640154 : FAILURE in 4m 00s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8b35581871a54e9f8d8c33ae112751df : FAILURE in 4m 10s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/67368fa0bd154482b8e5eb10746997ee : FAILURE in 3m 26s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/df47d57deb49495c8e679ad80a079ba2 : FAILURE in 3m 49s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/145a281c9484441a94137e8e40be3270 : FAILURE in 2m 58s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/7bd8db10fac7458da38071211abfa815 : FAILURE in 3m 37s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/ad8a2a09d75843c58240c88b82744433 : FAILURE in 2m 33s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/30039db6c1a248f0957ec07db878f15f : FAILURE in 2m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d759fe608c434718955ad18d512688d8 : FAILURE in 2m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/24a9e22042464ecf9eaec5d7b59c5c3b : FAILURE in 1m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c6bcf5a058c14f7184da041e06e10b3c : FAILURE in 2m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cfc10c8287a54825bc99b5d77294a0cf : FAILURE in 3m 01s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/97833f3ba1e0419c84c570aeba2f03ee : FAILURE in 3m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2ac412be9bca411a804d546c8897e623 : FAILURE in 4m 06s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0b0f429cea6b41f981c74cd744e73af5 : FAILURE in 2m 34s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b6ca2468dd0f4e37ac7bf093c3162bfb : FAILURE in 3m 28s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/8d67f900acde497aa533f7c7d2946a4c : FAILURE in 4m 07s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e3f6252c9d094fef965a83f3f2f16fd3 : FAILURE in 3m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0fa9afdc669e48e28b85f77af0065e29 : FAILURE in 3m 41s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/de45ec5bc0314c72b11a9207fa3c3afe : FAILURE in 2m 54s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/90f351f65e8b4d7ca822bf399e9dc3c4 : FAILURE in 2m 54s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/05317c12de1348e2b2db11a058431810 : FAILURE in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"c7cde494f44fd496ccce5c636e4b539c73c08e29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 14:41:07.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aed10b0e62ca417fae5da1118d11910b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/03d968bb6de8406ca80a496538257fb6 : FAILURE in 10m 23s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/d019e50d5bb84dfa980db10a9ef27614 : FAILURE in 5m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f272f968165346519d003df68ab9dd8b : FAILURE in 4m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4fa9a88da4b54092823f4d59cf5695f5 : FAILURE in 4m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18dcdb3a6e064664902c0f8e8942b753 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be08d251a6a34e03a7e19d006e501d4c : SUCCESS in 4m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/99bd7fd2f53a4ad38d2dac689c63fbcb : FAILURE in 10m 52s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/c6aa0d2674ba4993930cdded1b7c2f62 : FAILURE in 6m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bf593decb2ee4975a85b699412c423d7 : FAILURE in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"27070669095245cf71f694403cece5f7452faac3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 16:06:44.000000000","message":"Patch Set 44: Published edit on patch set 43.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"7e80d0153efc17792f2288592e22ca33fa907e18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 18:50:09.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20733b5b9b2e44749fb0317358c3c7e5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/731e06776d364a5cbb1c1b514f560f9c : SUCCESS in 54m 26s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/f73449d780f14e7fb098ed1dfb92932d : FAILURE in 44m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/88cea07ac1834520a3a93a1410553642 : SUCCESS in 1h 28m 53s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/3a16bdfbd28444eaaa7e6bd423b3182a : SUCCESS in 46m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4881f91ff2cd449a809c834d0f0f2915 : SUCCESS in 38m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0926c22665224c259c98176a8e8a290c : SUCCESS in 16m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6fef53ae1521484584c1c993e187afe1 : SUCCESS in 1h 09m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a8d92e52e858429ba2f48dc06c30b7a6 : SUCCESS in 1h 24m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3dd524d4a604ab0953c0aecf4046b12 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b030fe0904e64b67b7e84fb2fe973e67 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae7d058ea9bb4405a82852973b4999c9 : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfedaa2bbb48448b823ed523ae74ca4b : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a7d9e2c087487aaba86ace6da7019e : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d97e4ec61ea440cabd85dca7dc2eefac : SUCCESS in 3m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/44ef084c65f741ed8cea48f517584743 : SUCCESS in 3m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dec5dc94baf744e481b0defebc4dc79d : FAILURE in 5m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/78ec01ad2bf44a2c9183e58f88ee949d : FAILURE in 8m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6f1d640c446647268fa61011c8da4b45 : FAILURE in 9m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ea7127ea373a4caa8c1087c4e3016d9d : FAILURE in 9m 05s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/484367ded64842c2b6831a52e05f8d7e : FAILURE in 12m 30s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d32b9dbfa38f4179a7d9f8d3e4d3a4fe : FAILURE in 13m 36s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e219f601b79245e0b69cf48417be60a3 : FAILURE in 8m 35s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/d4b1bbbfecb84617b9984bab5093b6a2 : FAILURE in 9m 38s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7454a0d800de4a10bbec1bd3aafa340a : FAILURE in 6m 46s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/dc556a91f480467da6c908ffee3ca2b2 : FAILURE in 9m 07s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/3653850898d94afcb9ea0ce9623a64c8 : FAILURE in 6m 48s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9d25fa118fd24960adc7cd96da2c6b34 : FAILURE in 10m 37s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/ac20f5850e1845d3a2ff9ba33abad5a0 : FAILURE in 8m 52s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/078c274ff82e450ab3e17f5b92185dce : FAILURE in 43m 40s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/bb6171b0becd443ca790cf894febde72 : FAILURE in 56m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c1566e0c918c4c71ac9bc1a0bf38b83c : SUCCESS in 52m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aa0e93223bcc4467b861fe5e3f9126bf : SUCCESS in 26m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/04152c8b66034010b5320bfab5321350 : SUCCESS in 1h 04m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e67494585912498fa48a4900d6f4d004 : SUCCESS in 1h 35m 51s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f7cdc4b77dbd46d3b9356d5802ff5c9d : SUCCESS in 2h 30m 39s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2731110a35ca4986a51ff420771d9118 : FAILURE in 2h 17m 26s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d2843cef2ff14638a50ccbd8303d3d5d : FAILURE in 1h 04m 34s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/dca2bb9febac4d6992e0b599301133cb : FAILURE in 23m 17s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/abc442fbf6e94d2cbdfd4b66f0f2d892 : SUCCESS in 1h 08m 49s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/cdce23005e0a49dfb9ac6b4f53ad2d10 : SUCCESS in 1h 19m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ce50b35056a04f98b884bfc05c40f23f : SUCCESS in 2h 09m 00s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1d92c256848940f982e947fa58abcb5b : SUCCESS in 24m 44s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/4c76d8f2914d467d852da2798c465995 : SUCCESS in 53m 39s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/d78c656410844227a32eb82e2999c9ef : FAILURE in 57m 02s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"24e8b921a62ff8608ed02c57e473ec17d816b99d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 19:22:12.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c3e5e1c258a49f3bf76453b5eb03994\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/561f0707ab2048209f38530006b5b9cc : SUCCESS in 3h 00m 07s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/d1069e66d5b2407d895a99b03584ff39 : FAILURE in 1h 09m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/252f58a84204440cbe4410c654d17b10 : SUCCESS in 2h 23m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/261b4a240f0942e7bfd4778a4d2c8f72 : SUCCESS in 1h 16m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98ee5778f524405892934c77476aaa69 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7fe9feb077bb421d8985ec03569e6e15 : SUCCESS in 6m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/19b5b74a855c4bef9ce862fe9840ede4 : FAILURE in 13m 32s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/9942ead79b024f968fc86d237a7c438c : FAILURE in 1h 04m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2c1ca61e447b43f5a5f8ce43184e4a62 : SUCCESS in 1h 06m 26s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"8e2520e0a08b386de0220a67eee6469494464b70","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 06:10:13.000000000","message":"Patch Set 45: Published edit on patch set 44.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"249a94f116582fc8a1ef03d3842e5fdadc56f651","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 06:17:13.000000000","message":"Patch Set 46: Published edit on patch set 45.","accounts_in_message":[],"_revision_number":46},{"id":"4c8910a09e91d782f4b40fd7b762fa9b14b8a6dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 06:28:57.000000000","message":"Patch Set 47: Published edit on patch set 46.","accounts_in_message":[],"_revision_number":47},{"id":"8a7b2e5da5931111430f384b97c553879336b17b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 08:01:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":47},{"id":"bce586b88bd0d76526a4a3b68d75922f46391577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 08:03:31.000000000","message":"Patch Set 48: Published edit on patch set 47.","accounts_in_message":[],"_revision_number":48},{"id":"2e2df694053ab2818046bb540bd6f76488e9e592","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 08:04:34.000000000","message":"Patch Set 49: Patch Set 48 was rebased","accounts_in_message":[],"_revision_number":49},{"id":"cde19402f4a872f44994e6b2f88a298acbe046f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 10:04:02.000000000","message":"Patch Set 50: Patch Set 49 was rebased\n\nThe following files contain Git conflicts:\n* tests/templates/template_overrides.j2\n","accounts_in_message":[],"_revision_number":50},{"id":"cc764223a0c5b5731d66968c3da21986959f5556","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 10:05:21.000000000","message":"Patch Set 51: Published edit on patch set 50.","accounts_in_message":[],"_revision_number":51},{"id":"00c5fe235d19cfa7611c2f1875f25e915cf78165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 12:33:50.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da3e19db4514913b97bf8f053f55dbe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/de22737a7bee42289ea295543616c754 : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/3ef31c74efde4e33b65235abfa22e0e5 : SUCCESS in 41m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9c1fd428ea75450197964503e7e162d1 : SUCCESS in 46m 29s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/7a3c1072e94e44fdba21ab13413a141a : SUCCESS in 1h 05m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c50152f4c8534d079323551fdeafc515 : SUCCESS in 47m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/48284439350d4b67b711b14abb2f7d8d : SUCCESS in 16m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/10cbac275bfa40f7b40e4b6378e129fb : SUCCESS in 58m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7901867dd3804648afef7b8e875e3c26 : SUCCESS in 1h 31m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708d0d5c54c947f08d68d6b6e209d7c7 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7e2ac6ead164e2dbc2cf876f82d78c3 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18fa841135774258af2544b15de1c9a4 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/441a6743819a4640ab0e5cf901167cd3 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438e562ce7c44def87b8167ffec0c12e : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6456413c56924848a21943c3fdd5acaa : SUCCESS in 3m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/65b51dce84ce4820b2d26e20af5d3207 : SUCCESS in 3m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ec243fa90f094be0a1f964034f56ef65 : SUCCESS in 57m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b6b1adb04b6c4dcfb63ef01193a9b81d : SUCCESS in 1h 32m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0ce1bb3f79a54a7eaaa8c67b6a594668 : SUCCESS in 1h 43m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0f23f2da62704072b06b4b994b20c037 : SUCCESS in 2h 09m 54s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7fd7a831a8584e44b223c0c45aef0155 : SUCCESS in 1h 21m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c5e0fcd41f234492b6e02d1b73c81fda : FAILURE in 1h 25m 41s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/355273557f0b4c8e9bab16e783d47e09 : FAILURE in 1h 38m 01s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/d193848faa0d4ff69ebe27097a4a19df : SUCCESS in 32m 57s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ff8412815e65450d8b86ffaf2a2f386f : SUCCESS in 1h 22m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/edc8ced6faa24662be9f277858be66f5 : SUCCESS in 48m 51s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/28ebccc3d5d24e428871241ed97b250d : SUCCESS in 1h 15m 33s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/288887fa9fdf4797a47ada89200cfa5d : SUCCESS in 40m 33s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/fe019645692e473bace487f1a763b411 : SUCCESS in 1h 05m 28s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/f72d762045cb4a0ca7a398de995d0a25 : SUCCESS in 43m 50s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/acde3e0877124a8288b81f6018b9ba88 : SUCCESS in 43m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a9a160c61ad2403aa964e7882c0d1285 : SUCCESS in 53m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/07b9c805e9fe4d159011a0d046976c28 : SUCCESS in 25m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/797af99da58c4779965bc45860c3b003 : SUCCESS in 1h 18m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/92b4dbe24add43d1bebdc453f7108a33 : SUCCESS in 1h 49m 06s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e08c9f834fa944b89afe411b263873bd : SUCCESS in 2h 05m 10s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/05363ef9e88a49b1a75ad2ba6e1c6240 : SUCCESS in 2h 27m 18s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/19b3f56722b946acae1e90439f33043d : FAILURE in 1h 21m 26s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f239c62adc6d49fca3a0935f3d26b59a : SUCCESS in 28m 10s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c135d1b39f604258a10c65d52feec45b : SUCCESS in 40m 23s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/55167f65035943c3af57288dc4da604b : SUCCESS in 1h 16m 04s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8ee1a72e4f384e959ba9becd88e4af8e : SUCCESS in 1h 42m 59s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/54e63a6aa5ab4efc994a1ab0f196422c : SUCCESS in 34m 22s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/46ff31ead6124972a61aed554f7ea4fc : SUCCESS in 59m 37s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/5c7cd0367bc041a48830734b950e354f : FAILURE in 1h 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"752eef8d0620cd58c00a62c46746a83a58b2deda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 13:47:01.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e02d68d42674c449c3962d5f3628d8b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/00a3f333e2ec40efbcbf6d7ce7954377 : SUCCESS in 1h 08m 20s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5dcaed055f454345843342770bffe614 : SUCCESS in 2h 44m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/89d0ebfb5df740b2ab052090f368049c : SUCCESS in 2h 39m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f721a06d18246409a03d396c9675c31 : SUCCESS in 2h 40m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a54c22078824a0997ccf3024142634e : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8bace50be3cd4778a533aef61ca3f9f7 : SUCCESS in 4m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/20feecbf64d047a2b10d666f4df39b23 : SUCCESS in 1h 32m 31s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/85a9e72507a74eeb9133b16425d1c2b7 : SUCCESS in 1h 10m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/787793cdc2ef44919de30deb716f0409 : SUCCESS in 58m 41s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"a9a6470e587f724fc4777ead26ec085f656b05a9","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-12-04 13:53:42.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"b1a020871e3d49fac197a665f85ad7541b2ece55","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-04 14:07:10.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"cb4486cfc2b26f6e27a3cc66f1d3f5f8be7dc74d","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-12-04 14:08:30.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"ed2faf74ce8af78d2cb1dfbf0383267bee02e6d1","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-05 06:13:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":51},{"id":"2abd9e21a4081db55f64fac88a721d7bb95c532d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-05 06:13:38.000000000","message":"Patch Set 51: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":51},{"id":"0a809d27fe8095b20c4558e6210afa0d4fef5d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 06:14:08.000000000","message":"Patch Set 51: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":51},{"id":"9f37eac44742cb73f43d49920df8c41223ab3b23","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-05 08:22:55.000000000","message":"Patch Set 51: Code-Review+1","accounts_in_message":[],"_revision_number":51},{"id":"d66dcedb2a3ee0c8a933119e8dfebfd1f0865462","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 08:37:29.000000000","message":"Patch Set 51: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f50d7335484a9896d6fb94442f12b8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/80bbfb0296b1435fb457e2606379ef9e : SUCCESS in 38m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1f6158f5dd3e488c9caab8e3eee47601 : SUCCESS in 16m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef39f27bf1194d7da89a7e9946e7f316 : SUCCESS in 1h 25m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/51f19f34db6542e7bffb7c1e362bfba4 : SUCCESS in 1h 41m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24522bc341e452eab217a534e7365a2 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06a32ef5ab554d0c920c5e0f2aadbe59 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/019eaecbbbff4a5aa135d69a79fd0f98 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a8b437494a4242864f96fa97967ad6 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50f312c4f3274b7294b069b6145f12eb : SUCCESS in 3m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/96a4ac01cfdd4c23919bf31bc25e9ef4 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/042413652d0e4c4296be744cb5f2a0ae : SUCCESS in 52m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a7a9670194b0488696a42b8529286c2b : SUCCESS in 45m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d810a9843a0b46d09dc472fda5571ff4 : SUCCESS in 1h 22m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/741fdaaf7e7345b68483ff8588f661fe : SUCCESS in 2h 11m 24s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/dbad28a5b0d548f18a6515df4757fed2 : SUCCESS in 49m 26s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/77a7f39e685b42f1a5099968e9849581 : SUCCESS in 42m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a8b2f83ed1b8415baa58adaa3ff440fb : SUCCESS in 35m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e561c49515d24be0b81bfa63793c81cf : SUCCESS in 31m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/976029517b07486897603a71ee1df242 : SUCCESS in 1h 07m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8aadb27843df45938768498acc89db93 : SUCCESS in 1h 41m 03s","accounts_in_message":[],"_revision_number":51},{"id":"a49a25868b469fe47a12b05ac97dbe78592f1a46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 08:37:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":51},{"id":"6ec0bdba2d09a1879428fb7bd58695eee43660f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 08:39:02.000000000","message":"Patch Set 51:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d146ddb7f4d44f6bb43d479c4e02c22\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edfe1a5d5424defb00a4bcc936558bd : SUCCESS in 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b6803e9ae1a47ca8c1fe6fcf6c62962 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":51}],"current_revision_number":51,"current_revision":"584d57e7c378eb377e823b7cdf2c886ccf116312","revisions":{"564f3c94899a21aea4db8789e729a5036ff1601b":{"kind":"REWORK","_number":1,"created":"2025-11-18 11:21:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/1"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 11:21:06.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/564f3c94899a21aea4db8789e729a5036ff1601b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/564f3c94899a21aea4db8789e729a5036ff1601b"}]},"branch":"refs/heads/stable/2025.1"},"9821e5c1466cae84b2dfa058bdc22f78063ea9e4":{"kind":"REWORK","_number":2,"created":"2025-11-18 11:23:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/2"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 11:22:55.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9821e5c1466cae84b2dfa058bdc22f78063ea9e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9821e5c1466cae84b2dfa058bdc22f78063ea9e4"}]},"branch":"refs/heads/stable/2025.1"},"04b32d5bf85d4dc7d078b7e881d0c18412c21cb8":{"kind":"REWORK","_number":3,"created":"2025-11-18 11:24:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/3"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 11:24:48.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04b32d5bf85d4dc7d078b7e881d0c18412c21cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04b32d5bf85d4dc7d078b7e881d0c18412c21cb8"}]},"branch":"refs/heads/stable/2025.1"},"6e069e8320962000fb7951526e33a7e76569859e":{"kind":"REWORK","_number":4,"created":"2025-11-18 11:29:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/4"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 11:29:15.000000000","tz":60},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e069e8320962000fb7951526e33a7e76569859e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e069e8320962000fb7951526e33a7e76569859e"}]},"branch":"refs/heads/stable/2025.1"},"2405530a05f524ef1f5dbdbe2fbefd9513345235":{"kind":"REWORK","_number":5,"created":"2025-11-18 11:53:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/5"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 11:53:24.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2405530a05f524ef1f5dbdbe2fbefd9513345235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2405530a05f524ef1f5dbdbe2fbefd9513345235"}]},"branch":"refs/heads/stable/2025.1"},"e3f380f4234baaa77732e8a184abd27e8993241d":{"kind":"REWORK","_number":6,"created":"2025-11-18 11:55:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/6"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 11:55:44.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3f380f4234baaa77732e8a184abd27e8993241d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3f380f4234baaa77732e8a184abd27e8993241d"}]},"branch":"refs/heads/stable/2025.1"},"2c0c50eeb5e54d4b65a44c5f531e1223335a4356":{"kind":"REWORK","_number":7,"created":"2025-11-18 11:57:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/7"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 11:57:06.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c0c50eeb5e54d4b65a44c5f531e1223335a4356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c0c50eeb5e54d4b65a44c5f531e1223335a4356"}]},"branch":"refs/heads/stable/2025.1"},"306bb05e8318354a20c9506dc637d39cea53a528":{"kind":"REWORK","_number":8,"created":"2025-11-18 11:57:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/8"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 11:57:22.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/306bb05e8318354a20c9506dc637d39cea53a528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/306bb05e8318354a20c9506dc637d39cea53a528"}]},"branch":"refs/heads/stable/2025.1"},"bef1b730bae3f11f6c9538f00b7b257207d62b28":{"kind":"REWORK","_number":9,"created":"2025-11-18 12:00:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/9"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 12:00:36.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bef1b730bae3f11f6c9538f00b7b257207d62b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bef1b730bae3f11f6c9538f00b7b257207d62b28"}]},"branch":"refs/heads/stable/2025.1"},"13a24c94167c7ed80dfd859a9e9bbf9481d413c1":{"kind":"REWORK","_number":10,"created":"2025-11-18 12:16:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/10"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 12:16:27.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13a24c94167c7ed80dfd859a9e9bbf9481d413c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13a24c94167c7ed80dfd859a9e9bbf9481d413c1"}]},"branch":"refs/heads/stable/2025.1"},"01907bd32b9fbbfccbdfe6c6affc132d9084b430":{"kind":"REWORK","_number":11,"created":"2025-11-18 12:26:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/11"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 12:26:05.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/961237\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01907bd32b9fbbfccbdfe6c6affc132d9084b430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01907bd32b9fbbfccbdfe6c6affc132d9084b430"}]},"branch":"refs/heads/stable/2025.1"},"d5c92884d56189ab0a44248740fff44089c68ad6":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2025-11-18 12:37:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/12"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 12:37:30.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5c92884d56189ab0a44248740fff44089c68ad6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5c92884d56189ab0a44248740fff44089c68ad6"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"304ab06d5dcefe1ca79b87d4df8a154834e1003b":{"kind":"REWORK","_number":13,"created":"2025-11-18 13:46:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/13"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 13:46:39.000000000","tz":0},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/304ab06d5dcefe1ca79b87d4df8a154834e1003b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/304ab06d5dcefe1ca79b87d4df8a154834e1003b"}]},"branch":"refs/heads/stable/2025.1"},"b02d12caebb0fa5fd4d13369593900ca93aae7fa":{"kind":"REWORK","_number":14,"created":"2025-11-18 14:04:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/14"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-18 14:04:43.000000000","tz":60},"subject":"Move to Rocky Linux 10","message":"Move to Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02d12caebb0fa5fd4d13369593900ca93aae7fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02d12caebb0fa5fd4d13369593900ca93aae7fa"}]},"branch":"refs/heads/stable/2025.1"},"483fd1edf39713a6f669e7ad0c199b6c1c0d87b1":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2025-11-24 16:09:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/15"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-24 16:09:57.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/483fd1edf39713a6f669e7ad0c199b6c1c0d87b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/483fd1edf39713a6f669e7ad0c199b6c1c0d87b1"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"bfddca56c36f92439b8ad3e13325e7fb17ae22a7":{"kind":"REWORK","_number":16,"created":"2025-11-27 11:40:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/16"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-27 11:40:44.000000000","tz":-300},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfddca56c36f92439b8ad3e13325e7fb17ae22a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfddca56c36f92439b8ad3e13325e7fb17ae22a7"}]},"branch":"refs/heads/stable/2025.1"},"3a66edce562ec5723468e214c178ed1b8a4fe1fe":{"kind":"REWORK","_number":17,"created":"2025-11-27 12:29:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/17"}}},"commit":{"parents":[{"commit":"47313f61e3d110315c94f9de6b27a8df8aed7d82","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47313f61e3d110315c94f9de6b27a8df8aed7d82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-27 12:29:54.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a66edce562ec5723468e214c178ed1b8a4fe1fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a66edce562ec5723468e214c178ed1b8a4fe1fe"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"3aed94a1b120a89d24abbbbf4d21e249e908e6a8":{"kind":"REWORK","_number":18,"created":"2025-11-27 12:31:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/18"}}},"commit":{"parents":[{"commit":"47313f61e3d110315c94f9de6b27a8df8aed7d82","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47313f61e3d110315c94f9de6b27a8df8aed7d82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-27 12:31:10.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3aed94a1b120a89d24abbbbf4d21e249e908e6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3aed94a1b120a89d24abbbbf4d21e249e908e6a8"}]},"branch":"refs/heads/stable/2025.1"},"d3f3ea9a49f9fa909ed24bed807f310347687f3e":{"kind":"REWORK","_number":19,"created":"2025-12-01 08:13:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/19"}}},"commit":{"parents":[{"commit":"fb89e3a61a934018c971d4cba6558254db00ea3d","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb89e3a61a934018c971d4cba6558254db00ea3d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-01 08:13:58.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3f3ea9a49f9fa909ed24bed807f310347687f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3f3ea9a49f9fa909ed24bed807f310347687f3e"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"0b29002f72665d122805f1add8957ab30281214c":{"kind":"REWORK","_number":20,"created":"2025-12-01 08:15:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/20"}}},"commit":{"parents":[{"commit":"fb89e3a61a934018c971d4cba6558254db00ea3d","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb89e3a61a934018c971d4cba6558254db00ea3d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-01 08:15:13.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b29002f72665d122805f1add8957ab30281214c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b29002f72665d122805f1add8957ab30281214c"}]},"branch":"refs/heads/stable/2025.1"},"9bbd8c22ed80f71f3bf29a60514da5c0d059d3fe":{"kind":"REWORK","_number":21,"created":"2025-12-02 12:53:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/21"}}},"commit":{"parents":[{"commit":"a9418179e490fd2c1bfc12a180765d53a41854c5","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9418179e490fd2c1bfc12a180765d53a41854c5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:53:37.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bbd8c22ed80f71f3bf29a60514da5c0d059d3fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bbd8c22ed80f71f3bf29a60514da5c0d059d3fe"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"b6b67115ecf1337ce285db5e0725103e2a03516e":{"kind":"REWORK","_number":22,"created":"2025-12-02 12:54:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/22"}}},"commit":{"parents":[{"commit":"a9418179e490fd2c1bfc12a180765d53a41854c5","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9418179e490fd2c1bfc12a180765d53a41854c5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:54:20.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6b67115ecf1337ce285db5e0725103e2a03516e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6b67115ecf1337ce285db5e0725103e2a03516e"}]},"branch":"refs/heads/stable/2025.1"},"a4611dddb0da8095618ec75bf099f39358a335f2":{"kind":"REWORK","_number":23,"created":"2025-12-02 12:54:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/23"}}},"commit":{"parents":[{"commit":"a9418179e490fd2c1bfc12a180765d53a41854c5","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9418179e490fd2c1bfc12a180765d53a41854c5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:54:37.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4611dddb0da8095618ec75bf099f39358a335f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4611dddb0da8095618ec75bf099f39358a335f2"}]},"branch":"refs/heads/stable/2025.1"},"a043295e782e6d7aeae0a88656d4a4bfaf009eca":{"kind":"REWORK","_number":24,"created":"2025-12-02 12:55:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/24"}}},"commit":{"parents":[{"commit":"a9418179e490fd2c1bfc12a180765d53a41854c5","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9418179e490fd2c1bfc12a180765d53a41854c5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:55:28.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a043295e782e6d7aeae0a88656d4a4bfaf009eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a043295e782e6d7aeae0a88656d4a4bfaf009eca"}]},"branch":"refs/heads/stable/2025.1"},"f7bf21d8e1aef2e973480e75a0c1c57d544ba7da":{"kind":"REWORK","_number":25,"created":"2025-12-02 15:56:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/25"}}},"commit":{"parents":[{"commit":"cad00c5d12a673c6e82f4cd884e6357576b00ca2","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cad00c5d12a673c6e82f4cd884e6357576b00ca2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 15:56:08.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7bf21d8e1aef2e973480e75a0c1c57d544ba7da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7bf21d8e1aef2e973480e75a0c1c57d544ba7da"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"7a450851fdde909f7f4dd96b9e9de062d85af0fd":{"kind":"REWORK","_number":26,"created":"2025-12-02 15:59:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/26"}}},"commit":{"parents":[{"commit":"cad00c5d12a673c6e82f4cd884e6357576b00ca2","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cad00c5d12a673c6e82f4cd884e6357576b00ca2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 15:59:52.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a450851fdde909f7f4dd96b9e9de062d85af0fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a450851fdde909f7f4dd96b9e9de062d85af0fd"}]},"branch":"refs/heads/stable/2025.1"},"ecf1aeb220358f248874a1ee0cbdaa6e0b52bf60":{"kind":"REWORK","_number":27,"created":"2025-12-02 16:00:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/27"}}},"commit":{"parents":[{"commit":"cad00c5d12a673c6e82f4cd884e6357576b00ca2","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cad00c5d12a673c6e82f4cd884e6357576b00ca2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 16:00:42.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ecf1aeb220358f248874a1ee0cbdaa6e0b52bf60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ecf1aeb220358f248874a1ee0cbdaa6e0b52bf60"}]},"branch":"refs/heads/stable/2025.1"},"ea155363312d972ba089c37802c1abccee095552":{"kind":"REWORK","_number":28,"created":"2025-12-02 16:01:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/28"}}},"commit":{"parents":[{"commit":"cad00c5d12a673c6e82f4cd884e6357576b00ca2","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cad00c5d12a673c6e82f4cd884e6357576b00ca2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 16:01:26.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea155363312d972ba089c37802c1abccee095552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea155363312d972ba089c37802c1abccee095552"}]},"branch":"refs/heads/stable/2025.1"},"004545e2a3d7b6d754f63852aad7562aa2aadd23":{"kind":"REWORK","_number":29,"created":"2025-12-02 16:18:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/29"}}},"commit":{"parents":[{"commit":"cad00c5d12a673c6e82f4cd884e6357576b00ca2","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cad00c5d12a673c6e82f4cd884e6357576b00ca2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 16:18:25.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/004545e2a3d7b6d754f63852aad7562aa2aadd23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/004545e2a3d7b6d754f63852aad7562aa2aadd23"}]},"branch":"refs/heads/stable/2025.1"},"d67f1be6a35954e905f1b73a8963be3304d9c45d":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2025-12-02 17:24:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/30"}}},"commit":{"parents":[{"commit":"9d08b218e6acd532736d0e5c447c9571fd5a8e21","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d08b218e6acd532736d0e5c447c9571fd5a8e21"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 17:24:13.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d67f1be6a35954e905f1b73a8963be3304d9c45d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d67f1be6a35954e905f1b73a8963be3304d9c45d"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"5bd8aa631c1f57bb400a36d87c29719adcb19977":{"kind":"REWORK","_number":31,"created":"2025-12-03 10:54:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/31"}}},"commit":{"parents":[{"commit":"d9ea1c9a82ac42349e44985689c56079666cc5e5","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9ea1c9a82ac42349e44985689c56079666cc5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 10:54:35.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bd8aa631c1f57bb400a36d87c29719adcb19977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bd8aa631c1f57bb400a36d87c29719adcb19977"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"47737d42da3f0ca2fa114850c3e9d80a2680bdd7":{"kind":"REWORK","_number":32,"created":"2025-12-03 10:55:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/32"}}},"commit":{"parents":[{"commit":"d9ea1c9a82ac42349e44985689c56079666cc5e5","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9ea1c9a82ac42349e44985689c56079666cc5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 10:55:06.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47737d42da3f0ca2fa114850c3e9d80a2680bdd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47737d42da3f0ca2fa114850c3e9d80a2680bdd7"}]},"branch":"refs/heads/stable/2025.1"},"3b2710d7d4005cbe956fa7c79494e56156c70355":{"kind":"REWORK","_number":33,"created":"2025-12-03 10:56:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/33"}}},"commit":{"parents":[{"commit":"75038870806407717fcd3ba0c77778606c9bece8","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75038870806407717fcd3ba0c77778606c9bece8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 10:56:43.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b2710d7d4005cbe956fa7c79494e56156c70355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b2710d7d4005cbe956fa7c79494e56156c70355"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"95a1ef588d27392de0bd961e98224c3f8ec7759c":{"kind":"REWORK","_number":34,"created":"2025-12-03 10:57:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/34"}}},"commit":{"parents":[{"commit":"75038870806407717fcd3ba0c77778606c9bece8","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75038870806407717fcd3ba0c77778606c9bece8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 10:57:21.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a1ef588d27392de0bd961e98224c3f8ec7759c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a1ef588d27392de0bd961e98224c3f8ec7759c"}]},"branch":"refs/heads/stable/2025.1"},"8eb235e7adaa1e8ba36aeb62e29b49295d6ec8d8":{"kind":"REWORK","_number":35,"created":"2025-12-03 11:17:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/35"}}},"commit":{"parents":[{"commit":"75038870806407717fcd3ba0c77778606c9bece8","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75038870806407717fcd3ba0c77778606c9bece8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 11:17:10.000000000","tz":60},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eb235e7adaa1e8ba36aeb62e29b49295d6ec8d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eb235e7adaa1e8ba36aeb62e29b49295d6ec8d8"}]},"branch":"refs/heads/stable/2025.1"},"7edf54f08d424763079c8de60f937429189fff35":{"kind":"REWORK","_number":36,"created":"2025-12-03 11:26:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/36"}}},"commit":{"parents":[{"commit":"75038870806407717fcd3ba0c77778606c9bece8","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75038870806407717fcd3ba0c77778606c9bece8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 11:26:50.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7edf54f08d424763079c8de60f937429189fff35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7edf54f08d424763079c8de60f937429189fff35"}]},"branch":"refs/heads/stable/2025.1"},"0491102dad9136683dfd2347ec98f071c2261d26":{"kind":"REWORK","_number":37,"created":"2025-12-03 11:51:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/37"}}},"commit":{"parents":[{"commit":"75038870806407717fcd3ba0c77778606c9bece8","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75038870806407717fcd3ba0c77778606c9bece8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 11:51:19.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0491102dad9136683dfd2347ec98f071c2261d26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0491102dad9136683dfd2347ec98f071c2261d26"}]},"branch":"refs/heads/stable/2025.1"},"1c43f85fbd113eaadc850c94bf2af6609ebc06be":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2025-12-03 12:03:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/38"}}},"commit":{"parents":[{"commit":"6f6717a1fc8c257c80f4e3efd0c56cf60bbf4f90","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f6717a1fc8c257c80f4e3efd0c56cf60bbf4f90"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 12:03:46.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c43f85fbd113eaadc850c94bf2af6609ebc06be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c43f85fbd113eaadc850c94bf2af6609ebc06be"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"3884e73d40dc31c1758092b80cf458d0f7772f9f":{"kind":"REWORK","_number":39,"created":"2025-12-03 12:05:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/39"}}},"commit":{"parents":[{"commit":"6f6717a1fc8c257c80f4e3efd0c56cf60bbf4f90","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f6717a1fc8c257c80f4e3efd0c56cf60bbf4f90"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 12:05:07.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3884e73d40dc31c1758092b80cf458d0f7772f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3884e73d40dc31c1758092b80cf458d0f7772f9f"}]},"branch":"refs/heads/stable/2025.1"},"ed203a5191d7bf03378ae1d8593f230b0b7c915a":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2025-12-03 12:10:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/40"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 12:10:34.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed203a5191d7bf03378ae1d8593f230b0b7c915a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed203a5191d7bf03378ae1d8593f230b0b7c915a"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"d069f924b371080222849686d925a5b6532ca576":{"kind":"REWORK","_number":41,"created":"2025-12-03 12:49:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/41"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 12:49:54.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d069f924b371080222849686d925a5b6532ca576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d069f924b371080222849686d925a5b6532ca576"}]},"branch":"refs/heads/stable/2025.1"},"0f8387bd9bf746e32e97deb15141d7c636626baf":{"kind":"REWORK","_number":42,"created":"2025-12-03 13:24:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/42"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 13:24:28.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f8387bd9bf746e32e97deb15141d7c636626baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f8387bd9bf746e32e97deb15141d7c636626baf"}]},"branch":"refs/heads/stable/2025.1"},"fb5d3019d4cca14183472979385c5033cc91c734":{"kind":"REWORK","_number":43,"created":"2025-12-03 13:49:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/43"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 13:49:06.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb5d3019d4cca14183472979385c5033cc91c734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb5d3019d4cca14183472979385c5033cc91c734"}]},"branch":"refs/heads/stable/2025.1"},"1cd3a5cb163b1506b03605550576ccac66dbe75e":{"kind":"REWORK","_number":44,"created":"2025-12-03 16:06:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/44"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:06:44.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cd3a5cb163b1506b03605550576ccac66dbe75e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cd3a5cb163b1506b03605550576ccac66dbe75e"}]},"branch":"refs/heads/stable/2025.1"},"04d5a53acf5020dc62e624fa1532761761b76607":{"kind":"REWORK","_number":45,"created":"2025-12-04 06:10:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/45"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-04 06:10:13.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04d5a53acf5020dc62e624fa1532761761b76607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04d5a53acf5020dc62e624fa1532761761b76607"}]},"branch":"refs/heads/stable/2025.1"},"9a06788f93c8293349916cfdb00a0f953785789f":{"kind":"REWORK","_number":46,"created":"2025-12-04 06:17:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/46"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-04 06:17:13.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a06788f93c8293349916cfdb00a0f953785789f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a06788f93c8293349916cfdb00a0f953785789f"}]},"branch":"refs/heads/stable/2025.1"},"6823e6da4303f89f6f17f7a0e9d390ae3f568745":{"kind":"REWORK","_number":47,"created":"2025-12-04 06:28:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/47"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-04 06:28:57.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6823e6da4303f89f6f17f7a0e9d390ae3f568745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6823e6da4303f89f6f17f7a0e9d390ae3f568745"}]},"branch":"refs/heads/stable/2025.1"},"1ecd9e93eba6f54eb1ed38393380c7f74cdceaa6":{"kind":"REWORK","_number":48,"created":"2025-12-04 08:03:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/48"}}},"commit":{"parents":[{"commit":"19eb419ff2bd1ff9548b2005ac986d023b136f79","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19eb419ff2bd1ff9548b2005ac986d023b136f79"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-04 08:03:31.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ecd9e93eba6f54eb1ed38393380c7f74cdceaa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ecd9e93eba6f54eb1ed38393380c7f74cdceaa6"}]},"branch":"refs/heads/stable/2025.1"},"7091500757e45be567ac5f383b870897ea9bddcc":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2025-12-04 08:04:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/49"}}},"commit":{"parents":[{"commit":"5e02ed4d71c42795f80d1c9165647543ebfe9969","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e02ed4d71c42795f80d1c9165647543ebfe9969"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-04 08:04:34.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7091500757e45be567ac5f383b870897ea9bddcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7091500757e45be567ac5f383b870897ea9bddcc"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"c1e23bbeb99dd347faa525a851187e83b6a74b03":{"kind":"REWORK","_number":50,"created":"2025-12-04 10:04:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/50"}}},"commit":{"parents":[{"commit":"5d129e2870c92709a1e911f65433bcb1758de1e4","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d129e2870c92709a1e911f65433bcb1758de1e4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-04 10:04:02.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1e23bbeb99dd347faa525a851187e83b6a74b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1e23bbeb99dd347faa525a851187e83b6a74b03"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"584d57e7c378eb377e823b7cdf2c886ccf116312":{"kind":"REWORK","_number":51,"created":"2025-12-04 10:05:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/967507/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/967507/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/967507/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/967507/51"}}},"commit":{"parents":[{"commit":"5d129e2870c92709a1e911f65433bcb1758de1e4","subject":"Add support for CentOS Stream 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d129e2870c92709a1e911f65433bcb1758de1e4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 12:24:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-04 10:05:21.000000000","tz":0},"subject":"Add support for Rocky Linux 10","message":"Add support for Rocky Linux 10\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/962714\n\nChange-Id: Idd809f5efd2d308c1565d7635e67af0a431d18f7\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit d4b777bb060517dbb0b648d5d7d5c07336969866)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/584d57e7c378eb377e823b7cdf2c886ccf116312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/584d57e7c378eb377e823b7cdf2c886ccf116312"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
