)]}'
{"id":"openstack%2Fkolla~967657","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf","project":"openstack/kolla","branch":"stable/2025.1","topic":"update-debian-package-source-stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-21 10:36:03.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf","subject":"Transition to the new rabbitmq Debian package archives.","status":"MERGED","created":"2025-11-19 07:26:39.000000000","updated":"2025-11-21 15:17:16.000000000","submitted":"2025-11-21 15:16:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967657-update-debian-package-source-stable/2025.1","cherry_pick_of_change":967475,"cherry_pick_of_patch_set":6,"meta_rev_id":"8ef44d6c99a7b26af837f8a1ebf743022da6601d","_number":967657,"virtual_id_number":967657,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-21 15:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-19 18:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-11-19 10:20:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-19 18:19:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 07:26:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"CC"},{"updated":"2025-11-19 10:20:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-19 10:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-19 14:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8558e993dcfe87f770072545c557eb5ff45627fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 07:26:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/tests/test_methods.py","accounts_in_message":[],"_revision_number":1},{"id":"9a8e7a342b15ce5757866f60aab426bab307b1af","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 07:27:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2051e34f232d98ec72441a14fd1571b009318d64","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 07:27:44.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"21813b49a9419c97610c3060524e9a59cbd00ae1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-19 10:20:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c8ef97c8c9805d266278c5180f3d20974f814c16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:31:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e523a629a73e4e8797aa9f7a346da37e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aef9318346ff49a9858df346de62ad53 : SUCCESS in 3h 01m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba142671f5db4fe9adf2ef773bfe108f : SUCCESS in 2h 38m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4e474b90f90744578eef2cf50cb7d00d : SUCCESS in 2h 01m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0febf2b0abc644409450073ab749929d : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e1cac6be98342b8b9f359667afae536 : SUCCESS in 5m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f8489425f61443619899db072e861b08 : SUCCESS in 1h 02m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/83b2fd4678974314a258f977365584af : SUCCESS in 1h 05m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6604c29112e729b8a48e4213becbd2520a7dd92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 14:47:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd213599e0c7478bb017fe394f612c09\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3c1742248eb147188f2f24b2784db8a5 : SUCCESS in 38m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b91b5380939e4efc86b96ba1de8276d4 : FAILURE in 41m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/14d03864c8aa4d799a10ee4e179a4688 : SUCCESS in 40m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1ed6a774747a4ff4b3ca802d124aa0d8 : SUCCESS in 27m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/43dd5f0e11a5467aba564c53058e94e8 : SUCCESS in 57m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/faae2154face45cf9a64dfd0a5275c57 : SUCCESS in 1h 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b7492ed5ec48e68daaa2131a1d9607 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c44b34bf493f41e2ba5baac505a564b8 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b800b6a531c7485fb301d8caad6bdbfb : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d480688d7e0245a48d065a4a479ad015 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d16b71f0b043118d7f1cf7b2c82300 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dda4afc6fa594a429af10c92fe058250 : SUCCESS in 3m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ed10208cff254092b33cf034d6118f4c : SUCCESS in 1m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/529fdbcb0bfb41158e77032f15683e42 : TIMED_OUT in 3h 02m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3cdbbe2e4d574c16855e5ca5c43d1e70 : SUCCESS in 1h 22m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/52c2ef4d64834e8daf9ed540567eb436 : SUCCESS in 1h 28m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d595f4d2eaaf438eb7e8de43db026a9b : SUCCESS in 1h 57m 55s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/6db2fb42f5cb4d66bd3fdee5e3702033 : SUCCESS in 2h 15m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/242299f917a34fc1bbc7985d690fdf90 : FAILURE in 1h 06m 09s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/410b5f0ddf1f40a094b0519bde42a283 : FAILURE in 1h 21m 41s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/0bebb307489b4df9a2e00cae6f50d2ca : SUCCESS in 32m 21s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/68ee319e3de140fe8d2f758862e47491 : SUCCESS in 1h 22m 09s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/230eb4b5d4e14b08a03b0962c9e68528 : SUCCESS in 1h 18m 15s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/92bd1aa5bf47446bbe96988efc1adf0f : SUCCESS in 2h 02m 05s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/be70993c88ce4b90b9ff1216bf99a3b5 : SUCCESS in 26m 54s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1ec5f9140f3549aba0821ecc28953334 : SUCCESS in 45m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cb1bbf01d26944328c4e1484066070df : SUCCESS in 33m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e7307c5d13694d009c916e073c77ee7b : SUCCESS in 1h 06m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/433d6fd225b44255bbc0b13cca17f892 : SUCCESS in 1h 41m 17s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1fcb6bcbbc8d4deea2c3024cad3b949b : SUCCESS in 1h 53m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3c7daf0cd8604801ac0bc870311175bf : SUCCESS in 2h 45m 05s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9f69784ca5ce4b418fb51975255a813a : FAILURE in 40m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/09203844d7234adc9ec15f5acb59e967 : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/487890c140154bc5b77f479b8e8d17fe : SUCCESS in 1h 10m 26s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3fa4bac5f3d34a7582ac9e1222b0d085 : SUCCESS in 1h 10m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/495a1504a68b455091034312a6e26c3e : SUCCESS in 1h 41m 51s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1046d6e1308448edb764ad1c93bb2a16 : SUCCESS in 18m 30s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/aed51da1f6614c93b0eb149b258d1abe : FAILURE in 1h 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fce85f42173bf68850170734dd6e7009c5726cc3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-19 15:06:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb13138c04dc7129bbb720dd5f8311af82a52058","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 17:12:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd532fa11f484fe1ad8b0861d4a7b061\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4165126c75a84710b06f60e153072f93 : SUCCESS in 1h 00m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e37b5de97de5416088e0d5cf84c8dade : SUCCESS in 2h 00m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5b22c38734794f12bd049116c3d04e49 : SUCCESS in 1h 58m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49b4e4babd1c4f49bea8be99c725072f : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/58dc2079bae64de48e94859371158b4d : SUCCESS in 3m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/08949ee64f354cae884b0daed25a40fa : SUCCESS in 59m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/463009d80ea5425a8b964d9ffd6690cc : SUCCESS in 1h 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55d6db89791c960f3e9105869b53e6b9ae03eaf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 17:43:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7a8bcb5e4e4a6b88fe0d954c67c74d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/132c3ae6322c44c6b905394d5a2a0d2d : SUCCESS in 47m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fab3adb043f049aeadf819860d277bcb : FAILURE in 30m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/968312872497449d9adb61614472d454 : SUCCESS in 34m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/72f6cf33f1fe4027a275d95853a95dee : SUCCESS in 15m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e9583d243112447f8ba6ee5e2f534858 : SUCCESS in 1h 24m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a7f3cc6e93d0402bb0780a93689ee531 : SUCCESS in 1h 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4eb96f808e44101a68a29a25d7d01fc : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32de4d10ddb7440f8bad7ed5d990c977 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41b118727e044cc5826c0b597c7b72d1 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb141a169ae941d4b7801b76559db16c : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25658b7fb704714a7d8651b728c5201 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/594b550483c24340bbd60e25c1286636 : SUCCESS in 2m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f3a709b7e5af4cfcbe60c12a8ebbbae0 : SUCCESS in 2m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e816f2b7b4d74e0c8620883b5e045d76 : SUCCESS in 43m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/af2fdffbfa884495b3ede59922492ef4 : SUCCESS in 1h 00m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/220fda16569945668120de47683c14a7 : SUCCESS in 1h 31m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2a5bf80b49c647fa94c312dd6fd3a9a2 : SUCCESS in 1h 05m 47s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/80cacb82d631427585dc0bf2ff652a5a : SUCCESS in 1h 22m 50s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f9f8e322fb3947289f5d563a9271f18a : FAILURE in 2h 36m 04s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/05ac2ce8da16488195dc7876b0e727ff : FAILURE in 1h 14m 51s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/1545f031f8a9428a9fa0d7160cd61369 : SUCCESS in 17m 30s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a034813f4d814cc88b721e9652462d88 : SUCCESS in 1h 18m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e72c35df9f214cc285fcdaa7fd4de4b9 : SUCCESS in 1h 32m 39s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2e5eb1cefbcf450fb358f3bc648b74cf : SUCCESS in 1h 17m 52s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9e8d6d9722e149998320c0438264006f : SUCCESS in 38m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7dd16764920b47bfab95941f2b25a1ba : SUCCESS in 52m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/656ef5b3a24b449a859245a1ac74153c : SUCCESS in 15m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/19aef0bf984f4583b100110d8a056fff : SUCCESS in 1h 19m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b561d0acfbc34383b346e21cbf9b6621 : SUCCESS in 1h 51m 00s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fe0605e33a664c079bb3c18237bf24fb : SUCCESS in 2h 05m 55s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/babeea08acc445cebe3069763edda87b : FAILURE in 1h 00m 21s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/714762cce12a4131b582d9e7696bc10b : FAILURE in 1h 15m 14s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/663c58370c404dcbbb7b691fce4692d1 : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/77b47693b71844a8a837263cc75ef02a : SUCCESS in 46m 03s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9948c9b917064a998b6b18f739f55750 : SUCCESS in 1h 22m 44s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f4f34046eb14422f8c1acf814722cfda : SUCCESS in 1h 51m 27s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bf4bce117fe24c8287a72efc6d272917 : SUCCESS in 18m 24s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/fa39625a8ea244a49ad428dfd8cefd72 : FAILURE in 1h 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"26bef7885196f029ee32cd6479bfe0c7ea6d2f8a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 18:19:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d9ae5543f11b4e2baf80069c7a84e62b3c987de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 18:19:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"782b1869dba7951865dc8fefdd26e279d5eedd5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 21:31:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2759999dd67d4ba281ec8d77f6f1a1bb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d38b41bca4d44dd5b9f2e6d3c49bff6f : SUCCESS in 53m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0f26c70fcf5e45bea90425aca8143a30 : FAILURE in 31m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ab0daf2cb7344902aa98bc62f82fb99d : SUCCESS in 59m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e0827392c4944def84d89addb3f7fc1f : SUCCESS in 1h 26m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d545a2ab3d44a99f6cd440e3952557 : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21651676fc7a43598366638a036347bb : SUCCESS in 1m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ceaf0ec3c1a34fbaad0f74eec49b4c84 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26e2c4e2cfc4fb7bb3fb9ca9141778f : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fce651783a9d4a168ca01fc6e54c04dd : SUCCESS in 3m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b26fbb4b07c1408ab632bd1fd2f65115 : SUCCESS in 2m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/eb0849facd77495fabc53a9f61274c2f : TIMED_OUT in 3h 02m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4c9ead8f692e445493e90070dec4ea1b : SUCCESS in 1h 20m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d1a494f89a4249e2844903cf10bd7a86 : SUCCESS in 45m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a7a1f3bbad7247b7992d5aaafc4b7ff8 : SUCCESS in 2h 05m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/43998fd75182464892400a943446866e : FAILURE in 35m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/420ff6496f5d47bcb75167759420f31a : SUCCESS in 14m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cd883609de1a46a180d1a26e2722a6d8 : SUCCESS in 1h 06m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9eff7402651b499181166a93651b419c : SUCCESS in 1h 35m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7d5423c8578b2acb59a1c9c0fb3a1cac9ff7b01f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-21 08:09:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"083d3c5cb33303529cb7fdf5b80b58070d3381d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 10:34:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfd0e39e8468467486fb48033f3464a1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c16de70c301c4a52a930c51bc0340ba2 : SUCCESS in 49m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/def207f294be438b867ce6e002cb3842 : FAILURE in 40m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1648e520cf0e4ff4b115ab26f0f03836 : SUCCESS in 51m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4d98811e876e43b8961cde81b5783490 : SUCCESS in 30m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/54c48b6ba0e34240ba84ef2d1d2f2e15 : SUCCESS in 1h 10m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b5c5e347bc5e43c7b81c8974f4dbb44c : SUCCESS in 1h 43m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91567874ad74f979819053218a77887 : SUCCESS in 2m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2807914fc52e42d883a04064c1cb9cc3 : SUCCESS in 2m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe32c270398442b498a9b57334424293 : SUCCESS in 2m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/767474a80ee84ae0985f1c20799e8bb4 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4ccc3db93c42da888496f476a82ec8 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb372584f6f0478c810f4a4d89235c5f : SUCCESS in 3m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b4a7b8863aec44c19c4ebbdb2789d826 : SUCCESS in 2m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6368515a7a5941ecb375bed146b6f29c : SUCCESS in 1h 01m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fe884ca2f02b4139b0c15bd13e806f04 : SUCCESS in 1h 17m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6b41e1836e1e4cf5abc191ec5ff37ae5 : SUCCESS in 1h 17m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bfbf4c3a0be342b8b4536b679574c3ba : FAILURE in 5m 07s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/aa0bb10387cb4925a2e20ed7befdaa3f : FAILURE in 11m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7da7aae87a704967a11bd2fbcdd707f7 : SUCCESS in 1h 46m 00s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/0b259c46c8f54ccc8a5f242f26519114 : FAILURE in 1h 22m 07s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2d0c1f15fcfb4ef5a1d5c0b579ad685f : FAILURE in 9m 04s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0c56751903bf4606b02c98f4551d92ae : SUCCESS in 1h 06m 58s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4c47bca59fa04c7b8d10910ba7158eaf : SUCCESS in 1h 34m 49s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/bd8220128b7e405b977784582542639e : SUCCESS in 2h 12m 47s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0b126b2f82fe40b18592a6104c54cd78 : SUCCESS in 19m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d7494ddcc0cf4270894f80d10e03338f : SUCCESS in 55m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6688928adcff4913b5c1b8a10c98c076 : SUCCESS in 27m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/27610af0e26f4b98b0c24edca81ce453 : SUCCESS in 1h 08m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/553b611144314d1f94ec6e43e86bba84 : SUCCESS in 1h 41m 07s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/91b09cd53ca349b6a8d654800af188f4 : SUCCESS in 1h 58m 02s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5279de7505eb4036ade1c63c1467b481 : FAILURE in 2h 23m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/568c0a9855354a1ab488fe3f0378e64b : FAILURE in 1h 06m 20s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/345f854f3e02416f982e34393d98bdb8 : SUCCESS in 16m 24s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5a02e37dccec4a33aca751b346f49c5e : FAILURE in 9m 28s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ec2aa9bb1c6f4174a9af16bdcc52feeb : SUCCESS in 1h 15m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/73f20ceb52b641c6993cd7196f9a3d1a : SUCCESS in 1h 40m 37s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8642bcae6d714f22b9bf42aaaae54d13 : SUCCESS in 43m 37s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/edd4b2cef14840a9af94caaa77da65ca : FAILURE in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5467415a9ef30470bcb0ecac138f65c1de2a9e4e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-21 10:36:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ac23c5ee133f99f84ce08597233111aeb28d43c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 11:22:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abd34edf52f34a049c6285c237d10aa4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/460b34468b324496b6dc910502ac1f4a : SUCCESS in 1h 38m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d0da9da605f48c5a65c61433ebdebaa : NODE_FAILURE Node(set) request 343b875105b142c6875bf101977dd94d failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ee5893d79a4400aad657a607c90bb0a : NODE_FAILURE Node(set) request c8bb270f21424962a86ec7dbc4ff9e93 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43ef561b88da4efe8c96052ab82d5332 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0acf129f105f4442b25b69d82ad33473 : NODE_FAILURE Node(set) request 9382e873c9af4b5985fbdf17c5db8ac5 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/20a507daa4d7474fa7403db43e227c16 : NODE_FAILURE Node(set) request 77f1d2a3934a43ab9c0ca1ce9bdbd3a2 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e461731b1fa44b1d85befadef525e517 : SUCCESS in 2h 00m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b75d0868dfbe6ceb1e36b8cb9059b8a16584a8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 13:14:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d86722201234dd1ba049952eee10c89\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ee2f5e5a5d60488e94d593adca9d60a8 : SUCCESS in 55m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a661f055c24e4e24adea23c3b5845857 : FAILURE in 43m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/291617c99d984cd0bbdd21e150a44a3c : SUCCESS in 36m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1c1f7eb5276c42b89052480ca7bbbaf5 : SUCCESS in 28m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4ca3de252e584eababd0c3b4115d2146 : SUCCESS in 1h 00m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7f43456edf0d4c1c99ea5fc90d237782 : SUCCESS in 1h 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/886a6f41402444c697205ff2f2eacae2 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df976c0245954e348ec1ca5c3d0c455f : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fe3d468758b40ff864cab0c6c819185 : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b6de3b8e10d4324b83d32c85d33ba34 : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfee4ac222544bb7aaea6fc4b9b1507f : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b48dfa83874e44849e5f2abefe76178c : SUCCESS in 2m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa178659715c497e81b6b03673598e70 : SUCCESS in 2m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/18406153db5542d1b4b42ee43b907eff : SUCCESS in 58m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/098520fa1a88499891e6f4ad1cc00d74 : SUCCESS in 1h 18m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fc706ce6302c469ea6611e42cf9d38c9 : SUCCESS in 1h 02m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/16084b3359954e5695b3d50e31e45981 : SUCCESS in 2h 09m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/90c4fcc3dbf24dd2b7954c52959df44b : SUCCESS in 1h 38m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b362767f50db418caa41bac7884476f5 : FAILURE in 1h 38m 29s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b9ffd3bb11c34f0f8e8cf2c44b206452 : FAILURE in 51m 41s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e35da2810e5c4642a2a0f797e4a0c6f1 : SUCCESS in 34m 20s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/16d2a9afe72f422eac013059e45b2464 : SUCCESS in 1h 25m 18s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ce683144d263489586e6b93c88e31d26 : SUCCESS in 1h 36m 03s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/35ffeb0ff7d941f0be55e9eeab3477ae : SUCCESS in 2h 05m 24s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3c40bba256754b878856c46d98c84945 : SUCCESS in 21m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d7cb425ab9a4839990d205d9a51f358 : SUCCESS in 52m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/93f54f088ea24732b03a9934ae38ba0b : SUCCESS in 26m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2c667e42ff2444adbbb33467ea6a24e2 : SUCCESS in 1h 06m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a6de841cae0d449295c456aa73f9ae62 : SUCCESS in 1h 35m 10s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/eb1c662ebb9a4ea790169b86a859e84f : SUCCESS in 1h 51m 04s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e1bf38d653fe4e9198c839d59ecbd2f2 : SUCCESS in 2h 37m 23s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/aadb2e91127242d8bf474e77772dee87 : FAILURE in 1h 15m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5e4c83f40de34f0c8c1be1889dd959dd : SUCCESS in 27m 09s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/087e4217389846218f93af16c6266bf2 : SUCCESS in 1h 18m 45s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8203ab4edbcb4e79bf66cc1e79b81342 : SUCCESS in 1h 10m 18s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c31658674b464c39bf39c7022b8f4913 : FAILURE in 1h 38m 37s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/32363ea3e22f4581ada9980322873b99 : SUCCESS in 34m 32s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/884d3ec11b684fd29f5a0f0c69663ee2 : FAILURE in 1h 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1296c0e17ff97e5f05db026419e70c90d5a68b37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 13:15:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"874f0ed85248180a9d35ac4aa86fc5a26f698f5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:16:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e18332ce114ec99449a95982351f50\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/22c4857f542149b1b15973ffddfbcb0c : SUCCESS in 36m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/962ca1a2bf014231be35ec16e4cf1bd9 : SUCCESS in 29m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/980ff81ebfed4e10b75c6334903af220 : SUCCESS in 58m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/524a0be85d1a4f399ca61671fa6fcdc9 : SUCCESS in 1h 27m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f0113bab6a4f7f9fe5e607caf88137 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/987033a92dca481b9ac5d55ca09cab99 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc7cde9904c04f56b04b5d40032fefe3 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec3e2e8928b4883978cf216bbcb1c6c : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88d1c9a09d334506b57a5ececbc32742 : SUCCESS in 3m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ab2a701b3c1944009120d27237309a43 : SUCCESS in 2m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a6802be734e844cbbc7e12eb29bb4fa4 : SUCCESS in 46m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/86c9b31e589b48bea6dac03cdfed0e4b : SUCCESS in 1h 06m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c1f860d3ae64486581575f26c032b86f : SUCCESS in 1h 25m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0e2dedaa3b4b419ebb9783132742446e : SUCCESS in 1h 54m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/35ab4d707c7841a784e5c7e578b56ca0 : SUCCESS in 35m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/800b652cffea48c98d99077af5f17172 : SUCCESS in 13m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cd425e83dbc2409697c210344640591a : SUCCESS in 1h 06m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/aa8be5bae40e49439a0fb0b337d23cc1 : SUCCESS in 1h 36m 34s","accounts_in_message":[],"_revision_number":2},{"id":"1e75855791c643ddfce7b616dd66c8528e535a0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:16:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8ef44d6c99a7b26af837f8a1ebf743022da6601d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:17:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/622425974e534d6cb56c1f0a5a0c0f28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f52f22848a4723a7d472c57c8ca3a9 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cfc2a5d513e4e50bafbaea0f14917d6 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd501c51db412ba238aba5a2195a9827c9348c4b","revisions":{"1d05f216a87b2e35d037389467f6b5cb8008f7fd":{"kind":"REWORK","_number":1,"created":"2025-11-19 07:26:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/967657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/967657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/967657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/967657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/967657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/967657/1"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:45:29.000000000","tz":660},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-19 07:26:39.000000000","tz":0},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one.\n\nCloses-Bug: #2131736\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n(cherry picked from commit 0fc73a644bdb7e4f583ba5ebf7ee3fcd60fb8ffc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d05f216a87b2e35d037389467f6b5cb8008f7fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d05f216a87b2e35d037389467f6b5cb8008f7fd"}]},"branch":"refs/heads/stable/2025.1"},"bd501c51db412ba238aba5a2195a9827c9348c4b":{"kind":"REWORK","_number":2,"created":"2025-11-19 07:27:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/967657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/967657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/967657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/967657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/967657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/967657/2"}}},"commit":{"parents":[{"commit":"c3fa85b2e69e13ce07fb54bfb8752754bcc01121","subject":"[stable-only]: periodics: Move publish jobs to periodic-stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3fa85b2e69e13ce07fb54bfb8752754bcc01121"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:45:29.000000000","tz":660},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-19 07:27:09.000000000","tz":0},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one.\n\nCloses-Bug: #2131736\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n(cherry picked from commit 0fc73a644bdb7e4f583ba5ebf7ee3fcd60fb8ffc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd501c51db412ba238aba5a2195a9827c9348c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd501c51db412ba238aba5a2195a9827c9348c4b"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
