)]}'
{"id":"openstack%2Fkolla~967658","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf","project":"openstack/kolla","branch":"stable/2024.2","topic":"update-debian-package-source-stable/2025.1-stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-19 10:53:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf","subject":"Transition to the new rabbitmq Debian package archives.","status":"MERGED","created":"2025-11-19 07:27:50.000000000","updated":"2025-11-19 10:54:27.000000000","submitted":"2025-11-19 10:53:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967658-update-debian-package-source-stable/2025.1-stable/2024.2","cherry_pick_of_change":967657,"cherry_pick_of_patch_set":2,"meta_rev_id":"e33e2dd99fb85f660848b329c6b0ea6a9f706df5","_number":967658,"virtual_id_number":967658,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-19 10:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-19 09:31:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-11-19 10:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-19 09:31:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 07:27:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"CC"},{"updated":"2025-11-19 09:28:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-19 10:20:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"98d42eeaf38f9ab529632a89caf70064d91b8a93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 07:27:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"f91d86e430d9fddeaa372b650041ce167b8feaa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 09:28:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0ed9a62869e46fa9a7eda9ce627377d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/01c1e410ef784e88902c1cbbe5da5814 : FAILURE in 1h 00m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5b3204f9dfed466ca29eba50a22f602b : FAILURE in 1h 38m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ceb5b2a98dda48d0bdc64bbb9dea0e05 : SUCCESS in 46m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ca0959cf7ab14a379d74b24deaba23b9 : SUCCESS in 16m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/94208d25372e455898fdcfb766c322a6 : SUCCESS in 1h 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec1f6af9594e4799a8d338778e008719 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b01a66d30ac48b4ac330124810668a7 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/232ffff52e7b401ea344fc747f1ced8a : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/846f451f40824d4aadf4058c1beceaa6 : SUCCESS in 1m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d21866c351a249ce9870fa49b35f62a8 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/778ee9a9dd6e45db9ddbbf5c598a4ae1 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cecbc9d9cbc346baa6b557c4a84bb54f : SUCCESS in 3m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/da18d7b8bbd64452934ebaee33660bb7 : SUCCESS in 2m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/420ebfe98949456f879453b734beeda2 : SUCCESS in 39m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/afa5e1ba020b4462ba412844d0aed31e : SUCCESS in 21m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d11fc8c5207f4b9785aa311da707a489 : SUCCESS in 1h 39m 09s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d8789ae05dbc4e8e983ae997880774c2 : SUCCESS in 1h 56m 14s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3cc43203d0cf4a149f3523c85d1f6a07 : SUCCESS in 55m 53s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/6a39a33c4e66404f8d071cc9b7ed1aea : SUCCESS in 1h 11m 55s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9dc5a58226d74f9e9b5f5be2a01fc80f : SUCCESS in 28m 47s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e36e4419d55e443f83b317e2e3fe1cd0 : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/cfa7339c2261409381d12165fad07e1d : SUCCESS in 1h 13m 42s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7fef2d2dda3f49a1a8c126d9dc04bb47 : SUCCESS in 1h 46m 04s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/855ae930738748f3937cda4c03a2500c : SUCCESS in 27m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/904bb2a8c2c242c9bf1f6bf13ad0593b : SUCCESS in 46m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a09d7be5d6e244f3ac15368c85120ed2 : SUCCESS in 33m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9bbd7e7ba85c48f6b7a8c5c46ee5f169 : SUCCESS in 1h 27m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9034522661fa438a9b2d77701cf53671 : SUCCESS in 1h 01m 53s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/58df45b1917445669cdf4e473bfb2448 : SUCCESS in 1h 38m 25s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8bb67f2a270049fba0508e07037ade2e : SUCCESS in 55m 21s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b6252f6d4e0f4c9093bb810ba8c7001e : SUCCESS in 26m 28s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/03bf9a8c462840dfade2a2cb4e0fce82 : SUCCESS in 1h 12m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/66ea40e78636472ea7022bbea6395db2 : SUCCESS in 49m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/682aa6be111a48b4be0ae358f9befa08 : SUCCESS in 51m 21s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e745388e0e114b1baf29bd6ba6948fc6 : SUCCESS in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2dffb1b1189eca5ddbe1ad704000c60c36c6a2e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-19 09:31:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc6b21f14edc6078f22e2ea3c2dc0554def4b787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 09:31:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fa9ec33bca72931dda7e1bb63d91241ac2c49f6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-19 10:20:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dc197b41a06e6a36725bd7faa3a071c771875f0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:47:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a03f99758151401b9f189dc1bfdf73eb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/67d4620457534c80bd17e4d24774e2e4 : SUCCESS in 3h 00m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc964315c3b04a02920657704ecf846e : SUCCESS in 1h 15m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e1426a259a674770bc26c808213a3f7c : SUCCESS in 2h 55m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4dc890fa7fa41648bfc0d1735ab2fee : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e2a706cf1b04108b7f8d1e26fd57fc1 : SUCCESS in 3m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a8335767b4824d3bb0d6d75c41782e20 : SUCCESS in 2h 54m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/68872e8d3d9640b59d86fb720ec90dea : SUCCESS in 2h 55m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f80058aa96eb70ef22c83db1330fbf5b95deafe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:53:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2a05edfbb94d8b9c10dada301a9532\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/73ac5a158f194a2fa46297d4767b3184 : SUCCESS in 37m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4da8887257844a4c979e2f7e8049b8c0 : SUCCESS in 15m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bb6fa10c2b2a418eaec91d3ab57b73d8 : SUCCESS in 1h 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3829688f54a845feaaf7a2df73dbd90c : SUCCESS in 1m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08225c8f8f8745b2a3beed9f7a70adc1 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd022b76ddef47108b808a23b4304359 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ff0ca55bfd42ee92c9bfce07b646ba : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a845559582c543b2bad45a19eee435c9 : SUCCESS in 3m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8b71b830c6584fbd8ff34e1142d8f354 : SUCCESS in 2m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e2afe393c1454656a229f0bfc71aed7e : SUCCESS in 46m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ccccce8ede934415a8b9e74f1cfdc7cc : SUCCESS in 20m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ec8657c37ed54e58a540af024a00221a : SUCCESS in 58m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b36e5828f17241cdbda2182a58d3fa83 : SUCCESS in 32m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cb4b1cdd305e487ab855abe167bc6a0f : SUCCESS in 13m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fb570e004224435da113b0aa48b6d334 : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b8c4ded68a578cfe588cd17cc1d800163e9c6249","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:53:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e33e2dd99fb85f660848b329c6b0ea6a9f706df5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:54:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3466c1f768d44e8b7c8c75e2c82f6d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e765235c94f4c18984b58273dcdab91 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc5e1b17b8174b28af2371bf05ee5e79 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57f438d69362b28d03f190fa6046843c9578c8c1","revisions":{"57f438d69362b28d03f190fa6046843c9578c8c1":{"kind":"REWORK","_number":1,"created":"2025-11-19 07:27:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/967658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/967658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/967658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/967658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/967658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/967658/1"}}},"commit":{"parents":[{"commit":"b1ea365ae57aa85220a40379bdfffe35bc5cde4e","subject":"CI: Drop upgrade jobs - 2024.1 is going unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1ea365ae57aa85220a40379bdfffe35bc5cde4e"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2025-11-18 08:45:29.000000000","tz":660},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-19 07:27:50.000000000","tz":0},"subject":"Transition to the new rabbitmq Debian package archives.","message":"Transition to the new rabbitmq Debian package archives.\n\nRabbitMQ announced some time ago that the package archives were moving:\nhttps://www.rabbitmq.com/blog/2025/07/16/debian-apt-repositories-are-moving\n\nThe old location is no longer working, so this patch transitions to\nthe new one.\n\nCloses-Bug: #2131736\n\nChange-Id: I6ef0ed9cc40be42a0d98c6a4c83c5c5c26d9cdbf\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n(cherry picked from commit 0fc73a644bdb7e4f583ba5ebf7ee3fcd60fb8ffc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f438d69362b28d03f190fa6046843c9578c8c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f438d69362b28d03f190fa6046843c9578c8c1"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
