)]}'
{"id":"openstack%2Fkolla~968523","triplet_id":"openstack%2Fkolla~master~I60de8ef452adad4a9b0d974a2678264b91eb6e3b","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-26 19:55:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I60de8ef452adad4a9b0d974a2678264b91eb6e3b","subject":"rocky: Remove EPEL hack - 10.1 is out","status":"MERGED","created":"2025-11-26 17:27:31.000000000","updated":"2025-11-26 20:53:07.000000000","submitted":"2025-11-26 20:52:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968523","meta_rev_id":"553a81de665ec5d4d9e63bd1bc3904913a93bf0a","_number":968523,"virtual_id_number":968523,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-26 20:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-26 19:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-11-26 19:39:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-11-26 19:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-26 19:55:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-26 19:55:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 19:34:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 19:39:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-26 19:51:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"fdfc2682947fdcdb5d4b21647429e62c5cb923df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-26 17:27:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3db27a2dde8a21e3622ff5c3b29339482ca097dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 19:34:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab26efc3d0b44fcb5d4ec593535b530\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0005c8026b6d4e4d8d1e10991835812e : SUCCESS in 50m 28s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/b20b1365e49c444c8c82097efe21fc16 : SUCCESS in 53m 54s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/dca06669816f4ae1aa5b1cfdce786714 : SUCCESS in 32m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bd1eabdb53d74907ab72fe639ed805cc : SUCCESS in 37m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/43f0d0682acc419584da2724efa12ac1 : SUCCESS in 40m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10a487740ccf46f1b70848920854ef44 : SUCCESS in 53m 08s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/96b112dc56a94f00b2a431f4365ebaae : SUCCESS in 21m 46s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/fdecb18c12a9435cbc067d27eef11275 : SUCCESS in 1h 14m 42s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3efe9009158c4b12974bd590948907d7 : FAILURE in 1h 10m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/155e966ebdbc451c8d1698bc042b3f2b : FAILURE in 57m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/454c02a18863423495cd93f5576b6797 : SUCCESS in 2h 04m 46s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/6b6ae048aa354949b07c10d614cc02f6 : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/48e8c89db4eb44cab93ea6f0aec10860 : SUCCESS in 1h 10m 11s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/c101fa2e551c4dba91afa3e442a914dc : SUCCESS in 21m 57s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/c97b21fabfb84e5983a48e82e094e397 : SUCCESS in 25m 56s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/a39abba8a6014d2296fb5f44d1539e66 : SUCCESS in 1h 02m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/5762171470704e2498d1c602a4f5a12f : SUCCESS in 58m 39s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/4de7a6f324b14fb8ba42cb3032e53ac2 : SUCCESS in 49m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/f4d6609ce61c40cfafa71b5b197b88ad : SUCCESS in 56m 38s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/79414c1dced647d591bcce5ea63fb09b : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/9a2b3a75d4c343d39ac21c4fbd361c9e : SUCCESS in 1h 17m 44s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/15b161b7113345b59bc7635f175ae14d : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0044ec32877e4fe48f923132c266c9e0 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21ef628c8bf5480cabd6ebff1dbe8314 : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faa47f85c82c47849b87ef5fffe66ddb : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d5995683b2c41468adcb6025def7f54 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/605d51afecf9435c914b83c4f3553ebc : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2d7aa6e4c0f17175f6dd0eab9be857c1087ef3cd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 19:39:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fe002e85d264d18e7d861a39aa0244db94847e19","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 19:51:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e9322563ff9f4e4d2ac383fcae98386ee550fba1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-26 19:55:30.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6998331046004aa84205ef128b28a53c9837d08c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 19:55:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"710a367d2d6ac59d844c39843f74cd4d830571de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:24:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/855370a0e9f74fb1befb463b09b485c4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f227321e2f3d465f9c607304e3807897 : SUCCESS in 2h 45m 36s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/6d59030927104ba39187997514e41519 : SUCCESS in 49m 25s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/fd5e1528727245fe819f24c7936285cf : SUCCESS in 2h 36m 03s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b1b5815e2fc64a76928f288c1099d304 : SUCCESS in 1h 05m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d320b15040ba409f847af4eb0041cca1 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1473e94bb9784b0e91198e37254b4083 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/013f7a4d22fe404f8afb99f18a5731fb : SUCCESS in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5e1ab1fc81645853dfec89165d9299310e6886d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:52:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe511b02c7de4e2fb6ef6da8546dc0b5\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c80b4163a7984ea0b68bcb2f211527ab : SUCCESS in 51m 43s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/e5d6c2fec4a846c99f104ad62f11a1cf : SUCCESS in 36m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/75f2e5e851a945dbbb359acbb9ea01be : SUCCESS in 47m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/78558c15dbf242019e95d701bdbb975f : SUCCESS in 34m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20da1379c75b46e2a1135d4053454361 : SUCCESS in 51m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3925a030dc7049939cd8438a80801f6c : SUCCESS in 31m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/65cf376f9a4e43da993eff36a1585cd3 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f62d7dc196a74bf3823d3755a9cad85b : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3544bda26314e3791a93a16f6042095 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f9ecf819c7c4b33ae257c8a878a80e6 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ae3a1c66f847938e9269689417c963 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1a1a96f5091a63b3a38f8442e21feaa645678067","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:52:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"553a81de665ec5d4d9e63bd1bc3904913a93bf0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:53:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f18cf4b39c843899a9bbbd42cf7902f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63b5c4dc74874f9aa01385da9274ac8e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583","revisions":{"cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583":{"kind":"REWORK","_number":1,"created":"2025-11-26 17:27:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/968523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/968523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/968523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/968523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/968523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/968523/1"}}},"commit":{"parents":[{"commit":"0dedcb82e46b2ed05d8b53f7ddd8fb0ba19049f9","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0dedcb82e46b2ed05d8b53f7ddd8fb0ba19049f9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-26 17:27:21.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-26 17:27:28.000000000","tz":60},"subject":"rocky: Remove EPEL hack - 10.1 is out","message":"rocky: Remove EPEL hack - 10.1 is out\n\nChange-Id: I60de8ef452adad4a9b0d974a2678264b91eb6e3b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
