)]}'
{"id":"openstack%2Fkolla~969024","triplet_id":"openstack%2Fkolla~master~I9a82798f37d5cf5e172b11930de5d87d3c0d29df","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I9a82798f37d5cf5e172b11930de5d87d3c0d29df","subject":"designate: Run rndc-confgen on EL","status":"ABANDONED","created":"2025-12-01 09:43:51.000000000","updated":"2025-12-01 13:44:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84990d254c3a83e033b2b34d4aeb25df035b7cc7","_number":969024,"virtual_id_number":969024,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 10:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46677ea8373284bb882a62b999fc46c41f4f8573","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-01 09:43:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01e5ee415f2122f0e8dc11e65f13662ecccccbb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 10:45:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d550a2974f33454ab340c7c50194210b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f678b37121554b33bb550f41bc23e9ed : SUCCESS in 48m 25s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/6252ee5289c44cf0935d542a53082788 : SUCCESS in 46m 29s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/163fa663d5754949ad54fb9b5e6c8e07 : SUCCESS in 26m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7a53ac812596400ab2fd696a4f4385c5 : SUCCESS in 40m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/414f4f8cb80d4993a3c6b7cf94af972e : SUCCESS in 49m 38s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bc300ebbadc4b6681a772accfc42608 : SUCCESS in 37m 33s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c70aae21cdc4e68a1441cfd6c49db7a : SUCCESS in 31m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/67fe9a931399434380d254152b546463 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf2585c2f154faa9d1da8915e43fe2f : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0c65d79299e43d3a9e7b1176f54401d : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70ee884674ad4ac5aa40653be6eb9b2d : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cb8d54a59f544c9bb6c1d96053a6010 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e36ebec48c4354b71293a03af11493 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"40819977d6f1e83dcf98479458f88cf4589f061d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 12:37:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b0f6ccf0ace41379a1f9fa5e691eea5\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/918b07e5dc034b86955014493998d76d : SUCCESS in 49m 29s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/54d68c4629bf43c6afc03417a9861592 : SUCCESS in 2h 26m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/421c171aef4b4642953d52de0cf15071 : SUCCESS in 2h 51m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1757a2c193134d8683098c6bbaec26d5 : SUCCESS in 2h 47m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d88c0b877fc14a5d83e84b676e25f84f : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f37285a05ee44ab990d889d3d878d28 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d249ec2238964a7abe37940b9685b0c5 : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27e74bed05835c3342a3c0b665507b583ae7eb35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-01 13:04:18.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f747a38bc768de39f8b30b589e51e22c3d33a7c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-01 13:16:28.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"84990d254c3a83e033b2b34d4aeb25df035b7cc7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-01 13:44:44.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"491a798b839b5621f23c1035b08daecf9f08a498","revisions":{"e55e9946838da173c88b9b16b5f2f4554ac57a81":{"kind":"REWORK","_number":1,"created":"2025-12-01 09:43:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/24/969024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/969024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/969024/1"}}},"commit":{"parents":[{"commit":"cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583","subject":"rocky: Remove EPEL hack - 10.1 is out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-01 09:43:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-01 09:43:45.000000000","tz":60},"subject":"designate: Run rndc-confgen on EL","message":"designate: Run rndc-confgen on EL\n\nChange-Id: I9a82798f37d5cf5e172b11930de5d87d3c0d29df\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e55e9946838da173c88b9b16b5f2f4554ac57a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e55e9946838da173c88b9b16b5f2f4554ac57a81"}]},"branch":"refs/heads/master"},"6f196187a2ea1a64e32a0c472468a0e3afff8119":{"kind":"REWORK","_number":2,"created":"2025-12-01 13:04:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/24/969024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/969024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/969024/2"}}},"commit":{"parents":[{"commit":"cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583","subject":"rocky: Remove EPEL hack - 10.1 is out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-01 09:43:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-01 13:04:18.000000000","tz":0},"subject":"designate: Run rndc-confgen on EL","message":"designate: Run rndc-confgen on EL\n\nChange-Id: I9a82798f37d5cf5e172b11930de5d87d3c0d29df\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f196187a2ea1a64e32a0c472468a0e3afff8119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f196187a2ea1a64e32a0c472468a0e3afff8119"}]},"branch":"refs/heads/master"},"491a798b839b5621f23c1035b08daecf9f08a498":{"kind":"REWORK","_number":3,"created":"2025-12-01 13:16:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/24/969024/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/969024/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/969024/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/969024/3"}}},"commit":{"parents":[{"commit":"cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583","subject":"rocky: Remove EPEL hack - 10.1 is out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-01 09:43:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-01 13:16:28.000000000","tz":0},"subject":"designate: Run rndc-confgen on EL","message":"designate: Run rndc-confgen on EL\n\nChange-Id: I9a82798f37d5cf5e172b11930de5d87d3c0d29df\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/491a798b839b5621f23c1035b08daecf9f08a498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/491a798b839b5621f23c1035b08daecf9f08a498"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
