)]}'
{"id":"openstack%2Fkolla~969090","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I02f3adf9825636e2b81c50543d6f95ea986019b6","project":"openstack/kolla","branch":"stable/2025.1","topic":"cherrypick-add_valkey-2kollyzuqut","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-12-11 14:46:20.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-11 12:31:32.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I02f3adf9825636e2b81c50543d6f95ea986019b6","subject":"Add Valkey","status":"MERGED","created":"2025-12-01 14:33:52.000000000","updated":"2025-12-11 14:47:16.000000000","submitted":"2025-12-11 14:46:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969090-cherrypick-add_valkey-2kollyzuqut","cherry_pick_of_change":932174,"cherry_pick_of_patch_set":17,"meta_rev_id":"e80dfbf1187d85cacc531350fbec703248035aca","_number":969090,"virtual_id_number":969090,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-11 14:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-12-11 12:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-12-11 12:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-11 12:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 14:33:52.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-12-01 16:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-11 09:35:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-12-11 12:30:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-12-11 12:31:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b41deb17a0a3051df701d97e8b323517a135601e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-12-01 14:33:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f8d5e94ce34f2a3e6aeca0371f6f805a645bbc59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:50:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb268180d3bb4a1a9e281bf7a7b73c67\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/923e9d55ec884d749b3996a83c96b231 : SUCCESS in 57m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/541137d49e414eda9b32af3c6f724e74 : FAILURE in 34m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1fb2d475f5f041e7a7c3d2f8f1f91989 : SUCCESS in 54m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b5da14f221f344f6b0e3aa1ddd475dbf : SUCCESS in 31m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/29d4e15158134422a243a6f0c9614c9e : SUCCESS in 59m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/943ad30d8a2e4e5ea6d58a088b7d81f9 : SUCCESS in 1h 30m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fee0efd2e5bc43eba0626fa84095d9f4 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/742bfab472f048d6979f327cfb7a017b : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a129c00aa1d42bd829ba1463ae93300 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c400ccc55fd4095b219c120f4c32521 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bccd54464fa4f0c889a399444acd87d : SUCCESS in 8m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/01ab24bfda0a4f53a5974fa891984bea : SUCCESS in 2m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/190fe1e37d844b1bb65b0a8b202df4b0 : FAILURE in 55m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d584864c54f44f87b3fc0656387ca3e3 : FAILURE in 1h 37m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a31680aa07bb438ab1ef71f4d9f0af55 : SUCCESS in 1h 19m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c6ff7e8fe8ca4166b85b07cf7cb5e40c : SUCCESS in 2h 06m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/87fb2e7b57c54ba2b1eec55ca61515f2 : SUCCESS in 45m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e4f95c15a4c940b7a99778f65fe1a1b4 : SUCCESS in 28m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f678ecc0691045d4b327f0d64e54b8b5 : SUCCESS in 1h 05m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dd6a00b6d6d248e7902f8030fac715b3 : SUCCESS in 1h 37m 45s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/6a8713d982884e9ca88b392ad19fa560 : FAILURE in 1h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4af8b6fc9d6e993cd4f0fe4aa74f3b38c4df9e27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 17:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07407ed1115d4c86b271750f888b4e8a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/afc2c9dc9d7e4f32aacd5796889f2ed6 : SUCCESS in 1h 19m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d1ffba3b5cdc4bab913f9f241f72bc04 : FAILURE in 1h 26m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9cd89346d87f45e2a7bb03b150640348 : SUCCESS in 1h 08m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14eac95744f64da5a349c82309371203 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6d214cc30ef4249a621ccb4c858d874 : SUCCESS in 5m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ddd77aa8e3e64401bdce0dd0e40e4970 : FAILURE in 54m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f6b75baa8f4c4f7f9c2f4944bdb4cfb3 : SUCCESS in 2h 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"390373ba52e5b0e306446a6b6d336227dc4c503a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-11 09:46:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_35345\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}],"_revision_number":2},{"id":"33316fa13b5ab57db7dcf47f63952c61324d58fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-11 09:52:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dc5523a19a160bfdaa5c73198a9af2b2ec7c70e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 10:10:09.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f637942f0674d98b58205cb627d1c91\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c03904f60ca34c469417ae401d975433 : FAILURE in 5m 42s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5504cad416144ff9a7b687d46b15ecb5 : FAILURE in 2m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0d1da58f3cc34352ae20dbac86abbf96 : FAILURE in 8m 38s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/088fb0e241a24c588487171691f8806d : FAILURE in 4m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b4673118a6ea4900b8accf26d6bb74fe : FAILURE in 4m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/271b7fd887694a08b1699f8c01f59105 : FAILURE in 3m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/92c01bf52cdb48aa9fedb6f3b34492c6 : FAILURE in 5m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fa8df9ffcee4420283fcdd46c7639d5c : FAILURE in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017445d5b27a4d798287734ea300ce26 : FAILURE in 2m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9624774f6294cc9b1094d4180e1db84 : FAILURE in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34342c467f994c69a87e7525ba2b9972 : FAILURE in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d734747d831c4d4787f91cb11b7989bb : FAILURE in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e71548428014858b8d38924b1a8ad3e : SUCCESS in 5m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/34937992bab14b7eb588d9754a948d48 : SUCCESS in 2m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6327195042d84a10b6ce8cbe5ae222e1 : FAILURE in 4m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8b0b90d2eab7498eb19391cffac35676 : FAILURE in 5m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4c95cd143a714536a3ef654a1a85bff1 : FAILURE in 6m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7bfb554a1b4945539e9acc6fa2a16778 : FAILURE in 9m 36s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/09d3e8cb2d2347a89d852b866bd67b02 : FAILURE in 4m 15s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/aea475daf3e84ba192d25c8db72fa051 : FAILURE in 3m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/904a51b934f645ffb78d5e8fcab4a732 : FAILURE in 1m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fab76390c2b849e39e39f87f1bdfb2b0 : FAILURE in 3m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/57c1b9ae411746b89b4bb275df459f72 : FAILURE in 5m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/586281baf4624b5aa61748fb90a4677c : FAILURE in 5m 57s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/c744e70dbdd6409ab0e2c14e695e2a95 : FAILURE in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"74cfea86f52133c5c34c664208f2390f50ac9c44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-11 10:11:19.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"37cd53af2fc529abe8957a5e5b3b3edc52f32c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:11:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c02cb785b9b46c0876f09657579ece6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5fed3f9869fa457ca94a4ac87acfa2c9 : SUCCESS in 53m 15s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/d4f97d3c37854e58a4534ee631b95380 : SUCCESS in 53m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/558c66f5cb6a4ed0858a694ddabcb80d : SUCCESS in 1h 32m 25s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/02cc8035a7ad4ddcbc2002c4007edf21 : SUCCESS in 1h 08m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6f807c78080d444f95a14b7faf71629b : SUCCESS in 46m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1d0e5384ba6d4a408535345fc43fab52 : SUCCESS in 30m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/eca9d920d4d1407b9b1426e60663396d : SUCCESS in 1h 12m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ea127d45eb444bf5952dd5c5ab0ceef7 : SUCCESS in 1h 36m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b11c4f3dae9749e58b489fe0969957db : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09618cad315846da8e7de0697cdc5f1d : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0d92758912d4a65b805b2e51ed19b93 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf4caa4b38c344c3933faebe2a10065c : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b3b609c4534b3397503633292c024e : SUCCESS in 3m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0431e4c2278b4352b0011711aa4ba032 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3f7912757cec4e99a685e3e535666127 : SUCCESS in 46m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ac47f2b9fa064065bdc01f32e14813ec : SUCCESS in 56m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/96215b11ac864f6f8387ca5328407f32 : SUCCESS in 1h 28m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aff67d2f090144a8a13d85ad6a45b920 : SUCCESS in 1h 55m 49s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/ba5afb8ac96f4e46a96738746b0fcfcc : SUCCESS in 55m 19s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/8017ef669bd64b4582a48f2f9ef275bd : SUCCESS in 43m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/02d60c42a8174b76b35ba3f7847e3dc3 : SUCCESS in 38m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1883b73b67634a98837a328ca6d85012 : SUCCESS in 32m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bf6df580190d4cdf89165cb939cabf92 : SUCCESS in 1h 11m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3df94bc48a4b456182ad472dde28312a : SUCCESS in 1h 56m 32s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/cc010f2b9e224d39a8e92651729c8e1d : FAILURE in 1h 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2de9c4f44fd8c8ed08300a300474d48ca8708b11","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-11 12:30:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a43a77fe79714caecff97c6c1ef5b78708f34124","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-11 12:31:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"44010a982e4c61765dbd5da8046906ddd7630977","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:31:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"815ab73a58d64ada69e6f16c7e1cf62354896484","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:23:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fbaf363d27b48249eba5d21b8325f36\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/66a5c7984570496ea9ca9facc883fe56 : SUCCESS in 1h 05m 18s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/1d4583e1889b45328b57c0fa21631898 : SUCCESS in 2h 00m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f582fea1ea31460eb9f3781fd9b4c0a1 : FAILURE in 50m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/364f6ec7afd641318e23c23756efaf42 : SUCCESS in 2h 12m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c9e1d69d2af845f38bc1d98473078bf9 : SUCCESS in 1h 47m 46s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/69fca4f44f1c4a0a85af387eedbabd7a : SUCCESS in 1h 06m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dccc35af57f14d1abbee17b7c6a3b21d : SUCCESS in 1h 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5eab85fd10eb791fdb6b1367c1bd098965b7d98e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 14:46:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a93c6529770434fa323227e5e8fca3c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/111c93614cdd4d6f8789a67fe17fcc42 : SUCCESS in 34m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e562d3d8cf874cafb02646db781ff453 : SUCCESS in 29m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6bfd97f932164681bec7cfe4c1885e30 : SUCCESS in 1h 00m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b2f3dd4efca54e89939be62608892428 : SUCCESS in 1h 35m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5cf095581e4052a8639ee9f5c7bce8 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e386d79491b4cceb7a3d8fcaa7b2481 : SUCCESS in 6m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c27e417703994774bb28e6fc34f13450 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7709c84484b84b5ea2f6e5d996836eb9 : SUCCESS in 4m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8b2e31e8c1e24622947052b44bfed35a : SUCCESS in 2m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5b75b27c521f45c3b759ee7313047e74 : SUCCESS in 42m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb05a5388c6e46f99740c01cc1a076b2 : SUCCESS in 57m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/99696cda0a9c44dc8577a64ede112472 : SUCCESS in 1h 30m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dbf907a0601c446dbf83b004937616da : SUCCESS in 2h 09m 28s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/c9ca070ff9aa4e849e5ab7ffc8cd482e : SUCCESS in 47m 11s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/31f80def01b043648f34a95a56f0cc00 : SUCCESS in 53m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8c6e65d429624d8a9bb840b5c7f35225 : SUCCESS in 33m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8db9d4ed06a84a69a2dde695b1deabc5 : SUCCESS in 13m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae0bb06487184678b2be8d5e12d57362 : SUCCESS in 1h 10m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/94c5c068a90a4d9eadc0560ce5cca1c6 : SUCCESS in 1h 43m 36s","accounts_in_message":[],"_revision_number":4},{"id":"dfdbea238bc63d4314292268049b2eb3ca505938","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 14:46:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e80dfbf1187d85cacc531350fbec703248035aca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 14:47:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf56d2b0802147cdab32605b4aecac5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94cde5f6d4eb4825b48254d67390f996 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f3a5124f442397d06da73f820c45f4245cf1b8bf","revisions":{"73615eef42f737024711eb3d020ce15923360df3":{"kind":"REWORK","_number":1,"created":"2025-12-01 14:33:52.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/90/969090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/969090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/969090/1"}}},"commit":{"parents":[{"commit":"bd501c51db412ba238aba5a2195a9827c9348c4b","subject":"Transition to the new rabbitmq Debian package archives.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd501c51db412ba238aba5a2195a9827c9348c4b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-12-01 14:33:52.000000000","tz":0},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73615eef42f737024711eb3d020ce15923360df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73615eef42f737024711eb3d020ce15923360df3"}]},"branch":"refs/heads/stable/2025.1"},"b3c9c5900a3245dd8496468e86204224b2ea174c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-11 09:46:06.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/969090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/969090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/969090/2"}}},"commit":{"parents":[{"commit":"584d57e7c378eb377e823b7cdf2c886ccf116312","subject":"Add support for Rocky Linux 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/584d57e7c378eb377e823b7cdf2c886ccf116312"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-12-11 09:46:06.000000000","tz":0},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3c9c5900a3245dd8496468e86204224b2ea174c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3c9c5900a3245dd8496468e86204224b2ea174c"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"6968acbee0a6bbf4b432838e63993cbdfda17f48":{"kind":"REWORK","_number":3,"created":"2025-12-11 09:52:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/969090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/969090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/969090/3"}}},"commit":{"parents":[{"commit":"584d57e7c378eb377e823b7cdf2c886ccf116312","subject":"Add support for Rocky Linux 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/584d57e7c378eb377e823b7cdf2c886ccf116312"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-11 09:52:59.000000000","tz":0},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6968acbee0a6bbf4b432838e63993cbdfda17f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6968acbee0a6bbf4b432838e63993cbdfda17f48"}]},"branch":"refs/heads/stable/2025.1"},"f3a5124f442397d06da73f820c45f4245cf1b8bf":{"kind":"REWORK","_number":4,"created":"2025-12-11 10:11:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/969090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/969090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/969090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/969090/4"}}},"commit":{"parents":[{"commit":"584d57e7c378eb377e823b7cdf2c886ccf116312","subject":"Add support for Rocky Linux 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/584d57e7c378eb377e823b7cdf2c886ccf116312"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-11 12:13:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-11 10:11:18.000000000","tz":0},"subject":"Add Valkey","message":"Add Valkey\n\nChange-Id: I02f3adf9825636e2b81c50543d6f95ea986019b6\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3a5124f442397d06da73f820c45f4245cf1b8bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3a5124f442397d06da73f820c45f4245cf1b8bf"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
