)]}'
{"id":"openstack%2Fkolla~969236","triplet_id":"openstack%2Fkolla~master~I395fcae7fd39ea7488191690f994ad98a8184f65","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-02 15:14:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I395fcae7fd39ea7488191690f994ad98a8184f65","subject":"manila: Stop enabling kolla_el10 COPR repo","status":"MERGED","created":"2025-12-02 12:21:55.000000000","updated":"2025-12-02 15:53:32.000000000","submitted":"2025-12-02 15:14:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969236","meta_rev_id":"9bbaffa4a15eff6f32c452b706431ac9a8b8365f","_number":969236,"virtual_id_number":969236,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-02 15:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-12-02 14:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-12-02 14:47:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-12-02 14:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-02 14:09:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-02 14:09:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-02 14:05:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-02 14:09:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-12-02 14:47:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"2ff220d707e3684ec5d0db71a98846d7d5cdb06b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 12:21:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"561a2b7195184c44f8da39872d2eecdd6254d60d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 12:43:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"025d029938fefd8f11d0f71ef0c4ccddc3e279c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 14:05:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b44117ee10c4af4a3cc7e23e7ccb822\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f2c9b80d4b594b5098c15fece1d057f2 : SUCCESS in 45m 17s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/377bb7c06a674aa99cbcd832c514ab05 : SUCCESS in 44m 39s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/59aeaf15d97e4bdf8b7d487bb89d45cf : SUCCESS in 22m 54s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5304f2ecd9834842b43f09f3aeb07379 : SUCCESS in 36m 02s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a1de077a0e2844629458ae25b975df5c : SUCCESS in 1h 05m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f040eda2800243929e050f5d393ec024 : SUCCESS in 56m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ae9edeeedd7f413cafcfc34089c003c5 : SUCCESS in 28m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b85337cbdc40433dbd74a520fb195f15 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2032d992b5304707b81b4c0f0384e0bc : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/888b1072aeb9467097008652e09cf878 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38767f2c9ea2486987495072b3a63c2d : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/276e0780c31748039668052dce5aa8b2 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2473de91e24b1bb84561c8550de0ae : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":2},{"id":"aa1463938f83591128dc27a4a959f4c123d2166c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-02 14:09:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1d1d8cf613ec104d1961e7f8efd10e75e883cf0a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 14:09:23.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"135b6fee13cf2dc9d36307df0416962dfcfbdadc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 14:09:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef8608b8ff92fe28453ae39f8f7f24d9916ca48b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-02 14:47:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"87f76792e56da5acd3a02c55688fcd2a1ad6a06d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 15:14:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da00eeadeff84a0ca4d71b950e2faa7f\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c173f78ead0b4a48933c49569b6464cd : SUCCESS in 49m 41s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/2a63d41d82ab43079fa07d1a3c9c471d : SUCCESS in 25m 57s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a9a470ab3c244635a982c88bac3769bc : SUCCESS in 47m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/659705aa3995463b94f63c5aa2cfe2fe : SUCCESS in 1h 04m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4ae3663d2a448aaae13b624729b7376 : SUCCESS in 52m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3a1720973ce4b6d9e8ed0b76dae41c0 : SUCCESS in 25m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d26f9ba0bba94d67a530fec97f9b26ec : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00fa957b68f94e70895cd987e99d0326 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/966328906c614aa396ce73667cc61759 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6072648f81054c82b1816385f612623f : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fbc9f826835443385f249a40543f8d0 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":2},{"id":"84eac196088ada47b2b0f813b92333d767745ea7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 15:14:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9003e04a8f86cbf6a4f1641a91033cb1d0a1e67d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 15:15:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b715b9d8ee64818b8597d8fcb9629f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd663b8b04964377865115c02c86bdae : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2},{"id":"9bbaffa4a15eff6f32c452b706431ac9a8b8365f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 15:53:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cf09aacf024c36acc88adfe69864d1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/49de622707724a0dab1ae5a992cfbef3 : SUCCESS in 56m 38s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/ef66aa501f96471b9e5275ddd027d81e : TIMED_OUT in 3h 08m 06s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/9890e95eda3645909c4420d166e1c8c5 : TIMED_OUT in 3h 06m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b44d0712baae4a209dd6c47cf56a9a48 : SUCCESS in 1h 23m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a80411edc5ae4d4f961bd97a0a102266 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e0a75c86f4d4858a2c94b707142e771 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a21dfd2bb01c42c9b36890dcfea0146f : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14a7644a1e458dcfe8e928d9ea032b0465da08f5","revisions":{"ee20739a505cf4f0e93043aa2688cb1c0e3de21a":{"kind":"REWORK","_number":1,"created":"2025-12-02 12:21:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/969236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/969236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/969236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/969236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/969236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/969236/1"}}},"commit":{"parents":[{"commit":"cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583","subject":"rocky: Remove EPEL hack - 10.1 is out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:21:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:21:53.000000000","tz":60},"subject":"manila: Stop sourcing glusterfs from Kolla COPR","message":"manila: Stop sourcing glusterfs from Kolla COPR\n\nRH Gluster product is EOL and there is no reliable source\nfor glusterfs packages for EL10.\n\nLet\u0027s stop installing it.\n\nChange-Id: I395fcae7fd39ea7488191690f994ad98a8184f65\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee20739a505cf4f0e93043aa2688cb1c0e3de21a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee20739a505cf4f0e93043aa2688cb1c0e3de21a"}]},"branch":"refs/heads/master"},"14a7644a1e458dcfe8e928d9ea032b0465da08f5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-02 12:43:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/969236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/969236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/969236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/969236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/969236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/969236/2"}}},"commit":{"parents":[{"commit":"cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583","subject":"rocky: Remove EPEL hack - 10.1 is out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbbc1b0bca94e0f768be8433ba4bd38cb6f0c583"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:21:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:43:01.000000000","tz":0},"subject":"manila: Stop enabling kolla_el10 COPR repo","message":"manila: Stop enabling kolla_el10 COPR repo\n\nThis repo enablement is here by mistake, the glusterfs-fuse\npackage installation has been removed in the original\nCentOS Stream 10 patch.\n\nChange-Id: I395fcae7fd39ea7488191690f994ad98a8184f65\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14a7644a1e458dcfe8e928d9ea032b0465da08f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14a7644a1e458dcfe8e928d9ea032b0465da08f5"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
