)]}'
{"id":"openstack%2Fkolla~969259","triplet_id":"openstack%2Fkolla~stable%2F2025.2~I395fcae7fd39ea7488191690f994ad98a8184f65","project":"openstack/kolla","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-02 15:53:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I395fcae7fd39ea7488191690f994ad98a8184f65","subject":"manila: Stop enabling kolla_el10 COPR repo","status":"MERGED","created":"2025-12-02 15:15:58.000000000","updated":"2025-12-02 18:26:51.000000000","submitted":"2025-12-02 17:12:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969259","cherry_pick_of_change":969236,"cherry_pick_of_patch_set":2,"meta_rev_id":"2b15f137177acbd6b628787a22ab302cf890c6a1","_number":969259,"virtual_id_number":969259,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-02 17:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-12-02 15:53:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-12-02 15:38:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-02 15:53:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-02 15:38:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-02 16:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94a23a0a23cd42000378dcac91c55894f98ee2c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 15:15:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5c6a94896bfa3701810e52c907a3753f33f2105c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-02 15:38:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc8929ac127b0cb8bc1cd5e66587c06ffbb87956","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 15:53:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56653be215f571b2991d4cdab70e45606c1522bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 16:14:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0817d68b7549309614d0dfd6c43f97\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e7235487f2e94d8da12e7e038be4294d : SUCCESS in 45m 21s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/536459789643405197996c3aee2ba57d : SUCCESS in 53m 29s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c7906c26ab2c48dea18d61ccbcce6ff8 : SUCCESS in 30m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/697b26ce33074b09a161eb586f6905b6 : SUCCESS in 51m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9275b551dad144b798c8d0ab2f760a13 : SUCCESS in 34m 19s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58c52819a7314b398d143ebc086e7faa : SUCCESS in 51m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cc758f47523d4419b4f358e0c0c064f8 : SUCCESS in 29m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d2abb6e611f14455a0d36253f37a5fd7 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c51ce4797c64cdfa52e80c10ed8e3cf : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/635f2f6f875c4d96b3dff9b1e7738945 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/721c98b7fbc44669b46c488ff2079469 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4301c70e013f428581fc2cf4eebef736 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c665e2e668634dc4a1cc105843798197 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d019dd307cc34721a3aa5a045e77644b : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"97f1cc49ad57e2cbf83588d7a4ee4cd61386546f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 16:14:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4f8dd9782c1aa65078b4cdb338b359fb02466da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 17:12:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d14942fa56d8414da3c754f19a94692f\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/d410a028cbcf4d618eda7e7c42ec0e23 : SUCCESS in 52m 48s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/781ab81bacd645e097b0f00be8a04b1e : SUCCESS in 15m 40s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d72401bda46348ab8b3db84456d93e92 : SUCCESS in 37m 30s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ff32baeaf148444d922ab4e4b2d8e75b : SUCCESS in 47m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0def2189bc8e4b759abd91e21126f725 : SUCCESS in 52m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6c7fc0df0edc4e82aefc01b5e369b7c8 : SUCCESS in 21m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c04ff9dc82d84ba69eef4aedb3ef9fe1 : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f650266ca22d4293af48dd8cb4e5c14e : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fde258dd317b4ff5998ce404ec3d30b6 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65153dba01e14c22b5cac6ca068ee215 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb360ba571584ed09974fa172c298def : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ee444c6835f7a29b8fa174538dd5d04d0fe6e89e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 17:12:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f6d79c3bdb79451b326af6ab3aacec93e00db11a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 17:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414e52bfc0ed43309b9d6dd2221bd56e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898b455c66c34a8ba8ed613a68659ebe : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"2b15f137177acbd6b628787a22ab302cf890c6a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 18:26:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fccce07c34147c68c9f26666f02a6a0\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5482a12280874474967cb003953a2134 : SUCCESS in 56m 45s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/f13156b3718b4ff289f73c1d9f12a462 : SUCCESS in 2h 29m 01s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/81a7c5d68d70451c9207460792e6918c : SUCCESS in 1h 23m 53s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/877f92d63ee7486ea75a7148fe0a7259 : SUCCESS in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50a872f54425dbd7dc896c19e18f7e8e99e4e331","revisions":{"50a872f54425dbd7dc896c19e18f7e8e99e4e331":{"kind":"REWORK","_number":1,"created":"2025-12-02 15:15:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/969259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/969259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/969259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/969259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/969259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/969259/1"}}},"commit":{"parents":[{"commit":"a567182aa74d4be51315bf86dbc266ab9416cbf9","subject":"rocky: Remove EPEL hack - 10.1 is out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a567182aa74d4be51315bf86dbc266ab9416cbf9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 12:21:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-02 15:15:58.000000000","tz":0},"subject":"manila: Stop enabling kolla_el10 COPR repo","message":"manila: Stop enabling kolla_el10 COPR repo\n\nThis repo enablement is here by mistake, the glusterfs-fuse\npackage installation has been removed in the original\nCentOS Stream 10 patch.\n\nChange-Id: I395fcae7fd39ea7488191690f994ad98a8184f65\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 14a7644a1e458dcfe8e928d9ea032b0465da08f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50a872f54425dbd7dc896c19e18f7e8e99e4e331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50a872f54425dbd7dc896c19e18f7e8e99e4e331"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
