)]}'
{"id":"openstack%2Fkolla~969362","triplet_id":"openstack%2Fkolla~master~Ic40fecb87dd3eaf67eb1d3088978d462ff5b4320","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-05 07:39:22.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-12-03 13:58:35.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Ic40fecb87dd3eaf67eb1d3088978d462ff5b4320","subject":"redis: Drop redis and redis-sentinel images","status":"MERGED","created":"2025-12-03 09:32:01.000000000","updated":"2025-12-05 08:44:24.000000000","submitted":"2025-12-05 08:43:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969362","meta_rev_id":"1803a93921d881448a84f5b39f707cda2ea23ccd","_number":969362,"virtual_id_number":969362,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 08:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-12-05 07:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-12-03 13:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-05 07:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-05 07:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-12-05 07:39:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 10:01:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-03 10:31:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-05 07:37:58.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"02c80a6fa16274c945780fb13957ee475b22c4ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 09:32:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9da904382ab17535048a17da71e4e43fd2cd86c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-03 10:01:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a223f420122f512e8a07fbf814edae025d86cca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 10:31:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70b92569c90740ec97be0aee726ca6e8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7fd129f397dd456c9363bff67c801856 : SUCCESS in 40m 52s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/2065a1b455e347b6ba0ae0ea05b8c4cc : SUCCESS in 47m 56s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/0e7579fb38a74a95aaa7992c50c3632c : SUCCESS in 17m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/312c758b0ffe46f9b02e3cd27c65c282 : SUCCESS in 42m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/97e3ec2f2bfc42229840a995380a51aa : SUCCESS in 45m 41s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b794a4c3136d466fb43d48203f5079e1 : SUCCESS in 54m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4476231e66ac4d5cb61d053ab15266f0 : SUCCESS in 30m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/639a01b7d7074707b47927b563a90396 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0eaedb59acc4c01888f732ffc62dfd8 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d8ba7e5fe1c4e7d8397e77e07705bfe : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a49a9150f08444f0baebec988cec8861 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/073e26b1917448e4ae4ee4a662941d2f : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f27ecfd771294b62b42a42638e751e19 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/768efc9acac8455dae415748cc75bde7 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"742ef2091d5b86b8056e7676c10ea1b90b4ba698","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 11:06:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f046adb57317c9a3dcd959c7973488e974bf18c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 12:21:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f251c0eab8b74439a6884a3af4805855\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ef53cf8f5a3f4938a8a8794a59ae6954 : SUCCESS in 2m 28s\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1cee4944650f4a12af8c2ed0d42e43c9 : SUCCESS in 49m 14s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/e2fa97ee745f4d058c7a0c393006f3ba : SUCCESS in 44m 37s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/a02323508dd0400faf7e8bce5847de41 : SUCCESS in 30m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a437f19187a34605b4318dc9a9e7d0b8 : SUCCESS in 52m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/da23f5cf53f14c18980e7dc8e6268272 : SUCCESS in 1h 06m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ede33d606a4e46dd92c6616a8d1647e5 : SUCCESS in 49m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c20a3bfd0d174bb6be766cdb6e83fee8 : SUCCESS in 30m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5450c0d6f30540448eda6e1b509d4d57 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12a4c3516d84195af85f2c72d96fd43 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01aff49f517c432aa59eb9d5047827e9 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b87a7455eb42484fa53bc9db1abf6094 : SUCCESS in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0beac8f33f454920a9847d1acb5472b7 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2573582de32e45b4b7b443cfeea67295 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/998b1b87f19c460597b59bdb3b8b8304 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f1d62fd1763f5445bf6dffaa59d3651a2d0ffc4b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-03 13:58:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9c783f672a5ab0b1190f99c90887c97e5e859a40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 14:45:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a7e90dd193840cab9b06ea71c74193b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/ddb2676e4c3e402c82a577e3d7804545 : SUCCESS in 1h 36m 21s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/cc23f2215ee5499baa897bb0aeb5e48b : SUCCESS in 1h 03m 19s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0faadb0919b641f88b1aa966d6edcc1b : SUCCESS in 2h 38m 49s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ff7d3ab6d40742469f2a9746df1a2c40 : SUCCESS in 1h 07m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97ae56ddf65c4995b01b4393b5b169e7 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08507d337feb49fc90a260e185824c45 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89c58b700e7d4d66af8d0c89be3517dc : SUCCESS in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8302cc2f335e184d7b7cd8ed5da5d8eb294aadc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-05 07:37:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8b611b965673d97686140e73a5d1d56ff1e5c5cf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-05 07:39:22.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"497a99fb889c7ccbb99c3c31a7ade22992483db2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 07:39:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a8245cf2dd413b655f0e7d4f19025a98864f39d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 08:43:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ed19253629b4f4485264f9a0ffff319\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/db1def1f6d0b4be785885b665fde8b56 : SUCCESS in 3m 12s\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/1e97ec5cc73543048f2212fc2a3cbb31 : SUCCESS in 54m 29s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/96f15fea2a0b48da9c53f1b9aade053e : SUCCESS in 30m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a74998f80c2f4916b9102b833edf05c9 : SUCCESS in 51m 51s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8f8254da3c044e1680b5d8cb4afd5c70 : SUCCESS in 57m 33s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9470bc229b74652b7a5aaf67cbf7923 : SUCCESS in 58m 54s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e947c328f9641088a1344e404002810 : SUCCESS in 28m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/774adc45d28246ac8c7fe182b8ac4b89 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5de7101e9df44a65b8e50a134d50174a : SUCCESS in 1m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0485d60f775b46029e8883c6e6502340 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86e97ce0f4c842c78c7556a2f42d374f : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e2a9cc1d934dc79b0c4dff442e6bd8 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df4b00f549854a109c17e9be2bfb8fd0 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"34937085207a67d25a539996de84dfa09455132c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 08:43:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1803a93921d881448a84f5b39f707cda2ea23ccd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 08:44:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7eae64ca6a4695a544c541ea21d63a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/426ef5347764437591e8a25c75ff3cef : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a5ccce47bcf4fb8996da62e4bedaf28 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2da1676d7a861ff921936443861559f1b994f776","revisions":{"0068e803d3621150d43097d9fa056ce888b3abf5":{"kind":"REWORK","_number":1,"created":"2025-12-03 09:32:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/969362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/969362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/969362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/969362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/969362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/969362/1"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 09:31:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 09:31:59.000000000","tz":60},"subject":"redis: Drop redis and redis-sentinel images","message":"redis: Drop redis and redis-sentinel images\n\nWe switched to Valkey\n\nChange-Id: Ic40fecb87dd3eaf67eb1d3088978d462ff5b4320\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0068e803d3621150d43097d9fa056ce888b3abf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0068e803d3621150d43097d9fa056ce888b3abf5"}]},"branch":"refs/heads/master"},"2da1676d7a861ff921936443861559f1b994f776":{"kind":"REWORK","_number":2,"created":"2025-12-03 11:06:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/969362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/969362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/969362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/969362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/969362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/969362/2"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 09:31:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 11:06:34.000000000","tz":60},"subject":"redis: Drop redis and redis-sentinel images","message":"redis: Drop redis and redis-sentinel images\n\nWe switched to Valkey\n\nChange-Id: Ic40fecb87dd3eaf67eb1d3088978d462ff5b4320\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2da1676d7a861ff921936443861559f1b994f776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2da1676d7a861ff921936443861559f1b994f776"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
