)]}'
{"id":"openstack%2Fkolla~969492","triplet_id":"openstack%2Fkolla~master~Idfaddcba475a34423a53f5f11d0405c40056a227","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-03 17:12:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idfaddcba475a34423a53f5f11d0405c40056a227","subject":"DNM: iptables rocky debug","status":"ABANDONED","created":"2025-12-03 16:32:29.000000000","updated":"2025-12-03 17:12:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce28b8dc891af925c8b070dc1c86f3f84ac2894f","_number":969492,"virtual_id_number":969492,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-03 17:10:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 16:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"172a9ae699c5f504d0dda13ef1d54c29ff6a6add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 16:32:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"814fdbb967c7203d887fdf7449da2b4d5efe08f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 16:38:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6220368639901c72c84521f4c137d2a918662c1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 16:43:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c0db7197065c6c6b242ce6562086e69bb0d7e4ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 16:47:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0e659628bbc16da9a6552129ba81870efb2e4915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 16:52:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca88c125f6fd40ed9319a9f2455de44e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/307b9116e8a149829b7ddfe676d92ea9 : FAILURE in 3m 59s (non-voting)\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/098f71555ccb4d2a953e6dd7a3828a98 : FAILURE in 2m 33s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2e1489c413f640adb25fd2ddf1252b32 : FAILURE in 2m 37s","accounts_in_message":[],"_revision_number":4},{"id":"9e381d4c3912f9061357883ee1a1c0709df86bc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 16:53:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"21c12d45c78e0e030f86fe622802a380d3516421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 17:10:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b55d0ff1ca714faa865ee261214958ee\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6c17b0692e8142028375b96626870d0a : FAILURE in 4m 11s (non-voting)\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1623f0791a36432a8ebda8c7427a7b6f : FAILURE in 2m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a93129b814884504be51c84d9adff936 : FAILURE in 2m 38s","accounts_in_message":[],"_revision_number":5},{"id":"ce28b8dc891af925c8b070dc1c86f3f84ac2894f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-03 17:12:53.000000000","message":"Abandoned\n\nserved it\u0027s purpose","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a930904c8722d5a2d9dc1a8307a8a01a144e7d3e","revisions":{"10dc21956f89d73e977afba5e9bc04208016ab9c":{"kind":"REWORK","_number":1,"created":"2025-12-03 16:32:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/92/969492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/969492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/969492/1"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:32:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:32:27.000000000","tz":60},"subject":"DNM: iptables rocky debug","message":"DNM: iptables rocky debug\n\nChange-Id: Idfaddcba475a34423a53f5f11d0405c40056a227\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10dc21956f89d73e977afba5e9bc04208016ab9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10dc21956f89d73e977afba5e9bc04208016ab9c"}]},"branch":"refs/heads/master"},"9ed136185b3d93b9f5d3e6ad86ac1314c8177f9e":{"kind":"REWORK","_number":2,"created":"2025-12-03 16:38:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/92/969492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/969492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/969492/2"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:32:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:38:01.000000000","tz":60},"subject":"DNM: iptables rocky debug","message":"DNM: iptables rocky debug\n\nChange-Id: Idfaddcba475a34423a53f5f11d0405c40056a227\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ed136185b3d93b9f5d3e6ad86ac1314c8177f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ed136185b3d93b9f5d3e6ad86ac1314c8177f9e"}]},"branch":"refs/heads/master"},"9b3ef7009db1a243faebb4774a607fec7b03e6ee":{"kind":"REWORK","_number":3,"created":"2025-12-03 16:43:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/92/969492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/969492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/969492/3"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:32:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:43:35.000000000","tz":60},"subject":"DNM: iptables rocky debug","message":"DNM: iptables rocky debug\n\nChange-Id: Idfaddcba475a34423a53f5f11d0405c40056a227\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b3ef7009db1a243faebb4774a607fec7b03e6ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b3ef7009db1a243faebb4774a607fec7b03e6ee"}]},"branch":"refs/heads/master"},"ca20825d3c384a740416a5ab784dc741bfe91054":{"kind":"REWORK","_number":4,"created":"2025-12-03 16:47:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/92/969492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/969492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/969492/4"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:32:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:47:16.000000000","tz":60},"subject":"DNM: iptables rocky debug","message":"DNM: iptables rocky debug\n\nChange-Id: Idfaddcba475a34423a53f5f11d0405c40056a227\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca20825d3c384a740416a5ab784dc741bfe91054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca20825d3c384a740416a5ab784dc741bfe91054"}]},"branch":"refs/heads/master"},"a930904c8722d5a2d9dc1a8307a8a01a144e7d3e":{"kind":"REWORK","_number":5,"created":"2025-12-03 16:53:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/92/969492/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/969492/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/969492/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/969492/5"}}},"commit":{"parents":[{"commit":"b013f9ad1dd7a70fc4607a363edeae085d08adf3","subject":"Merge \"mariadb: Switch EL to mariadb upstream repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b013f9ad1dd7a70fc4607a363edeae085d08adf3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:32:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-03 16:53:09.000000000","tz":60},"subject":"DNM: iptables rocky debug","message":"DNM: iptables rocky debug\n\nChange-Id: Idfaddcba475a34423a53f5f11d0405c40056a227\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a930904c8722d5a2d9dc1a8307a8a01a144e7d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a930904c8722d5a2d9dc1a8307a8a01a144e7d3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
