)]}'
{"id":"openstack%2Fkolla~970830","triplet_id":"openstack%2Fkolla~master~I5607081701b59ac0ca58cb6c285769209a009355","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-12-15 15:20:45.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-12-15 16:17:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5607081701b59ac0ca58cb6c285769209a009355","subject":"CI: add rocky-10 kolla-ansible jobs","status":"MERGED","created":"2025-12-12 12:38:57.000000000","updated":"2025-12-15 16:19:23.000000000","submitted":"2025-12-15 16:17:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970830","meta_rev_id":"c169bc18c4dd886d4caec204a88085a214d45f9f","_number":970830,"virtual_id_number":970830,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-15 16:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-12-15 14:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-12-15 15:20:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-15 14:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-15 14:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 12:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-12 12:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 12:47:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-15 14:56:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d48e0cd2a29d68a29da62ffc258e001d1267faf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-12 12:38:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1054259cf3c28f4c066a52f0313d2d624ee7109","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:39:16.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  Job kolla-ansible-rocky-10-upgrade not defined\n\nThe problem appears in the the \"kolla-scenario-aio\" project-template stanza:\n\n  project-template:\n      name: kolla-scenario-aio\n      description: |\n        Runs Kolla-Ansible AIO jobs on all Kolla changes.\n      check:\n        jobs:\n          - kolla-ansible-centos-10s\n          - kolla-ansible-debian-bookworm\n          - kolla-ansible-debian-bookworm-upgrade\n  ...\n\n  in \"openstack/kolla/zuul.d/scenarios/aio.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"7e7f43a76c8d7a869c90f16597053c1538d22c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:39:16.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  Job kolla-ansible-rocky-10-upgrade not defined\n\nThe problem appears in the the \"kolla-scenario-aio\" project-template stanza:\n\n  project-template:\n      name: kolla-scenario-aio\n      description: |\n        Runs Kolla-Ansible AIO jobs on all Kolla changes.\n      check:\n        jobs:\n          - kolla-ansible-centos-10s\n          - kolla-ansible-debian-bookworm\n          - kolla-ansible-debian-bookworm-upgrade\n  ...\n\n  in \"openstack/kolla/zuul.d/scenarios/aio.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"3b54adf644c94867333af14fdc7acc8f2f3d629a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-12 12:39:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"98b9802529c267dfb913d144805a46fd5dd7e02d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-12 19:48:31.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"786b8c15fcbbb53d955eac396d37bc47bc5fcb4a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-12 19:48:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe55a90a103fde3105595f4ccf7e7a76b9361beb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 20:26:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75ece421e44541f19b67a00ec5b383d5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/210b9f5960f34841ae6de35d1673d281 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e82c835369df464794a820b70351068b : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f8d3df242ac845ad929182556a351350 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85288a6d98610d43ad8287952cf6a48c18712966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 22:59:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a519b18b1d441e6bb2a48ee57bc8055\n\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ad93035e195f427f8cb9dfeaa7c249a9 : SUCCESS in 1h 06m 05s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/73d65d95879c49ed9bf8ae0af4d702cb : FAILURE in 32m 28s\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/4f679a790b024d7ebb6f6920eaf59ff9 : SUCCESS in 27m 35s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/0c4ee26e2a3246fa8b66cabf70ad8c75 : SUCCESS in 43m 43s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/98e6a474653d473782d6afe42a1f50b2 : SUCCESS in 1h 38m 32s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a9c6728199bd44fa97620a696f3fd0d7 : TIMED_OUT in 3h 06m 50s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/33d7f33e1d394c2e8fade183308f60b8 : SUCCESS in 1h 11m 04s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/87871f2d0e5c45e2a520821a50806f44 : SUCCESS in 31m 28s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/9d0610b4bd02421bb5a4f7637c459751 : SUCCESS in 32m 55s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/274774f37c324d998fedffd52b116441 : SUCCESS in 1h 30m 17s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e7e17d511330474bbc7418fe0ebf53c4 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48de577318a84ac2a2378efe65e53cb9 : SUCCESS in 2m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f25a3ccebe84a19a18213acfd58d1b2 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60c161cf128c43c0b3317a817b105592 : SUCCESS in 5m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84cba4b412f94632826826c3e0ed0f4b : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a14384f1c8c4400cb0c05ef1945c2432 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"591edb10fa8a531b76bbf8ee33916e65f80890f3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-13 17:16:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d338c7508871abab4fec038a1f219f4e2f77f5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 17:25:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84949e0cdeff419397e93809e46c428a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc153eb8ea5440a8b78eb703d5695446 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8cd5872cd87341dcbaebc23b76e7c8a8 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6cdb5961f58145c1b74cac56e432c26b : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d4af0b189f5a4fac3763477750eccd828a38f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 19:20:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58785e5b0d7d4411a785401bc7f0b362\n\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/18291319602a49b69c56b2a2e42f00c0 : SUCCESS in 1h 02m 51s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b21a21aac81443bea4733c787027d256 : SUCCESS in 53m 36s\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/10ba21a4ea3b48118a3ff823a0413b3c : SUCCESS in 13m 29s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/434edf87c3e1454cabe230851829f3bf : SUCCESS in 1h 24m 58s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/a8177e411fd04eb09084c9a1f24eefdd : SUCCESS in 51m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3fb8ca23551441d19abccf487aeace6e : SUCCESS in 2h 00m 04s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/492e3d6785e54d84a182618b41844a92 : FAILURE in 52m 07s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/91a696f7fc24404cbb3451ee95daec36 : SUCCESS in 24m 08s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/69b2c1aa58aa42969cbac118e28f1eaa : SUCCESS in 1h 11m 09s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/86a733ec0cb7477280ca8038fb569836 : SUCCESS in 1h 19m 28s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/edca057e1f5644998bac8b2717d78651 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27a3759d5a17426784c5d7cd362f62ae : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11cd9c43e7594f7f98c738295839cd3d : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c8e20a9625240e7b42ae202989d44ee : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeeaf78c71c54e47a3ae040b880c4cf6 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6f82967ff64457a27de5c0056f71b2 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b7b684dd6e4cfc27f98f037833b1c9654524f61a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 14:56:38.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f24e8dcdd80eaf5f631d3b1464941f325fc44832","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 14:57:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a6d061c6a810f2b24ff281e9cee69ed144b355e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-15 15:20:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"52026ef02f02230f9ea31152cf0243bf048f03ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 16:17:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77acbd3161e24c099b0e3c3ca8f4b175\n\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/dada4022baf74bccb92f515c4a5141b5 : SUCCESS in 58m 04s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ba6b9051f8ec49bdbb231096bb666d30 : SUCCESS in 1h 19m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5265f7e3cb4243cf8a19281591fd9d20 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0214358502a94b5bafe5e20e777b9605 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7db76d3b78c84393921c0968b3ca6598 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/944b7bb816ce4123b7b15deb5dab24bc : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4ebaee3d3364ab287a8086ce9fa4c24 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c5319b6011aa119bc909e584d738d77b655430af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 16:17:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c169bc18c4dd886d4caec204a88085a214d45f9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 16:19:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e21634bf6f44c419c7a7cf7581b6bc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0197a3447948ceaecf37e02d4c16f3 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7401179ab66a30319520a3806ab6f04dd4d41e8c","revisions":{"7401179ab66a30319520a3806ab6f04dd4d41e8c":{"kind":"REWORK","_number":1,"created":"2025-12-12 12:38:57.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/30/970830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/970830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/970830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/970830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/970830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/970830/1"}}},"commit":{"parents":[{"commit":"96b54022857ccef8dc6bb11789abe32119827109","subject":"Add genisoimage package to nova-compute-ironic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96b54022857ccef8dc6bb11789abe32119827109"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-12-12 12:38:23.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-12-12 12:38:52.000000000","tz":60},"subject":"CI: add rocky-10 kolla-ansible jobs","message":"CI: add rocky-10 kolla-ansible jobs\n\nChange-Id: I5607081701b59ac0ca58cb6c285769209a009355\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7401179ab66a30319520a3806ab6f04dd4d41e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7401179ab66a30319520a3806ab6f04dd4d41e8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
