)]}'
{"id":"openstack%2Fkolla~970855","triplet_id":"openstack%2Fkolla~master~I8156e2c2a57d820afdee0b76cd5a050ad682fb38","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-12-16 12:20:52.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-16 10:02:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-12-16 12:20:52.000000000","reason":"Change was submitted"},"2271":{"account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"last_update":"2025-12-16 12:20:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8156e2c2a57d820afdee0b76cd5a050ad682fb38","subject":"debian: Switch to deb822","status":"MERGED","created":"2025-12-12 15:12:09.000000000","updated":"2025-12-16 12:21:48.000000000","submitted":"2025-12-16 12:20:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970855","meta_rev_id":"6a329cde58d8d550398398df1757c2a05bad61ea","_number":970855,"virtual_id_number":970855,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 12:20:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-12-16 10:02:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-16 10:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-12-16 10:02:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-12-16 06:05:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2025-12-16 10:02:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-12-16 06:05:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 16:41:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-14 22:27:28.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2025-12-15 15:33:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-16 05:58:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-12-16 10:01:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"3460c6cfd5ef295bbcfd606e94811b16f88af6cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:12:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3027ccc98e3a77505eeb4adc071d296fb28134cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:15:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f08f3472919e10f39ebadf44faf9536a6df145e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:23:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7b1f261c850bcfbc34dc3037923791daa1ca1071","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:33:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"91be99f0f0c3b589709e4db77ee03a240dbc3b4e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:36:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"3aec390a529cbfb391640374825170745ababe28","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 15:44:09.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2d151ee3d11cc622836fbde22c6a95797fff6c18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:41:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7b6eb32978c4e258b7a0686fdf22b9d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/333ab7e668fa40dcb70a49c93daa1085 : SUCCESS in 50m 07s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/6ddbc4207b4e40c2a838746ee1699f94 : FAILURE in 5m 49s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/048468068a5c4abda7c638ec33596037 : FAILURE in 4m 50s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/231b986cb05843b4812e8d014c85ccb4 : SUCCESS in 52m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/38b1d192f0234eaaa09503489b9d5b78 : SUCCESS in 44m 44s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7539a0b206ea4585a9045c59cf501d5a : FAILURE in 5m 12s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/53ddbebe3a60422ead239b86cc1463c8 : FAILURE in 3m 37s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/5c458d113e824709860d833b5bb4d700 : FAILURE in 15m 01s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/86e395ba812b49cfaa7baeb345cb4aff : FAILURE in 15m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/eb1e7de8bc58457e90be985f955066f3 : FAILURE in 15m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/40c1a42f393d42ee84a576b26603b554 : FAILURE in 15m 23s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/f675ecf27cda442dba8f35eaa956cfb2 : FAILURE in 13m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/1826f1cbe9cd4a5db9716102063723ab : FAILURE in 14m 27s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/2a66527a451a443ba9a7f600007548fb : FAILURE in 11m 36s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/469fc18476c24f5bbae1ea882d4a852d : FAILURE in 6m 00s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/fa5487c3c46445f083f3432886371efb : FAILURE in 7m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/97a8e1bf304247c18756bb442bf8b738 : FAILURE in 13m 22s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/ae88fba3ca8440e3a5bcfd777330dc96 : FAILURE in 6m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/a90ef6673b244135bd785f861e96f68f : FAILURE in 7m 26s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/157d77367c5c4510bd78a3e401d23aa4 : FAILURE in 9m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/67a2f8a063dc4b719c3d1b5a9e0ccc53 : FAILURE in 8m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/10235e483a5a4d4ab630cbbdd7da9123 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/844138a7cc854a6db56527e0acf69f46 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de7db04f54db47628a5fbc68c174d24a : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf08160b3e4b492db6693c93f7dd4315 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb0fe75643e34e028524e25704162e63 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6741719cb3d34485915c1414c4bcd3d0 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/316a50e516274235974ece8c87fa8534 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":5},{"id":"706b9f09caef24ab51603b0e6efaaf0437ee6fda","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 17:33:26.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"561fc28447f8d179a74ff65e9b869d316123b3a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 17:36:18.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"c534750ad628283e257d0f5b2576c95ab790db17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 19:02:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75ddaa33633b45ff89eef18d935c3f0d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/14bbc9e3e7a044a49950f55761bdbac8 : SUCCESS in 56m 53s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/afdc2409976b4089ad25575eec4efc5c : FAILURE in 9m 06s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/55bb5084eec2413bb4cb30a1432ec938 : SUCCESS in 1h 21m 24s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f5a1e6ae34cc46fca6ab6f619f3ea5b1 : FAILURE in 6m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8169865991e1441eb9fef564f85b1cf1 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f1473f3af00c4923b1f0a9f92b66a24c : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/256e50b02cbe45b1b10c06785f063519 : SUCCESS in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0b90b03943159e2b1802246cf193f57855d3a73e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 19:15:06.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"cfb3f0e78a6170bbae7bdd798e88322d37106967","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 19:27:41.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"eb81d2cbc0f009acb6773b322b3d37da671d86ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 20:26:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2036dfaaab244c88f668790c0fe7d9e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2bfae6077950459dbdce18fe75bfc316 : SUCCESS in 45m 53s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c89d05a2ef3e42f496fe4374106e7d4a : FAILURE in 5m 55s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/b68e9965ae0b42eebdc6bd0da285467a : FAILURE in 5m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7d75717933da4a20accbfd7e264edddd : SUCCESS in 49m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d9b3848e83124b22ad222b0d78efb866 : SUCCESS in 48m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46cbe57380cb4b8c8f08a7abeb9d65d9 : FAILURE in 5m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7168817e6ed450c81c1c0e28f9d36ba : POST_FAILURE in 4m 29s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/d5b6b7a745d84ba9a13d637b7774da3d : FAILURE in 10m 56s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/822610b98bd94349a5bf1efd0b1c7471 : FAILURE in 9m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/25f7b9e8e16c423090c4330c3ca21a9e : FAILURE in 8m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c0973a2b05dc452bb89556a7c796d95a : FAILURE in 9m 12s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/3cad1358e4f346a7a42694a005ec3494 : FAILURE in 7m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/b1ca87b690c34d32b283eefb94f7812e : FAILURE in 7m 31s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/5dce780579ca496f8b72f420e98e0b95 : FAILURE in 7m 37s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/2bc234dbff454030b0871cded190bcc7 : FAILURE in 8m 34s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/8114af8f87004038908ab13e316bc560 : FAILURE in 13m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/2a88c64575784de5a326d82b792ee7ca : FAILURE in 12m 18s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/8e515669802b43779b58c92174fe9b30 : FAILURE in 6m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/0cedcc7f0da84fffacb6b97242bb18de : FAILURE in 7m 14s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/f055d43eacf1498e8680ad628d5f06ca : FAILURE in 7m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/5db2db1d76e14c729f5f1fdc7d350c27 : FAILURE in 7m 42s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ec3e922912bb4decaff7cd037ed717af : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2c8fada6874f5b90718c67e24322b8 : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34f5195268fc416fad5a3ddcc8e74d0d : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83b8faa03a8044ed803f109ac198d84e : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/745e0c49ffba452c9f11d4018b650215 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b06b6ca3cb44517935b1beb49a536db : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b557856a7b147baa44b0d66f669e329 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":9},{"id":"fa75bcb103cd9c42022e4910d33dc0fe74fcfdb8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-12 20:44:11.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"172565e537bec0423564590d75a989b1b85a8d44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 22:15:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a07c6094640a4772acd6a8eaac3f8219\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/6888d2dc49674d44b209a4a3ed91d956 : SUCCESS in 1h 28m 42s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/bdeec04cde9c4d56a88f1433aae9343b : SUCCESS in 56m 33s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/76371450e76846098ee5a1956f99dbf0 : SUCCESS in 1h 28m 15s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cbab52436e664c1fba732956531e8d30 : FAILURE in 7m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b6d62d40099246409cc214c2abffd61f : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b9e774bfff04511ab8e3a1aedbac157 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d4991ccf693d4dbea27e217d5e14bd56 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c7521cfacd5884a98ba482353535e69d4331389d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 22:35:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cfa2167140d4234abdfb0c1bd94aea7\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b26f908bdcba476183510a1a7119f46c : SUCCESS in 52m 21s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/dc2b458e4b8446eb807c6df6af7787c8 : SUCCESS in 50m 42s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/e70735fd54244af18abd47079cf819a7 : SUCCESS in 14m 55s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8536270f607e461e97dd9aa05281525b : SUCCESS in 49m 30s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8bfdc39afede4273adafb3f42a8e1329 : SUCCESS in 36m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e22f2d7ad80240a89cf40ab83f77851f : FAILURE in 4m 17s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d9cd59e6b6e14095bbc58f9b9271ab21 : FAILURE in 6m 15s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/496f94b388d04685ab76469c7048de57 : SUCCESS in 1h 15m 43s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/70c731c0a0d640feb1843c2bf7940318 : FAILURE in 1h 49m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/7064f8b805d74ccfada18cbb5c30da67 : FAILURE in 8m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/959b3c5989c64939b0fa1831db8847c6 : FAILURE in 8m 34s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/fc7e7b1cdfdb480482ab2e369593a41c : SUCCESS in 1h 00m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/6a0f5b4c4dba49bfbd8ffb48445ea67f : FAILURE in 10m 20s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/c6bf38558a7a4d9a812968bc981d5248 : SUCCESS in 22m 24s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/249fce4c1b88450aa8c18b90a6f44f15 : FAILURE in 7m 54s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/d43176d4d670464abc5358b645d726d0 : SUCCESS in 29m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/5c6629570c3248bb97b1c6c72c49e315 : FAILURE in 7m 48s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/ea5c98453eaa47e2861ddb869e3f8065 : SUCCESS in 47m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d88a6439785d4d319b0d96050c241391 : FAILURE in 7m 45s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/8c72f16570b04e3e9fd4f563f4357d92 : SUCCESS in 1h 12m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/c801b6171bc2414699d2d16ec91281cd : FAILURE in 8m 35s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/44bfa58b851845cfbaadedf8eb840b0a : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be0eb0825814f53a86e77efa68562ad : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/585eceafcaae4f168047d8a91de74774 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ccf01f329334293b49cc66dfe3312da : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ed74664e40d4d3ead4833e29a10a6c1 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68fac1a06e9472893ed1da2c2d292cb : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b167eb23eaa46e6ae3cd41b60fc47ab : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":10},{"id":"71f51b53becae98f9c12b1555ff4a28f0c19d4c9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-13 06:04:17.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"18798b3a54ac45d0ec5f1f35cb1983941f7f48e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 07:37:19.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8ccecaebd224c4ab064c9f2e55cca8e\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b393f22fba6840b2b709b5e8fd6d9d44 : SUCCESS in 1h 31m 05s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/cc8fca7358154403a559def1d90c3b0c : SUCCESS in 1h 21m 40s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/72085080e00c47b795435773a2fa7241 : SUCCESS in 50m 45s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/97a48ba4de404792a597705b52b79a44 : FAILURE in 9m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59a15b34bd3e4f27ab410f63b086d2f9 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/48414d795d5f41af9ff1c5d9583d1d5e : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f170c90a70b49b28646d1669189e3af : SUCCESS in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"821689d82c8487a5a3b9cff2b6bdf18b65ef1ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 07:51:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dc329824bd74b59b5f1a763226c6d1b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1aad79ebdad34f58a13e3f7a9d78061e : SUCCESS in 41m 30s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/0a4e50fce8c644e4acb91ec9bda3d60b : SUCCESS in 35m 57s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c424c271d6b84975bd9b019045338109 : SUCCESS in 16m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ede7a5a5f7b649f0aedd9b92cc434e69 : SUCCESS in 39m 50s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/188364277948413cb2d2778c0b9a90eb : SUCCESS in 37m 21s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed15040da0b044e2b097640bd86f78b5 : FAILURE in 4m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45a643bbcb8f44c4b15c53e1ceb1759f : FAILURE in 5m 09s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/5c66b3f7d017411ab3b9fbf454eb942b : FAILURE in 1h 11m 32s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f461d6e2a93d4687950b9eef685b3229 : FAILURE in 1h 10m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/4cf252ee01af4433a1aa0415b546cbb3 : FAILURE in 6m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5e3df7fba6e14af69c104d59a011081f : FAILURE in 9m 54s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/f8d407414f3d4411a0fcc2c11a4cbd45 : SUCCESS in 1h 42m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/9828c8ae35bf4bbab8bd76087ff870fc : FAILURE in 8m 26s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/577e24c6ea9a48dcb176e689cdf4961c : SUCCESS in 26m 22s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/8bcab1be008b4b828aef3df4dce291c1 : FAILURE in 8m 07s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/324130520ad0481db54de8db433c9fc7 : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/765ac68bc59b4b17aced8527292dc47b : FAILURE in 11m 26s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/9705c3eb106b4a718c4bc86fe333d52f : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/59add0e772d94b31a63d720af103a8e0 : FAILURE in 8m 29s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/4b6dd23154f24947a4e152610a8bd474 : FAILURE in 59m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/3b2d73e44d3d4646b41ff9a6137668bf : FAILURE in 10m 36s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4ef2b3abcf5f4accb28d2b9235531d31 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1216d144517b4de48e6b302ffb9e86c3 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2ada6b32127408e9a6e15d6b83f29b8 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3cdbd4eee68472ea2e36a4e1e1715d8 : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a0d0770757948f99ca73848c02d96c8 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce448dbad45945d1aebb5817446860ac : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a81b79df80f4d8d9e5859ea2d85485b : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":11},{"id":"df03c78fceb91286bdfac8e7873c605aeca04073","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-14 13:36:26.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a51f8be005675abcde40023eb1faadea6cca7615","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-14 13:42:41.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"8cdb65da2921c1e2b3cc95adbfa874e59caa78fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 15:20:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64d3958310a34ef7b645f702946cb778\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8402ad5ead4b4b1fbbd4c8baab8c2d9b : SUCCESS in 1h 36m 12s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/9f77a005a2a046de995d63c3494b4ac3 : SUCCESS in 1h 21m 41s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/34f078ac107943edbb7643926b37ccd1 : SUCCESS in 1h 00m 17s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cfda7ff1455f4bcabc23ffe044636795 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8bba565ef2a04afe93a152f8db075355 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9b319906d8a840749b9a5feca3fd85c6 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/829f45fc914c46cd9746563ad3d58b76 : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2fcafa36ee5935ce2f0a96f6532173ae498639b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 15:39:05.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e13ad9b15fa84ff196da5666251701e3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/fce16a9107554c89b65201bea626cb0a : SUCCESS in 48m 06s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/14d8b29f69c04960b32a04b97636e06b : SUCCESS in 34m 00s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/4e42185c83e84afd90446e23b422220a : SUCCESS in 14m 41s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c30c92b7fd5d498eb242d638011c9e68 : SUCCESS in 39m 36s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4feecf4990fa4929bf1403aa9bd2eaf4 : SUCCESS in 34m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dd8272e2b67450f847ae7cf9c8bd961 : SUCCESS in 36m 28s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4325026e981e4a439306bf1334247c03 : SUCCESS in 14m 03s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/5a205e03aa7e4990b5481b3aa33fe1c3 : SUCCESS in 1h 21m 43s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/631affcba9e043fa82f375e084d210a9 : FAILURE in 1h 52m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/154bc54321a24062a4fe23133e1f1009 : SUCCESS in 1h 34m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/837cd907518a442786401ded21d68e9c : FAILURE in 1h 02m 54s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/78408b35f754433ca3b38d612bc3f67c : SUCCESS in 1h 09m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/9076d64561b74495aa82cf33ce6d1fc8 : SUCCESS in 1h 10m 18s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/800488ec2aa14742bdaa206ce13b2602 : SUCCESS in 30m 08s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/10e54cf5aa5c48afbcd47093f0911efc : SUCCESS in 32m 34s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/1559b98a542243ec9d304cb24e513703 : SUCCESS in 41m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/4998db802d504c43ac069fcc9da6b974 : SUCCESS in 1h 12m 12s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/dfa80c558e2940059413833599dc62f7 : SUCCESS in 1h 09m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/5403eb8c9eef484e85dc78d8c9ed7dd3 : SUCCESS in 35m 54s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/72e8c3de237148eabcb602f52734e931 : SUCCESS in 1h 06m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/6fcd2ed2d20149b2bf144cfdf02b20fd : SUCCESS in 1h 21m 34s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b26b71a96821422cab664eaf33e854c6 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a4070d72844ca1aba7522df361c4c8 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0d66a9ab0614b9b8e3ab4e72ca57a74 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fef789866b441a3bf0dd49874e77ab7 : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd1cffaaf5d04a78936393dd879daa0e : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b060ab3d1e44ce88c5994a4e79bed61 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a2df30478cf4cc693dcd030bf56a266 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":13},{"id":"4f80e472216bda6c7aabfbbc55a468083e558acd","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-12-14 22:27:28.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d2d3590a0ae13b2551f44d25b104e1797fa7fa2c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 07:14:07.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"34ffbb75270b90422bca30a8b828fb00d51681f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 07:27:44.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"8c69f2c12045d46a7f4ea4d11c9e3fa0bfb316e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 08:00:32.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"84db60de62c04c7c05f7c1d94270a94e652ce776","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 08:05:42.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"22e6c0eebf86c4fdcd1cdead47ceeae0b28f3b1f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 08:17:32.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"698db1d00a76daca998757fd06305d44de435b23","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 09:11:55.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"0db48c8fb40b892c0fd21abce76cbb7d896f4844","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 09:24:14.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"c7caf6b5f8fa93cad00ed80eaa3ec7a2c346a931","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 09:51:12.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"1ca83e29140a33f20698a9ce67c3722a6a05b27e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 11:46:20.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7504f294d044708ad4030fa04d6a4a2\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/95ad4007a3fc479f976a0d8a4c40523e : SUCCESS in 43m 02s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/b2f87ef852314998877d16ea64946a83 : SUCCESS in 48m 12s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/95d8ff67f5e54d0ba25eae26c7859d3c : SUCCESS in 16m 34s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/dff4ac8ba9d84c05992dcc9a9963b1a4 : SUCCESS in 48m 12s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/83de33530dcf4016ab1e64bf9611d8cd : SUCCESS in 1h 03m 55s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72c7321b5a1543968f1f5de0c1f48819 : FAILURE in 37m 25s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe7a4276b11348e780cfe0cbf3048754 : FAILURE in 26m 17s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/c5464f74ae1d4d9489a90f1afea37989 : FAILURE in 1h 01m 59s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/52ccf90a574640f28296d155bca25524 : SUCCESS in 1h 51m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/526e94e3aef345e68b5a8f3446dfd155 : SUCCESS in 1h 26m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/97ee4513cb69446cb3f08651302839dd : SUCCESS in 1h 53m 58s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/e00c6d25555945d7b6ced5266f044a51 : SUCCESS in 1h 00m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/9361f09ab1a249db9b92b6da2365ae6f : FAILURE in 21m 28s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/f54b0823df0048238883ea81e5627b8d : SUCCESS in 22m 41s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f51cda5a3a8f4ba29731377db3e941e7 : SUCCESS in 25m 59s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/0bc1ff0baf2d4a85a70f56f75410e728 : SUCCESS in 58m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/76ad6c31009c46f7ac8e02219e06bf8d : SUCCESS in 30m 41s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/8ae6a2bfce304be98b851f77da1cc6de : SUCCESS in 51m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/9e49c168571a44729a07c816b2a2b30d : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/c9fd7d49e9194ad896059db46b05871d : SUCCESS in 1h 08m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/49bb7e94d6594f309ff22b5bc74898b0 : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/362590d3e0b74728ae75daa96613b7c6 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/601a0aaa96b64aafbea4f172521c2ea6 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16fe7d236af74cbba29f8609e6d69f9b : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b6aeda002164c87b35856907e896d2e : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e39df03ee09841a3b17f21bb40b93b56 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc6858300ec47568d357f496d2575c8 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df0c4750ae6a4a359afd27109fa092d5 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":21},{"id":"0c5eecf3d60d7167b740c37a358799e08ff16168","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 13:09:37.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0dd6c12547d43c7b4f3c3215863b4cf\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f1620f965a4642869df7e68d8fc953d6 : SUCCESS in 1h 58m 37s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/486900e527df469b96b9adfc47fdec59 : SUCCESS in 49m 40s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/9ca8f9a416c344efa3f3a352f4ad2483 : SUCCESS in 57m 53s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/aeafb23f14714a5c82da03be45b299af : FAILURE in 1h 49m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4daeb5041df4f28a944c0fa851efe4b : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be9ff82d4f5d4ab1b16f2ae542800a56 : NODE_FAILURE Node(set) request 0f5841b429e547958ae9eff2ed07b5c9 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1ecaf20e301543928e1fcad2ed024368 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"9f448aa92e94f53f2a06984c85cb4cd4b3290332","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 15:05:53.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"c66a04cff2a181786076c38b60058f0b6f6e34e8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:15:10.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":22},{"id":"7954a27d07d250c00fcab92f65084c234fae3fb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:15:20.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"19294a46b682cb1f59e143eccb5e7d9833452cff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:15:47.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"8f8978fb764be80faff3406c850ac2668e68aefd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:16:27.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"91b39fbc00a8357f33116f8c4f63be6fc66c1b51","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-15 16:26:50.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"ce523e72e25147b42f5b9835c901b6b6d190a283","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:30:07.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"69213bf1f36e673a2f0b5a1398de60cf3ea6edbf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-15 16:32:55.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"24c32cd1949444bd151db4f76376019aa5167197","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:35:04.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"58f06f9137d39b8dd5397b564c025fbbaec2f8ce","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-15 16:36:32.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"90beaa3b653776d8af518bb7b94808687aa99ebd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:37:23.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"adf418c576efd2ded2f11a9eac84aec2a91135b7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-15 16:38:45.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"f6b384bf073aacca7dbabe4fc9774fe225b6056a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 16:47:30.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"906e3bd45f7d1a8aa5dbcef37096366ee0f3f2ac","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-15 16:55:03.000000000","message":"Patch Set 25: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"cefdfb33cbe060ae36597a48052afce7a84747cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 17:54:53.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcee1aad50564e2baa734ba5f3278c75\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7bb79d01e6be401aacefd8f913bd0881 : SUCCESS in 1h 31m 46s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/9735dc47a78c4e7fa51ad013b47fbc57 : SUCCESS in 1h 15m 40s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1f34880940764c328b9bc8d0873fddbc : SUCCESS in 1h 06m 27s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/94cb59f2dd844c8bbe9c142b9fd5561a : FAILURE in 48m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/798c3331752b423baec1fdbc4b03b576 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91691d5618884af2a29dcca797af0449 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fdaf41de631345e49b5f7219561bbcd5 : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"0bc3fdf407b1d2d42a266bffcac520ad823e079d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 18:16:30.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c9679d85ed43f280f67a8490c4b990\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b3d2bb64f82041bc8748da3ec6d89b31 : SUCCESS in 42m 08s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/8660cc27bd7c47daba7f169a6f7a3687 : SUCCESS in 54m 40s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/8c563dc220a24e7cb54993f3b24a4244 : SUCCESS in 29m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0bfb8eb14d83424ab3864540609c902f : SUCCESS in 47m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/900500c5b6b740bb99ddbbf189188e3f : SUCCESS in 38m 40s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72d1f039df77466e9af3c9d5cfef9519 : SUCCESS in 49m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a35ccc2e0c054d6ea25d4931262855d6 : SUCCESS in 28m 51s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/b4dfb6ddbf1a426f8da819606a3c80a7 : FAILURE in 1h 12m 10s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1a6124fbe21f4d71871e9945000cf73f : FAILURE in 51m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/5e8f7028413e42c0bde389f2cd195fba : SUCCESS in 1h 30m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4f518a0fc93942b7b6ee1ccdec723d7e : SUCCESS in 1h 58m 12s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/44dd7563d89f4ee49121e86d4a6e8b01 : SUCCESS in 1h 05m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/fa369c8bf6934ab493f0ea6dfc823aa3 : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/eb010096f377409dbd2c872c927b9cae : SUCCESS in 24m 49s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/9a5b9604c9eb4f3491094a41418a6acf : SUCCESS in 30m 51s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/5f858821b10b40b3a0cee70ae115e856 : SUCCESS in 1h 05m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/a9ea76d0ae074b8384dcc2b6c5ae1910 : SUCCESS in 1h 12m 43s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/2afa636e53594bb19408407c7197bef1 : SUCCESS in 52m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/6c77d0de3bb641edba51b56b65e37cef : SUCCESS in 58m 42s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/5e6e8dff8c204901aafb29e23aa31341 : SUCCESS in 1h 12m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/50227cd51ae9419c86c3022994c2fc31 : FAILURE in 1h 11m 18s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1373d350c42440e9b7b891fb5bffe190 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532399c390a14474a667d934d71c6d9c : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ca49643367c4a329404aa92274373f0 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/74353c98c97d430faec73bcad68fb891 : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4f1ee0d255545ebb55295d7b98dfb63 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b3f5e4c8294faa9ac3023ef16785ca : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db2d0834e73e42cf9ee95b88b64440d7 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":25},{"id":"05f9cb49bc22d184b4ddf234309ba1618ede44db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-15 20:51:02.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"4add1f88e01df1aa6cb8a35bc2cfe150f80d7d5c","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-12-15 21:51:00.000000000","message":"Patch Set 26: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"c3ac85fab78d5b23b028cc7a44052bd5bb53899e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-15 22:38:47.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"db13e34e7be691de8c2a53b7e1ecd44679b92c18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 23:02:11.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba27f8a6352748deb095d176376dd6f6\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c947275db7e04d828a47d31a4a1fc207 : SUCCESS in 2h 08m 47s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/b19609bb04224407b188e2739f410382 : SUCCESS in 1h 53m 14s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/58a79e35a541435791d47da07b9dbc4d : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/25d261a4ff6e41f0b5c6a1efbe5d8532 : SUCCESS in 2h 05m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36db860053f648348ce32d83b388f15d : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52a290bd699346048257f3e588f8964b : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb5cd650044f4a539b383890c686fc56 : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"f92fe541fa8240598d970f4d1545802f2de4ff54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 23:25:34.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5836e4b8d554a809e81dc372319de4e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d27cdc79a4be45b09c15c396d71882ed : SUCCESS in 45m 49s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/b248e11724164a70b3952d5e872a6bee : SUCCESS in 34m 49s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/20416508433e411bbab42b6d7258e0b4 : SUCCESS in 29m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d0319702bb0445368a5f7ebde362eba7 : SUCCESS in 47m 12s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8a8357ca52864881920eac352e6f56c3 : SUCCESS in 39m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7862d6c10aaa4addad41d7802ab861a9 : SUCCESS in 34m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe51acd4699847e19787f99b8c42c47e : SUCCESS in 27m 07s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/18d8e396cfa8432f8bc760d999398918 : SUCCESS in 1h 51m 58s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5ce98faad6054cbdb46d638a9140f6d9 : SUCCESS in 2h 23m 47s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/1909b38965df47329719ce66022cf834 : SUCCESS in 1h 57m 05s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/def24f8ac4ea459993758ecb2b1faa45 : SUCCESS in 2h 30m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/8ee3e5b4ac9c40d08984d3e057ce29c3 : SUCCESS in 1h 52m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1dc60eb8802b4452a0cde9f49e0d5290 : SUCCESS in 1h 51m 49s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/bbfdc53b2d1c49dba2181740ed15b537 : SUCCESS in 1h 07m 53s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/6f4021dace10498680a9803cfe527abb : SUCCESS in 1h 07m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/d97c21cae097439996d6a67e4dad0017 : SUCCESS in 1h 32m 26s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/6b9cc57a8c714ba9b21e22e145338706 : SUCCESS in 31m 36s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/85be0dcfcd0d45858de3e06ee9d85f27 : SUCCESS in 25m 16s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5c1c6ae2cad64306b09dd09c42bf2714 : SUCCESS in 28m 26s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/87d7d81b3a12486baddff1eafabcae96 : SUCCESS in 1h 03m 40s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/f1508cd1fe894d3a81aedb1ac912bdf5 : SUCCESS in 46m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/74fa97865de047afb222a9ed73264479 : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/39cab09c0e994b2caa57f91bafbcd635 : SUCCESS in 50m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/4fede28e2d80472fa3957a59e94ae117 : SUCCESS in 55m 48s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/30176b2f7f8440f3aaa78fd33d32b673 : SUCCESS in 1h 10m 49s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b16f03f3ee224499b32563f16bf6511c : SUCCESS in 1h 17m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/cdd2176e32184477a545c9da939d9a14 : TIMED_OUT in 2h 32m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d0c56a969bbc46d48321c62ad9133b1a : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b235e4d581854d2d972024585e63c7f1 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a20032c213434266a1a5ea9abbb1015d : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/302305a696aa4f5caef2a8834a8bb462 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c770e5e6e5ac4f1bb1cbb1d9e4c71a37 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4dbee38f1d45c394bbc840040df4d8 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d22864a7fbb4676adc60064a1a84044 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":26},{"id":"6588519e5840d0edfc4c14d3c0523433b903497f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-16 05:58:15.000000000","message":"Patch Set 26: Backport-Candidate-2 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"68d989d69946cc3771c4ee932531a29c17a847b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-16 06:05:28.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"9273ede0ae926cf15ca7c2c43378095761bf6d01","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-16 06:05:45.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"9f4f9b76beaedea9d2a250d705f40f5739e99027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 07:34:53.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2393fed9181c4582b72c8f676c899b09\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/0fc7e232b3564efb995cfbf8b5fcd3c3 : SUCCESS in 1h 27m 22s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/2e911d4aa52c42c9a42a3339c0d7a568 : SUCCESS in 58m 55s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/5e0e17383cd84ef6a579505d628437e8 : SUCCESS in 59m 31s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7e70ad96303d49aab8f4edc3787d2120 : SUCCESS in 1h 23m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1cb3d985370a4b4991df4f5a09770d32 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a95250e92ec4b57a28b0f0772576575 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/baef1f5770c648ee9ef6c8f66ca1ab0b : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"589bbc3477e4359c16b3c20509ff4952f4c869c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 08:41:01.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79b296a1b28d4fdbb170e5270bcbf1e8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/91876ed225754873a2da06abafc13b5a : SUCCESS in 41m 47s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/06a8173573ae4675a826482238a991e7 : SUCCESS in 34m 12s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/991dfc7ce1c242428b00b697e4c13f2d : SUCCESS in 14m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b491199d8acc444d81198c37ba7da36a : SUCCESS in 45m 36s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9b4d0b078d56470eaf621f08f4297f3c : SUCCESS in 1h 04m 59s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cbe3584957b4bffbc82ed412aa1d025 : SUCCESS in 37m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/082a93eab9fb4a6887cb4f527e123224 : SUCCESS in 27m 18s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/2a390307787e4b9db6e93fc44c8f0738 : SUCCESS in 1h 52m 53s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9e95b8a3ae794d20b49401b1acdddb5b : SUCCESS in 2h 26m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/baa91a8b62b0489195f3cfe349be840e : SUCCESS in 2h 02m 52s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5de5eab0642448848ae8301634d0f64a : SUCCESS in 1h 21m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d1d5e3430ccf4c6aa87f24a182309967 : SUCCESS in 1h 56m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/bc8e07a73f4546c9981fb18a57157f05 : POST_FAILURE in 2h 31m 10s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/62b8c026cee3446eb5625201ce5c0e5c : SUCCESS in 1h 23m 09s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/a8aa2cbf7b5d441cb4f81c84acbe4468 : SUCCESS in 1h 38m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/bd48249d925b472b8b11aa7b793c37da : SUCCESS in 1h 39m 54s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/b65d8da70a7f493ca5fbe9bec780f200 : SUCCESS in 26m 00s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/de72dd5ba0bd4211a90ead7d00b49ae1 : SUCCESS in 16m 38s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/45c249d23ad6478d94d710a28847e7b4 : SUCCESS in 23m 11s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/8485ebf666784e3bbf54b3ef602db9e3 : SUCCESS in 1h 02m 20s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/cd612c23f5044a2c8e3dba4ed003333a : SUCCESS in 54m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/ee0cd99a7bc54f618377b99604dc88b2 : SUCCESS in 31m 46s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/d11fe21b65614cd9b67e0de30814c8ab : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/57086307a68d4573864120f506bc50cc : SUCCESS in 54m 04s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/3fff1d4855644afa8a994c6c469e3656 : SUCCESS in 1h 06m 46s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/0e54eab2ed224ef594b75fb965d132e3 : SUCCESS in 1h 49m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/3efd377f28e44d8a8cf96185a0f9f1cd : FAILURE in 1h 36m 30s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/63f0de8af16443eb82e4478411027dc3 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/685c05f123c746e180605ee59ad95078 : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c5a253331ef43479db9d31b3b2100c9 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a65d67671c264ca38ede3b569ba119f2 : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2a5eda34014439bbc90dc47c1162b57 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88af47ec24134fa98f79368ab8fa90ba : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/049987b1c1174074906a31689436e6f2 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":27},{"id":"7783c47a88933e923d72d022de709ad4ba698c8c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-16 10:01:03.000000000","message":"Patch Set 27: Code-Review+2","accounts_in_message":[],"_revision_number":27},{"id":"cf5c1638d5e582ff2aab63e579bb95c9bbf77183","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-16 10:02:48.000000000","message":"Patch Set 27: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"45ac417bb047717afa0b9959778bb75b6dbddc36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:03:22.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"6b1b730ef9f71707fec56a2c1945b533ed8cff74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:20:51.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2516f0867eb94aaa8eb5084b9ea8384d\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/bd0038ef0499491f85c41fab12ad6cdd : SUCCESS in 33m 55s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/0daf63ffc5d74144816277d18b88941d : SUCCESS in 28m 48s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/16f0aef44b024d7986f38610d2b49edf : SUCCESS in 48m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dc602247873e4934b4f4c62d515a8cfd : SUCCESS in 48m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d9225c3f3d74fb191a145614c446f91 : SUCCESS in 45m 38s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/942793fe22724286920b9817983049b6 : SUCCESS in 26m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/52f5e9371fb14dca8d1fd4cf112bee0f : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23b5f76f261a4b55a888fbad82270796 : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63d53f511d414064bac070e4c179b532 : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a704e8b1e90415e8059bb9d2474d4f4 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b168e9b44f4e21b76d11ad1fcb22a9 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/749989661e9e4cb1bf2bdc3082512895 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":27},{"id":"b1786f5244cea6ec67f1bc2daae36319eefba32a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:20:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":27},{"id":"6a329cde58d8d550398398df1757c2a05bad61ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:21:48.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d7462591a54738931045985070a513\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1516279d1f6d41ec9ec2916308862073 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f1708212681417299aa8b43181efb5f : SUCCESS in 38s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"a7ee5ccb2b551af5be80c0f082e503f6cbd54096","revisions":{"7787350e3e400d683f3e4c78d5cc202a7cc61fff":{"kind":"REWORK","_number":1,"created":"2025-12-12 15:12:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/1"}}},"commit":{"parents":[{"commit":"96b54022857ccef8dc6bb11789abe32119827109","subject":"Add genisoimage package to nova-compute-ironic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96b54022857ccef8dc6bb11789abe32119827109"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:12:06.000000000","tz":60},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7787350e3e400d683f3e4c78d5cc202a7cc61fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7787350e3e400d683f3e4c78d5cc202a7cc61fff"}]},"branch":"refs/heads/master"},"fa281a2c16ce4aaca817eab9cad354a60c2b059c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-12 15:15:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/2"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:15:14.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa281a2c16ce4aaca817eab9cad354a60c2b059c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa281a2c16ce4aaca817eab9cad354a60c2b059c"}]},"branch":"refs/heads/master","description":"Rebase"},"0f65bd828bffb113b35a0feb8d30abd8cae69746":{"kind":"REWORK","_number":3,"created":"2025-12-12 15:23:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/3"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:23:07.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f65bd828bffb113b35a0feb8d30abd8cae69746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f65bd828bffb113b35a0feb8d30abd8cae69746"}]},"branch":"refs/heads/master"},"917657601d7af59d8fc0d1b6822f273b7b32ef5a":{"kind":"REWORK","_number":4,"created":"2025-12-12 15:33:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/4"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:33:54.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/917657601d7af59d8fc0d1b6822f273b7b32ef5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/917657601d7af59d8fc0d1b6822f273b7b32ef5a"}]},"branch":"refs/heads/master"},"5f643ce9ab263ef723e483a6489886242f99561e":{"kind":"REWORK","_number":5,"created":"2025-12-12 15:44:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/5"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:44:09.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f643ce9ab263ef723e483a6489886242f99561e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f643ce9ab263ef723e483a6489886242f99561e"}]},"branch":"refs/heads/master"},"8498319cc08d150033c55593ad0494427788f7ea":{"kind":"REWORK","_number":6,"created":"2025-12-12 17:33:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/6"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 17:33:26.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8498319cc08d150033c55593ad0494427788f7ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8498319cc08d150033c55593ad0494427788f7ea"}]},"branch":"refs/heads/master"},"2c1578cd2ab625795e165a44edcdd11dedcd4f2f":{"kind":"REWORK","_number":7,"created":"2025-12-12 17:36:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/7"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 17:36:18.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c1578cd2ab625795e165a44edcdd11dedcd4f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c1578cd2ab625795e165a44edcdd11dedcd4f2f"}]},"branch":"refs/heads/master"},"f2af0b2b93e340d83ce8e5d4c03856dcae567756":{"kind":"REWORK","_number":8,"created":"2025-12-12 19:15:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/8"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 19:15:06.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2af0b2b93e340d83ce8e5d4c03856dcae567756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2af0b2b93e340d83ce8e5d4c03856dcae567756"}]},"branch":"refs/heads/master"},"21de9a9da96e4fd32105baf9b300f9cf40cc9f1f":{"kind":"REWORK","_number":9,"created":"2025-12-12 19:27:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/9"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 19:27:41.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21de9a9da96e4fd32105baf9b300f9cf40cc9f1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21de9a9da96e4fd32105baf9b300f9cf40cc9f1f"}]},"branch":"refs/heads/master"},"1e3c95e86cbcbdbd9079c7771dcdfb8a107b9e58":{"kind":"REWORK","_number":10,"created":"2025-12-12 20:44:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/10"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 20:44:11.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e3c95e86cbcbdbd9079c7771dcdfb8a107b9e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e3c95e86cbcbdbd9079c7771dcdfb8a107b9e58"}]},"branch":"refs/heads/master"},"7e07d22066d328cfb70326978339e5e39bf18b2e":{"kind":"REWORK","_number":11,"created":"2025-12-13 06:04:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/11"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-13 06:04:16.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e07d22066d328cfb70326978339e5e39bf18b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e07d22066d328cfb70326978339e5e39bf18b2e"}]},"branch":"refs/heads/master"},"42fa761bbd0fd84e409aa0578cf84130d7778c2e":{"kind":"REWORK","_number":12,"created":"2025-12-14 13:36:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/12"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-14 13:36:26.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42fa761bbd0fd84e409aa0578cf84130d7778c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42fa761bbd0fd84e409aa0578cf84130d7778c2e"}]},"branch":"refs/heads/master"},"23e0bd27955c34a278527b648cfd84527162d32a":{"kind":"REWORK","_number":13,"created":"2025-12-14 13:42:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/13"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-14 13:42:40.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23e0bd27955c34a278527b648cfd84527162d32a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23e0bd27955c34a278527b648cfd84527162d32a"}]},"branch":"refs/heads/master"},"f05a2fddcbcd1a7e344c851332e1236d5292be82":{"kind":"REWORK","_number":14,"created":"2025-12-15 07:14:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/14"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 07:14:07.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f05a2fddcbcd1a7e344c851332e1236d5292be82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f05a2fddcbcd1a7e344c851332e1236d5292be82"}]},"branch":"refs/heads/master"},"9bb4d7ac220a698046d7a9290f9206cf5b58c78e":{"kind":"REWORK","_number":15,"created":"2025-12-15 07:27:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/15"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 07:27:43.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bb4d7ac220a698046d7a9290f9206cf5b58c78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bb4d7ac220a698046d7a9290f9206cf5b58c78e"}]},"branch":"refs/heads/master"},"375857283cd2a708c72691559872588032f62394":{"kind":"REWORK","_number":16,"created":"2025-12-15 08:00:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/16"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 08:00:30.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/375857283cd2a708c72691559872588032f62394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/375857283cd2a708c72691559872588032f62394"}]},"branch":"refs/heads/master"},"81ac011f3b0e15c48af802285125897efe49d9c0":{"kind":"REWORK","_number":17,"created":"2025-12-15 08:05:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/17"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 08:05:41.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81ac011f3b0e15c48af802285125897efe49d9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81ac011f3b0e15c48af802285125897efe49d9c0"}]},"branch":"refs/heads/master"},"f021704b137e1338515a692a5dcc1b8a45ee4abf":{"kind":"REWORK","_number":18,"created":"2025-12-15 08:17:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/18"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 08:17:32.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f021704b137e1338515a692a5dcc1b8a45ee4abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f021704b137e1338515a692a5dcc1b8a45ee4abf"}]},"branch":"refs/heads/master"},"af79831989f876c8c3f0a3aeae671e5a428f20b1":{"kind":"REWORK","_number":19,"created":"2025-12-15 09:11:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/19"}}},"commit":{"parents":[{"commit":"7487ac609f09fa55a0b9a77c0ae55752fd102a0c","subject":"setup.cfg: Drop Python 3.8 and 3.9 according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7487ac609f09fa55a0b9a77c0ae55752fd102a0c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 09:11:55.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af79831989f876c8c3f0a3aeae671e5a428f20b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af79831989f876c8c3f0a3aeae671e5a428f20b1"}]},"branch":"refs/heads/master"},"e937a8a71d32516e60656247f17dbcfafc77f049":{"kind":"NO_CHANGE","_number":20,"created":"2025-12-15 09:24:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/20"}}},"commit":{"parents":[{"commit":"35f7b3c3426662877d4f8fc2e3c517697ffba6d9","subject":"setup.cfg: Update Python versions according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35f7b3c3426662877d4f8fc2e3c517697ffba6d9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 09:24:14.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e937a8a71d32516e60656247f17dbcfafc77f049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e937a8a71d32516e60656247f17dbcfafc77f049"}]},"branch":"refs/heads/master","description":"Rebase"},"b44f5df4f7dea7dcafc5ca9993110f057b9f0901":{"kind":"REWORK","_number":21,"created":"2025-12-15 09:51:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/21"}}},"commit":{"parents":[{"commit":"35f7b3c3426662877d4f8fc2e3c517697ffba6d9","subject":"setup.cfg: Update Python versions according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35f7b3c3426662877d4f8fc2e3c517697ffba6d9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 09:51:12.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b44f5df4f7dea7dcafc5ca9993110f057b9f0901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b44f5df4f7dea7dcafc5ca9993110f057b9f0901"}]},"branch":"refs/heads/master"},"160ae4ab83ca4504de0d5e83463e182a535b58de":{"kind":"REWORK","_number":22,"created":"2025-12-15 15:05:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/22"}}},"commit":{"parents":[{"commit":"35f7b3c3426662877d4f8fc2e3c517697ffba6d9","subject":"setup.cfg: Update Python versions according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35f7b3c3426662877d4f8fc2e3c517697ffba6d9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 15:05:52.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/160ae4ab83ca4504de0d5e83463e182a535b58de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/160ae4ab83ca4504de0d5e83463e182a535b58de"}]},"branch":"refs/heads/master"},"f8ebe0973a9ddc31591bf336aac3e6e709aded71":{"kind":"REWORK","_number":23,"created":"2025-12-15 16:15:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/23"}}},"commit":{"parents":[{"commit":"35f7b3c3426662877d4f8fc2e3c517697ffba6d9","subject":"setup.cfg: Update Python versions according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35f7b3c3426662877d4f8fc2e3c517697ffba6d9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 16:15:20.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8ebe0973a9ddc31591bf336aac3e6e709aded71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8ebe0973a9ddc31591bf336aac3e6e709aded71"}]},"branch":"refs/heads/master"},"3ec171a8cc4a5c84570210276d31279a96108106":{"kind":"REWORK","_number":24,"created":"2025-12-15 16:15:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/24"}}},"commit":{"parents":[{"commit":"35f7b3c3426662877d4f8fc2e3c517697ffba6d9","subject":"setup.cfg: Update Python versions according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35f7b3c3426662877d4f8fc2e3c517697ffba6d9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 16:15:38.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ec171a8cc4a5c84570210276d31279a96108106"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ec171a8cc4a5c84570210276d31279a96108106"}]},"branch":"refs/heads/master"},"09a907808417a5d77c35b9c7616fc0ce8875baf3":{"kind":"REWORK","_number":25,"created":"2025-12-15 16:16:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/25"}}},"commit":{"parents":[{"commit":"35f7b3c3426662877d4f8fc2e3c517697ffba6d9","subject":"setup.cfg: Update Python versions according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35f7b3c3426662877d4f8fc2e3c517697ffba6d9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 16:16:26.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09a907808417a5d77c35b9c7616fc0ce8875baf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09a907808417a5d77c35b9c7616fc0ce8875baf3"}]},"branch":"refs/heads/master"},"807ed9744f35cd7ae8d6795ae6692bcf98da185c":{"kind":"REWORK","_number":26,"created":"2025-12-15 20:51:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/26"}}},"commit":{"parents":[{"commit":"35f7b3c3426662877d4f8fc2e3c517697ffba6d9","subject":"setup.cfg: Update Python versions according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35f7b3c3426662877d4f8fc2e3c517697ffba6d9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-15 20:51:02.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/807ed9744f35cd7ae8d6795ae6692bcf98da185c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/807ed9744f35cd7ae8d6795ae6692bcf98da185c"}]},"branch":"refs/heads/master"},"a7ee5ccb2b551af5be80c0f082e503f6cbd54096":{"kind":"REWORK","_number":27,"created":"2025-12-16 06:05:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/970855/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/970855/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/970855/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/970855/27"}}},"commit":{"parents":[{"commit":"35f7b3c3426662877d4f8fc2e3c517697ffba6d9","subject":"setup.cfg: Update Python versions according to Flamingo PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35f7b3c3426662877d4f8fc2e3c517697ffba6d9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-12 15:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-16 06:05:28.000000000","tz":0},"subject":"debian: Switch to deb822","message":"debian: Switch to deb822\n\nChange-Id: I8156e2c2a57d820afdee0b76cd5a050ad682fb38\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ee5ccb2b551af5be80c0f082e503f6cbd54096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ee5ccb2b551af5be80c0f082e503f6cbd54096"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
