)]}'
{"id":"openstack%2Fkolla~971681","triplet_id":"openstack%2Fkolla~master~I3451940c90180ba91759c839987434e94251e32e","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-31 12:07:07.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-31 12:07:07.000000000","reason":"Change was submitted"},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-03-31 12:07:07.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-03-31 10:00:26.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I3451940c90180ba91759c839987434e94251e32e","subject":"Add OCI standard labels to container images","status":"MERGED","created":"2025-12-23 09:33:11.000000000","updated":"2026-03-31 12:08:08.000000000","submitted":"2026-03-31 12:07:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971681","meta_rev_id":"10d06040e177e5d50715d8590d7ebfb5d28e16d8","_number":971681,"virtual_id_number":971681,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-31 12:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-31 10:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-31 10:00:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-03-31 10:00:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-03-30 18:47:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-23 10:55:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-23 10:56:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-14 06:29:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-01-14 10:10:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-30 15:26:52.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"1029ee2a1fd634f1bddeb5c23379ac70f3f59095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-12-23 09:33:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7b82ecf07484478083d9b632371b5edcc8cd2eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 10:55:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63b09d596d134438a13507b532531850\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8d28ffa1996a43d58ff6b28605c0f9fa : NODE_FAILURE Node(set) request 1a694be770e64706b53831ff9d51f6bf failed in 0s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/38063a00642f4df488284ce7092f44fd : SUCCESS in 1h 19m 53s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/18555b47ac0c4f978d2e678758856b05 : SUCCESS in 1h 20m 16s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f43a30dd49ff4204b0bbc2f425e5bec8 : NODE_FAILURE Node(set) request 51b7e7d944b041e6a2c45134f6ebfaba failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d15d53a313e74d5d9e720502f8494665 : NODE_FAILURE Node(set) request 25bf0520cda64e11bbf12c08cfba6a2c failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e97972e35af42eba8e26949b995b206 : NODE_FAILURE Node(set) request eb9337e64e4e4aebbfe91fa6e068e72f failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b64a48622ecb4e87a93703198330190d : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"559ee84ecaa959f443287374d9174b333a37a5dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 10:56:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af2edf6d90054df0af3a2f017739687d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3386b4f1b2c6475bb31a34a3748e2014 : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/665f6abc7ba5467faee5b8790559786c : SUCCESS in 1h 04m 16s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/0ce6ab5b78be464b9ee3adba17b150b5 : SUCCESS in 25m 56s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ad708afc9db54577a2ac51ef2dbd4065 : SUCCESS in 1h 15m 11s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5751edd196d94030a3d524bea8b60356 : SUCCESS in 49m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8da34a454e594c05af9679cf6bfac52c : SUCCESS in 1h 14m 34s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7722a8a857bf4a589fcbd19642bfa213 : SUCCESS in 14m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8db2573e422f4f6f9b16ad21ee67ab02 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4039ec2015f34c5aa282415cfc56b078 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d312ab53c67145cfb94f7f3cd09cbdbd : SUCCESS in 2m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a17a62dab60741b0a333094655752626 : SUCCESS in 7m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5319b196e96b4015924dde276d91c997 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc3ae3d037548c1a066757b6188479b : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21fc6dbdbb2d4ee5b143306e5564e237 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cb6bb9afa86e77d5ce1f1514c6e102486e0cb97b","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-12-23 11:05:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f27837536173de885d1860fa5ccba28da65e9266","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-12-23 18:20:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0b496172d8e392e624b2991a231cbf5faf36c91","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-14 06:29:08.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b7c2f597b932e492da96c2649a9008882c4fc74d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-14 10:10:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"399bb130d6d74e2a187fabb0597d5926b5d2b739","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 05:43:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6351a3e4da481eb7f1961f929904e03e6b702ae9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-25 08:03:31.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e7734d14371aabc8396eca2e15232f59230d5a8e","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-25 08:03:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2154ffb2e6b60e823a76c7f0729fb0d752df7db7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-25 08:09:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f67a072c0f24c40f71a724f16e9b4fe8e937bcb7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 08:12:14.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"67e2c76d790beda30d7b2d3d4e87b7b0b213f786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 09:40:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff08b9786cad492eb5259632f5013f14\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/83109804cc4746de85f3d2956b5afcd0 : SUCCESS in 1h 12m 28s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7418edf8b42a4e59bcb4988f4d0f7d6f : SUCCESS in 1h 15m 06s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c3100b20b6d84f8091f4275660c241b1 : SUCCESS in 28m 59s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/115d988c746e4e73b9d5c53b3c47d3d2 : SUCCESS in 1h 10m 23s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/741bbd1df9c5432cb96a47b0e2579114 : SUCCESS in 35m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/222a54009a3541438c28c7acc8d5cdbe : SUCCESS in 1h 30m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/08fca6a4b6974cb684013653aa353a51 : SUCCESS in 26m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3c66a7550d12425ebd269ba7e9b38f4a : SUCCESS in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a0173967b54320bb451827d96e2ace : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/306fa572307b47769c537f811109c13f : SUCCESS in 2m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d5c0cbb2f3a43a6808b2d14f5cc8b50 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96b0cdb2b7654fe080706a488958cff1 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3f89dbb530248999d8adc0ac349a600 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2af1c9aa3cf42b5b1766178d768aae6 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cf0c93ffae744838e5976ae9e3c67a6 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3a5bcb080effda3c6748108e913abd2f8b9c1e29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:48:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a235240bd44e7eba9c496759f96ebb\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/3534b117242f44d8bba7274741fe5156 : TIMED_OUT in 3h 05m 02s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/db6803ec89044918b9546651cf620531 : SUCCESS in 2h 18m 18s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/da34cba47b8a4c5da64ed4ed9256aa08 : TIMED_OUT in 3h 05m 54s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/a817bdb58d014e91900aad8d844ced28 : SUCCESS in 2h 18m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94f16d04f84542bcb30c731f873c6f3b : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b7f2359f78b4b6087b62c1b3b7b1ef8 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/14d54481574c4ee19b11d28c1042a773 : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bcab08cb187f8f50daffb69bc2a7694a514768c6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-30 15:26:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a913a4502cb13d610273ff38e7cb3f804e6eea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-30 15:41:55.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f5066437e37b5e972c1d48057843f4ef6257a026","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-30 15:44:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47affb2c8fae05c5fd0938dbbf8216527902ec45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:13:49.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f07aca9976ad4fce9dc61be47d5b2fff\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/bbece33fd27b439593f5c1c8dcf4b4dd : SUCCESS in 1h 06m 25s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/fe0e653121d14d9d855192960516f5d3 : SUCCESS in 1h 18m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a98331ed218c4d82a93332ca53f47ac7 : SUCCESS in 27m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9be8c1d5cd4d4d50a77d15c0e04c5704 : SUCCESS in 1h 30m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a5eacbc69cde463799f0b08d634e7771 : SUCCESS in 1h 06m 19s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/148abd8a144e4149b2f56cae07fd8fa6 : SUCCESS in 1h 25m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1e639401a7974e82a02eae777a773e3a : SUCCESS in 31m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f13ca1d56d5247ee9127c1d5922742f6 : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fba19112f2e41c8b99f3be12b03d32d : FAILURE in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/873fc9c446334415b063bf2da0c7f40b : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a641b44e7f34444db9ebb4df559820d7 : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/748cc575a5494b0da4bb66eb6d862397 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6f3888100a40dab2917bc2a12188ec : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb77f69f76f4f38bd1363acd6a5d681 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a8ed82a92d04ac2ad65c4e185d00e31 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"3a4f88187c0165654ccbbfa20909d6e4165e334f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-30 17:30:21.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a6d0bdc4f6ecbaf3437d42d2487424a0d7afa60d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 18:43:16.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f8ba23392df4cea9f2a5248b79ba0a3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ea1cca9ec6bc4f5d85625b04d92a167b : SUCCESS in 1h 08m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a08d11ed136e483ca7471789e679f667 : SUCCESS in 58m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ee16efe900c74161bcc77f36c9834414 : SUCCESS in 24m 01s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/413d4043795d4ff48226c88764223e88 : SUCCESS in 1h 06m 28s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/01abba7300b14d178298aca675eb7f3c : SUCCESS in 38m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfffa495c084408fa73ec8e77365a1ab : SUCCESS in 1h 11m 18s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/42d2d6c1815346f4b1be2274a591acbf : SUCCESS in 14m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ca7a7b7b27174157b265ee7645153dd9 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660fd6c9600241b5bddc12d78324b60f : FAILURE in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2a590092ba24338a0f2fa2014a40d86 : SUCCESS in 2m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/166a820fc234495fbb490d788d311ce0 : SUCCESS in 2m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7b7d88ad6ed9402098a922f17051d4b6 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef29e0a2430441c688412d23f77bf999 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82b343dce68483ca014026831f492b0 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b5e8100c3e24c99b65a3bce66f81aad : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":4},{"id":"cfd4aa992313740ae75df013ebe5c8dd58b1b784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-30 18:47:32.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c7d0fb5ec30748bc6cd3049e7e134b8b8e61b118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 20:11:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69bbb1fbb7a34d568b07da0ac9a6d311\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c92b8409577246f9aab681c53d8e435f : SUCCESS in 1h 15m 26s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b52a2fda42034f27baaee6dd3d0fc54a : SUCCESS in 1h 03m 54s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/26d0dd82183e430a8d86f92592f3b6cf : SUCCESS in 15m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c05c22d1af1641488d2b91598eaf2683 : SUCCESS in 1h 13m 38s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/512f529052d647a4b7774562912ac89d : SUCCESS in 35m 23s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4586419f355044f9ba37fce9e39f65a2 : SUCCESS in 1h 19m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e48ad44e0df348a5bf60c37da9537714 : SUCCESS in 23m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/25f0059eb0954134b3e6cde0952df4a0 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953df7b77c3344d28a5d943bfd5043ad : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51ab7ada35924d5298a2cde7f8c6e123 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2923164576e94b13b3457f1db3304440 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73c36efeb44747aba5fe96aaba87fcbc : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/badf3e9ce5734d97bd66c8f3b87d4ea9 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c37554029b458eac7b8de1e1f063fa : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e92638db58754dab992ac0c1ccd0fabe : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":5},{"id":"e6b4e898d5aa5807baef5b052ab2054c5cbf0373","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 21:55:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de5fb91f4c7944d8953ac09bd8857822\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/bfa1ce397e57466a93b03e2b3bcf2091 : FAILURE in 7m 35s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/be8e4678616f4650878d52534602e2ec : SUCCESS in 1h 25m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/511b58b242c64a16ad92898299adb709 : FAILURE in 6m 44s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1b0e538f007342629eb675daefc7f53b : FAILURE in 7m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f5e75d94cddf4c2893de3140d1bc07ae : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d574b2dc41394082b29077fb98e4bd02 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d55e1bbaa1fa4ae48b9f3be080cebd4d : SUCCESS in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7040aaf1976f15f2823af3673f14568e240627e8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-31 10:00:26.000000000","message":"Patch Set 5: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1a0990e8d7e8b303aa98251fb7f68ea3472109e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:00:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"83db1b853cf729ae693a55b31dccc8a73a76475e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:07:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95958f17eb62456abc69de2835f22610\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ccc3618139cf4e1ea541e5618df7faf3 : SUCCESS in 1h 09m 04s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f9e61553747d49d681b4c43687522582 : SUCCESS in 26m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/6ffdddde854349238610d58f0c336239 : SUCCESS in 1h 07m 34s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b69bba1813524156a401e709dc694feb : SUCCESS in 42m 55s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03eeaacc04904e51abce743b12a94a87 : SUCCESS in 2h 01m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4a3d2d01c60f4bf78ba9853cfd6e5aa3 : SUCCESS in 22m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7c57a90aa65b4b7abf993d42857da919 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a1264386514d83ba545d16006d300b : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c24a9ac659d944b28e61e4f959ad232b : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fdeac776424436796583703d2845c37 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd494ce84714b618792143189e70f1c : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f843bdcdce59483e9984e3397081fd2d : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":5},{"id":"fc094bb3a85d22116a591b351f60efb0a03d689d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:07:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"10d06040e177e5d50715d8590d7ebfb5d28e16d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:08:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d807ac1170048188c8fc7f98833853a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ca0f88117a4e8288e50baad0467929 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9aced35e60444158fa7f9ab9d6cc84c : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cbca935b87b4606534fc08d2cae80d864af51c7c","revisions":{"b2b9fde0c2dc5cbe74aedfa70ecaccf254e4a10d":{"kind":"REWORK","_number":1,"created":"2025-12-23 09:33:11.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/81/971681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/971681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/971681/1"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-12-23 09:30:02.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-12-23 09:33:01.000000000","tz":60},"subject":"Add OCI standard labels to container images","message":"Add OCI standard labels to container images\n\nCurrently, Kolla images lack standard metadata, making it difficult\nto identify their build source, creation time, or exact version solely\nby inspecting the image artifact.\n\nThis patch introduces the addition of Open Container Initiative (OCI)\nstandard labels to all generated images. The build process now detects\nmetadata (utilizing `direct_url.json` for pip-installed packages to\navoid subprocess calls) and injects the following labels into the\ncontainer runtime config:\n\n* org.opencontainers.image.authors\n* org.opencontainers.image.created\n* org.opencontainers.image.description\n* org.opencontainers.image.licenses\n* org.opencontainers.image.revision\n* org.opencontainers.image.source\n* org.opencontainers.image.title\n* org.opencontainers.image.url\n* org.opencontainers.image.version\n\nExample output:\n    \"Labels\": {\n        \"org.opencontainers.image.authors\": \"Kolla Project (https://launchpad.net/kolla)\",\n        \"org.opencontainers.image.created\": \"2025-12-22T14:02:47.015819\",\n        \"org.opencontainers.image.description\": \"OpenStack Kolla keystone-ssh image\",\n        \"org.opencontainers.image.licenses\": \"Apache-2.0\",\n        \"org.opencontainers.image.revision\": \"96b54022857ccef8dc6bb11789abe32119827109\",\n        \"org.opencontainers.image.source\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.title\": \"keystone-ssh\",\n        \"org.opencontainers.image.url\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.version\": \"master-ubuntu-noble\"\n    }\n\nThis enhancement improves the interoperability of Kolla images with\nstandard container tools and registries that display OCI metadata.\n\nChange-Id: I3451940c90180ba91759c839987434e94251e32e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2b9fde0c2dc5cbe74aedfa70ecaccf254e4a10d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2b9fde0c2dc5cbe74aedfa70ecaccf254e4a10d"}]},"branch":"refs/heads/master"},"9ebf07bc3a445a01a3aab0849df2843ee516757f":{"kind":"REWORK","_number":2,"created":"2026-03-25 08:03:31.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/81/971681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/971681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/971681/2"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-12-23 09:30:02.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-25 08:03:28.000000000","tz":60},"subject":"Add OCI standard labels to container images","message":"Add OCI standard labels to container images\n\nCurrently, Kolla images lack standard metadata, making it difficult\nto identify their build source, creation time, or exact version solely\nby inspecting the image artifact.\n\nThis patch introduces the addition of Open Container Initiative (OCI)\nstandard labels to all generated images. The build process now detects\nmetadata (utilizing `direct_url.json` for pip-installed packages to\navoid subprocess calls) and injects the following labels into the\ncontainer runtime config:\n\n* org.opencontainers.image.authors\n* org.opencontainers.image.created\n* org.opencontainers.image.description\n* org.opencontainers.image.licenses\n* org.opencontainers.image.revision\n* org.opencontainers.image.source\n* org.opencontainers.image.title\n* org.opencontainers.image.url\n* org.opencontainers.image.version\n\nExample output:\n    \"Labels\": {\n        \"org.opencontainers.image.authors\": \"Kolla Project (https://launchpad.net/kolla)\",\n        \"org.opencontainers.image.created\": \"2025-12-22T14:02:47.015819\",\n        \"org.opencontainers.image.description\": \"OpenStack Kolla keystone-ssh image\",\n        \"org.opencontainers.image.licenses\": \"Apache-2.0\",\n        \"org.opencontainers.image.revision\": \"96b54022857ccef8dc6bb11789abe32119827109\",\n        \"org.opencontainers.image.source\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.title\": \"keystone-ssh\",\n        \"org.opencontainers.image.url\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.version\": \"master-ubuntu-noble\"\n    }\n\nThis enhancement improves the interoperability of Kolla images with\nstandard container tools and registries that display OCI metadata.\n\nChange-Id: I3451940c90180ba91759c839987434e94251e32e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ebf07bc3a445a01a3aab0849df2843ee516757f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ebf07bc3a445a01a3aab0849df2843ee516757f"}]},"branch":"refs/heads/master"},"35701c29eb6c1c927048a8a191d80aa4eb4f04be":{"kind":"REWORK","_number":3,"created":"2026-03-30 15:41:55.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/81/971681/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/971681/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/971681/3"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-12-23 09:30:02.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-30 15:41:52.000000000","tz":120},"subject":"Add OCI standard labels to container images","message":"Add OCI standard labels to container images\n\nCurrently, Kolla images lack standard metadata, making it difficult\nto identify their build source, creation time, or exact version solely\nby inspecting the image artifact.\n\nThis patch introduces the addition of Open Container Initiative (OCI)\nstandard labels to all generated images. The build process now detects\nmetadata (utilizing `direct_url.json` for pip-installed packages to\navoid subprocess calls) and injects the following labels into the\ncontainer runtime config:\n\n* org.opencontainers.image.authors\n* org.opencontainers.image.created\n* org.opencontainers.image.description\n* org.opencontainers.image.licenses\n* org.opencontainers.image.revision\n* org.opencontainers.image.source\n* org.opencontainers.image.title\n* org.opencontainers.image.url\n* org.opencontainers.image.version\n\nExample output:\n    \"Labels\": {\n        \"org.opencontainers.image.authors\": \"Kolla Project (https://launchpad.net/kolla)\",\n        \"org.opencontainers.image.created\": \"2025-12-22T14:02:47.015819\",\n        \"org.opencontainers.image.description\": \"OpenStack Kolla keystone-ssh image\",\n        \"org.opencontainers.image.licenses\": \"Apache-2.0\",\n        \"org.opencontainers.image.revision\": \"96b54022857ccef8dc6bb11789abe32119827109\",\n        \"org.opencontainers.image.source\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.title\": \"keystone-ssh\",\n        \"org.opencontainers.image.url\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.version\": \"master-ubuntu-noble\"\n    }\n\nThis enhancement improves the interoperability of Kolla images with\nstandard container tools and registries that display OCI metadata.\n\nChange-Id: I3451940c90180ba91759c839987434e94251e32e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35701c29eb6c1c927048a8a191d80aa4eb4f04be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35701c29eb6c1c927048a8a191d80aa4eb4f04be"}]},"branch":"refs/heads/master"},"b5f0c139cff25495bcc6aa054e118c2e63345c11":{"kind":"REWORK","_number":4,"created":"2026-03-30 17:30:21.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/81/971681/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/971681/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/971681/4"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-12-23 09:30:02.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-30 17:30:17.000000000","tz":120},"subject":"Add OCI standard labels to container images","message":"Add OCI standard labels to container images\n\nCurrently, Kolla images lack standard metadata, making it difficult\nto identify their build source, creation time, or exact version solely\nby inspecting the image artifact.\n\nThis patch introduces the addition of Open Container Initiative (OCI)\nstandard labels to all generated images. The build process now detects\nmetadata (utilizing `direct_url.json` for pip-installed packages to\navoid subprocess calls) and injects the following labels into the\ncontainer runtime config:\n\n* org.opencontainers.image.authors\n* org.opencontainers.image.created\n* org.opencontainers.image.description\n* org.opencontainers.image.licenses\n* org.opencontainers.image.revision\n* org.opencontainers.image.source\n* org.opencontainers.image.title\n* org.opencontainers.image.url\n* org.opencontainers.image.version\n\nExample output:\n    \"Labels\": {\n        \"org.opencontainers.image.authors\": \"Kolla Project (https://launchpad.net/kolla)\",\n        \"org.opencontainers.image.created\": \"2025-12-22T14:02:47.015819\",\n        \"org.opencontainers.image.description\": \"OpenStack Kolla keystone-ssh image\",\n        \"org.opencontainers.image.licenses\": \"Apache-2.0\",\n        \"org.opencontainers.image.revision\": \"96b54022857ccef8dc6bb11789abe32119827109\",\n        \"org.opencontainers.image.source\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.title\": \"keystone-ssh\",\n        \"org.opencontainers.image.url\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.version\": \"master-ubuntu-noble\"\n    }\n\nThis enhancement improves the interoperability of Kolla images with\nstandard container tools and registries that display OCI metadata.\n\nChange-Id: I3451940c90180ba91759c839987434e94251e32e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5f0c139cff25495bcc6aa054e118c2e63345c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5f0c139cff25495bcc6aa054e118c2e63345c11"}]},"branch":"refs/heads/master"},"cbca935b87b4606534fc08d2cae80d864af51c7c":{"kind":"REWORK","_number":5,"created":"2026-03-30 18:47:32.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/81/971681/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/971681/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/971681/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/971681/5"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-12-23 09:30:02.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-30 18:47:21.000000000","tz":120},"subject":"Add OCI standard labels to container images","message":"Add OCI standard labels to container images\n\nCurrently, Kolla images lack standard metadata, making it difficult\nto identify their build source, creation time, or exact version solely\nby inspecting the image artifact.\n\nThis patch introduces the addition of Open Container Initiative (OCI)\nstandard labels to all generated images. The build process now detects\nmetadata (utilizing `direct_url.json` for pip-installed packages to\navoid subprocess calls) and injects the following labels into the\ncontainer runtime config:\n\n* org.opencontainers.image.authors\n* org.opencontainers.image.created\n* org.opencontainers.image.description\n* org.opencontainers.image.licenses\n* org.opencontainers.image.revision\n* org.opencontainers.image.source\n* org.opencontainers.image.title\n* org.opencontainers.image.url\n* org.opencontainers.image.version\n\nExample output:\n    \"Labels\": {\n        \"org.opencontainers.image.authors\": \"Kolla Project (https://launchpad.net/kolla)\",\n        \"org.opencontainers.image.created\": \"2025-12-22T14:02:47.015819\",\n        \"org.opencontainers.image.description\": \"OpenStack Kolla keystone-ssh image\",\n        \"org.opencontainers.image.licenses\": \"Apache-2.0\",\n        \"org.opencontainers.image.revision\": \"96b54022857ccef8dc6bb11789abe32119827109\",\n        \"org.opencontainers.image.source\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.title\": \"keystone-ssh\",\n        \"org.opencontainers.image.url\": \"https://opendev.org/openstack/kolla\",\n        \"org.opencontainers.image.version\": \"master-ubuntu-noble\"\n    }\n\nThis enhancement improves the interoperability of Kolla images with\nstandard container tools and registries that display OCI metadata.\n\nChange-Id: I3451940c90180ba91759c839987434e94251e32e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbca935b87b4606534fc08d2cae80d864af51c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbca935b87b4606534fc08d2cae80d864af51c7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
