)]}'
{"id":"openstack%2Fkolla~971693","triplet_id":"openstack%2Fkolla~master~Ie72a6e16f0f732007b175b9dc866489fa5505dd3","project":"openstack/kolla","branch":"master","topic":"kolla-ansible-2-20","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-01-08 12:38:17.000000000","reason":"Change was marked work in progress"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-01-08 12:38:17.000000000","reason":"Change was marked work in progress"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-08 12:35:58.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Ie72a6e16f0f732007b175b9dc866489fa5505dd3","subject":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"","status":"ABANDONED","created":"2025-12-23 10:28:51.000000000","updated":"2026-01-09 17:41:48.000000000","total_comment_count":2,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"revert_of":971614,"meta_rev_id":"f370d59d69780879b4d861e14967a633db62aa00","_number":971693,"virtual_id_number":971693,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-12-23 10:28:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-12-23 10:28:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-23 10:28:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d54b64b72bc2dfa914dd1b8dcd0b4a78e292477e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-23 10:28:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3867afcaf8ffe8d31ee0bcd9fd4f2708d107631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 11:28:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23c2e9ccbecd43c3aeea878f44c40d2c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2207e7c90bb543f8a6fb697dd8a550ae : SUCCESS in 46m 42s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6befc556766840199e65062cfc090e81 : SUCCESS in 52m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7fc91e222d724bc0b53336c19498ef40 : SUCCESS in 28m 50s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/32e2b60c24ad41bfa498dac13b19e3f6 : SUCCESS in 40m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b4adeea2eeb547d5ba7fae965f833042 : SUCCESS in 58m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cea0b45acef14a9696d23d208878ab7d : SUCCESS in 48m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4d09fe841c0b46b2a007ade5614a47d6 : SUCCESS in 28m 52s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/93f159d5162f4734b6733703b2a07e4f : FAILURE in 13m 39s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8e907bd14a394a659ba0bce78cd5bf8e : FAILURE in 22m 56s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d2780fd452834fdbabe120951c26b64e : FAILURE in 48m 25s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/836694a38ab1448189bf568cc9b69369 : FAILURE in 24m 46s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d091ae2d17404a84ac09b3a8d6a0bba6 : FAILURE in 53m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bcb9a6539374dacadd862e53ff475f6 : FAILURE in 24m 05s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6643a18a2e8a4e1c836da55c86111632 : FAILURE in 51m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/891239024c7d47f589cd4879d5adbe7a : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a674dd0f50cd4238833b30476b8a43d1 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44b776551eaf40a6a1d93cf2089fb92b : SUCCESS in 1m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6850afae68024ae1940eb08903809439 : SUCCESS in 5m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8412f7d5f504bf1bf6cedd04b012a95 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ae0e2a242d422886329bcb04518651 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f011ec831f54470a953e29fc6c41714b : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c2bf5e590d8452b29a5fecc9e572f165f9477bcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 12:11:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f70ce741719944dfae12b7271a300b18\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/6592b62a370645d3ab1047eae2b8bfcd : SUCCESS in 1h 40m 24s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/562da0fabaaa4674b76c04b0dd067ac0 : SUCCESS in 1h 01m 51s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/34fde3dab07c47bfaf99fea285431620 : SUCCESS in 52m 33s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b54a00784c7146cb99857ff68228f7a8 : SUCCESS in 55m 58s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7c5e41975efe4deabe110bfd96ad9781 : FAILURE in 40m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35be80415b9b4a31a4b7d22c46e79b6b : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d58ad01fe5fc442a89dbfbfc44d6c85c : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/82ab8f4e07dc436583a1ceab97bb74b9 : SUCCESS in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd2f8651af6239c6f77c053250d5f4953b60a3e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-30 06:23:49.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e578c84ccc21479bb33b8551702fb46564a763c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 06:25:56.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/971606 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"2ee3afc970301702386ada480c983416e3de8085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 06:26:08.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/971606 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"3d4f7ef86059d5034793fcce0ab0311dceaa4085","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-30 06:26:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9189d41849fd907843c0fe049a9acf4ee3aa9b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 07:36:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/632600cb43da46a5889715933d545ac9\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/79fe2ebf9d234509a416afa22a543d9d : SUCCESS in 38m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b61c229010b84e09bda8fc087e55c197 : SUCCESS in 47m 40s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/cc5ae165f94d45b7a267c406dbd5c03a : SUCCESS in 26m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/06426fb52b684bd5bd159b57b9dfc938 : SUCCESS in 43m 34s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f2d6cae5ecc6417f9ac711b2512b5c2b : SUCCESS in 1h 04m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84d565abe0674d0094fd2cb38c82c4db : SUCCESS in 49m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c518a8440f814b95951dabfc054b8146 : SUCCESS in 28m 20s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2fd9f1d388a64df8b02d6909cb82a419 : FAILURE in 21m 19s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ce95b16f99c840cdbda3494ea7dc1266 : FAILURE in 29m 25s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/8312f6075f8d40a590f052e1bbfe5cd9 : FAILURE in 47m 12s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c1521b09c6c04c159592da800df0819b : FAILURE in 31m 27s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1655b79f69994b219d398b3bf7b531cb : FAILURE in 1h 03m 39s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06e6c1f868634977849baffec400b7ec : FAILURE in 30m 38s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1683d3f6ebc143daaf6b4c98a57656fe : FAILURE in 1h 02m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8bf20c8157254dc88013cd338a9c3c0e : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e466af31f0764bff9cd2ec221c9a72f0 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98c6cf8214534289b78e46e77a43bde6 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff6560f5073b4dc49fb956ac023639ea : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05547b74f15744ba9ca6df9db9195664 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7693365f03da4efbb91e657b331d65e2 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11507b944f57490798d1db3bebc78559 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"58804dcbee3ecc15f1b9035f17733a3b139d0aa0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 07:54:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8c75ff2af646e595e60305d2745cba\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/55140e4321f04476b1c2f1fa64bde3dc : NODE_FAILURE Node(set) request 48b03ee074224c859321752dc730ea90 failed in 0s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6e36aca2e772480d830d0101a5a4debe : NODE_FAILURE Node(set) request f72749a149da4697b47262855bce15bf failed in 0s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1191a672fa764e40bfecba14a66ceab2 : SUCCESS in 1h 27m 46s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/bd1387206944401ea24e107b9f6549d4 : NODE_FAILURE Node(set) request 44ccf31c259d4d90b003813e9e5a1c23 failed in 0s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/c9495e7de3c04a959c143433699604bb : NODE_FAILURE Node(set) request c7276dde4bb64115925c5a05e62a1266 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/576e70136378430bb82b4e09de5467cd : NODE_FAILURE Node(set) request db9a152bf0fd42729886e7aae65577fd failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5771e15219bf435ca9ccac07d614b3e1 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7748b20702d54e5688629b0f154379d4 : NODE_FAILURE Node(set) request 4118964928304eb7b5fff97770f2a3c6 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c4006e8fdb8ec05db45dffbbd6f31f1467804f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-08 12:01:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5caeffb1b42b0e5752165800d47ddf365f3b5015","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-08 12:35:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c17cfdc70535d1326e9b9b0a8fe8f5ac56b0517d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-08 12:38:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"b80bbfd0f171366325410879918442081fd82908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:07:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dce190277a54151806128eff746b445\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/14b1d8da25434a5bafbccd8487cc24fa : SUCCESS in 55m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/52b64e775c894b859f4f8b3b5b44fe8b : FAILURE in 52m 58s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1b5d1c8b47f34c29b09112223d0018ba : FAILURE in 26m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/984fc528d5494c46acef5914c1d67154 : SUCCESS in 36m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0eb7e7aedc1b447f8329e1f95474a10f : SUCCESS in 35m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78834e66360b4a8780942b6b7e726e67 : FAILURE in 32m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2178b2056d2947a6b21cc92562eccdaa : FAILURE in 14m 02s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c7760da1de0345fdbc7f92b8ea711bc3 : FAILURE in 14m 46s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/d2855de270154c77adf97e1da1be49e7 : FAILURE in 5m 59s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/c42f08ecc2de41edbb9c1629a487104b : FAILURE in 6m 38s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c6320dbdb69b400d9c42e86dfbf69de8 : FAILURE in 23m 06s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/327411a2cae24425b465b28443e13081 : FAILURE in 6m 09s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e633ec91af224ba28d041ec2457c20cb : FAILURE in 5m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/29585abc442147b59df46ab9d5eb68b4 : FAILURE in 5m 44s\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/b58e0ad8d05e4849aa76a8a0f2d46570 : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/afc577afd8d446ac83f9a41694f17cc7 : SUCCESS in 28m 13s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/04e9aca89aa64382a6d0d67e4349be4a : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb9822d55d444568a4c2c4ca9650b64 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/714d5ba278624518a7cd190ac004cc0f : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afd7f65281f54c5f8015ebde2807d157 : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7271b638d95f491bb75d9900a026af7f : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c1f6d16bb2470b81012ad6a47d3a7f : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/385c35568d3a484fbcd6b77349e1b801 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"a7bc080fd0aaeaa8bfc7f74e027587f22358d837","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:37:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3354739d89c483f9526541ed2921434\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7fc1f4837d734ea3a411c52c1a5154e0 : SUCCESS in 52m 20s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d1b44140bc5f4f06818ccef9873de5bb : FAILURE in 1h 09m 56s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1d219acab59c4683a7b038b6b699d64f : SUCCESS in 1h 32m 42s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/931dcbfe4ee644da87338f7475f3f095 : FAILURE in 55m 50s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ecb057e54bc44a5ea0376338ffa01d34 : FAILURE in 45m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d587b3db32044196b3ee788ffa0ba3eb : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9b94816c0ab94a998b6f5a19ee143053 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/abceae97a8114f0181c1df4c4695df9a : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c45c08721a62e7981d721ab29230f8839068f0d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-08 14:47:40.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d0a66258c8207aa28637c18f5209e146e44f79c2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-08 15:32:55.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f3e272d2783a690fe0a15f95140f114e0c6c3b4d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-08 15:39:04.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"d357cee1019b0f9a843b6618a60025109ec513df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 16:44:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df6ac0ac42e8451cbd238da431bbe41f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e58fe8954ae549b680297073465186fe : SUCCESS in 46m 51s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8a57a3d535204312a5e51576964c7119 : FAILURE in 38m 55s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/05bc0bd58ea34087b2668220b1bf4741 : FAILURE in 24m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9bb397026d5043f9b4dec26f5b3bfe91 : SUCCESS in 36m 38s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e2ea6b749a9e4386bf3cfac63ae9656e : SUCCESS in 45m 20s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7289f0c05a8465d9d5d9233544a8021 : FAILURE in 42m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/596dd7dd45ec46449bdfeefdb56a5650 : POST_FAILURE in 29m 52s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/32f6766bb54648fb8d8aa510ac8fc4f5 : FAILURE in 24m 37s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/eff1282d4cdf44c6b3a428beb982637b : FAILURE in 18m 32s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/23dce4ec0cdf4236b0f25f7bcdc14f64 : FAILURE in 50m 24s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/876449466854449cbcb40a7e55cbce28 : POST_FAILURE in 21m 54s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7393ea0cb16c4555b9ad3e5e59eda32f : FAILURE in 51m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cef97cca46dc453eaa4aad50fb5a9d88 : POST_FAILURE in 20m 00s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/91d50d4fa98744b6951c65765711ff1b : FAILURE in 51m 00s\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/b44163f689f8442296440af3ddc11276 : POST_FAILURE in 13m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/6cb320ae26f14c579ab7edefab09584b : POST_FAILURE in 28m 14s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f4284fe5b2d74ba48dfeb7edec1e8125 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83cb2aa4fc8e4d74a99394925f85508f : POST_FAILURE in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb1faaff30b041caa14eb764c1851031 : POST_FAILURE in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abea92184da74e33b4a3e578b18c737d : POST_FAILURE in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4c666e2e7f4422b979463f0c2433e8d : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f692a196e0304223b82b0a673cb9d721 : POST_FAILURE in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4c77c9a28504c1c996d093f3f818e99 : POST_FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":6},{"id":"b5bcfca3debf0350a2a2986240b641ee573d85a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:08:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da51b0c95f2c4016b6302c8a650617b9\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c1cd1d324dba4378abd075e65ca3546b : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a73a29f52cfe47e1808289390abd7dab : FAILURE in 1h 17m 12s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/924a00285672470d85d6baf111bc166d : SUCCESS in 56m 44s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9b0d5e07d3a049a8a0f54039c7a2972f : FAILURE in 1h 01m 53s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a5107f535309405591594920073db1e1 : FAILURE in 34m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5efb3d7be144440d8187050a86ea0d9b : POST_FAILURE in 5m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/033d18f06f3e410c9ce95f585224a818 : POST_FAILURE in 5m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/92e8b44ca1184d689b02abcf14424b1b : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b3cc5331f0c2613854774137e8d73018fa755336","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-09 17:39:32.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f370d59d69780879b4d861e14967a633db62aa00","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-09 17:41:48.000000000","message":"Abandoned\n\ncontinued in https://review.opendev.org/c/openstack/kolla/+/972944","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f1cbec7682ab34eacceafcb243f1956b5b74192e","revisions":{"2fd206942359bc2cdf3340142bf5a63be361dac3":{"kind":"REWORK","_number":1,"created":"2025-12-23 10:28:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/971693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/971693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/971693/1"}}},"commit":{"parents":[{"commit":"8aa631d40c0960a257382c50c16dfb45b07d19c1","subject":"Revert \"kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aa631d40c0960a257382c50c16dfb45b07d19c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-23 10:28:51.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-12-23 10:28:51.000000000","tz":0},"subject":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"","message":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"\n\nThis reverts commit 8aa631d40c0960a257382c50c16dfb45b07d19c1.\n\nReason for revert: Take 2\n\nChange-Id: Ie72a6e16f0f732007b175b9dc866489fa5505dd3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fd206942359bc2cdf3340142bf5a63be361dac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fd206942359bc2cdf3340142bf5a63be361dac3"}]},"branch":"refs/heads/master"},"d57c799f83a3c2517175741c60e5747102508dcc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-30 06:23:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/971693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/971693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/971693/2"}}},"commit":{"parents":[{"commit":"8aa631d40c0960a257382c50c16dfb45b07d19c1","subject":"Revert \"kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aa631d40c0960a257382c50c16dfb45b07d19c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-23 10:28:51.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-12-30 06:23:49.000000000","tz":0},"subject":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"","message":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"\n\nThis reverts commit 8aa631d40c0960a257382c50c16dfb45b07d19c1.\n\nReason for revert: Take 2\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/971606\n\nChange-Id: Ie72a6e16f0f732007b175b9dc866489fa5505dd3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d57c799f83a3c2517175741c60e5747102508dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d57c799f83a3c2517175741c60e5747102508dcc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"69ae7b261320b448d93d216dada12633f5362a34":{"kind":"REWORK","_number":3,"created":"2026-01-08 12:01:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/971693/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/971693/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/971693/3"}}},"commit":{"parents":[{"commit":"8aa631d40c0960a257382c50c16dfb45b07d19c1","subject":"Revert \"kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aa631d40c0960a257382c50c16dfb45b07d19c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 12:01:37.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 12:01:43.000000000","tz":0},"subject":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"","message":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"\n\nThis reverts commit 8aa631d40c0960a257382c50c16dfb45b07d19c1.\n\nReason for revert: Take 2\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/971606\n\nChange-Id: Ie72a6e16f0f732007b175b9dc866489fa5505dd3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69ae7b261320b448d93d216dada12633f5362a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69ae7b261320b448d93d216dada12633f5362a34"}]},"branch":"refs/heads/master"},"c8ff3cf43d733b8c89b0148eddd936528d02b7ae":{"kind":"REWORK","_number":4,"created":"2026-01-08 14:47:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/971693/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/971693/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/971693/4"}}},"commit":{"parents":[{"commit":"8aa631d40c0960a257382c50c16dfb45b07d19c1","subject":"Revert \"kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aa631d40c0960a257382c50c16dfb45b07d19c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 12:01:37.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 14:47:40.000000000","tz":0},"subject":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"","message":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"\n\nThis reverts commit 8aa631d40c0960a257382c50c16dfb45b07d19c1.\n\nReason for revert: Take 2\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/971606\n\nChange-Id: Ie72a6e16f0f732007b175b9dc866489fa5505dd3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8ff3cf43d733b8c89b0148eddd936528d02b7ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8ff3cf43d733b8c89b0148eddd936528d02b7ae"}]},"branch":"refs/heads/master"},"3092ccf81fa6c6a5d8812a79c398d0ec9f95400f":{"kind":"REWORK","_number":5,"created":"2026-01-08 15:32:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/971693/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/971693/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/971693/5"}}},"commit":{"parents":[{"commit":"8aa631d40c0960a257382c50c16dfb45b07d19c1","subject":"Revert \"kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aa631d40c0960a257382c50c16dfb45b07d19c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 12:01:37.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 15:32:55.000000000","tz":0},"subject":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"","message":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"\n\nThis reverts commit 8aa631d40c0960a257382c50c16dfb45b07d19c1.\n\nReason for revert: Take 2\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/971606\n\nChange-Id: Ie72a6e16f0f732007b175b9dc866489fa5505dd3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3092ccf81fa6c6a5d8812a79c398d0ec9f95400f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3092ccf81fa6c6a5d8812a79c398d0ec9f95400f"}]},"branch":"refs/heads/master"},"4484409001f5ba59f6dace38cac5b22121b14172":{"kind":"REWORK","_number":6,"created":"2026-01-08 15:39:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/971693/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/971693/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/971693/6"}}},"commit":{"parents":[{"commit":"8aa631d40c0960a257382c50c16dfb45b07d19c1","subject":"Revert \"kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aa631d40c0960a257382c50c16dfb45b07d19c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 12:01:37.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 15:39:04.000000000","tz":0},"subject":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"","message":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"\n\nThis reverts commit 8aa631d40c0960a257382c50c16dfb45b07d19c1.\n\nReason for revert: Take 2\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/971606\n\nChange-Id: Ie72a6e16f0f732007b175b9dc866489fa5505dd3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4484409001f5ba59f6dace38cac5b22121b14172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4484409001f5ba59f6dace38cac5b22121b14172"}]},"branch":"refs/heads/master"},"f1cbec7682ab34eacceafcb243f1956b5b74192e":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-01-09 17:39:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/971693/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/971693/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/971693/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/971693/7"}}},"commit":{"parents":[{"commit":"8aa631d40c0960a257382c50c16dfb45b07d19c1","subject":"Revert \"kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aa631d40c0960a257382c50c16dfb45b07d19c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-08 12:01:37.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-09 17:39:32.000000000","tz":0},"subject":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"","message":"Revert^2 \"kolla-toolbox: Bump ansible-core to 2.20\"\n\nThis reverts commit 8aa631d40c0960a257382c50c16dfb45b07d19c1.\n\nReason for revert: Take 2\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/972696\n\nChange-Id: Ie72a6e16f0f732007b175b9dc866489fa5505dd3\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1cbec7682ab34eacceafcb243f1956b5b74192e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1cbec7682ab34eacceafcb243f1956b5b74192e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
