)]}'
{"id":"openstack%2Fkolla~972434","triplet_id":"openstack%2Fkolla~master~I85e35fcba454c519c12c7ed90210b7dbd46c1677","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-07-23 06:44:22.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-07-20 11:58:14.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-07-23 06:44:22.000000000","reason":"Change was submitted"},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-07-23 06:44:22.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-07-23 06:44:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85e35fcba454c519c12c7ed90210b7dbd46c1677","subject":"Add monitoring for external component versions","status":"MERGED","created":"2026-01-07 09:06:10.000000000","updated":"2026-07-23 06:45:24.000000000","submitted":"2026-07-23 06:44:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972434","meta_rev_id":"69791836a5241f1dee643aac22c4a4c822fb9557","_number":972434,"virtual_id_number":972434,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-23 06:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-07-23 06:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-23 06:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-07-23 06:14:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-23 06:14:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 10:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-04 14:11:23.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-03-31 10:55:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-06-25 14:31:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-06-25 14:45:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-06-25 17:04:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-06-25 19:51:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b15bb5c5903d0a7b984efe150404f7bc9b94d217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-01-07 09:06:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1a3fe16cd647be1269c82c7e6bc1a0f061ffc05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 10:05:49.000000000","message":"Patch Set 1: Verified-1\n\n(27 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b45109196dab4395b63bfce1a9674a14\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b8d341a047d8449a9e004a857ba991f7 : SUCCESS in 40m 54s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9c9be90222644ab5830ae513fcfc5abb : FAILURE in 49m 16s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3d920125adb344289483b78fbeb7bbc7 : FAILURE in 13m 43s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bbe710a97f3a41ff906ec8a4faaf7acc : SUCCESS in 47m 29s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/aa97cc03e10b4415b301ce15093316f1 : SUCCESS in 34m 46s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/def8e695994b4102b9ed2d1ff934b1c5 : FAILURE in 51m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8fb06250f3a841d7b7d2676d0131db2b : FAILURE in 14m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/91913fb21d3a47daa58bf75393783ec3 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04fef944e3ec4046a420052141087a7b : FAILURE in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab1eb4111f5f4097bec34096c2373267 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01cf2944078749c79f1b9be812f3894f : SUCCESS in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a570cf5bb3804ec793bebf49d21b0b30 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28347d42adde45148b4651e9dab3e080 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c18c6950ccec4e858899bf8f15fefeb8 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2fe0a7861acc1d1a8a260d292f5c595593048b4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 10:36:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4286c419d594607b1a2fd3be85c3714\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a20b48e08aca4a3ca7a223ce842659c8 : SUCCESS in 57m 40s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/18ae48af119546a4b9499ac0eeca106c : FAILURE in 1h 25m 37s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/857b3800bf434a24822149f3d85b2dfe : SUCCESS in 54m 48s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/742d151736f748b3b2f7a53f192014ba : FAILURE in 1h 07m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3037e89b30174fb293a81d2bc368020b : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/20fa4c5384dd47acadda4e71f145a873 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7eb82477446f4dc0b2df80dd9c1143ef : SUCCESS in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"512ddb0e720441088878858b8649256d9c719e28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-01-07 10:38:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b214b56feb30e28868304d99a1e720c4a2b531fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-01-07 10:41:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3b066bd66f99cb1822909076b61e1ab89ec29cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 11:36:07.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a43cf9aea37b461ca0025dfc1717cad6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/18757b5f43c7432fa28fa6b057d32414 : SUCCESS in 49m 38s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/dd0a5e5b68da44839713a074a3ee543e : FAILURE in 32m 29s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/86e0e44b8bd5490fb776447085d8295d : FAILURE in 26m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4da546a6ba8542df8fd635a3843c4e06 : SUCCESS in 52m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/648c110350cd4c1b8bd760d811743fb9 : SUCCESS in 37m 23s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00612be073be46b5b1b088d81f8b2e87 : FAILURE in 47m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6554be0ab9204014ad935055610c3e0b : FAILURE in 29m 27s\n- kolla-monitor-sources https://zuul.opendev.org/t/openstack/build/7c9a7f13eeaf49e8badd87a9f98585f0 : FAILURE in 4m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b55ebfa514e44e28abc71f34138d1622 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a08b9d0f090540869d25e040bfa6f80d : FAILURE in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b20ddf7f97345f8b263b45d442c48f4 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9be5b97e377a4491bde9dbf9b6ebc10a : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9949f350bda645298130c9b55972fea5 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8030e96cd66a4d5ab723cdfdf40b9c86 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a47861f71e24f41a4c81601f14ccc73 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"35c80d8bb193fef4e7168521d487a6ffb7f8cc28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-01-07 11:50:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"af0e678f125edcdbca23ceb1aa50564a49f0657c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:46:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b37a0b6d7d7b49a2ad3b443f5759520f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/9119b791020e4a0aa13c746dadfe37e2 : SUCCESS in 43m 53s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/fff7685e53344216b7d217bb8a9893e9 : FAILURE in 37m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f3ca8944ef7749fa981dde1e7f0b81d6 : FAILURE in 27m 20s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1eebaa542e624062a18a69d2cb5a7613 : SUCCESS in 48m 21s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/560a84a180014602b618d1294052f9e3 : SUCCESS in 38m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d20b75f04faa46229e725dae0d009eb9 : FAILURE in 49m 34s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8a80f1caad474c0894ef8a9dc31dc65c : FAILURE in 12m 54s\n- kolla-monitor-sources https://zuul.opendev.org/t/openstack/build/e5f08ca5cd2f44f4bf84576b0ad0e2b8 : FAILURE in 3m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/305c787bfcda45c78e96952e67003933 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0975b722a3b9477b8ad5fa576f58a707 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b4a619e34754812be54a359fba635f5 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f559abdf64be4ba091680b39211b8a0b : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55db10ae692d4e0cb220ffae848d2b13 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5acc93b66fd1471496e67ff6f5f0db13 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ff260de3dd94f9497a9dcdabe609247 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":4},{"id":"2e2496cd89eb01e86e2feb40927c08fd9b996597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-01-07 12:55:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f87ddc08d8045d45b255f62530e9f3019dd2cc8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 13:56:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1ac5989310b49f7926b54781d540dbb\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/defea68314304a7bb2981b58461bdc0a : SUCCESS in 37m 29s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4ebbb20582a74f2b88f1508ec46b56ec : FAILURE in 50m 29s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/534231a11cc949d599cf683d730e1b05 : FAILURE in 12m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f9de9e9e75e7403bba917bd4e3733a60 : SUCCESS in 46m 07s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/65052feb66f441e799d1bd6c4bad1e1f : SUCCESS in 47m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31f24b768e524a89b415c5c13a9eb232 : FAILURE in 49m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d708cdd7b0640df96b8baf4819cce09 : FAILURE in 30m 13s\n- kolla-monitor-sources https://zuul.opendev.org/t/openstack/build/7b84ec8b25104f5ca414035828e07b3d : SUCCESS in 4m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d64cf6b1a0b44b94ac5f792fd6203454 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0278f9e9714c10bfb4bffd06e0a0e6 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cb8d06706e846529fa1c3061dd8426c : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b613c660094548278362d4d52c91c75d : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/264602050de542dbb24ce6da4d3914b4 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a106078ce34c56bf47fde0580e3a75 : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0846d6a2522a458cb15a57b3980990c2 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":5},{"id":"cc87e96ea2695aa76e97abd4ea414be24b3e081b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 14:37:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/680035065bdd469693b8896d6b4456d8\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c7f0e478bd1b457daa70b1df3e4171a4 : SUCCESS in 1h 02m 19s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2dae4810ca574086941249c76a18fec2 : FAILURE in 1h 19m 16s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ed25d542a7e9485b8b5dd4261f7eb447 : SUCCESS in 1h 02m 06s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b05ec12e2a914f92af0594d31d643423 : FAILURE in 1h 00m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a475d05a24ac43d4bdab67728e83edbc : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c91725bfa64847a69cf8d10961883b39 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/40fc6986f74345aab02471e455812312 : SUCCESS in 5m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2ff07f20fff4b5a942c1473bd62db46d6383f420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-01-12 12:40:35.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f7dab91a0cecafcad39130e37e36f6059a752930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 13:53:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2517549da90a4996a1b874a89ef0a30f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/af6f8827c2064ae88d32a453c6244c54 : SUCCESS in 43m 23s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/26377222451b46a5b397ae66bd808f34 : SUCCESS in 51m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f35f6079ae12466eaf0a6733d2744db3 : SUCCESS in 27m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bea487189a104715807ac97a2b8a82c0 : SUCCESS in 44m 02s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/07895082ce954aa8b74559fa62236cba : SUCCESS in 1h 07m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80be4355d54c41589adbed1424399d81 : SUCCESS in 37m 21s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2605a67a1c704c2eb62ed512a61402f3 : SUCCESS in 24m 04s\n- kolla-monitor-sources https://zuul.opendev.org/t/openstack/build/eaf693cdace34e948d24d3d5da936313 : SUCCESS in 3m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/da9c17708a1c449aba063eceb00efbd7 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a9f8cef99c43dd91b3b7fd799757ec : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1f5da17ca91429ba8113cb6f08c70fb : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03592230549045e887f36cf416e92212 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c1bc9f3db734df08562b0a649cf05ce : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4428fe35119d4597875a3d46eb6a3d6e : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfaca105650440b58e373ef674b5b349 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":6},{"id":"73fc3983058b29846100a7f6208ca9244002f8cd","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-01-12 13:58:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"eb69ae7a8b9672b52798dc741f0e78d43a4172b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 14:27:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37a6c8cbaf554e9b825be335093b2dd0\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/151c44ca5e67488dad309758293ca031 : SUCCESS in 1h 38m 19s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/311f577fc9a7469c9944ab9a752fc3b8 : SUCCESS in 1h 06m 20s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/91b63fcdd61f4e44904a27364506fa90 : SUCCESS in 1h 26m 54s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4cb8f084e2294529b993e7d974612d5c : SUCCESS in 1h 01m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3bd51a97fc554190879cfb9e80f7a69f : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7218576e1e054320b3fd04326d775b54 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d93a61423ac54b3eaab424dfb32491cd : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bce1891e9b2291da873263bcba4626d8cfdc7a11","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-04 14:11:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8ba09c4eadb74994a36a039cec25c6f3c913140f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 15:13:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9bf3710e02445fb9bcec1de6027be67\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e51287c256824b59b7a7e925871049f0 : SUCCESS in 53m 33s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/14e4e4d7d8dd4a37bb5722fd10430a0d : SUCCESS in 57m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/98fd3930ba644d93a3a2b8b482654eab : SUCCESS in 27m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f06e28777d134881a0c585919e6db020 : SUCCESS in 50m 34s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9fef3ce82b8b458f86b9397e758fc7d4 : SUCCESS in 59m 21s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/973e4aa3a15244a293e661ec3a382dc1 : SUCCESS in 47m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8268d13a32f840eab9d025215c1a79bb : SUCCESS in 26m 00s\n- kolla-monitor-sources https://zuul.opendev.org/t/openstack/build/55e23974efb44ef9a2874399efe07ddf : SUCCESS in 4m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b50a0629bee649ac913afff45f3f09ba : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab7ead7eeb34841b37ba07aae9746b9 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e85e6a198e654fc39e801438d6a2b419 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46f9c265aff642eab4b4fdcb3b429b05 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c3b070a58284497b354da41bd949697 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bedb6f6712649fd8906ca0ab7467792 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e141067918c40c69d7f42d24132d121 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6858b8d1747b443abf49e4a086d62b96 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":6},{"id":"51a12763804b11ae9bc41a37185b08eef34531d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 17:21:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/828fe01fcd6347b088cc789bdeba965e\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/de5fc0ebed09410e9daa67ef59585b31 : TIMED_OUT in 3h 04m 18s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/661bc7c99293493cabbaba6ecca28490 : SUCCESS in 3h 00m 18s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/246a6325ed3d48fc9b65fdb858940e1f : SUCCESS in 54m 17s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/46db45ef269e47698508d3878ff9b530 : SUCCESS in 59m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ecffc3696b6146038e33e038a39b0b20 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4f3cf04f49644c15a341822ed4a1220d : RETRY_LIMIT in 2m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b8d7853bb1de4b4db4f4951d2ae30f87 : SUCCESS in 6m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"80d325fda13ea83d58759d7e1d3cd76d6b789c59","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-08 06:36:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5db02b307cddddec2e3f44c04aac49d165b626b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-31 10:56:07.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_37306\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"_revision_number":7},{"id":"dab53f2bc24b9e4051762dbce4c1668b877cd0c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:02:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c64a64f5fa914ddfbc61ceb8b6508526\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c542ff82d23f45b1b7a29f2416f9c911 : SUCCESS in 46m 49s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7cfe164f12dc4acfa063ed303862a727 : SUCCESS in 42m 08s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7f1bf502278941c0affbb1d6de2b4296 : SUCCESS in 22m 04s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e8f07d4b5f38468e87ad3dcefaeeb025 : SUCCESS in 52m 21s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/483ec4bc25dd4d838e3f3728940f81ff : SUCCESS in 1h 01m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69736369b3334218a59561cd46c0ac06 : SUCCESS in 58m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9afee27acbeb4c91ad6136db3c113ace : SUCCESS in 13m 51s\n- kolla-monitor-sources https://zuul.opendev.org/t/openstack/build/094522463d22428ba45e8f06e2f95d2a : SUCCESS in 5m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3c6dc7c862fc4fd8a2b65bb1d1043c82 : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfec64c145544c3bf456dd7d79df160 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5acb428137ef4419ba758d9ba98e5d1c : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f898b46d6684aa5956c410df139667e : SUCCESS in 5m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1527d9132211426ea51c16f5651d6b3d : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bd4cc9c91f34489940d58fb892ba340 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f06fef8421264dc68d30e640e06d37ef : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aff9bb4030e4e63b01a6a215b8f3030 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":7},{"id":"f680ef08ae65bbd2fd56396e58c0e6e1884707b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:21:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b83730e2abc4f2e80954b3fa8bbc72d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d787172a3bab4eb5b389ff4cafbc291c : SUCCESS in 1h 00m 24s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1482f424a59d444ca054214dc394ca61 : SUCCESS in 56m 09s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/041926fe5e794ce6ba9b216dc31dd61b : SUCCESS in 51m 57s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/aa6a93fbf1ea483a956144284640daac : SUCCESS in 57m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eecc2cddcce94b01b0b145e3eda63308 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1edff5aee42b4807bdc9caf5bc3cc197 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ff328ee70e1b48babaa2daf5f2e20d76 : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ce5ac4b605883913c785ef8da6b2f93314f90610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 08:51:44.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"d0a2825e4f144c8f7f16eae2f4f78c1d3be2439f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 09:27:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c985f27bfd2b4019a2192935a6fe4a29\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e84f86aafcb74c25b6d7304c480562a7 : SUCCESS in 24m 48s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/048722f2756f4525a8854d37f6a2ec16 : SUCCESS in 25m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/19a4e9d182344e91b684fcb97c28760b : SUCCESS in 27m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d2a0a2eaeb5f491ea0a3e32763d27c09 : SUCCESS in 22m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/20627975c74043059978936a6867737d : SUCCESS in 32m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca8bc74cfdd04d8cbad1fe70c5de8eba : SUCCESS in 25m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ccda6ea7936e4c2e93eee89d0689a0fe : SUCCESS in 21m 34s\n- kolla-monitor-sources https://zuul.opendev.org/t/openstack/build/3f352797680b4ab98cc7ee03df0b0bcd : SUCCESS in 3m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6bda3f6224f64060bab97f02a64e0dbc : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec64374e6804fc097b7a276d7bf2466 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/478a2dbb92e948c5ba55f57221fd0dd1 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec13373d93d149c3b8d432042d7e3c8d : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce55a93854ea4efe9ac2f9cfdbaffff9 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34ca07bbe5bf4c258ef1ce05c0122663 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feaa49fdada84c00b0d88791445a0270 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90c3fb2b850c4675bb4c7fb04d21c2a2 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":8},{"id":"eafaf45e034425d17871c6f7552127172b44e37d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 09:32:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1655b5a8b69b4dcbac546d174e047de0\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8f5f72e4107e40dba96946b4adaba05d : SUCCESS in 34m 45s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/16fc2a19fd8341a9921a2924b7627096 : SUCCESS in 34m 01s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/2f77bad67ecd412f8891872078bdc556 : SUCCESS in 34m 12s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9c88b9b63b5841b4b0bd2df2cc0b61bc : SUCCESS in 35m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ed25ddb86324f7dba3c85f035277fef : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a1f854130f548b5a840f6c76e869542 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e6a021c8ecf94b2c80422dd6ec00ce36 : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8643f296053bd6d24b349cc14f2e7850cd98efb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 13:57:14.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"0519d7b8c65f3ac64f86ba3df3703a51dad20b8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 14:20:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6b72cccb405c612512de23b21fa5f830488f9d2e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-25 14:31:16.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6dd9f93d2ee2a5c4455f35ddc203f67896f5822e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 14:45:33.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6f2fd9c42989abfeb2d0c364c60b422302960b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 14:45:52.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"eb8453033f58f9a3ecada5f9cea378f627efada2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 14:46:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ab58142406024c1560bf50a7f78c323e3b47c206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 14:47:27.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"2833ee70d2bb28cb18e779658010acc869807c7f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 14:50:04.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"845c54bc39b933d10338a17bcabdb4415f44bcda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 15:39:22.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c290bc2b28764e9ebb8afac32e7b4773\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e46ab97c00e84a4f855b4f385f2b57d3 : SUCCESS in 32m 15s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b43bfd2ec6604fd7ae338c41a5cbdb3a : SUCCESS in 34m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/9c036314abcb4bf6a88626982071a452 : SUCCESS in 44m 49s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ff11fcb6bb5340c0a41e4c19c3fc0189 : SUCCESS in 46m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b399d380300419999302f7c584dc0e3 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3f333ecb65db469e99e53a8aacd9f1f3 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0a1068ec978244efbd6d1efed8ae051d : SUCCESS in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7fa18b61082258aa960ca3b1fd58f450053338ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 15:45:07.000000000","message":"Patch Set 13: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1c888ecc52849cda2fce18d3c8a3de2\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/80267f76a1db4fcfadabb801da44b9ec : SUCCESS in 28m 06s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e6f8572adff54f69870f7351fd4bb68c : SUCCESS in 24m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7fb3bc850c8b47a2ac8ee70fe4291d8d : SUCCESS in 30m 31s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/61154e94e7cb48fdbefb00b87e17dbce : SUCCESS in 29m 54s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/18738c4b7ff14cfa88e0b09dfe27dec3 : SUCCESS in 55m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e71bb93d0af74b1e9996ff288011bce0 : SUCCESS in 36m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ae2285b82a141d2a19d1a264811e633 : SUCCESS in 15m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/46e0969e0d7c44c2be9a2716d37490a1 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79fe980e1e74133ad6f0bfe9f1f4e8f : FAILURE in 2m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2e525d2a2d74189bc860240d2bf0f70 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd4dee9437d44b7a87dc8450e67b3160 : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ae8d70bfcc54bab80c59e4944e27723 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/101545aeecb9467990818793207d4ebb : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf65875195848e1a67d3c2555f40502 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/430d3fea7a9f4092864f037fecdf831f : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":13},{"id":"8f858bf77c6976afc0cebd959c7393ed08a46dc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 16:39:45.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"7f3280d9db7abac57f51365cefa308d3a027651e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 16:43:01.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"71098b9b689a73c5a5bb77309e7a69ca336bcb8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 17:01:49.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"907b5512f4bcb0e4b4f10ba55984ea395d8a9f96","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 17:04:00.000000000","message":"Patch Set 16: Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"f0cd902e7d6e8fc4d0a87329845d63a28f015f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 17:36:28.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62a464a829d841e2852ff02fb1f78fb6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e26c953c03e0463c951116cd79fea06d : SUCCESS in 23m 41s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e07414bf15ec40c6bb1cf0b86403b462 : SUCCESS in 12m 08s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4ce3d76581904e51a6078fe508242619 : SUCCESS in 13m 14s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0ed523f8507d406a9fddc553f7e4047f : SUCCESS in 18m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/aab1ea8f27f241ad8d31eac9e3383419 : SUCCESS in 33m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b994ea3f7dcd48e991bdc84bae334f05 : SUCCESS in 24m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bf060ec8dcc743bf80a77357c97f5ae0 : SUCCESS in 13m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d076fc6227c349e6abd94773e0434c3e : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/460a774e662e4eed9ba20c55d6a4879c : FAILURE in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1d17ca3ea674b0f83069373d9a70f2c : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f05eef8da62d40878e12e992e4bad9cd : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54248b0b84864558a821f04a8e4a5b14 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b162dd3d59b94a8790d79f3cb1da66eb : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d5fd70e12c4ca69f98346d881b4946 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2c33717bc2e46ee8aed7766b0702f2c : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":16},{"id":"5c06be948663706ca8e9dbe840b3bb3395baaeb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 18:32:28.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"efd6757d3c22bc4ef2dde668f664b073ef701b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 19:11:26.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9a43e77a7b147c992a7d0fb6fbeb3d3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/15247f4d849a48f4beccbeb0d439064b : SUCCESS in 14m 08s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3088f4f052cf43dea6a0babd03631ce6 : SUCCESS in 25m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b1d9629b6a554005a315181760dbb759 : SUCCESS in 12m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ea8916019ccc4cdc95c5f71db0fbc120 : SUCCESS in 17m 48s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/02470df6ad7c440c9f7305216a4c93d7 : SUCCESS in 36m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe6ea1202c584d90b210399ca295375c : SUCCESS in 14m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8fb0a5f0c8fd49209bd68c513ef2bcf0 : SUCCESS in 31m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3883d195407c4ce9af5d422f9e2cfb3d : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/620c29be9a2e4aeaaa74171e96b9520c : FAILURE in 2m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a9c099a758c4a568eb0ce96f67b5704 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eecc9c1f8c2a401eb5c38e7c94a2f431 : SUCCESS in 3m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/712bb1f4dcb945c4987179e3a6898e1b : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55c140517ecd403eb73842c073da0e98 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91edc4e9cdc0442e8e9e9cac42cc39b5 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8564f9e52f034eb5958af7ba73366999 : SUCCESS in 9m 21s","accounts_in_message":[],"_revision_number":17},{"id":"750a5dcb7409ce171ea0f0f90372ba441650815f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 20:52:09.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/121ff335c03f477db018b0b8217cb473\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/6d21547c20594babb84a23c3f11d0b83 : SUCCESS in 27m 29s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/986675f09aa84169a6c697d67e25fe67 : SUCCESS in 43m 20s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ff2b2ab6b18b4da586198ac0a316a37b : SUCCESS in 34m 26s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6bece2efcc214302aea3af9d89a14b24 : SUCCESS in 35m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cc7f2f5866254e3dbc24528b7e29bb4a : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a11ac2641c6f4b858e98632bf8f7a579 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ea6393871e0d4fbda274b2af070014b7 : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ad9606a4650bec8417b43ac007d26c1e1144d06f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-26 05:22:20.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"e07489948adb3640aaf6f070db08d4ddde90ebbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-26 05:59:57.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":19},{"id":"1fc66d316b31fd29c8555a25c878ca1f301473f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 06:45:34.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba034e21fd5a45ffaf11053fa867cb78\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/06762b237368422188016f00514f4bcf : SUCCESS in 22m 02s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/27f67d46441a4a1dbb38d1acfe3c9dce : SUCCESS in 19m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/51ff7f54755c49b5b3da211630987ed7 : SUCCESS in 27m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bfd5577be16641318c01448222d9ef79 : SUCCESS in 25m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e8100c2f1f624b1b949e25c746fa2f5e : SUCCESS in 44m 23s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f33686dfe4464ebbb419e6123488629a : SUCCESS in 26m 33s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/81b014dce05c472087bb90b2f9339647 : SUCCESS in 30m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/780aca0bcace4f8c9b461d7b0ebbb3e9 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/681c3d31d1194a009d1171c62110f018 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18cb845f520c448e966ae077076fc79c : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65061fa23b3a4b80a53730fe312fb1aa : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c4ef63857ec4d0c8d8e4466efddf10e : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d53120c95e44187859ca639a81d0e3e : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c81aa9436e54b1d8957c9ddb0e7fd87 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dc779719ac64633bedb6c9c1a6eaa19 : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":19},{"id":"cf952fc232552bb635ba601b0ca88f6d44f82553","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 06:56:22.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7c6b510bf694f24995f46292cb156af\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f3cdaef0c73d41e28f1a17a64773f319 : SUCCESS in 48m 06s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8472425bd869498bb3b4e397dec4f002 : SUCCESS in 53m 44s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/104e386c45f949e28d03816c4c68c00c : SUCCESS in 49m 24s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/64ba91bb69cb4348b543582b805e95e2 : SUCCESS in 51m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/112ec0d117f540ab9bdf178257a40d4d : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/458f262f4bf7489ea29e62b68315fed9 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da7f8819cab0482187cbce0a613db6e3 : SUCCESS in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"62ca61af8d4cbc149a0c43b5a7c2e8843c147d8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-26 07:19:55.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"5ed5df20526f163977caf92553f2b2dc69802c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 08:27:30.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1de622adb84463f8de52b30193a1dd0\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/361b144e77df4a64ad88ed0798b7e623 : SUCCESS in 15m 40s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/70df1f54761f4272b40d089d25e47ce1 : SUCCESS in 23m 50s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/6ed459c3cec24b3ab4570a290fabf888 : SUCCESS in 24m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0567045b54bf46b2a55beae5ff032daa : SUCCESS in 18m 28s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8ed355111a014013b5c02e42812fb7cc : SUCCESS in 1h 00m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7a4683505a147448b9bbc469d417c9f : SUCCESS in 15m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f55337c2c370496cba45dbbaf7b37954 : SUCCESS in 17m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/355ca90fb97b4a8380949b89c42a0931 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b33702f01cd440fab83a678f309464a5 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4238b805514a4015b50b267ae17bb774 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb0345c65c5044a78ef19518460d5673 : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ee58eb401be418eb3164008d0a035d5 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8f43833832a4b4287e4ca637f50cbef : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61dceed9557041909b70565c759ec2b1 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18eb1d2c605e460a9e82f6f488352c53 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":20},{"id":"e3464cebbf22b5dc0e371c5840240f4ffa6c2095","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 09:19:12.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8b2948b41a8489a8fe59eb5fad1ab46\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e3249277c4c645d7a86081332662c1df : SUCCESS in 33m 14s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/cd932fe8001f4c9fbca81a144a758730 : SUCCESS in 36m 50s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ccf15083e0bb4403ac5414bff5f8ec6c : SUCCESS in 42m 01s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/66a90b37cbc3434a88f2a78409f7d320 : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d36a57605e404d7b8359b2e8f2b7c82f : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/56c595c48b6b4fe19ebe9ac660e1a23f : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9ce482a7612249e08e6d667aa00aaa31 : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"44fa5e2716087ae4aab31ecd8968f27c5666ffd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-17 08:31:24.000000000","message":"Uploaded patch set 21.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"b5febf946787e9b40c6093f45879ab2090fbd899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 09:28:47.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea88dc0c4e644e61b19cb8f91e756080\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/51a8d135914c421b80035838d16199c1 : FAILURE in 55m 22s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/535515111baa4be98a9be1f6c2af7afc : FAILURE in 29m 59s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/175fbd7657a3491fafa03099149347b0 : FAILURE in 31m 05s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f43858fe6fec4c6d8eb48d0228bc7b95 : FAILURE in 17m 36s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1e16b49163364226ad8bacde90c0bc6c : FAILURE in 34m 41s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1994f24a0bb842328d4ee1a966db29a6 : FAILURE in 30m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/407ab836b24a4dc7ad0e45744dd0267a : FAILURE in 31m 59s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2a1b8815b70549aea37e039abbb6a152 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96eed794245347aeaf9d9988e53497a9 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c98d7bff5fc40e1b838b62ef2405c80 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0806f6775dce4951b42974bf6e8c745d : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ee11c3665034b30ab39db558796fc94 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36b4a1dba1c949de90e74f9cdf1f7410 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705dac9646714ceba723daf9104a9fc5 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5116f518bbd2468090a711d5fa56f836 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":21},{"id":"23050f87bb2f3a8237cd017f4e15f39efc2af58f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 09:56:41.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5756a527f14a47a7a9d16b9abccd1363\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/eb2c6c8fc0f842738f0e97b17357560a : FAILURE in 52m 53s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b6ca310e2118467299e901f10a6d5c92 : FAILURE in 50m 51s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/7230b99fb5de43328a131a6eda7ad21d : FAILURE in 54m 14s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/925dc9e41ddb4b15a508a85e513eb462 : FAILURE in 52m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b5f4bbdf8ce4242b016f3ce4bd9b1c4 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8ab775a23204075a7970d1988f8e161 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fa88deccb2014ae3a73c0465835448a7 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"27f44e73e931a4678aa179390d91d60ebdb4251c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-20 11:58:14.000000000","message":"Patch Set 21: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"42720bb5fa65d1434fbc4bd0aa49cec8284ad4e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-20 12:23:36.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"1a5653fe98ad855b5a8a2e48e933e688f3c6740e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 13:33:37.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0ff9d6548cd45e3ac2672d3de2b3873\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/5edc55da3a594181872e6a520e119a1e : SUCCESS in 55m 19s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/876f9e00564543b8b5a6bc636b731aa9 : SUCCESS in 24m 01s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/29c4c9e597b3498ab5308e3ed26595ab : SUCCESS in 36m 20s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/be2aee70bf8d47c7b2719a424f9dc400 : SUCCESS in 20m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ad486ce87e594877a9ccca4b4dcd071d : SUCCESS in 51m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77c03bf2a7b54676862dfe9377935c84 : SUCCESS in 32m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/72a9f170f3a442ddb4aacd3d466cbf19 : SUCCESS in 28m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1a68a80f1e1745428ec7e6c0dce6df52 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf71ec45d02491e93ef8bfabd7c2bde : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/001bfa472cb94c33a56553e05908ce51 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c48a2ccd5374004b439f315d0c1005d : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/80bf2044107c42a8ae39cb552c200ea1 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c60280eb3d9a44d6aa87775262e440fd : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16dc943b2b6348a88aa9b5f113b09657 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87629afbf2374e43a9244d24b5090b9c : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":22},{"id":"405c50beaf794e2c70b31372bf3f6b298fd2bcb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:52:00.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0a7f02267d54430913b9e3eedffd0c5\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5d5ea9f6a57e4d3aa9555bf1d12505df : SUCCESS in 28m 58s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/ec54218cfe984d38959428b33cdd0fc8 : SUCCESS in 42m 37s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f95bcf2b6f0b456ea8819ea4f18a0691 : FAILURE in 40m 04s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f02eb1284ee24a6cb9158ad6251dfaa2 : SUCCESS in 38m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/525f4bc1fbf44c009003e1913c4fc6ae : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e1967a724c9d4b57b782e1bd174e94ae : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5fec4516914d407394824cf8509afb5e : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d68a953c11b81ddb12e8d8ba6161f8fc669078e7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 06:14:20.000000000","message":"Patch Set 22: Backport-Candidate-1 Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"710b4e33818be4471dbd4246b7b7fd1732c48232","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 06:14:35.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"05fd1b39f8884584b82fad9e39201e9e89e5dbd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 06:44:21.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d43d46c02044e90ba216105473e7dcc\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/20ac8fa3023042e6801ec0943262d235 : SUCCESS in 24m 54s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/43553bc8285f400d84ac18fa816876f9 : SUCCESS in 23m 54s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9c9ae632971b4f3db5de0984177c62bd : SUCCESS in 16m 30s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dd9b008781f243a98ef7ea733ed4f737 : SUCCESS in 28m 11s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c28bb230aef4bdd9fbaba6e4bc54522 : SUCCESS in 15m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/18bd09a928f049eda94c81b5a2bfa7c3 : SUCCESS in 24m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4debc48c1d8f450c9b22ab8b044680ce : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a924b731f4f14499812b2a3e9c40ab6e : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99c99838367848a1bb7b5d9eb1c59bf6 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c0b0b8efc5c4774a9120b287afdca34 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71f4b16169249219f6f8923caced9e2 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba3b6ee0a0fa4c70baaab000bee1fff4 : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":22},{"id":"23131041657f900d1e97de3f009015655e7b89ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 06:44:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"69791836a5241f1dee643aac22c4a4c822fb9557","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 06:45:24.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee157c216ec646e9961f421e7a6f92b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d304ce6ba02741148e259a3a7b2f18ef : SUCCESS in 45s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"c7b75c5c2c6b0575d54cf97096cadf3502c18873","revisions":{"4e2695d3892152396e8554b57214b39cd2e2aac7":{"kind":"REWORK","_number":1,"created":"2026-01-07 09:06:10.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/34/972434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/1"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:06:00.000000000","tz":60},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool and a Zuul job designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nKey changes:\n- Created tools/check_sources.py: A script that queries GitHub API for\n  the latest releases of components and extracts SHA256 checksums.\n- Created tests/playbooks/check-versions.yml: Ansible playbook to\n  run the check and save results as a Zuul artifact.\n- Defined kolla-monitor-sources job in zuul.d/base.yaml.\n- Added the monitoring job to the periodic pipeline in zuul.d/ubuntu.yaml.\n\nThis helps automate the identification of outdated components and\nsimplifies the process of updating versions and hashes in Kolla.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e2695d3892152396e8554b57214b39cd2e2aac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e2695d3892152396e8554b57214b39cd2e2aac7"}]},"branch":"refs/heads/master"},"9acc8852d0820f6b5c18b23823bdc9b82926d143":{"kind":"REWORK","_number":2,"created":"2026-01-07 10:38:43.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/34/972434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/2"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 10:38:33.000000000","tz":60},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool and a Zuul job designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nKey changes:\n- Created tools/check_sources.py: A script that queries GitHub API for\n  the latest releases of components and extracts SHA256 checksums.\n- Created tests/playbooks/check-versions.yml: Ansible playbook to\n  run the check and save results as a Zuul artifact.\n- Defined kolla-monitor-sources job in zuul.d/base.yaml.\n- Added the monitoring job to the periodic pipeline in zuul.d/ubuntu.yaml.\n\nThis helps automate the identification of outdated components and\nsimplifies the process of updating versions and hashes in Kolla.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9acc8852d0820f6b5c18b23823bdc9b82926d143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9acc8852d0820f6b5c18b23823bdc9b82926d143"}]},"branch":"refs/heads/master"},"f2b84a75d046c8fc77cba67590492a48f5d8e4a3":{"kind":"REWORK","_number":3,"created":"2026-01-07 10:41:16.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/34/972434/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/3"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 10:41:06.000000000","tz":60},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool and a Zuul job designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nKey changes:\n- Created tools/check_sources.py: A script that queries GitHub API for\n  the latest releases of components and extracts SHA256 checksums.\n- Created tests/playbooks/check-versions.yml: Ansible playbook to\n  run the check and save results as a Zuul artifact.\n- Defined kolla-monitor-sources job in zuul.d/base.yaml.\n- Added the monitoring job to the periodic pipeline in zuul.d/ubuntu.yaml.\n\nThis helps automate the identification of outdated components and\nsimplifies the process of updating versions and hashes in Kolla.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2b84a75d046c8fc77cba67590492a48f5d8e4a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2b84a75d046c8fc77cba67590492a48f5d8e4a3"}]},"branch":"refs/heads/master"},"a7a026111a8bc15944b07a233709db5952984666":{"kind":"REWORK","_number":4,"created":"2026-01-07 11:50:24.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/34/972434/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/4"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 11:50:14.000000000","tz":60},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool and a Zuul job designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nKey changes:\n- Created tools/check_sources.py: A script that queries GitHub API for\n  the latest releases of components and extracts SHA256 checksums.\n- Created tests/playbooks/check-versions.yml: Ansible playbook to\n  run the check and save results as a Zuul artifact.\n- Defined kolla-monitor-sources job in zuul.d/base.yaml.\n- Added the monitoring job to the periodic pipeline in zuul.d/ubuntu.yaml.\n\nThis helps automate the identification of outdated components and\nsimplifies the process of updating versions and hashes in Kolla.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7a026111a8bc15944b07a233709db5952984666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7a026111a8bc15944b07a233709db5952984666"}]},"branch":"refs/heads/master"},"19c3dc54ee36dfd89ef2de304b177bb458197ba7":{"kind":"REWORK","_number":5,"created":"2026-01-07 12:55:26.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/34/972434/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/5"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 12:55:11.000000000","tz":60},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool and a Zuul job designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nKey changes:\n- Created tools/check_sources.py: A script that queries GitHub API for\n  the latest releases of components and extracts SHA256 checksums.\n- Created tests/playbooks/check-versions.yml: Ansible playbook to\n  run the check and save results as a Zuul artifact.\n- Defined kolla-monitor-sources job in zuul.d/base.yaml.\n- Added the monitoring job to the periodic pipeline in zuul.d/ubuntu.yaml.\n\nThis helps automate the identification of outdated components and\nsimplifies the process of updating versions and hashes in Kolla.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19c3dc54ee36dfd89ef2de304b177bb458197ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19c3dc54ee36dfd89ef2de304b177bb458197ba7"}]},"branch":"refs/heads/master"},"fb58c221fadf240e2f6a3ca0f18d20dc657932d5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-12 12:40:35.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/34/972434/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/6"}}},"commit":{"parents":[{"commit":"2aa6b33b5823df3e3a6937ab696dc22057c57479","subject":"Merge \"Update repo GPG key for influxdata.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2aa6b33b5823df3e3a6937ab696dc22057c57479"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-12 12:40:24.000000000","tz":60},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool and a Zuul job designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nKey changes:\n- Created tools/check_sources.py: A script that queries GitHub API for\n  the latest releases of components and extracts SHA256 checksums.\n- Created tests/playbooks/check-versions.yml: Ansible playbook to\n  run the check and save results as a Zuul artifact.\n- Defined kolla-monitor-sources job in zuul.d/base.yaml.\n- Added the monitoring job to the periodic pipeline in zuul.d/ubuntu.yaml.\n\nThis helps automate the identification of outdated components and\nsimplifies the process of updating versions and hashes in Kolla.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb58c221fadf240e2f6a3ca0f18d20dc657932d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb58c221fadf240e2f6a3ca0f18d20dc657932d5"}]},"branch":"refs/heads/master"},"687f432be73914333d0ed387475fd0a421af51a1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-03-31 10:56:07.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/34/972434/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/7"}}},"commit":{"parents":[{"commit":"d215d4ba30518ef2bd0297cba3c8e21f2afe5caf","subject":"Merge \"Update Ceph Squid RPM repo to centos-10-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d215d4ba30518ef2bd0297cba3c8e21f2afe5caf"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-31 10:56:07.000000000","tz":0},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool and a Zuul job designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nKey changes:\n- Created tools/check_sources.py: A script that queries GitHub API for\n  the latest releases of components and extracts SHA256 checksums.\n- Created tests/playbooks/check-versions.yml: Ansible playbook to\n  run the check and save results as a Zuul artifact.\n- Defined kolla-monitor-sources job in zuul.d/base.yaml.\n- Added the monitoring job to the periodic pipeline in zuul.d/ubuntu.yaml.\n\nThis helps automate the identification of outdated components and\nsimplifies the process of updating versions and hashes in Kolla.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/687f432be73914333d0ed387475fd0a421af51a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/687f432be73914333d0ed387475fd0a421af51a1"}]},"branch":"refs/heads/master","description":"Rebase"},"839ecb499e419cb7da247c129e524ce492726afe":{"kind":"REWORK","_number":8,"created":"2026-06-25 08:51:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/8"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 08:51:41.000000000","tz":120},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool and a Zuul job designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nKey changes:\n- Created tools/check_sources.py: A script that queries GitHub API for\n  the latest releases of components and extracts SHA256 checksums.\n- Created tests/playbooks/check-versions.yml: Ansible playbook to\n  run the check and save results as a Zuul artifact.\n- Defined kolla-monitor-sources job in zuul.d/base.yaml.\n- Added the monitoring job to the periodic pipeline in zuul.d/ubuntu.yaml.\n\nThis helps automate the identification of outdated components and\nsimplifies the process of updating versions and hashes in Kolla.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/839ecb499e419cb7da247c129e524ce492726afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/839ecb499e419cb7da247c129e524ce492726afe"}]},"branch":"refs/heads/master"},"359eeab4c0281cb4c8002c9a64578251bce1eccb":{"kind":"REWORK","_number":9,"created":"2026-06-25 13:57:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/9"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 13:57:07.000000000","tz":120},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/359eeab4c0281cb4c8002c9a64578251bce1eccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/359eeab4c0281cb4c8002c9a64578251bce1eccb"}]},"branch":"refs/heads/master"},"b44d46aa93b51b34bcc270a32c31f70ee139aa88":{"kind":"REWORK","_number":10,"created":"2026-06-25 14:20:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/10"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 14:20:07.000000000","tz":120},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b44d46aa93b51b34bcc270a32c31f70ee139aa88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b44d46aa93b51b34bcc270a32c31f70ee139aa88"}]},"branch":"refs/heads/master"},"59e4b065b1527e33cb0e30a793acf0dbea792b74":{"kind":"REWORK","_number":11,"created":"2026-06-25 14:45:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/11"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 14:45:50.000000000","tz":0},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59e4b065b1527e33cb0e30a793acf0dbea792b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59e4b065b1527e33cb0e30a793acf0dbea792b74"}]},"branch":"refs/heads/master"},"47d62a2c2ce2d8f2a8cb400ab29a5f7358127632":{"kind":"REWORK","_number":12,"created":"2026-06-25 14:46:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/12"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 14:43:57.000000000","tz":120},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47d62a2c2ce2d8f2a8cb400ab29a5f7358127632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47d62a2c2ce2d8f2a8cb400ab29a5f7358127632"}]},"branch":"refs/heads/master"},"daa89e20f45af533f44b08021412158aa81499d8":{"kind":"REWORK","_number":13,"created":"2026-06-25 14:47:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/13"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 14:47:26.000000000","tz":0},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daa89e20f45af533f44b08021412158aa81499d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daa89e20f45af533f44b08021412158aa81499d8"}]},"branch":"refs/heads/master"},"47383d4decfb5cc66001fd072cd5d01208cda2d0":{"kind":"REWORK","_number":14,"created":"2026-06-25 16:39:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/14"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 16:39:45.000000000","tz":0},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47383d4decfb5cc66001fd072cd5d01208cda2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47383d4decfb5cc66001fd072cd5d01208cda2d0"}]},"branch":"refs/heads/master"},"ac360d7fd39d0e6b8724a9a31fe244b0eb0903b1":{"kind":"REWORK","_number":15,"created":"2026-06-25 16:43:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/15"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 16:42:55.000000000","tz":120},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac360d7fd39d0e6b8724a9a31fe244b0eb0903b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac360d7fd39d0e6b8724a9a31fe244b0eb0903b1"}]},"branch":"refs/heads/master"},"fd93c68516b91859323a78aa23e5d43409c979cf":{"kind":"REWORK","_number":16,"created":"2026-06-25 17:01:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/16"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 17:01:41.000000000","tz":120},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd93c68516b91859323a78aa23e5d43409c979cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd93c68516b91859323a78aa23e5d43409c979cf"}]},"branch":"refs/heads/master"},"f29d450f09f7850331212a597378bbceac5fd54e":{"kind":"REWORK","_number":17,"created":"2026-06-25 18:32:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/17"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 18:32:28.000000000","tz":0},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f29d450f09f7850331212a597378bbceac5fd54e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f29d450f09f7850331212a597378bbceac5fd54e"}]},"branch":"refs/heads/master"},"7b7624a26db00b6c9c8239b781e60752caa8faed":{"kind":"REWORK","_number":18,"created":"2026-06-26 05:22:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/18"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-26 05:22:20.000000000","tz":0},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b7624a26db00b6c9c8239b781e60752caa8faed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b7624a26db00b6c9c8239b781e60752caa8faed"}]},"branch":"refs/heads/master"},"3353f90eb5983d455ef1857016ca4e564ff7e85f":{"kind":"REWORK","_number":19,"created":"2026-06-26 05:59:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/19"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-26 05:59:57.000000000","tz":0},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3353f90eb5983d455ef1857016ca4e564ff7e85f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3353f90eb5983d455ef1857016ca4e564ff7e85f"}]},"branch":"refs/heads/master"},"fd20ea5d1e819bd45c2a85543f64878ca40a6178":{"kind":"REWORK","_number":20,"created":"2026-06-26 07:19:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/20"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-26 07:19:48.000000000","tz":120},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd20ea5d1e819bd45c2a85543f64878ca40a6178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd20ea5d1e819bd45c2a85543f64878ca40a6178"}]},"branch":"refs/heads/master"},"87dc3f7718e4a08768e420bfc47706893db21edb":{"kind":"REWORK","_number":21,"created":"2026-07-17 08:31:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/21"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-17 08:31:22.000000000","tz":120},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87dc3f7718e4a08768e420bfc47706893db21edb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87dc3f7718e4a08768e420bfc47706893db21edb"}]},"branch":"refs/heads/master"},"c7b75c5c2c6b0575d54cf97096cadf3502c18873":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-07-20 12:23:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/972434/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/972434/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/972434/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/972434/22"}}},"commit":{"parents":[{"commit":"6709ee49acf21dc858501e099c5e012b3426b16c","subject":"Merge \"ironic: stop using driver-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6709ee49acf21dc858501e099c5e012b3426b16c"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-01-07 09:04:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 12:23:36.000000000","tz":0},"subject":"Add monitoring for external component versions","message":"Add monitoring for external component versions\n\nThis commit introduces a new tool designed to monitor\nversions of external components defined in kolla/common/sources.py.\n\nChange-Id: I85e35fcba454c519c12c7ed90210b7dbd46c1677\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7b75c5c2c6b0575d54cf97096cadf3502c18873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7b75c5c2c6b0575d54cf97096cadf3502c18873"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"108660d64406067b5d066c612f5de882d45b0e24","ours":"87dc3f7718e4a08768e420bfc47706893db21edb","theirs":"6709ee49acf21dc858501e099c5e012b3426b16c","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
