)]}'
{"id":"openstack%2Fkolla~973051","triplet_id":"openstack%2Fkolla~master~Id17aa4946617c851780d424e7e053dd79196be6f","project":"openstack/kolla","branch":"master","topic":"kolla_drop_influxdb","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-01-22 12:57:37.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-22 10:44:06.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-01-22 12:57:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id17aa4946617c851780d424e7e053dd79196be6f","subject":"influxdb: Drop support","status":"MERGED","created":"2026-01-12 08:48:45.000000000","updated":"2026-01-22 12:59:46.000000000","submitted":"2026-01-22 12:57:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973051-kolla_drop_influxdb","meta_rev_id":"676a31368e85445bb1a8fa72e4ffc47e2f700c22","_number":973051,"virtual_id_number":973051,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 12:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-01-22 10:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-01-22 10:44:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-22 10:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-01-22 10:10:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-01-22 10:10:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-01-22 10:05:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-12 08:59:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-12 10:09:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 10:01:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9b1641eceeca173cc7a1e6f4fa5fcfd137fd6165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-12 08:48:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ab59c2ee99129ecacf2ffd11ce9acbb1be63f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 10:09:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f2732bf5f2b4f02b982c7e4bfcfe397\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e516c68e149940f5a6f1a2192e54b678 : SUCCESS in 3m 18s\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/22883449ee984521b2918972ff2f59ab : SUCCESS in 46m 40s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a9d41b7e0c9348fa9f48bbe7027eba70 : SUCCESS in 33m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d9693b35ce2b4b1cbfa825c020d4dbd5 : SUCCESS in 25m 55s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0c022a5244d64a498a848b4c17649a64 : SUCCESS in 41m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/104ab2d3ca784991be6988b6db698bed : SUCCESS in 51m 52s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44955e6fa52f44b09d3d8207123410c4 : SUCCESS in 52m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/96da3499754f4a45b3c16e5698b4d807 : SUCCESS in 31m 48s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c89f35086ca44d1bb3925e60d21ef1df : FAILURE in 55m 31s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/dd6e118c30fb48c8836bdd89d86149cd : SUCCESS in 52m 31s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/01a58b8e51e44b2fbf96817b24c35ef5 : FAILURE in 38m 50s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8c59294287634f4e8c074a1b6724dd6e : SUCCESS in 59m 08s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/156b27d3122f4ded832490b9c3ff4534 : SUCCESS in 1h 18m 10s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ea28b3b27e84a28928d6606410765da : SUCCESS in 57m 16s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/687ae3f84726418fa1a8e17836c144db : SUCCESS in 1h 17m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0273d297f14d49bc95f20b5ad5eee990 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53feab28402476394ea3ab12e4617d3 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e02ef54f91b413291e44995b66964a0 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/530dd7d6c0a8453c87703a9c7572ceeb : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25c2b06571cf4f05b740f33cd5ebc052 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb129cf54a4d425d8423dc655b497ec7 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f59e5140e424422cb1f3202255a2665c : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4ab6b965b37c46a1f5ce8aa8ade4f4e89080daef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 10:09:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29a0cef7b3cf494ea93aa4a2c8bf43b8\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/88046d35da534a1d9238ef685c6a9ba1 : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a378395630ce422e92004dcf222ca2e6 : SUCCESS in 1h 13m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6fa30b8b98a246f2bd2a3ced46704313 : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/319b4f5d81db4dbb8aae8274fb93858a : SUCCESS in 1h 10m 19s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b6c78835004f42999b063966dfb37752 : FAILURE in 1h 17m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f84c39fc33534ccb84d8b9bba1e9538f : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/35b4f709fb854cf4b75c716c7bdd9081 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/87683183462c4c7e8cf977888cc87c77 : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a2d740c4c8c55cdba38e469242f71ad0d29dcf2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-13 17:32:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36aedf675edb1b89836ac0c964f4bc06a087deff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-13 17:32:46.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"49aa309407e2635c3c0b33536043e9fc14e818c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 18:54:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ba9b90ec8d841ed9bbd8fd3c022e96f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4cd5d6d942d7480a826d6a2d97eeae41 : SUCCESS in 5m 25s\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/92901deb7b054b57b8b17fb25027b4b6 : FAILURE in 46m 21s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5ee658c4678f43d98a395427560b588b : SUCCESS in 53m 43s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/494aa8b91f7246818d54117c1d94cb9a : SUCCESS in 12m 47s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a627284a3b474fcab92ba780b7c54187 : SUCCESS in 44m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/739e2b7334934b6082d52a5d8a7f5653 : SUCCESS in 52m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a713da8710541ecb347d07ad070efc8 : SUCCESS in 49m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/717d0df8d01a4dfe959f83deec3c80cb : SUCCESS in 25m 48s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/60453978b6c347c79dd7800eeee4b7bc : SUCCESS in 38m 44s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0a9208b3fa5e43e38138ada844ae9e8c : SUCCESS in 55m 20s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/1f65ae5121134ffca162289472af211d : SUCCESS in 1h 12m 38s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/15003f864ec24bee96d24daae47de20d : SUCCESS in 1h 00m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/77717031fb75472386cb132fc3357bb6 : SUCCESS in 1h 16m 11s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f1a51abcaff420880aa8134997aad1a : SUCCESS in 57m 06s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0e1d6fffcb954d0eb9a6c04a6428394d : SUCCESS in 1h 20m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4604f556e2324de0825f38395e2072d6 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b02a2f9961e4dab848a3c2328c0c540 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b6423676ad348b6b9cda67a61a24847 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32c2467f3a9e4a42a5aad4c58f1d3ee8 : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7f44dd370564fe182bd70e0ea9a95d8 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7778bef9bfe14a35bb17c06da274daae : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2272509a4b4d4287a8ec16d516992c3d : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d6efd5a2b72e184459f44e4c6cfd684cd7652b10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 19:19:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdf77a4bb889412e8b65ab40e64c2164\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/66197a0c28624cdcb1bd50b640de1142 : FAILURE in 54m 50s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d7ad4f6cb9ea416bb2b89cfff8012a11 : SUCCESS in 56m 02s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/a539894fb2ab4cc6add4b10de2519b85 : SUCCESS in 1h 02m 38s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1ad836b94340465282fe8c807c30ff95 : SUCCESS in 59m 13s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/f9ffb52076e640c28988f171868de8fc : FAILURE in 1h 41m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f1e76655ddd4031831f9e31c340b484 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1b083597a824ec6966205f913d4760e : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a53c6554f1d40f48c9cac8d1ba98894 : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cf554ad820248fd92e57c7f13028645d80712dc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-14 10:05:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7fd3d932fc1c8a944ca862658c3471e414573682","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-22 09:49:36.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae50c207f3b2c997a092b56b5ed157868e10bef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-01-22 09:59:40.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d53f0ea707d8937593fbb44898b5548139aba145","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-01-22 10:01:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ce56177598e8fd47dd27aa426cba8e870677ee37","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-22 10:05:40.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5f6e9cc00b4e71beffef0499c120ca80467dadb4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-22 10:10:37.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"17bbf95d68cb1b9e9dc9581e172a0c8c335f2c1f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-22 10:44:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e6739e63784da5c907c7722d415e1fb3f18eac3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:34:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af35f3dd456e45d2b71de89b9af7f72a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e3f6b5566af44ef2b7b5a638c752a6a7 : SUCCESS in 2m 45s\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d3937a98208f4cd99525fb263b094ab9 : SUCCESS in 54m 44s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6d49e021067140038f5f24cc42c618d6 : SUCCESS in 52m 06s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/df6a9978b59346e2b3958097f6900e98 : SUCCESS in 12m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d2f41843634041fc9d2a7c45edcea36b : SUCCESS in 39m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/247a08ec1bf44deb9e67477152ce0e51 : SUCCESS in 35m 40s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/816187c21ac64fb3b1ee284ce164037a : SUCCESS in 53m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/80f904a63fba469fa012f9371cae860c : SUCCESS in 27m 02s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ba98128567164b60b992af31a0f05c27 : FAILURE in 1h 13m 57s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/1cfe581ae72e4edf9cbb38506aa3ca06 : SUCCESS in 1h 05m 36s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/8ff26324eadd4ec6a0b738460250a4b5 : SUCCESS in 1h 19m 29s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b86cbb83b5f0441ca541bd5560ead473 : SUCCESS in 1h 09m 38s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/873ef8572ffe472184928e6cb44db9a7 : SUCCESS in 1h 32m 18s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de33a6de0c6a4c48ab50581b25173eb3 : SUCCESS in 1h 08m 41s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6efaa7c146b8492ca8720ca82be81ba7 : SUCCESS in 1h 26m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/007d44853e2d4d4eaff98d541bfe6caf : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d24fa786b7a4c4899cb4ed8b5d7a034 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c73b6ef1b7d417a8852c6fa4617fa6b : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/380bd4d75df948ee87c58ae8961933d8 : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/639cfe9a3ef64c15aa265e13e218717b : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c4e27f49ee4193a5a60e1e10bca7e3 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b0e43bf6c3d4dd2a71fc7ee1b4d6a92 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"86394c25ea5521d76648497e8c1af2eced27d843","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:34:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6e480f8df058ab3943f9233824f5b979c2c47eb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 12:06:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b284115dff0d44b899bf95a431d083ff\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2c9d432e3976404fabfe2c790f10fca2 : SUCCESS in 1h 29m 32s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6b135abcfb5d4692801b28f99e5697c7 : SUCCESS in 1h 58m 19s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e56d20f50fcc47ac97da694d1a929040 : SUCCESS in 1h 37m 07s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b8fe8192293e490a82d7de6c3c487359 : FAILURE in 8m 22s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/24acdd237fd241abbf8c3ef408202ad1 : FAILURE in 1h 56m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d808baa594e49a184bd473bf5babf1f : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5841b11bf6c74d2c99457ca09b3513c3 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e336409225904f58a940bf5d4310b133 : RETRY_LIMIT in 2m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2563ffcb2a44d6f751e54d26485572ec3399b160","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 12:57:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f64447bccf674409acdc42a3954eb962\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26ab143deb5d41a7b14b99e561b8ec19 : SUCCESS in 3m 41s\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/56a4178efa364dec842de37cfb30ec4f : SUCCESS in 42m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/95d63318b6fc440cbcebed8d97ad2b8c : SUCCESS in 24m 39s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/981cbaac5c8748c9b3ecc83aac4aa4c6 : SUCCESS in 49m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/88cbb464f3fa4c2e96c1e7f8b1409194 : SUCCESS in 1h 06m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/184039b34cce4b59a50c0b7e061d5834 : SUCCESS in 56m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/44904c56c1f34da38f1ce540e7ce15a2 : SUCCESS in 19m 27s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9d70f34e956f4f548d0f53f5881874b6 : SUCCESS in 54m 23s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/24cb7ac4968148f39c78da0da7a88e1f : SUCCESS in 1h 12m 09s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a91f3570cef146abae2820c54e91045a : SUCCESS in 59m 54s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d577c846c7934b59a469b9a52ea255e1 : SUCCESS in 1h 21m 59s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a442fe746734444863968d2edfe1858 : SUCCESS in 56m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7594261d387640d7a33e18bb10dcb008 : SUCCESS in 1h 16m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4efe0ba52d4a4d86a620c0928beecde0 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7f108493c1649be802dd04c0a1c25c1 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0945383d9f6e4b9697cb805380640ef4 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/611142170cb44cae969cfbe250461d68 : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ed89bbe590d4e27ad4ffd336acda15e : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bdffc0448d64151829e8ea214db7361 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":3},{"id":"13cde397955c26ff27dad183f2e210c679f712fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 12:57:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"676a31368e85445bb1a8fa72e4ffc47e2f700c22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 12:59:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6dfb1e5b1d419b964baaf2caee7b1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e68e93db4476496b9583271a108dd6af : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2050091c9774313a5b9e0dab28652aa : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"77ff5b2621a31f9644d38dcd63ed89a00ff25997","revisions":{"eb7152c5f8e83abcec37a4247cfb81a753941b3d":{"kind":"REWORK","_number":1,"created":"2026-01-12 08:48:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/973051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/973051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/973051/1"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-12 08:48:12.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-12 08:48:42.000000000","tz":60},"subject":"influxdb: Drop support","message":"influxdb: Drop support\n\nChange-Id: Id17aa4946617c851780d424e7e053dd79196be6f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb7152c5f8e83abcec37a4247cfb81a753941b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb7152c5f8e83abcec37a4247cfb81a753941b3d"}]},"branch":"refs/heads/master"},"a87c9af840ece7bc49ff97e72895737bc753fb5b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-22 09:49:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/973051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/973051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/973051/2"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-12 08:48:12.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-22 09:49:36.000000000","tz":0},"subject":"influxdb: Drop support","message":"influxdb: Drop support\n\nKolla delivers InfluxDB v1, which is EOL (unless you have paid support).\n\nThere are no plans in upgrading InfluxDB to v2 - so let\u0027s drop it.\n\nChange-Id: Id17aa4946617c851780d424e7e053dd79196be6f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a87c9af840ece7bc49ff97e72895737bc753fb5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a87c9af840ece7bc49ff97e72895737bc753fb5b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"77ff5b2621a31f9644d38dcd63ed89a00ff25997":{"kind":"REWORK","_number":3,"created":"2026-01-22 09:59:40.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/51/973051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/973051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/973051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/973051/3"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-12 08:48:12.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-01-22 09:58:48.000000000","tz":60},"subject":"influxdb: Drop support","message":"influxdb: Drop support\n\nKolla delivers InfluxDB v1, which is EOL (unless you have paid support).\n\nThere are no plans in upgrading InfluxDB to v2 - so let\u0027s drop it.\n\nChange-Id: Id17aa4946617c851780d424e7e053dd79196be6f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77ff5b2621a31f9644d38dcd63ed89a00ff25997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77ff5b2621a31f9644d38dcd63ed89a00ff25997"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
