)]}'
{"id":"openstack%2Fkolla~973202","triplet_id":"openstack%2Fkolla~stable%2F2024.2~Ib7c2d8419932e049db871053c3b4faad791c3566","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-01-13 17:59:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7c2d8419932e049db871053c3b4faad791c3566","subject":"CI: Fix publish job secret aliases","status":"MERGED","created":"2026-01-13 08:20:44.000000000","updated":"2026-01-13 18:01:15.000000000","submitted":"2026-01-13 17:59:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973202","cherry_pick_of_change":973200,"cherry_pick_of_patch_set":1,"meta_rev_id":"83b547a7963831f7d2da112080530e735b286b2d","_number":973202,"virtual_id_number":973202,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-13 17:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-01-13 17:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-01-13 17:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-01-13 15:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 08:20:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-01-13 08:21:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 08:23:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-01-13 08:40:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-01-13 09:13:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-01-13 15:04:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-01-13 17:43:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1a58dbc05033b283e8d2a9bcb6d2d5480448e9f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-13 08:20:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.\n\nThe following files contain Git conflicts:\n* .zuul.d/debian.yaml\n* .zuul.d/rocky.yaml\n* .zuul.d/ubuntu.yaml","accounts_in_message":[],"_revision_number":1},{"id":"395d229eae1d564925c619edd6e749aadc1cc94a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 08:21:03.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.2.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/debian.yaml@stable/2024.2\", line 71, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/debian.yaml@stable/2024.2\", line 75, column 1","accounts_in_message":[],"_revision_number":1},{"id":"fd13ac0dfcb214d66ea1c11a7740b9d8e840d768","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 08:21:17.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.2.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/debian.yaml@stable/2024.2\", line 71, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/debian.yaml@stable/2024.2\", line 75, column 1","accounts_in_message":[],"_revision_number":1},{"id":"f316939fe02a6c4849a07d45ba7bfbd990c880f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-13 08:23:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ea2f3371ae0fd2dbd7b35bb3540ef56108eb5e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 08:24:19.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.2.  The\nproblem was:\n\n  Decryption failed\n\nThe problem appears in the the \"kolla-publish-debian-dockerhub\" job stanza:\n\n  job:\n      name: kolla-publish-debian-dockerhub\n      parent: kolla-build-debian\n      post-run: tests/playbooks/publish.yml\n      vars:\n        publisher: true\n        kolla_registry: dockerhub\n        kolla_namespace: kolla\n      secrets:\n  ...\n\n  in \"openstack/kolla/.zuul.d/debian.yaml@stable/2024.2\", line 51","accounts_in_message":[],"_revision_number":2},{"id":"7bc5b76db724f24eb5d632d1bd9f58a615e4b473","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 08:24:32.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2024.2.  The\nproblem was:\n\n  Decryption failed\n\nThe problem appears in the the \"kolla-publish-debian-dockerhub\" job stanza:\n\n  job:\n      name: kolla-publish-debian-dockerhub\n      parent: kolla-build-debian\n      post-run: tests/playbooks/publish.yml\n      vars:\n        publisher: true\n        kolla_registry: dockerhub\n        kolla_namespace: kolla\n      secrets:\n  ...\n\n  in \"openstack/kolla/.zuul.d/debian.yaml@stable/2024.2\", line 51","accounts_in_message":[],"_revision_number":2},{"id":"a14fd63aadc638dbe3dbcc953a8f0ac8d9823221","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-13 08:40:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5ffdaebf77ae0fa3c311467832ef217f1e4eea55","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-13 08:45:00.000000000","message":"Patch Set 3: Review-Priority+2\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"ceef7cb2ba4db87c02a22678e7dc2a06f0bbe841","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 08:56:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d70eafc6536541d0830346edbb1bae81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d77df2aa2cc4b83a6b1c6df3b67dfb4 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbd1e9b8caf14edb933ecd4ba4c722fa : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37e20c3c8cac44998a6720e4896b1921 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bae3ed7f26b1436eb69520597021dba3 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/127db7f7f9ee4cc69804d9e84c2240b1 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c666d104c8045748bd6b701d3dff077 : SUCCESS in 3m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b9bbc44354114515b9cd969133a1690e : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"be35521973fb647cab755c6d5a59e9c1bab609c9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-01-13 09:13:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3fc078429b8e6681e59afe10e9d9ee33647286a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 09:14:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bb3a93eaa5d6d8b3628a249722e522c30e60eaea","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-01-13 09:14:40.000000000","message":"Patch Set 3: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"cb9bc4628308b42d1a46b87dad7744651dfbe4cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 09:14:42.000000000","message":"Patch Set 3: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":3},{"id":"fdcc8b30afa49e0fc058d8e69de050a402d6273c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-13 09:28:45.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"59754846d667fda6edc7dbe904522c3e617c367a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-13 15:04:13.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-2\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2efec9d098d199b0715ff9bde36431e39eeab326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:17:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bec2d1d17f8a444baf058a44bc38d647\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28beed7a45694452bfde317bd836cf56 : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e4af50063c84c75bd4433e369f0965a : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/637f686ffd3f4d618648427f8fa85383 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f52139ba0bc4436ba70a98aa2b82631 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31164d6c017345fa815f1a837e3e42df : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4329ed3ba44ad7a97baa2a9d28031e : SUCCESS in 4m 59s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/be2f53dc21c547dab72610fd3e6be4ce : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":4},{"id":"38529014a9a6af46bc6415c22dc72c5721f19e94","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-13 17:43:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7e1c75d0e99ccb3004937b9167db1e819ff48666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:44:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b92d29b3e03af67e0770a930aa198495c7e006c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:59:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f22dee6e22466491315a243c044c33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac75503cfcf4c128868cbad684ecf00 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75f88d7bb50f46948597ea8b934f9a19 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aab151a44e3243f78d64e1a4973abf32 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb7665421a14092a8547bdd68e2a091 : SUCCESS in 4m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c6d2fb4519cc4fa3b429fea2bf386948 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":4},{"id":"cb349128b8c904dfe288b3abae757c02e068166b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:59:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"83b547a7963831f7d2da112080530e735b286b2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 18:01:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df97c98aec44e96b4e4592a8e11207b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af65bdf74da7443cbb2942088fb7c6b5 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a43fe438adf1c3db1e6a383b116205190f3a2a2f","revisions":{"ffe25b0212e80c1fbc51dc4fc8798423e89c81f1":{"kind":"REWORK","_number":1,"created":"2026-01-13 08:20:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/02/973202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/973202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/973202/1"}}},"commit":{"parents":[{"commit":"57f438d69362b28d03f190fa6046843c9578c8c1","subject":"Transition to the new rabbitmq Debian package archives.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f438d69362b28d03f190fa6046843c9578c8c1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-01-12 10:56:15.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-01-13 08:20:44.000000000","tz":0},"subject":"CI: Fix publish job secret aliases","message":"CI: Fix publish job secret aliases\n\n[stable/2025.1] adapted to old job structure\n\nChange-Id: Ib7c2d8419932e049db871053c3b4faad791c3566\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 5a47700917e51ce4f3fe5bc4b4197f09bf0fce32)\n(cherry picked from commit 41c09022443910920ee688369edbc8a308dc7386)\n(cherry picked from commit 6e6b3315d2ef739fbb6d439c026b4fcfeb606773)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffe25b0212e80c1fbc51dc4fc8798423e89c81f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffe25b0212e80c1fbc51dc4fc8798423e89c81f1"}]},"branch":"refs/heads/stable/2024.2"},"8eebf96e4c9a3e628be609b267806d3604352e61":{"kind":"REWORK","_number":2,"created":"2026-01-13 08:23:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/02/973202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/973202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/973202/2"}}},"commit":{"parents":[{"commit":"57f438d69362b28d03f190fa6046843c9578c8c1","subject":"Transition to the new rabbitmq Debian package archives.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f438d69362b28d03f190fa6046843c9578c8c1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-01-12 10:56:15.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-01-13 08:23:02.000000000","tz":60},"subject":"CI: Fix publish job secret aliases","message":"CI: Fix publish job secret aliases\n\n[stable/2025.1] adapted to old job structure\n\n[stable/2024.2] fixed merge conflicts, dropped old dockerhub job\ndefinitions that were no longer used\n\nChange-Id: Ib7c2d8419932e049db871053c3b4faad791c3566\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 5a47700917e51ce4f3fe5bc4b4197f09bf0fce32)\n(cherry picked from commit 41c09022443910920ee688369edbc8a308dc7386)\n(cherry picked from commit 6e6b3315d2ef739fbb6d439c026b4fcfeb606773)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eebf96e4c9a3e628be609b267806d3604352e61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eebf96e4c9a3e628be609b267806d3604352e61"}]},"branch":"refs/heads/stable/2024.2"},"b8b6812a88e2da3504f5390c1ccb6023e488717f":{"kind":"REWORK","_number":3,"created":"2026-01-13 08:40:11.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/02/973202/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/973202/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/973202/3"}}},"commit":{"parents":[{"commit":"57f438d69362b28d03f190fa6046843c9578c8c1","subject":"Transition to the new rabbitmq Debian package archives.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f438d69362b28d03f190fa6046843c9578c8c1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-01-12 10:56:15.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-01-13 08:39:44.000000000","tz":60},"subject":"CI: Fix publish job secret aliases","message":"CI: Fix publish job secret aliases\n\n[stable/2025.1] adapted to old job structure\n\n[stable/2024.2] fixed merge conflicts, dropped old dockerhub job\ndefinitions and secrets that were no longer used\n\nChange-Id: Ib7c2d8419932e049db871053c3b4faad791c3566\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 5a47700917e51ce4f3fe5bc4b4197f09bf0fce32)\n(cherry picked from commit 41c09022443910920ee688369edbc8a308dc7386)\n(cherry picked from commit 6e6b3315d2ef739fbb6d439c026b4fcfeb606773)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8b6812a88e2da3504f5390c1ccb6023e488717f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8b6812a88e2da3504f5390c1ccb6023e488717f"}]},"branch":"refs/heads/stable/2024.2"},"a43fe438adf1c3db1e6a383b116205190f3a2a2f":{"kind":"REWORK","_number":4,"created":"2026-01-13 15:04:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/02/973202/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/973202/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/973202/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/973202/4"}}},"commit":{"parents":[{"commit":"f1da7130abcdb8c6b7469782800df5bf66a0ee21","subject":"CI: rotate zuul secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1da7130abcdb8c6b7469782800df5bf66a0ee21"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-01-12 10:56:15.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-01-13 15:02:32.000000000","tz":60},"subject":"CI: Fix publish job secret aliases","message":"CI: Fix publish job secret aliases\n\n[stable/2025.1] adapted to old job structure\n\n[stable/2024.2] fixed merge conflict\n\nChange-Id: Ib7c2d8419932e049db871053c3b4faad791c3566\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 5a47700917e51ce4f3fe5bc4b4197f09bf0fce32)\n(cherry picked from commit 41c09022443910920ee688369edbc8a308dc7386)\n(cherry picked from commit 6e6b3315d2ef739fbb6d439c026b4fcfeb606773)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43fe438adf1c3db1e6a383b116205190f3a2a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43fe438adf1c3db1e6a383b116205190f3a2a2f"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
