)]}'
{"id":"openstack%2Fkolla~973417","triplet_id":"openstack%2Fkolla~master~Idb44a06cc8e44652ac8958a71be48f1920105f89","project":"openstack/kolla","branch":"master","attention_set":{"34212":{"account":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"last_update":"2026-01-14 18:48:35.000000000","reason":"A robot voted negatively on a label"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-24 12:25:23.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Idb44a06cc8e44652ac8958a71be48f1920105f89","subject":"Add purity_fb to python deps for manila-base","status":"NEW","created":"2026-01-14 17:21:04.000000000","updated":"2026-06-24 19:42:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a904f454ae5343e9042129db4ef8543f7cfaa49e","_number":973417,"virtual_id_number":973417,"owner":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-24 12:25:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 18:48:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd5e45edbc54fc65c1248794b715198bf4a642ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-01-14 17:21:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3154a278afccc934cf110b6617c01a9adcb4ac6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 18:48:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed16fcdeaa09491c8dc7a687fafcd1b3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/435be20a4d1e4768906dd881fb663cd5 : SUCCESS in 45m 26s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/640c6b0add8145059e5e16bbcfb2716b : SUCCESS in 40m 09s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5f6bd3ac68874d32b8418ac83f3151c1 : SUCCESS in 25m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b2a8064b57144073829a22d276494ff3 : SUCCESS in 47m 36s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/36326f40ab494b26aebdced43845280b : SUCCESS in 48m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae8e2fc83aaf4f69b8191103bc4386e7 : SUCCESS in 48m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c11119b4216048b2bb99840b91c3c1d9 : SUCCESS in 28m 52s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/257db426abab48269ebc65bcc1a24edf : FAILURE in 1h 15m 30s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/cdc1239881aa4fadabd0e8806645868e : SUCCESS in 56m 00s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/519edf3613374fa48a0ff6c650435b46 : SUCCESS in 1h 21m 05s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/25a65b858e2747fa8f81b33a309f3a5d : SUCCESS in 1h 05m 22s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/4028697128104fa1abf17e59483ee4d3 : SUCCESS in 1h 26m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7c60c9f8daa4af0b84f846e03125bfe : SUCCESS in 1h 03m 14s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f492ba8b84294d5693a8b5a5be7b590f : SUCCESS in 1h 24m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4a62fb98ff4047019d9f0583230cb665 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73fc0fdbc4474a93873f0d2e574cce00 : FAILURE in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8c7019fa65342b8b6b6e9b818d95df0 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b49e659a001e43e0a44b4126e8cfcc52 : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35d942b0de244998b9fa797095b3a498 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3bcd972f29450c8600c2bf1eefeeb4 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f5c568d69734eb78c5fb553ca17e8de : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d0e0fcc54d717e19da9a09087bc916b9a283bb41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 19:00:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094c1fe384734b02a4335d12b101a1f8\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1e5edebcaba248fe9b058276fba61ea5 : SUCCESS in 1h 37m 36s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b97a9e9aaef1436ebfb1da350d7905e6 : SUCCESS in 59m 52s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b1439e8656c348f699e9eecc6516673b : SUCCESS in 56m 46s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d7c741391c304a4ba08022cb4685da07 : SUCCESS in 55m 18s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/379932728f7f4e49a0f618eb384773fe : FAILURE in 1h 30m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d1c040890a9245238396f6d44c01df7c : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/03db71d6932e44af827d1a193443d703 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fb2ddbfa87844568b4961939ce6cef05 : SUCCESS in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a99d927e743b973382ba2ba4bb6b66fe1a1c77b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-07 19:51:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_34212\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"}],"_revision_number":2},{"id":"2f6485cef9c40bcbf222d97ea319edd1f84c764c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 21:53:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78df0b293f77443db468615b3717f044\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f43540e0ab8e401c97175b65aa4b20ab : SUCCESS in 1h 25m 38s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f7683783f8164f619a8a71b14fce6d11 : SUCCESS in 1h 16m 32s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/63a2e70ff3c740d080658aa8c9de41fa : SUCCESS in 28m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/eeba9211853d47aa8624438b21fff3ad : SUCCESS in 1h 23m 48s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/fda2447d2db44e369cfd1c2f2d25d829 : SUCCESS in 41m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b6b2ac89a494e878db4e9724d5f1687 : SUCCESS in 1h 18m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/03b9c4684d3f49ec801c407b12a5524d : SUCCESS in 26m 46s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/69e25ba140bf4845ba17296411b68be0 : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5fa51177b74f44bd8850910827196df7 : SUCCESS in 1h 22m 32s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/acf57c0dfbb84257b5489a94361541a7 : SUCCESS in 1h 19m 05s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/222cbaa30be741e9be0d7cd0f2b4b547 : SUCCESS in 1h 53m 07s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8965d2909b745d7b0ca61af81e03bc4 : SUCCESS in 1h 23m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1f2a4e7fa7ff462d9132afcde04ca07d : SUCCESS in 2h 01m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/66c53dcb5d0d4791bcb317cb42de89f9 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eabe79b012bf44ec8e28a81bebe622a5 : FAILURE in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c25ad5faa50247f7b778d34eca74c158 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5da09fc8db8946ae8638740ec925fcba : SUCCESS in 2m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8000fcbad5e34101832aeb6a5d3dfd77 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40767308d243473bac33cb40e9488767 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0bde4ebf454a14873367624fc28f78 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2493599539144fbda02e87864daa1746 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0827a9bd50473e0574fd8accd3b8d05da42aee49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 03:28:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9e96af0ca21475db89ee99b3a723d25\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1970e333ef10459db2d36e54ddc85c64 : SUCCESS in 1h 44m 41s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d42f45e9ac304fb5bee56c978ea788ad : SUCCESS in 1h 51m 20s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/efa16e8698a84b85aa33a6b4938612ca : SUCCESS in 1h 56m 13s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e05e99d8835e40aeb2cf8b295f6c2dc0 : SUCCESS in 2h 10m 39s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/3b5ad114692e4f6e8e6d5976c47a224b : TIMED_OUT in 2h 32m 46s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/0d0624e23211427b91bbfdcdbf82c01d : FAILURE in 2h 37m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c340c847e31d41b9976c3de904e34d1d : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/066ada870aee417f8e52776c7a9a310d : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6d102634852c4d4f95c57b9381889f01 : SUCCESS in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fd60ae14d16490312eb12ba0a3b759a3fbbcc5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-24 10:34:03.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1b5359ca06daad99beff1f5615ddcfdddeb546ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:25:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46451a307f9c49a28b7e6a32254dee96\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/631e0c0f6ae24afa988bde1e9ede007b : SUCCESS in 31m 33s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3b1afe616a974d8e9c89784dfba97340 : SUCCESS in 31m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2a9cdc85f5fe4762b274eee040f02444 : SUCCESS in 28m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1b54396f680e42d0912f4b32ea2a009c : SUCCESS in 16m 30s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a2c978c476524355b5c1b769ecf94c00 : SUCCESS in 32m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e2ef99860a94eab8d7d0ac09494f752 : SUCCESS in 27m 22s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/49aa1d375a2043908c93c934fe791757 : SUCCESS in 33m 25s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ca3aae0130434bebabca1ca040563bf2 : SUCCESS in 1h 08m 41s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/cdb3f30d911c4ddcb56ff29c2f23cf3c : SUCCESS in 1h 09m 02s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/f3c0967d67aa48b1b1fa7953b55a0edd : FAILURE in 27m 00s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4779279fcc4c4d2eb1efd31270b932bd : SUCCESS in 1h 09m 49s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c6c439a30e2b41d5bd0c6304141cbe05 : SUCCESS in 1h 41m 30s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cefc463880e4ffdacfa4f9d28ea8689 : SUCCESS in 1h 12m 13s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2f1560e8c169488980dc2fb9a5b09fdb : SUCCESS in 1h 46m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9f959ad346db46e0a1b30593ed7eb6d2 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036c8ee95ee74dca8b162695da4359a5 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86042edc64e044bbb91c6c02a587eea0 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdbae4fcb4994f94a0285bed7bd2cef3 : SUCCESS in 2m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f60157bac6ef4aab955afd869fd0ecc4 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3d78abe0754b64bfaf7738bc538fb2 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4314de835f3d46c29d1c66709634da0a : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e8eb31c2e2a475695fdca448583e6aa : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":3},{"id":"a904f454ae5343e9042129db4ef8543f7cfaa49e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 19:42:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b09b98aad54448a2e5d252f88d20ce\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/98137d759c074dba891d7ad78fd610ab : SUCCESS in 38m 03s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8876e48273aa440cad96500e242ad717 : SUCCESS in 34m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8533195806e7463c98b39b80d3f7e580 : SUCCESS in 31m 05s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4506600fb3e046b889d381ed28644761 : SUCCESS in 1h 20m 46s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/aa20dd8b79774f61ba8fd3e848ddbcc2 : FAILURE in 1h 06m 28s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/fe830f761364478da893e35993d4dfa7 : FAILURE in 1h 37m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63015359ca254b2c98165b65cabe37b2 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/685de32a42b040418393cbd93e55f39c : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/00ba82cfe5734464b15565ffecf47cce : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aac6defa2d9b9ff2d866fa7fd437f3933f8263ec","revisions":{"d1b52e8edbcd504fac0f0f689fd711ab5f4e5486":{"kind":"REWORK","_number":1,"created":"2026-01-14 17:21:04.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/17/973417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/973417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/973417/1"}}},"commit":{"parents":[{"commit":"9a09f82646c577c4b484eeaac7bf56d8005da11b","subject":"CI: Add ^docker/ trigger for aio jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a09f82646c577c4b484eeaac7bf56d8005da11b"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-01-14 17:07:36.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-01-14 17:20:50.000000000","tz":0},"subject":"Add purity_fb to python deps for manila-base","message":"Add purity_fb to python deps for manila-base\n\nThis ensures that the Pure Storage FlashBlade driver\nfor Manila has the correct Python dependencies available\nwhen configured using kolla-ansible config-overrides.\n\nCloses-Bug: #2138391\nChange-Id: Idb44a06cc8e44652ac8958a71be48f1920105f89\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1b52e8edbcd504fac0f0f689fd711ab5f4e5486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1b52e8edbcd504fac0f0f689fd711ab5f4e5486"}]},"branch":"refs/heads/master"},"db27e30100114a58f4e03887336e6485e125de20":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-07 19:51:24.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/973417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/973417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/973417/2"}}},"commit":{"parents":[{"commit":"cbdbd894b06b049e36ecd4fab4d0ecc3a49e0c00","subject":"Merge \"masakari: Add crmadmin and cibadmin to monitors sudoers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbdbd894b06b049e36ecd4fab4d0ecc3a49e0c00"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-01-14 17:07:36.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-06-07 19:51:24.000000000","tz":0},"subject":"Add purity_fb to python deps for manila-base","message":"Add purity_fb to python deps for manila-base\n\nThis ensures that the Pure Storage FlashBlade driver\nfor Manila has the correct Python dependencies available\nwhen configured using kolla-ansible config-overrides.\n\nCloses-Bug: #2138391\nChange-Id: Idb44a06cc8e44652ac8958a71be48f1920105f89\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db27e30100114a58f4e03887336e6485e125de20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db27e30100114a58f4e03887336e6485e125de20"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9a09f82646c577c4b484eeaac7bf56d8005da11b","ours":"d1b52e8edbcd504fac0f0f689fd711ab5f4e5486","theirs":"cbdbd894b06b049e36ecd4fab4d0ecc3a49e0c00","merge_strategy":"recursive","contains_conflicts":false}},"aac6defa2d9b9ff2d866fa7fd437f3933f8263ec":{"kind":"REWORK","_number":3,"created":"2026-06-24 10:34:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/973417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/973417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/973417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/973417/3"}}},"commit":{"parents":[{"commit":"cbdbd894b06b049e36ecd4fab4d0ecc3a49e0c00","subject":"Merge \"masakari: Add crmadmin and cibadmin to monitors sudoers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbdbd894b06b049e36ecd4fab4d0ecc3a49e0c00"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-01-14 17:07:36.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-24 10:34:03.000000000","tz":0},"subject":"Add purity_fb to python deps for manila-base","message":"Add purity_fb to python deps for manila-base\n\nThis ensures that the Pure Storage FlashBlade driver\nfor Manila has the correct Python dependencies available\nwhen configured using kolla-ansible config-overrides.\n\nCloses-Bug: #2138391\nChange-Id: Idb44a06cc8e44652ac8958a71be48f1920105f89\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aac6defa2d9b9ff2d866fa7fd437f3933f8263ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aac6defa2d9b9ff2d866fa7fd437f3933f8263ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
