)]}'
{"id":"openstack%2Fkolla~973868","triplet_id":"openstack%2Fkolla~master~I76f91c0f59a2a623308267afa7810314294fa064","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-21 10:47:32.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-01-21 10:47:32.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-21 10:47:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I76f91c0f59a2a623308267afa7810314294fa064","subject":"Remove remaining mentions of Docker Hub","status":"MERGED","created":"2026-01-19 16:21:36.000000000","updated":"2026-01-21 10:48:24.000000000","submitted":"2026-01-21 10:47:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973868","meta_rev_id":"3c240b1f6f970d317ed22745bdfd3f29bc03081b","_number":973868,"virtual_id_number":973868,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 10:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2026-01-21 09:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2026-01-21 09:24:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-01-21 09:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-01-21 09:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 16:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 19:05:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-01-19 19:05:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2026-01-19 19:05:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-01-19 19:05:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-01-19 19:05:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-01-19 19:05:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2026-01-20 06:09:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"747f3535480e4ee1cd9763c72d211101683c597e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-19 16:21:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bda5c90e399582b010430302e0f922939695d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 16:34:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5df1550c9614cfea18621c1dfcd1f15\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9f4e188fa5c7454c9063e029e6326d1b : SUCCESS in 3m 05s\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/321be50fd2fa4cb38a718fca89b20654 : FAILURE in 4m 02s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8e0bfe1dacc54313b8433e3b1ea2186e : FAILURE in 3m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1171522e4c0242d69813b7ed9b8dd3c4 : FAILURE in 2m 46s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b6d326f2e35d4b75b7dfe423f9542746 : FAILURE in 5m 51s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ece3e9a9f8f04d20806c5ed6fd439c73 : FAILURE in 2m 38s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbb48f6c833d49d8a3bd5dd247e98f2a : FAILURE in 3m 35s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee3f0d8a5bd843cdaaa2bffe141bdcea : FAILURE in 3m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a06ab7947fc8402e9cf7d587aa25efed : FAILURE in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d851c796bea04cf7a03657059d7d9e9a : FAILURE in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2b83d3630594df191830d67ce0612b9 : FAILURE in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63540bac35a544b2b418eb29623c13b1 : FAILURE in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3fec2fab8134313ad67c52a9d0e78e5 : FAILURE in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8e84a655ea450eab7801643ed2114b : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"779772792798814e2836fa6911d8a7bdb1c527aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-19 16:36:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5e4569045de746281bd72337d5fca26d99d442c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 17:36:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489e2f26961f43fa98434048e35f9864\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b563f217e6f74de28cfb862232f174ec : SUCCESS in 2m 37s\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4d4c4be7a64a4416ad89fa0bb1314eee : FAILURE in 53m 05s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/bd177977e4414005b518f7a492c391dd : SUCCESS in 45m 31s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/fdda97e7d9f64f71ab69cd3ce763f928 : SUCCESS in 12m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bcc950dc49cd467788f205589d3c6af0 : SUCCESS in 48m 49s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6713d3cfa53e4ed8a0b2449f860591df : SUCCESS in 36m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96d0ca074c7e4081a301c3fae9cca94d : SUCCESS in 50m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bc5277b90eb645f8908a358e22be8da1 : SUCCESS in 26m 27s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/75b9a6e50e1c4d718575de294b75da2e : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721c5f71b6204e4784d633723c338ea0 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca24d26c6d284e4292952d3aa6bcedb4 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23793003196b4ba6a395fb3837f14ca4 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/622659b2e4fb4262bd8b8aa4997deb23 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0aeb348eefa41b68c36e1947b0a23ba : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"0df61cfd5eb2069b01006570f80925ca3d7a13b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:17:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73e18b851e74ea49b675b30c216cc6f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/729c65f03b1647a6bbd0f9615850e766 : SUCCESS in 1h 26m 49s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/22acf50418cc4e268e73656cfb1174a3 : SUCCESS in 1h 38m 07s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ed325b6840aa4e53b3807dc5d45e76a2 : SUCCESS in 1h 41m 20s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7c4b5cddcc124701b7f0d1131735d18f : FAILURE in 8m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8670139291614c2aacb285f7abc8d78d : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e210ab35c4d4faeaf953cf55652f157 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/81c4d740ddbf414fa5fe9f3c942092a1 : RETRY_LIMIT in 2m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"63e4d833d2df7e43fb69de1497f1291b3e247e6d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-20 06:09:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e852d56ede38f57e5c9cd54957c7df4734bd24bd","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-01-21 07:33:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e0f98d42d8a1df387e1bc8c227948c14d50d2f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-21 08:13:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9775db80d38ee6818b7e97d004d0afac0cf5ff4d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-01-21 09:24:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"01f6c80bfbcab6c245749792636b4738bc66b6d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 09:34:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04381cbdc45d446691cd44b5bab2ecbb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c3ddd7783f7349fa9bd6ef74166898e6 : SUCCESS in 1m 28s\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f794969ee68a4e43870c5033b0315bd8 : SUCCESS in 1h 16m 21s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/fac2abd38b2543c9b1dc0619708c0ac7 : SUCCESS in 43m 31s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/602aa1fa65e44230b09586af26fff3c9 : SUCCESS in 13m 28s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c7173d91a0ce489e9e5b7cbad5294fb1 : SUCCESS in 37m 44s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/976f4ab18bf3474fa92aebda42700ce5 : SUCCESS in 35m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cccefd83da7e4c0facdc1c28cebcfec4 : SUCCESS in 37m 07s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f126aba7ce314a508c9cfd31f06afc5e : SUCCESS in 15m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/72d3d271ba784f70b073fa6c78e40e87 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8d9fc9a8573453c901a1e0796230c92 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6df486a9062c4381969e680b09380988 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01954bdcf555445aa060a98952285697 : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/979dd0b3bb3c46efbed7aee534d02655 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1e7016e28f4330a50f16b5d87da629 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e92304396676cf3fb1a874a0584490daf2a19870","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-21 09:42:32.000000000","message":"Patch Set 3: Backport-Candidate+1 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ef24425b9695d9a769664bd5a9749e5b38b13bd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 09:42:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4c06df1daae9fa3daeef00002f9a168393be7ffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:06:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/763f6056ac9844a18af66d70375dcdc4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e2de2b64fbfe4cfdb59d3c2ba0de4dd2 : SUCCESS in 1h 41m 50s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/44d1c9ef3ab84ab0b22ac9349c0e952f : SUCCESS in 1h 50m 14s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/c15624fcb1304b0dac307e8f5fc60c42 : SUCCESS in 1h 41m 32s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cd8aef25f44d4c9094d84c743f3b40e5 : FAILURE in 8m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/008dabe3c56f47a3a1d3606cfd8787dd : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22ede55e53ca49eb889a4771d16a0bc6 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/14fe0e31e97747a2a39bb8b0096d1d48 : RETRY_LIMIT in 2m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dbd7d7f7bc9b504f967e1f3a1f7cd467f440d34e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:47:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa0aab7ea0f4420aaf3f32deacc15d9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/59ee95698830435fb738d5e77b2b2b2e : SUCCESS in 2m 23s\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2121e868722048fa992386d229689c88 : SUCCESS in 46m 45s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f05ba442e8fa4e2db51c6ff1a5555535 : SUCCESS in 29m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9fd8d49bd0374877bc3fd130ff1d3c66 : SUCCESS in 59m 23s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/192a90e80b7e42cbaa18199be1434f96 : SUCCESS in 1h 00m 14s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94a2cf28d2784ac7bc64fc9c4296b4c1 : SUCCESS in 45m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b94aac5afa04d75a2bdb5b36f15542c : SUCCESS in 32m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5286055b058a4870b5268582fafa61dd : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a101d13bf4e4034afa88b51c95836ea : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bc11c69db0f4cf396eac4eae9fb6228 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95780d123bce417c95cdc0ba724d75cf : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acf3adea3cc421294fd0c16b2655dac : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":3},{"id":"007706c9d99763b5f8221fec1ffcb5394d6aa9bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:47:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3c240b1f6f970d317ed22745bdfd3f29bc03081b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:48:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/821a266527f141819b99ab905e9fb9fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eabb8824a72a4ba596f90478100a9a2a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9805dffa61c7c1582ef7f7307d8475972dd9b515","revisions":{"2a87adc7f0ea8baf5d25b6d684622b4c2c6dedb4":{"kind":"REWORK","_number":1,"created":"2026-01-19 16:21:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/68/973868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/973868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/973868/1"}}},"commit":{"parents":[{"commit":"a590ac5a5e94355081bd7192e776b6b5ac2851c2","subject":"Merge \"bifrost/kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a590ac5a5e94355081bd7192e776b6b5ac2851c2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-19 16:21:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-19 16:21:24.000000000","tz":60},"subject":"Remove remaining mentions of Docker Hub","message":"Remove remaining mentions of Docker Hub\n\nChange-Id: I76f91c0f59a2a623308267afa7810314294fa064\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a87adc7f0ea8baf5d25b6d684622b4c2c6dedb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a87adc7f0ea8baf5d25b6d684622b4c2c6dedb4"}]},"branch":"refs/heads/master"},"b1ec1b6ee5eb82dbe30e2569e59942cd43516b32":{"kind":"REWORK","_number":2,"created":"2026-01-19 16:36:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/68/973868/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/973868/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/973868/2"}}},"commit":{"parents":[{"commit":"a590ac5a5e94355081bd7192e776b6b5ac2851c2","subject":"Merge \"bifrost/kolla-toolbox: Bump ansible-core to 2.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a590ac5a5e94355081bd7192e776b6b5ac2851c2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-19 16:36:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-19 16:36:08.000000000","tz":60},"subject":"Remove remaining mentions of Docker Hub","message":"Remove remaining mentions of Docker Hub\n\nChange-Id: I76f91c0f59a2a623308267afa7810314294fa064\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1ec1b6ee5eb82dbe30e2569e59942cd43516b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1ec1b6ee5eb82dbe30e2569e59942cd43516b32"}]},"branch":"refs/heads/master"},"9805dffa61c7c1582ef7f7307d8475972dd9b515":{"kind":"REWORK","_number":3,"created":"2026-01-21 08:13:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/68/973868/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/973868/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/973868/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/973868/3"}}},"commit":{"parents":[{"commit":"f179e83e55320464dc66922fcb0aca88463696cc","subject":"Merge \"Re-enable SPICE support on Debian.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f179e83e55320464dc66922fcb0aca88463696cc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 08:13:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 08:13:07.000000000","tz":60},"subject":"Remove remaining mentions of Docker Hub","message":"Remove remaining mentions of Docker Hub\n\nChange-Id: I76f91c0f59a2a623308267afa7810314294fa064\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9805dffa61c7c1582ef7f7307d8475972dd9b515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9805dffa61c7c1582ef7f7307d8475972dd9b515"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
